)]}'
{"id":"openstack%2Fswift~839442","triplet_id":"openstack%2Fswift~master~I02ba67e77e3378b2c2c8c682d5d230d31cd1bfa9","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-05-05 05:42:28.000000000","reason":"Matthew Oliver replied on the change"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-05-27 18:32:46.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-05-27 15:58:48.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I02ba67e77e3378b2c2c8c682d5d230d31cd1bfa9","subject":"replicator: Log rsync file transfers less","status":"MERGED","created":"2022-04-26 20:45:12.000000000","updated":"2022-05-27 18:32:46.000000000","submitted":"2022-05-27 18:32:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"839442","meta_rev_id":"6a5d9b9fd8d7f1b302af0a84012d3d21435e5496","_number":839442,"virtual_id_number":839442,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-27 18:32:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2022-05-05 05:42:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-25 17:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-25 17:22:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 00:06:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 13:52:46.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2022-05-02 05:46:15.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2022-05-05 05:42:28.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"b47525707d1a82c4db7fb3fb0d0578e4695f0384","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-26 20:45:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2ac48bfe33a7928b910c786e1b28258746baa72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-26 21:59:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0a0f369b890883d939195c530995338098da4056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 00:06:46.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a001f14f73d49fca12236b8f7df16f3 : SUCCESS in 11m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/70e7fc2c56404594b58acfd2a2d60ed3 : SUCCESS in 55m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d6043e8edd324e72a64b2f15a981dcb7 : SUCCESS in 52m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7814c3c120184417b9433c444a7c9867 : SUCCESS in 56m 54s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9d2ca899ae7b48b5ab538758ed265e28 : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/033d946cfc074dc1a4b09d5c1a3cda65 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d6d3aa3a27c9411199e2d541bfca4925 : SUCCESS in 21m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/55ea147bfdf945f692f9f9c02b9a7dd0 : SUCCESS in 7m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bb6c5cccaa164f51b4fa8600cba642da : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1a5b39f3712d4e40b11c358a6899fc40 : SUCCESS in 17m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/85a45520657d40f3a551701de100ad1a : TIMED_OUT in 1h 31m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7d7f303b1b8d48cdba1d4318caf2e42a : SUCCESS in 16m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ae63054d0fa44346bc3c6515e752021f : SUCCESS in 18m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d77b1268790a4efdadc331e5822eb086 : SUCCESS in 22m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/13c01952232b48fc8eeaf97795336035 : SUCCESS in 21m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/563612aa61e346649ab386b96dff44c6 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9cf1d0cce1b74ff2a079b955586a1d25 : SUCCESS in 20m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6f57549331c54b5eb5216eb012f4b62e : SUCCESS in 18m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/24af1eafe50b4b74801ffee3ce435d59 : SUCCESS in 7m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/06aa8202c4bc43ce940a370ec96a6ff7 : SUCCESS in 19m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1c0905f82f95415da3adf9bceefcd7c0 : SUCCESS in 57m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/db12f1321c6d4c678020ab20f9f9c010 : TIMED_OUT in 2h 01m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26bd32ee8d2d4805807c04712ff7099e : SUCCESS in 33m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd343dd97ea44e16b8f3c3bee11ab923 : SUCCESS in 32m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02f665acca0b447f851e8d973dc82231 : SUCCESS in 16m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ddb67707b94a9ca7c61c5584a423a5 : FAILURE in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b782f4473354ac2b4d0f7731a3fdbac : SUCCESS in 22m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/497a43723d3b417e9a45a84830f5a072 : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":2},{"id":"6248de65888ae96150bfdc34f6286eb3bf52a66c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 01:03:20.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aea11583a0c74e009a3eac04c923883a : SUCCESS in 52m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4bbfb7502c2a44e59a56e202274c4bef : SUCCESS in 58m 25s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/5bf8a558521a4aa2aca3779c060e74f9 : TIMED_OUT in 3h 01m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4043bf62b34a4251bf794264ff209faa : SUCCESS in 1h 23m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb39a530feb64547a1d0408640d1b6c2 : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":2},{"id":"ac2200ca41e4c20965cbbe51f067e7d09128afb0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-27 13:52:46.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cf0e462c9818de1e97ca323dd1a181f84eac3670","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-28 19:35:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"afbea1e77d7db5749bab3efce05cde57ad4b5730","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 21:45:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac2a5176bfad4ab9b05d812f0e4c97d5 : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/08cebef87e45477d87e33db2941fc238 : SUCCESS in 1h 04m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/29b7578d475a45ecbf1cc8c6edaf5e8b : SUCCESS in 46m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eae89d13a98d4984bd7734b7c453eb9c : SUCCESS in 48m 31s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0b1c7f51f1354eef848f74826bd55a67 : SUCCESS in 20m 06s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a4cd836269184e87ad1e17c118c15776 : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f1290f44e4354847afc117f988c94413 : SUCCESS in 25m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b8d97585aea94f3aa197f05da69d2e6b : SUCCESS in 7m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2e202978af40445c9e67aa6aea50acd8 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/42c1496329f34327a223ef4cc636b409 : SUCCESS in 17m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a4bc090c24846e0aa9496e28034d03a : SUCCESS in 19m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5a0d52f52552426d921420cc58e98f8a : SUCCESS in 15m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d246e3f0f49a40669634084afb94852d : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6a06d1ceb09446939b431a8d364acd09 : SUCCESS in 25m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0895e5165af649e2a10d0f75078928ae : SUCCESS in 16m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/30505e0d56654aaf87ee47a45be3b0a1 : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c553a777bf554e79892e51a74c4f4eb5 : SUCCESS in 17m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/65707e5804be4087ba1293b9d5e00f50 : SUCCESS in 49m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/41b6dfd9a92e4b5d8e918f80c940e13b : SUCCESS in 7m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a07c2f578f3d4350b9093a3870ba9213 : SUCCESS in 19m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/158b851bd9f64ee3bf6234a577716043 : SUCCESS in 47m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8214de06f3f4b9d85ae5a703fb239cc : TIMED_OUT in 2h 01m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d00cea8f0d443469d1a820e4f80b483 : SUCCESS in 40m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fc4f88f34e7f4879b5ebf28d2e861813 : SUCCESS in 1h 11m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb0806627bb84fb9b618e6b417687995 : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08ea30d68e47425b8174c24742106f20 : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2419219d1da64a96bbabaa40e2d44b47 : SUCCESS in 19m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8b276c1733a42bf8c9fbd6231e51c4d : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":3},{"id":"c316c10a75f54bab584c44c1bdaca0c847964375","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 22:39:22.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2fd09b00aaef409e8a21f2868294569e : SUCCESS in 40m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d57224ce6f84d13a5093fb7ffd0a5cc : SUCCESS in 45m 18s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/64b2718fa163457b8c3776670052a72d : TIMED_OUT in 3h 01m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b272ab3bafab4ab7978f044f413aa07f : SUCCESS in 45m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e597e524ae4443aa84ee3bcd558b42e7 : SUCCESS in 40m 36s","accounts_in_message":[],"_revision_number":3},{"id":"d2d3e881ea439df6460c78f32723cd39ce5a16f5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-02 05:46:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d2df3f8ba3350f924e1bf95128a3b6260c901f2d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-03 17:34:15.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e21795a074ecf5422ff3113f04f674b2e3af4629","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 18:29:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b619ffe4d7c4999a3e50fb0af5ca69a : SUCCESS in 51m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5a67585a7114c2a9b0bac63d04223c3 : SUCCESS in 43m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/09a07ebcc1974b6ba1a708f149a73730 : SUCCESS in 45m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/98f7c0c2bb80475f9855049d569fadc9 : SUCCESS in 53m 36s","accounts_in_message":[],"_revision_number":3},{"id":"f0e463417123578944af88bacc562674974d799b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 18:45:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9342bfd3a6c54d8499c686b283ef9975 : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/8a0d847b79044981bc1c7dbf9ba5e0b0 : SUCCESS in 52m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b72c37d7f4384cf9898db3f05f8c2540 : SUCCESS in 46m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe671d1f4c5e4ea29e751e68657ee4ab : SUCCESS in 47m 11s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3bf367bd336547d482f9b1a0919385cb : SUCCESS in 22m 12s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cf433fe31837449aa2d17487fadbc17d : SUCCESS in 22m 52s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/30372bd8c02a42e0b44e74f3206ee464 : SUCCESS in 21m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d48754b5c8494dfb885936df2e1287f4 : SUCCESS in 7m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ced326a1b20947ea998f09c9427afeb8 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3af9f6ff2c554b50990eb2363a04a07d : SUCCESS in 16m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/20e9617762604f96ab04f1755f72ab93 : SUCCESS in 26m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9dc1c7664b904c37bb3a1b8a8c941407 : SUCCESS in 19m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4279fdce76e2401e955efb99024c2e92 : SUCCESS in 19m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f093ced965c24b65ba98dc081bb53781 : SUCCESS in 17m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/40dc85b725a84b069f4d6465bafbcbcc : SUCCESS in 19m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4071248ef754478a98a0a451d8feb7a9 : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b1173df2128a42edaefe6d6bd3e732ee : SUCCESS in 14m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/26f69b8079a44b638d53aa52addbe7ee : SUCCESS in 40m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4720975392a648f98ef17b6ba3d1261e : SUCCESS in 6m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b0d5261f599643dfb8b28879cb33532c : SUCCESS in 15m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9ac92ed7eb7f42babbb854e6998d1222 : SUCCESS in 59m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f5a753e503240e3b26b7b176b53ca70 : SUCCESS in 1h 04m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a98b8e08f924e48b353616d56fa07bc : SUCCESS in 41m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c67aead9c484022beee6adb31473f6b : SUCCESS in 49m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4961266710454b0292cb0df98a8ce879 : SUCCESS in 16m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1496d250b7e143cdbc4c153eca246228 : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cc0e90bb5d634140a3f95615b9320697 : SUCCESS in 26m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea0ee50b69804e03a8d476bdff75b606 : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":3},{"id":"5a698be17935ab01fba0411966485e857717547e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-05 05:42:28.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"da6187cc4401eff6d2b11a35676da1e5f51839f1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-25 17:22:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"553cf4decd3c24e273ccb2b2ae01699eb1bb7bd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 17:23:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a992dd3802d2f5ed2e4336be3786bea72b932a83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 20:46:33.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e04187186e324d7499fb41f42c75f4be : SUCCESS in 8m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/3584d3994cc345f487caf6e7f3331bb2 : SUCCESS in 44m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/57e4b895a56b46689a8827db47bd7ec5 : SUCCESS in 38m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c3f306439da4239b7f9487572f18b76 : SUCCESS in 42m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5c7af61ad5ec402fbc5c7c27691ff72a : SUCCESS in 7m 19s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1d78b6ed62a44b02a9562eb8dc9a15d5 : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/01abee97412049caa3d8cc153d33483c : SUCCESS in 17m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bb8a5e2fdc9f4ba49f5b40feea42e2e6 : FAILURE in 3m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5ecb9b3e0a50496e830895a220eb3705 : SUCCESS in 14m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f43ba9eda4db4a72bcd8729040ab3d44 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8b19cecd3fec464b8943f3f39887b42a : SUCCESS in 15m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4bb0bfeaa9a44d4fb6b362ebaea6a199 : SUCCESS in 20m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fb326ade82e7400db92f3ece9c44c984 : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1908e164d4bc4e6b872028e7128987b2 : SUCCESS in 20m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/47c26999e17942f391f2ef5fd8419a2d : SUCCESS in 14m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1d49745d23624b5a9d8fdff94cbf80f8 : SUCCESS in 6m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/39b4eb69d7e54dfb85e6cb9366fbf1b2 : FAILURE in 57m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac6e888711a64329951a4a066967bfe0 : SUCCESS in 54m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e4271e428ae4eaa866aa19c2ba6e6f6 : SUCCESS in 49m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7eaf85e6846e432781962177ebc275e0 : SUCCESS in 43m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1925dad03cd94d09bc1a88c86fc6615e : SUCCESS in 16m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8614c11910424dd6ad9305ba3c1be1bd : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/83a2dbcbf1f54578a3513f57df5f1386 : FAILURE in 8m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/365d92656d6d48e2905b0a45df7dbbac : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":3},{"id":"9d01e4bb5856119fa7aba36714262e9531768fe8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-26 19:46:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9c9639c508f97a1ef03403dc6cb1dde1e956c5ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 20:37:32.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2eda6ac63a5a4eff897a1aae2b878e3c : SUCCESS in 44m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0d726ebd924b4968a001c24689778a87 : SUCCESS in 29m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/40d83e9db69a442b98d1040b8af13d54 : SUCCESS in 34m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e1eb1640a3e42d98822192c00cf9687 : FAILURE in 49m 03s","accounts_in_message":[],"_revision_number":3},{"id":"556d65349346726df9fab9a76c06ee9a7b7fb444","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 20:44:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fef5f1283b344098b26f42b000b83c3 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/dd95f2199bcd4217b942a753f16bf793 : SUCCESS in 50m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/855a7a76f8524ea2b9429a19d63f6140 : SUCCESS in 46m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd4a789efbd0452296b409b67c5d938c : SUCCESS in 43m 53s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/15535cef626d47829d6b04687651e749 : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1beb5ef6941d430db5ca8eb0ca5c98ff : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/37fe5928a5e046a7af17503f0d735196 : SUCCESS in 21m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e5710fb9b624235846bebba4c18123f : SUCCESS in 7m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d532c860842c47769e918cd3812830b3 : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c1cefd2faab4031a94dbdc614f55258 : SUCCESS in 17m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/41abf90078354c5fa0434df05669efbc : SUCCESS in 16m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8548cc7d4f68425e9bb94ba67974f031 : SUCCESS in 13m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4677187a52da42a0ace4fe0e18f78667 : SUCCESS in 20m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2673d1557bf34bbf8978ce2b98364e71 : SUCCESS in 14m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da2ddaf44a6d47ac83053ceda1ebbf4b : SUCCESS in 25m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/177384da77344bab8af5a22c7f48fb64 : SUCCESS in 19m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/980b030f22964b97849ae1064fab16c4 : SUCCESS in 13m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6333834671b54b7887fb61ef486adf0f : SUCCESS in 20m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c2c85a7de9044e61807c4ea32f61a9b8 : SUCCESS in 5m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a75d9574adfb47c7ad0c2bb63dbcfb68 : SUCCESS in 19m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ce6231419ffa4f15a595289ec0b8b63e : FAILURE in 48m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a1f5197960144889baabb5ae90214fb : SUCCESS in 55m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76712f9f86ce4c07811e00106fbba522 : SUCCESS in 45m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee4e3df269774addace83be62690f803 : SUCCESS in 34m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da526cb24f214bea826b536187d2f7f7 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d62a29d6f8744936a60fd597bf10ad8e : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9e429c1da79a4554ad2418ff36ef51fe : SUCCESS in 27m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3049551fe8a24c2d9d8350caf97941c2 : SUCCESS in 40m 39s","accounts_in_message":[],"_revision_number":3},{"id":"c28f17f12b166600f2a635ac2d34c0a865e79528","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-27 15:58:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8da63147a51de8f31da7fb58dca9c24dd83251c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 17:08:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbf186226c2e4c99b7b5f951fd7b530f : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/ddbc5622094e4b0fb4f60d5af1776a17 : SUCCESS in 59m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/80fc208a382a4cb9aefe942a78ae4674 : SUCCESS in 1h 01m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0788005eee94c4099b1a47b7d01bd61 : SUCCESS in 56m 00s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/807a19785cd143aa8b392eafb8b04141 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9f67e37d32ef4b53be99663c6711ac27 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/bb152d6aa29744039338ec0f5ebfbb46 : SUCCESS in 18m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/73447312ff7d4e5fa17c98f9af11baaf : SUCCESS in 7m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/de58f5883ed7405383292b8262c3fe57 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/02a136b15c834cc29616b4177fbae21b : SUCCESS in 15m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b1229e11a2534627a0994063315451a5 : SUCCESS in 16m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b2909d49492840b684ed6e2178851b5b : SUCCESS in 13m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f2c77ed25b28444ebf490766085cc3aa : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/87ca83e704af4937bda091bcfd876db9 : SUCCESS in 15m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f7c3d513b0df4e978410af5c25d0dabb : SUCCESS in 19m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7e5142c3f9cd4baeb7cf3253e80aee09 : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/55f1450824ec4a2b99f8b013880f345e : SUCCESS in 16m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/49f2fba8ec7a42fa9a38a25ed78afa8f : SUCCESS in 16m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f5e19aa38b474b528482aac392279fa8 : SUCCESS in 8m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dfc028eef1014187b17c58a95f8be53a : SUCCESS in 16m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5b8e219beaeb4d8881d54b716b46e9b2 : SUCCESS in 49m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/be2f954b467846d19cde7102c93f3028 : SUCCESS in 1h 00m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7d6f16d1445442a48c77076851b733ed : SUCCESS in 43m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/00441f8d71fe48f59448884288831e24 : SUCCESS in 46m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d5137a66a7948468384768d4ed380f4 : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae140073e3341eba27069409fbaac1c : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e6669ce2c9cb4c88afff754e1af5872e : SUCCESS in 19m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a44f533568944ba888e39361b7967193 : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":3},{"id":"aed70e6dd909f11d5226d96506a5e49be6a19fbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 17:09:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b8e226b4d8e9f66f9b060fa5a091bec8ff427509","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 17:37:20.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/479a43ee702b41088ea2c484492a0dfb : SUCCESS in 1h 18m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9763b9ca4f294a759d42037f334fc36a : SUCCESS in 1h 01m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/43136ba762b14c2fa00b14544656a4eb : FAILURE in 1h 16m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c6e76ab28d249f3824d73704cf63b34 : SUCCESS in 1h 07m 16s","accounts_in_message":[],"_revision_number":3},{"id":"10f3885784572bd9af23a2448e71a1ed2fda9e1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 18:32:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/376e56ff95af4997a8a5ebc1285cd973 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/79c2495e62ff446a8f112aa8f19cb63f : SUCCESS in 1h 00m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f78a4a7aa2be474791436ab18613564b : SUCCESS in 46m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d85f84e64884e4884d22585c5f4fc13 : SUCCESS in 53m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/42fb9181b89f4a4fa23615d827dff003 : SUCCESS in 10m 05s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1eff656891394ff89126c5312de79dd8 : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/15742342809140909425256f86ed4b57 : SUCCESS in 21m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8964f38a8dae4813a860859aabb048b7 : SUCCESS in 17m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2ad1d14ed48d42998a609deed92935ce : SUCCESS in 16m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f603b5fd0d984851bc6301c451eba25d : SUCCESS in 17m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/050ea275d420417892cde58914f42e9f : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bf1dabba982c420ea589f796e8db5c35 : SUCCESS in 18m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ecd9a3d5267b4abda707bb2dfa9c136a : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8af1b99ebe6b46ff86cb25527b31eda2 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/317d5be1538047f5b1d249b735cd04a3 : SUCCESS in 16m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/388e990d88be43f189de4c96bd1fb390 : SUCCESS in 7m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d95861eb10eb443b9f7b41e7d020ecae : SUCCESS in 58m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/94fc34c207a44db587dc823266b0a96f : SUCCESS in 59m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/19253b799076434281177eb2ad595e8f : SUCCESS in 42m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/11de2a9e777343bc9754264e91858ff1 : SUCCESS in 43m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a32b48b0bd224b4e8f07f4ab72de34f1 : SUCCESS in 18m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6ce0129c4c0418da40448affa0f17f9 : SUCCESS in 6m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cc3269d63c0b4301b2775f92790fd13c : SUCCESS in 22m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/622333e697fb4e7485978ce9b56c6ed3 : SUCCESS in 46m 46s","accounts_in_message":[],"_revision_number":3},{"id":"6a5d9b9fd8d7f1b302af0a84012d3d21435e5496","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 18:32:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7e691768171be9ccf0a1f6f5ac7630e83b5c9344","revisions":{"747e99896ca5ca497fa7c27e89015242a1a3732b":{"kind":"REWORK","_number":1,"created":"2022-04-26 20:45:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/839442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/839442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/839442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/839442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/839442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/839442/1"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-26 19:10:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-26 20:43:27.000000000","tz":-420},"subject":"replicator: Log rsync file transfers less","message":"replicator: Log rsync file transfers less\n\n- Drop log level for successful rsyncs to debug; ops don\u0027t usually care.\n- Add an option to skip \"send\" lines entirely -- in a large cluster,\n  during a meaningful expansion, there\u0027s too much information getting\n  logged; it\u0027s just wasting disk space.\n\nNote that we already have similar filtering for directory creation;\nthat\u0027s been present since the initial commit of Swift code.\n\nChange-Id: I02ba67e77e3378b2c2c8c682d5d230d31cd1bfa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/747e99896ca5ca497fa7c27e89015242a1a3732b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/747e99896ca5ca497fa7c27e89015242a1a3732b"}]},"branch":"refs/heads/master"},"addafdf973b7d749a3c94819a3547a4ceedf4c3a":{"kind":"REWORK","_number":2,"created":"2022-04-26 21:59:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/839442/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/839442/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/839442/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/839442/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/839442/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/839442/2"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-26 19:10:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-26 21:55:52.000000000","tz":-420},"subject":"replicator: Log rsync file transfers less","message":"replicator: Log rsync file transfers less\n\n- Drop log level for successful rsyncs to debug; ops don\u0027t usually care.\n- Add an option to skip \"send\" lines entirely -- in a large cluster,\n  during a meaningful expansion, there\u0027s too much information getting\n  logged; it\u0027s just wasting disk space.\n\nNote that we already have similar filtering for directory creation;\nthat\u0027s been present since the initial commit of Swift code.\n\nDrive-by: make it a little more clear that more than one suffix was\nlikely replicated when logging about success.\n\nChange-Id: I02ba67e77e3378b2c2c8c682d5d230d31cd1bfa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/addafdf973b7d749a3c94819a3547a4ceedf4c3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/addafdf973b7d749a3c94819a3547a4ceedf4c3a"}]},"branch":"refs/heads/master"},"7e691768171be9ccf0a1f6f5ac7630e83b5c9344":{"kind":"REWORK","_number":3,"created":"2022-04-28 19:35:21.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/42/839442/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/839442/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/839442/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/839442/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/839442/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/839442/3"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-26 19:10:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 19:35:00.000000000","tz":-420},"subject":"replicator: Log rsync file transfers less","message":"replicator: Log rsync file transfers less\n\n- Drop log level for successful rsyncs to debug; ops don\u0027t usually care.\n- Add an option to skip \"send\" lines entirely -- in a large cluster,\n  during a meaningful expansion, there\u0027s too much information getting\n  logged; it\u0027s just wasting disk space.\n\nNote that we already have similar filtering for directory creation;\nthat\u0027s been present since the initial commit of Swift code.\n\nDrive-by: make it a little more clear that more than one suffix was\nlikely replicated when logging about success.\n\nChange-Id: I02ba67e77e3378b2c2c8c682d5d230d31cd1bfa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e691768171be9ccf0a1f6f5ac7630e83b5c9344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e691768171be9ccf0a1f6f5ac7630e83b5c9344"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
