)]}'
{"id":"openstack%2Fswift~839448","triplet_id":"openstack%2Fswift~master~Ic8d0915235d11124d06ec940c5be9a2edbe85c83","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-05-13 14:32:03.000000000","reason":"Alistair Coles replied on the change"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-05-13 16:44:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8d0915235d11124d06ec940c5be9a2edbe85c83","subject":"memcached: Give callers the option to accept errors","status":"MERGED","created":"2022-04-26 23:24:55.000000000","updated":"2022-05-13 16:48:53.000000000","submitted":"2022-05-13 16:44:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"839448","meta_rev_id":"63c81a3ef8d69671db30b3f04c0328ecd6d9e21e","_number":839448,"virtual_id_number":839448,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-13 16:44:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2022-05-13 14:32:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2022-05-02 22:52:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2022-05-13 14:32:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 01:28:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 09:30:37.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2022-05-02 22:52:14.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"b6937b65b6a980e051371dd899ff670fa08c6e23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-26 23:24:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4093bfdfa1cc6d289dce61980074f9f54bc1b3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 01:28:02.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/699419de683848e6b59b38129e4cbfbb : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/67af7d22ed624f51aec9353862b3dc67 : SUCCESS in 1h 21m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a25196862e14f1b9ded809938ea2d59 : SUCCESS in 52m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef1c5cb1e2934422a6ab5f3df21b258d : SUCCESS in 59m 05s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/76e4aed4b2db4b0085d5931fb987f46c : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/db399c92795241568c219018294b56b2 : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/10425daa0ad449f381e7e29fcf9c3d09 : SUCCESS in 19m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c0b346812434ede836e1c9bd1a2d16d : SUCCESS in 7m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/058c83c1b9e64c10a95d89a09fb85fa7 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2e8f955f7bcd4a9da14e158506b9dc39 : FAILURE in 16m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b370456ee5144ba09e0b5e4a02cb51a5 : FAILURE in 18m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/727b53f3354342e4b33741974a017f8d : FAILURE in 15m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8015b9f3c0d84edeaae65782d0739673 : SUCCESS in 37m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d43439c3515747e1b0a3a37bbc78d8eb : SUCCESS in 18m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c5ad0220a8194b93aeb01360e8021fde : SUCCESS in 22m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e122206792f74fc0ad83bbf40e367cc8 : SUCCESS in 19m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a7af106e1252402a9c9ea1a98ad1f097 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e6f243bc833d44f38ac0759fdc140d67 : SUCCESS in 17m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d85b2b5068f347328022720c4002996c : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/64707fff01cf4d878259516d0a3922dc : SUCCESS in 19m 46s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/327e4f5a25ec42f7a87d7c22d7662666 : SUCCESS in 58m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/94b343698c49450d8aac4c0d6988279b : TIMED_OUT in 2h 01m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/99d2f0aa32ee4e259054f2cda7675903 : SUCCESS in 43m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5765a9c6d4fb4ee983c5ab06083cecac : SUCCESS in 42m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4d2b1eacad741bd905fc7f7b653183a : FAILURE in 20m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e894ad7d9f654de28268e331ded41f18 : FAILURE in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/61d129b8c8f4489eb3a3112b839ed1c0 : SUCCESS in 27m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca3fe91a74d1472b910ee599a1d48cec : SUCCESS in 52m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a53a45d598e6d2b6c8137329ace993c960c8870d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 02:28:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0d5e7656b4841b69cd18ba95a140582 : TIMED_OUT in 1h 31m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f8de1dd733b49d4a0a4afe146e5702d : FAILURE in 43m 35s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/1ac9d0fef69e4837bbe66fe9a2196755 : TIMED_OUT in 3h 01m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c34b3f906c1544989d659b09ca65bed8 : SUCCESS in 1h 11m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/41049cfae838473a88e6a8caf249ee06 : SUCCESS in 47m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b624299aa8d34e6903297036ef1d1e9fd336fdf6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-27 09:30:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"42670166aa6f84c36c1449d118dfe5cc499b4e62","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-04-27 13:25:10.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2c0782cb2ae5565958b7c16a975a4f1e41faf22d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-28 20:59:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"74b120fa910e47410e71d98250c906cbd1c60a06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 23:03:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ae33d8756d9482e8ced7937265a687b : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/c386dda1d8224441958cb55a735d8bb3 : SUCCESS in 57m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d42b8447f33c436998daa316004aa73c : SUCCESS in 40m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f760683ff6544162b74beeea8906d30b : SUCCESS in 45m 08s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cb23c82260f94b3b9764421abaeee78a : SUCCESS in 22m 59s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c1e4ec8647c44044b9edf99fb043d805 : SUCCESS in 24m 11s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9b11690ffdaf46799cbab464bdc6d945 : SUCCESS in 23m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e831f7203ae944578fccd33a9d93a205 : SUCCESS in 7m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/435d3d296919496a96a0f1cc78247bb3 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c97a55668711450886c175c18ebea919 : SUCCESS in 21m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/431b500707504c2cb1cd7717fb8d62e0 : SUCCESS in 21m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c3a59b82dc614105a47c386c96d356a8 : SUCCESS in 20m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7ccb3513520d40eabdbdbec2dc96899c : SUCCESS in 24m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6ee38b46c9ed4972b68c774342d9ac08 : SUCCESS in 21m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eeeab6919e2b4b28b0a102cc535f188f : SUCCESS in 25m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bf58e9502d2e43cc956e7cf53ea18829 : SUCCESS in 19m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3bd3ad26042c438ea9b2a8e24f782e6e : SUCCESS in 21m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/404eb20d4fe443a39ca1861171bb5738 : SUCCESS in 22m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a1dee0d60c074792bfcf85d97b757d9d : SUCCESS in 8m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c26131e296d64f94a7c7670f7468cbe8 : SUCCESS in 19m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c544f8cf5d854e85a1899a2cff2d5574 : SUCCESS in 58m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b3f288390c14855967fd37f52b9cfb1 : TIMED_OUT in 2h 01m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e1adadd41b14423a05972c3094b911e : SUCCESS in 41m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f1391ed4a61c46ed8e0573571451195b : SUCCESS in 43m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84ebdbcca3844bc582dd58dd75f19150 : SUCCESS in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77e3e5075da74081980a08f784fa0b35 : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/af215a045aae4bd489445a865d6c205f : SUCCESS in 25m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d25f4fcda3e54460ae1a5262ca900705 : SUCCESS in 44m 49s","accounts_in_message":[],"_revision_number":2},{"id":"41a9f426b4490d3dfbb4733085b6f4c7e2156af2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 00:03:38.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3f9262563624ed296612d137ee22282 : SUCCESS in 59m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c46861a3f6fe4942b624a8aefe4fc84a : SUCCESS in 49m 49s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/47b7c835ad634903b799d6b6d20247c0 : TIMED_OUT in 3h 01m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6013832320e470ea8161470cd283de0 : SUCCESS in 48m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/30a92a5be49d4d5a8a24edf06ec92121 : SUCCESS in 1h 01m 42s","accounts_in_message":[],"_revision_number":2},{"id":"c16482b1319321c898426626638bc9c90bf77cab","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-02 22:52:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3f8d639b6ce15f4c9bf3f1680d2707c309548a59","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-11 20:58:58.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9a7231104b90ab2802e0ed63ba923a08b336543f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-13 14:32:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"296673d39fb3856d6aa62b848cde484d58be8a36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 15:41:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65ca82eea2e84446a66fa0dc1c2e6b6f : SUCCESS in 9m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/b414a65345b3406ea6f2059a145c206f : SUCCESS in 55m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a39e2b7ff8af4fb19fd67f92fc5fa882 : SUCCESS in 47m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0434efcad82b46ca9cb6c657926e465a : SUCCESS in 48m 24s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e45aa72f29154346af64110d40481a45 : SUCCESS in 28m 35s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ab9e3918a8334b279fbec46e08f2df1b : SUCCESS in 29m 41s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a9ee5df1a414433e9b82bf77c01c7ad1 : SUCCESS in 24m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/04647c8c578b4d1593e3b86b5eabb741 : SUCCESS in 8m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d84f87cd2b054b969def0ea126c1671d : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/60f8f19fbf1f4056b1ec24ac22e11f24 : SUCCESS in 20m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/158d58fc65474e118092ec05f9edaa8d : SUCCESS in 15m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a59b07b678cc42c180f59da29b3b6988 : SUCCESS in 16m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ea66589b04254300abcb6450e5b067eb : SUCCESS in 19m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bcc41aefcd924aefb2aeb7c391b7c294 : SUCCESS in 20m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/83e2a550b72a41468993cea443ba1f59 : SUCCESS in 20m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1d9dfd5190e84f7bacac16af1c7d0794 : SUCCESS in 19m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/89ab43bd06964f4883dc652ed5879e98 : SUCCESS in 20m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cbcf5d8214cb4707b355ffc651672e30 : SUCCESS in 16m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/053d11add3894fdbbfd1840b32e9ee33 : SUCCESS in 6m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c7db8dc7a26d4cb4a4fbe61f3741bba4 : FAILURE in 19m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/21b07c6074c54a8986aa7db3a359ac8c : SUCCESS in 48m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b517c4833a564fe5bf2257ec59e13755 : SUCCESS in 56m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a2639a0b2ce4be6b42cbdbaf385020a : SUCCESS in 39m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c83d1932b3954aecbaef249dbf95f7de : SUCCESS in 41m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3e9f95a16654405b1e0968d70c37fb8 : SUCCESS in 19m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc2b1e4781d24bf096c8f7da1e4424e0 : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/47dd8b30220045428d7f2c08c136a438 : SUCCESS in 26m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59217433d4cd4caf86baae45a3e0eeac : SUCCESS in 1h 00m 48s","accounts_in_message":[],"_revision_number":2},{"id":"3e568e47263012793808778adf60e87786ac433d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 15:41:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"558e7237f392c4cd0aa9b2d59d38ce89d9a46634","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 16:44:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77245f85351447c8ad7e804edda09641 : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/c2e5b7fa447841df92dbc0a8e3f1af15 : SUCCESS in 49m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/27284ed98458415cb9bdd1d03924fe64 : SUCCESS in 50m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8113d20bf3b84b6a8006989ec5aa39ca : SUCCESS in 39m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bf2deb1d5ea34742b982b05cd2e3ce29 : SUCCESS in 7m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fd5a0c996d9042deab6362f3b2307565 : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8dc80d3ffab845d790c6beb9800a6c15 : SUCCESS in 13m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/61db9dba849049368d91f7179e10c045 : SUCCESS in 12m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/01888da277924689b5c0c4d9370aa86f : SUCCESS in 13m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0a83a727712346b1a6827df7d0167861 : SUCCESS in 27m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c912d116b7e645f58aa500c8aba720a4 : SUCCESS in 27m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/96cd168cbacd4afba632ae35b3b470df : SUCCESS in 20m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/91a4e4562a184633b930df7075492a40 : SUCCESS in 20m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c7c3bfdcb1ee4c4d981cc48fe4140097 : SUCCESS in 17m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/500a226da34c4eff8aaa7154b95aff34 : SUCCESS in 20m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8fd16e974b0d430b936d555298412a23 : SUCCESS in 6m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2066172550754244918b7bed1b7aa50e : SUCCESS in 1h 02m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/251c8ba8407f446593c6eac7f6eb0eee : SUCCESS in 56m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/852fd1fc750544648c2169306f4f4a1c : SUCCESS in 46m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e0b1f9b7b12644c387a1ada831750f38 : SUCCESS in 41m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1faa67121934341ab4cb6ffb5541d82 : SUCCESS in 16m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a0de0b7488041ed81d2c4be3badc419 : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0ad5a463e8b441a4a92e6c5381f6d5d2 : SUCCESS in 26m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71ea444dbc774ff1baea47a8d9dd6759 : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":2},{"id":"83991640318a99b7e12352f4da8cbf33bc8f1547","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 16:44:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"63c81a3ef8d69671db30b3f04c0328ecd6d9e21e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 16:48:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdcfd63fe757452f948201a4edd922bb : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4f34fb43e5b449369390ea6dda3887a6 : SUCCESS in 49s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/527684ee6a9b41c6b83b9343ece4e080 : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9bed525bfbe3ec05cb45c9835333fe45f8e04b40","revisions":{"46a49bf2e43cfca59184f6a10410205bc7b21e6d":{"kind":"REWORK","_number":1,"created":"2022-04-26 23:24:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/839448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/839448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/839448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/839448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/839448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/839448/1"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-26 23:12:49.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-26 23:24:46.000000000","tz":-420},"subject":"memcached: Give callers the option to accept errors","message":"memcached: Give callers the option to accept errors\n\nAuth middlewares in particular may want to *know* when there\u0027s a\ncommunication breakdown as opposed to a cache miss.\n\nUpdate our shard-range cache stats to acknowlegde the distinction.\n\nChange-Id: Ic8d0915235d11124d06ec940c5be9a2edbe85c83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46a49bf2e43cfca59184f6a10410205bc7b21e6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46a49bf2e43cfca59184f6a10410205bc7b21e6d"}]},"branch":"refs/heads/master"},"9bed525bfbe3ec05cb45c9835333fe45f8e04b40":{"kind":"REWORK","_number":2,"created":"2022-04-28 20:59:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/839448/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/839448/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/839448/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/839448/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/839448/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/839448/2"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-26 23:12:49.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 20:20:44.000000000","tz":-420},"subject":"memcached: Give callers the option to accept errors","message":"memcached: Give callers the option to accept errors\n\nAuth middlewares in particular may want to *know* when there\u0027s a\ncommunication breakdown as opposed to a cache miss.\n\nUpdate our shard-range cache stats to acknowlegde the distinction.\n\nDrive-by: Log an error if all memcached servers are error-limited.\n\nChange-Id: Ic8d0915235d11124d06ec940c5be9a2edbe85c83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bed525bfbe3ec05cb45c9835333fe45f8e04b40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bed525bfbe3ec05cb45c9835333fe45f8e04b40"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
