)]}'
{"id":"openstack%2Fswift~839649","triplet_id":"openstack%2Fswift~master~Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa","project":"openstack/swift","branch":"master","topic":"delete-timing-breakdown","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-08-11 18:26:14.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-08-20 22:12:43.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2026-03-05 00:44:13.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2025-07-30 18:44:13.000000000","reason":"Reviewer was added"},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2026-06-02 14:35:58.000000000","reason":"Reviewer was added"}},"removed_from_attention_set":{"12050":{"account":{"_account_id":12050,"name":"Charles Hsu","email":"charles0126@gmail.com","username":"charz"},"last_update":"2024-11-19 00:34:43.000000000","reason":"Change was marked work in progress"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-08-20 22:13:01.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa","subject":"replicator: Add sync_batches_per_revert option","status":"NEW","created":"2022-04-27 23:28:52.000000000","updated":"2026-09-03 18:05:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":204,"unresolved_comment_count":16,"has_review_started":true,"meta_rev_id":"bf4be5c9e1516c8397518c93318e5035b7acd380","_number":839649,"virtual_id_number":839649,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-03 17:52:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"CC":[{"_account_id":12050,"name":"Charles Hsu","email":"charles0126@gmail.com","username":"charz"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 01:33:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-23 01:58:44.000000000","updated_by":{"_account_id":12050,"name":"Charles Hsu","email":"charles0126@gmail.com","username":"charz"},"reviewer":{"_account_id":12050,"name":"Charles Hsu","email":"charles0126@gmail.com","username":"charz"},"state":"CC"},{"updated":"2022-06-27 21:55:22.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-06-28 06:58:47.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2023-08-17 16:15:03.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2025-07-30 18:44:13.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"},{"updated":"2026-06-02 14:35:58.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"REVIEWER"},{"updated":"2026-08-20 11:50:34.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"b86cd3611ce2e5304449e49ea0d3c9c77f5fd98e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-27 23:28:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efdedb3f24878265526ce399868bb7c64df067ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 01:33:28.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f67a735c1f4a433e8ba2be17cff2d3e1 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/358359556df64892afba4705648c25a6 : SUCCESS in 1h 02m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/182e8e3de8854a959934efc9ccd4cb4a : SUCCESS in 51m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc9e66ff39884ec8857fa9a5fcd50ad5 : SUCCESS in 51m 44s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1dc49b6ac27e4725ace007a3944876d7 : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/749a641f774043fe8de4802e0d33ed60 : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/85648dd95ca7461997560937384060b2 : SUCCESS in 20m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e79336e0d80a465fa2466b19543e65b5 : SUCCESS in 7m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ddf9ff0796394a668b426ba665c4dffb : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f15aebc972764b939d42e7a36bafa021 : SUCCESS in 20m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/52ae594d0c47432698143530ed660bb3 : SUCCESS in 13m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9b643e08166c46f1b01d0598c3341fbc : SUCCESS in 36m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b3f7c7fad0564698a05dd9842ab24242 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/05438dc94fe049bcaddf62a373143ccb : SUCCESS in 14m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f9a33ea3223940c7b869e93e15a11949 : SUCCESS in 16m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/12604c5d2976492aa3ea86d91b4ad791 : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6c17b6a9778e40c8b5b31001be7aa331 : SUCCESS in 14m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8960342281c2457891887d397b2b4094 : SUCCESS in 13m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fe25ce47b6904c81a0432a94c483c0ef : SUCCESS in 7m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/166aade7436740fdb9b55f39fb22c1b6 : SUCCESS in 16m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/907d66fb40984d01b005bc7c84b95f7f : SUCCESS in 50m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4835beaa9121418eb6ef9dcf85acb664 : TIMED_OUT in 2h 02m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c57ef486b6d443c6802bbcc37dd0cdf2 : SUCCESS in 27m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/801ba630127840acb00f4bd7bfa73776 : SUCCESS in 45m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5ce395d2a46400a98169b0a93f2c6a3 : SUCCESS in 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3200aa06c3542bb945d65648e497fdc : FAILURE in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e70c9b1d75c4caaa55138bb997fb4ee : SUCCESS in 16m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1af1b4fd315442acb66893c9e2d9fac9 : SUCCESS in 45m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d95b3624c6e09a1cf9823905d055840ee7b212d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 03:28:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bac0de38554943718aa0aa3ccde472c8 : SUCCESS in 42m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29ada90a8f1948af9312bb3f753cbeb1 : SUCCESS in 36m 05s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/dd4e2cff42044681ac2652ed39d2b626 : TIMED_OUT in 3h 01m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d75a5dd1d2a94edd8c7bcbf855104630 : SUCCESS in 50m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1e838d93e7c408c8272f621d42f8964 : SUCCESS in 47m 24s","accounts_in_message":[],"_revision_number":1},{"id":"939513cd0e20261b938769f285069da36003a31e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-03 18:03:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9a476fcd8798d213719596490231e751dda286a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 19:12:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ec66aea246e443882d8d3d19334bf69 : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/3bb432ac6dfb422fbaeb50e042ea4dc4 : SUCCESS in 1h 01m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a274109aef37487da2f3c1e59fb46b63 : SUCCESS in 48m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/148eb09736944f3a802292f0043ec5b2 : SUCCESS in 49m 16s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a24f18ed9556429d9b6e129f91729c9b : SUCCESS in 16m 56s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/749982d344d64aff95352e7b61426037 : SUCCESS in 18m 35s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/161841ad666f4cc69ddf5a5e7caabac9 : SUCCESS in 20m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/236b29627fa744f0b2a1996bdcd22cc7 : SUCCESS in 6m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2d0f36c882744204844c9fad913e107f : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ac1bf32a85c24a4ba08dcb04b5863af5 : SUCCESS in 15m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f5e15b1a851b4d288d1a03ada3adec23 : SUCCESS in 15m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c9216f7f2a74409aeb7614c2529c926 : SUCCESS in 18m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6f1e63b798b44c3784e08d7ca6cc8387 : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7d7561304fd84578a81b33a64292027d : SUCCESS in 17m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/55d433cd8b4d49b3a1cd4d7bf1a4a658 : SUCCESS in 17m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5acc48f387d24db1a3227f7c9181dbb6 : SUCCESS in 21m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c54fabaecc81422b8f931216d8820427 : SUCCESS in 16m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/87283e85b241419083015ea59d169eb1 : SUCCESS in 15m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cd479ee8bf894a2297d54a76ff328665 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aff64d199a184622bb871405b0c53778 : SUCCESS in 17m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0ed3c8cfc95b47c2921f960b952b1cf0 : SUCCESS in 50m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/39e2c923e5c3463a8e8fe81ace76cd47 : SUCCESS in 53m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3ca991034db430fbaaaeb71a545c32b : SUCCESS in 44m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a7dfa386d3cd43868ff18e10a2d6688f : SUCCESS in 41m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/456e727235024e378722b7c072adcd05 : SUCCESS in 14m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be7d0f2669ce4dccbae5da3ef822449d : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb8e36ad31bc4fcaa326f466f7f9130c : SUCCESS in 20m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4ed2a4abe0c4c59bc8b07e34c25130f : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":2},{"id":"c57d4781fcf91e7807cac88bc6fedef9cec0b916","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 19:53:28.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a1587e76b2a49e8913d0f79536147b2 : TIMED_OUT in 1h 31m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e06e1afe49154573a83b0e81c5365bb7 : SUCCESS in 44m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc9a9086ccb94bcc8039fa35322f7808 : SUCCESS in 50m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8333ab3aa8884c9dad9c446ef66a3577 : SUCCESS in 1h 06m 18s","accounts_in_message":[],"_revision_number":2},{"id":"f5ff65584d5a2e384dd9a58d0810167a3236d370","author":{"_account_id":12050,"name":"Charles Hsu","email":"charles0126@gmail.com","username":"charz"},"date":"2022-05-23 01:58:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"04b44b3265aad3ee583e4fa62cc19d15243e39e6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-23 16:23:44.000000000","message":"Patch Set 2: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6853bd0bccc5563efcaad2d77530380d97b699f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-28 06:36:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0ac4b369c694dbe241d64aa8df333549e6b7ce8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 07:22:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7742a68274854335a325cf9399bbfc44 : SUCCESS in 37m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e6b0385c1adb40c08d97478263b23872 : SUCCESS in 28m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/62acbc56afcd4518b84a900f6258ef2d : SUCCESS in 43m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1df86114dce64b1bb1c4b83153b26f01 : SUCCESS in 44m 29s","accounts_in_message":[],"_revision_number":3},{"id":"887050f809d29347c008e280a01780c667832a53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-28 07:44:27.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45c0a749c71240d9b9f2f42bc6d9bc32 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/b2b4c19236af4de5930f96ff24a991d5 : SUCCESS in 1h 02m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/96f7ad1a97d44f34907f953701850614 : SUCCESS in 43m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9234b0ffa4846c48a66a89a22553f16 : SUCCESS in 46m 04s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b88d8ad1a06f4e55a87a79ed73dbf7eb : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/15f5fd2f4c79446caa989946e84ee558 : SUCCESS in 21m 30s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e8c193e6c6de4f23a4050660971eebf8 : SUCCESS in 23m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b0fd4c69f63044b8a73c0eaa10cc5dc9 : SUCCESS in 7m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dbc5fa702e554e8f86abce11103c093f : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/517619d022674bdab308a1f21b98c643 : SUCCESS in 16m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e7132f09d42348ad823d3496dcb8e88a : SUCCESS in 19m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b46599514d934d0eaec282e9d5e42d9a : SUCCESS in 15m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4a2ce2ab75bd438f9ed7f475b4df1c7c : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ff24179d368f4ba9a2aa1cb786138592 : SUCCESS in 18m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3acc5c9e6da94a94ac5178f4b7acbc7b : SUCCESS in 21m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b1aa6bb6f42b44d6bc8653fd023067f7 : SUCCESS in 19m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/80472f17e7ff44e2aa9695fd211bc834 : SUCCESS in 17m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3d830c972d5e42108fb2d6ae033804a9 : SUCCESS in 18m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/92e3e56255924303b96154395dc6dbc8 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/534bf7ecda9444f3a1c6ecda4c1a1118 : SUCCESS in 18m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/570a4d1d8db048058a8011ea8f37d5d5 : SUCCESS in 56m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/851d706e0e4142368cb2863ee6a5a30b : SUCCESS in 1h 00m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c56141740c854fa38e16623f15d7abb5 : SUCCESS in 40m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/651b1f1d339a4c55808fd5bcb9c14108 : SUCCESS in 39m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a7a8cb4fb324b108aa3c014e7ca80a3 : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fd11f97203f435b8414d7cf2a3cfb4b : FAILURE in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e44375e55f14af19bbd4543b1ca3e8f : SUCCESS in 24m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8a84aa65bfe486daed93c850db84378 : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":3},{"id":"e3d8e535bb59f98c81da6220a2353c06407d64c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-31 15:24:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c12d1cdcb2fb004c9c88957d72e9e4bbb113f5a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 16:16:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/716998ca34334f168411dda6c75dfc34 : SUCCESS in 35m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04b4ffbd83034181a221b4594cbafe46 : SUCCESS in 39m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8a34371fddf4fdb9606c5717f94e1cb : SUCCESS in 48m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce64b18ca7e64dbd8ad7da034a15313f : SUCCESS in 44m 37s","accounts_in_message":[],"_revision_number":4},{"id":"2e5751866d93284e7b2dc161820c4c808a9c3d8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 16:46:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6578a17fe9494dad8cb61ad8b794628c : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/b17aa0417eb442f79323123f00b04227 : SUCCESS in 1h 12m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/60c7f4b267184748921983069848f80c : SUCCESS in 1h 01m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ac10580626a4b7eabd963c65c3aeb8b : SUCCESS in 43m 49s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fb094fb61a0945df9065531c374370d0 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1230b93029a147a8859a144c664b7ce8 : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ad25197d7ebd4ebca993e5c72759d79e : SUCCESS in 19m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bf518e27073f4d22823ccd7a6f086161 : SUCCESS in 10m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2c48a98a3f884fc1b6b26fc0a85c02fc : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/025ce78d2371490f8ecb37ac55ff1ab7 : SUCCESS in 14m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2f1aecd1bc134bbcbf09bb980b8e9879 : SUCCESS in 13m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/518448d57dc14ef191dc2f4cede33182 : SUCCESS in 29m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9a2729e43dbc41668223e95f0982321f : SUCCESS in 35m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eac1b71954bd42eca488d3ab64359c2d : SUCCESS in 35m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c1447e31c2a942ccb020ea2dfefc7481 : SUCCESS in 16m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7505a8bae3dc4632932669c592a49fb8 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/59d217f8b147458b8394fbc5cfdc1cdf : SUCCESS in 20m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a0ffbe4045864d7b8ae6da7a2ef6cce6 : SUCCESS in 16m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ba3dcd3192fc466d8ee20d6eef5d86c7 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ed27a21058ce402bb01881d9d93aa414 : SUCCESS in 18m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c77a8007293d42269d90f04f9d8bcdf9 : SUCCESS in 57m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0039af6e7d6d4e26be255be2e80ff450 : SUCCESS in 1h 15m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/955a329dad1a4d7c96d01be87fb1b976 : SUCCESS in 40m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f805d43bf37a44ec867df053ef8c9d7d : SUCCESS in 51m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a223f54f94784593b3493a345698a229 : SUCCESS in 19m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c56223f95e450a99422808f4e319b2 : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cf5e1c6ebcf644df910f92b04a0fce4f : SUCCESS in 25m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4b7f1590dc648a9bb6ae4cee15fbde2 : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":4},{"id":"cf42dbdb5ae3aebad94e1e5ec0ec9d0af4bf513f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-24 16:04:44.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"878d2fe04eb1b803a8a5ec7da7d9f31d55176ce7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-24 19:17:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e012514f992506c4f24222c05936124796824aee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 19:56:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6853876dfded4039a46eb0d73700d53b : SUCCESS in 31m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/805833dad3b140daa8f1688e0881cbe7 : SUCCESS in 30m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5789a5553d924077a6a626a23f0bc798 : SUCCESS in 31m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d76b247b76a74b9b8b3934b7b21eb9e1 : SUCCESS in 29m 36s","accounts_in_message":[],"_revision_number":5},{"id":"b6402ab7dc35cc10b0d18fe95d81bf950236a062","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-24 20:21:01.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ccec0bfbb9147b563cd41d80659641f88aff4075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 20:25:08.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95d93fb64fbc4b628b26cb5514b0fed3 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/ce77e5f6a1b34b0da9a7bd607764184f : SUCCESS in 45m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7499c2db140d43e6bd3d52048837f87e : SUCCESS in 38m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79101d1013e54dc1aca6697c91e280d6 : SUCCESS in 38m 20s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7a8eccece22e4e52b7f5dbc923a1f19e : RETRY_LIMIT in 4m 45s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/31a575a25785495ba2298a20027529d3 : RETRY_LIMIT in 4m 51s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6e117e7c7b4d49649e97f43a3ed05e2b : RETRY_LIMIT in 4m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1f98f6056f064748ba4fd303426dada1 : SUCCESS in 6m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8c4a7801bbb44ade8df15fa9d202fbd0 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59e9e77c90644eebb99ed179f4f0afb9 : SUCCESS in 13m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bf4381915a1e4ef6bc1917104db15527 : SUCCESS in 13m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0084e85ff1f14a7a9d9cc513b07ef394 : SUCCESS in 13m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3cbb7410f2b248019d251acbd16b5304 : SUCCESS in 13m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ee0901198d634ae2bd4aa29aeac32528 : SUCCESS in 16m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/14d541b0899d4fba940336bd149175cd : SUCCESS in 14m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2a34aba98191435882eb4b83c06495db : SUCCESS in 11m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4173451ab9a34c73961be8c282b8e71b : SUCCESS in 13m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8379105da24a4f3085ece090606aacea : SUCCESS in 12m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b53a47416407468494e5fd9b6b6125fb : SUCCESS in 5m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/822cb910c0814147ba8c810fb8988160 : SUCCESS in 13m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6882ebf0a3a442f9afd817c569e44f71 : SUCCESS in 6m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6f06e740dd384a72ad8763e0dbc80c41 : SUCCESS in 40m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/75abf6a2d8e7423daa4b77d38c5d6887 : SUCCESS in 52m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6134eac7ba4f450e85259adc5f9be07d : SUCCESS in 32m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2391ef4729af40988f552a4972ce0408 : SUCCESS in 33m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e77eecc526404d13a61cb60822f67c69 : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6118fca35bb461e843f089aa19cebe4 : FAILURE in 5m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/135a2f6abd504921b1e2e3cdcf69d82f : SUCCESS in 21m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cbce8841867c44df9e515f7caad7e9a1 : SUCCESS in 42m 01s","accounts_in_message":[],"_revision_number":5},{"id":"b2d7a6d57ac5a9f78f7aeb83424aff02a3bb9151","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-24 20:44:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d6e3cb95f8db76060f70bdd052f2d4b3b9e50e01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 21:19:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d37c417b8d494b6b800b3d19e6aef50d : SUCCESS in 27m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/602a76d4eaa844b0882dc8d5e9c002ca : SUCCESS in 28m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8ffb9950ac740609d3e6fd499a6082f : SUCCESS in 27m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/51ea2dcd05f54293b4627efee84d8284 : SUCCESS in 27m 21s","accounts_in_message":[],"_revision_number":6},{"id":"b7d70a14a35578b30afa351f08837edad1b2cb8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 21:59:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd5451c716e4d0a880310070b400c6f : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/a69f64ecda2741efa62d682136b87fbc : SUCCESS in 1h 07m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aac35f196ba845deadd1f15ab6413414 : SUCCESS in 40m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a06c30c38d6743248dec186f82e847c7 : SUCCESS in 47m 18s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f14981813fc247ecaa9522d89a226257 : RETRY_LIMIT in 5m 11s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ebafbfa43c2a4c1c94e075a47abc6d24 : RETRY_LIMIT in 3m 47s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5c0a803513144858969c1205f48c3862 : RETRY_LIMIT in 3m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/238db03748dd4217a53fd9bd711f12f8 : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a7ab448f13064e49854d7ecfb78ea596 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/05162bdb276b4c228a0aaa41ae844ccd : SUCCESS in 17m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/270fcdc295684c98b09970f16da124ba : SUCCESS in 19m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d022a123d9824a379959bbae87ed8204 : SUCCESS in 17m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4e73e7d93385404792ecd91b3a851eeb : SUCCESS in 22m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/56a61b7d2aa246f3a5d73672756f8da0 : SUCCESS in 17m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6a57b8295e7545fda2114535990c05ff : SUCCESS in 18m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5702c88723954b5a99e631ae0c094cfc : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a5bf8c7e84384a1486319a5fb509fbe2 : SUCCESS in 19m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/be85d458d067472bb0ad83396b993bba : SUCCESS in 18m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dfe1cbbc99e840059c25c7f87160e262 : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f446452f43574aa1942803f5cf4d4325 : SUCCESS in 15m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2c8a883358f04139918af65ba66ef786 : SUCCESS in 6m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1e7d45fdc03e4c5d80c511a543987510 : SUCCESS in 54m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4a3be8e62184d16a9e2fcc5a63aa142 : SUCCESS in 1h 07m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3e00299017bb43c1a78e32dd3ca7de9a : SUCCESS in 45m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5d59873418bd42f8b8f3a13f4c8f84b8 : SUCCESS in 45m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e6832165be5490784a07208a4d50ebb : SUCCESS in 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a04484828d894210b12410a5ed11f8a9 : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d59cf760ddb545998ab5f7caef346d8c : SUCCESS in 20m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea2aaa727a43430abb193be1a1e6bbdd : SUCCESS in 46m 54s","accounts_in_message":[],"_revision_number":6},{"id":"e39711a9f0884527eb757c2c45624b3f80696bfa","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-06-27 21:55:22.000000000","message":"Patch Set 6: Code-Review+1\n\n(18 comments)","accounts_in_message":[],"_revision_number":6},{"id":"74d66d24c580a196546efb2193d4cb48e91fd5ae","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-28 06:04:45.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8e0d85f5fd6d574ac1b9744d1d3f4c7628d2cec0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-06-28 06:58:47.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"e655bfd048635910aa089ff8cfc0bb6a33eea507","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-28 22:27:30.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ac6eebf43deb143a4fd4bfbb7b01476bdf849c18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-28 22:28:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"66f604dd31bc8fb01cd10b29baf66803100e5e6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-28 23:11:40.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"3631758e860986a5028c505603575067f069575c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 00:10:42.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d881cad19a54a62a92f98bc04489a9c : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/5c66f9ed68d5404a8a9f2e255ffc829d : SUCCESS in 56m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f4734d5e9ab7467085bedd777d32457f : SUCCESS in 30m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3110aca3ee9f4e7ca8082de2c3b0645f : SUCCESS in 35m 20s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/57d77e6572b145b48aa93624377a8d90 : RETRY_LIMIT in 4m 33s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4ac2c73800c14b589baab44d4b553e4e : RETRY_LIMIT in 4m 33s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f4f800b22ff1424984436bcd36453335 : RETRY_LIMIT in 4m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a28d0416379b42cabf9a9f8ca6bcaa11 : SUCCESS in 8m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/57f2e019df8646c882ddf9b66a59e8a0 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/76473ca436f74099bb8982c7764d9f39 : SUCCESS in 15m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3e60d98963894636bf7c207797e1d2e2 : SUCCESS in 12m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e356bc425331442a8f89bfc4e3ee83f1 : SUCCESS in 14m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9e75bc5228e3440382fa5b7b95db1e01 : SUCCESS in 13m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0ee749dad77147548e7b83afe80300c8 : SUCCESS in 17m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9199c0b301ee4e9f8e46c79ff85ab930 : SUCCESS in 13m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4a672b3aa5d94c22aea898a48083a061 : SUCCESS in 16m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f730f2963b7c4ec98d449788df784357 : SUCCESS in 13m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7374f98f893d441395b37e2bc14c84f4 : SUCCESS in 16m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f836fe97f324d1b9eb6b98acce4c809 : SUCCESS in 6m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8eb8b1f2b06d4ba58cb44d1577570c98 : SUCCESS in 17m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/99bbc0ecd109432b88f158de5698e64a : SUCCESS in 6m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ec23e3d3a0df4cdabf73343c974db5fc : SUCCESS in 49m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c4cbff7e68b4f11918419c60d2e2d98 : SUCCESS in 48m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c50c414c7a340a49b08e3a78576a601 : SUCCESS in 40m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6f6b2b143a83412c983a2854e482cdad : SUCCESS in 27m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49cce420aa5146f0aea8d44042e81c2c : SUCCESS in 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47622a4c48264e09b5c39c7b7e91389c : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/730e43a184364fef8dd2783f4990becb : SUCCESS in 26m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1eee92cbcae44f68b20bdfd2e51c8389 : SUCCESS in 38m 59s","accounts_in_message":[],"_revision_number":8},{"id":"4b74f4c3b4d45fd10d6658a5cf5efcfdfb1a7224","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 00:12:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49b12b77d6124a4c9612814ff5e9df4c : SUCCESS in 28m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/db038a0671ed452ea8bad240fcacef68 : SUCCESS in 47m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3a8de44078943628a8f38e68ab3f453 : SUCCESS in 59m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/54fa25673651426298d034e1d01fc069 : SUCCESS in 29m 22s","accounts_in_message":[],"_revision_number":8},{"id":"c66c35393a18a03576bea2e5f7c54899e99d7c38","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-20 05:09:07.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f60e12b383ff7f4afd6dc587344d5f5ec09f1bb7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-22 19:50:55.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c773a96401221acdada8af7c5df1d182672c5cb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-22 20:06:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4f0e59f9d803ac9ccfb61c50952a1e5dd2870ede","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 20:58:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4ab1cba7c684ed3bbfedca7fc7464de\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c5fd4e14525439f90c78d19749a773f : SUCCESS in 36m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd368a0dc58e4cae903ae815904c59b4 : SUCCESS in 50m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d05fe17ea6634480969185af5bf11258 : SUCCESS in 36m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd696f3e51ff4a1f8d897c985fb0fa78 : SUCCESS in 35m 12s","accounts_in_message":[],"_revision_number":9},{"id":"7a15a52f862ee8b214558749582e73182edf7a48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 21:15:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da2e3c88d76c484bba1fac32b0883cce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b600c5a979840cd97056e6a7bcf337d : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/c4ab5e1a07f54f8e96ac07e8ac0f3b9a : SUCCESS in 46m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6dc80210663e4df09a2acf61d2f6f625 : SUCCESS in 38m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d05cd75af3de4b3a9e8ad338d3e0c1fa : SUCCESS in 41m 04s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/56333074adf94ba181fd80920cf382eb : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b8a16841daf0428a90a35ac99c314114 : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/93a6328920144a1b8192ad1cbdebb55a : SUCCESS in 18m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b425adf5fa8430db6cdab5632da6011 : SUCCESS in 9m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cd6fe1d956e140ab9dcc35eb8098a10e : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/42426ff0ee854be0a076f2efbb5028d5 : SUCCESS in 16m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/44da5dd65aa64270b77b98d0728a9d6d : SUCCESS in 18m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f2e723da87054fc6acad40509c41ce9e : SUCCESS in 17m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eed149ce9f724a7c935118b020be5948 : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/911f1f29e12e453cbbf58d498e40eca9 : SUCCESS in 18m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f2f824ba670e451dad60a92cb3db9880 : SUCCESS in 14m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eb7f5b0942264603aa9f0bdf7ecfa226 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/607b7dcae5ca4ed88b76591f5027848e : SUCCESS in 13m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4e7416b351ce43c1811c50591a3b15e1 : SUCCESS in 13m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1442bf9b9fc5494fb6e4b4425d581aa0 : SUCCESS in 8m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/911431c2d0a9452a9fd8347de0249055 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e5d4ef96533a4a6b9128947fdb3f8719 : SUCCESS in 8m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e23a42cb497f472080947626ef5d449e : SUCCESS in 1h 02m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/73a566e431264efd84ac5161abce7340 : SUCCESS in 52m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3ce30d77025141d3b7239630541a329d : SUCCESS in 48m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d789ab10be3a4aad8463d6e291bff61a : SUCCESS in 35m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f8e4436f40a491ca53fdb312ccb60f9 : SUCCESS in 17m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44413acecf034556a0666aa99a391447 : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec8dfe3b00d34e65b8bdb2f5aef5b335 : SUCCESS in 24m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a18beb00a4d84ee6a7c025dc48310503 : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":9},{"id":"12b004b4e6c6cb518827d1d278f436ba2223e8a6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-23 04:40:41.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a23f0dd78586cc8cad4dda0f3a7067987a2cdd62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-24 22:08:26.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d105576acd796a4999c46b8c5eebad3ab8a3037f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-24 22:24:08.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"6c5c7eb9748190ab05360aaa4317149899c708b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 22:59:44.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f80a744d8fa8418ca84c6aec62e9027e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/48de07c447ae4dd380407bd782a7f083 : SUCCESS in 38m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f3645274e0249aaa0d453848332aa14 : SUCCESS in 42m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/59e427c87ed54b4aba034f3767216458 : SUCCESS in 43m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b00875397226481c93a098eab45a4500 : SUCCESS in 49m 08s","accounts_in_message":[],"_revision_number":10},{"id":"512a0c9a89624a5aa8736d5591190cdc02051c65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 23:20:21.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21ce7b817fe7482bbdffe0f69a7dce49\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/234fd8347ed14273b76e20c3c171bc87 : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/a7277ee55d7f46139f172d9fe7e9f6ec : SUCCESS in 47m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aab6c2546453412fb3d0fd93802cfdd3 : SUCCESS in 43m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a03aeed3ff974f04b949caa9dbddbb2e : SUCCESS in 46m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03652ee601be48c8bcc707fd8b5a3708 : FAILURE in 18m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6fe91720616f46c38d1978e46ee9b4c4 : FAILURE in 22m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd2f482c24de4fc1a4b608dbdefd6c5a : FAILURE in 19m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a7e29659dda24c4185603830b1584fe8 : SUCCESS in 7m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1d433f50adf447eb9f50355e2b9d3b3d : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/26e82c8bd16d4a25927d8fc1901111ec : SUCCESS in 15m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7fd3d1d5454a488b8ac57ab4590d7cf8 : SUCCESS in 20m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d1eb7a37786948119da25bec4c017ccc : SUCCESS in 14m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e5b371e6f9694d858b7a268a65884feb : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a63cba2626a540099fb85ebbca047cf2 : SUCCESS in 19m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c512c41c58d4f9992c5a753baddee74 : SUCCESS in 17m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3b6ef6dcdea54394a1548b9034a23d17 : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e3c3ef95039e4fc6b54ff86f928c1b0c : SUCCESS in 18m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c5c43754916f4678bd311a2c55e373c0 : SUCCESS in 16m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/09da9e0260c34261ae26614107e9045c : SUCCESS in 7m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/988653c47bea42398c5719d0ff9c7d03 : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/976ad2b2bd3d41c29a64bfbf046f2489 : SUCCESS in 6m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ec3cc9b27b674d2d9d742bc1050ad877 : SUCCESS in 52m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4f6f5f5c8d44ecebc83a492ecb8dd55 : SUCCESS in 57m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e705668705e4410d89dc742ac0b07b4b : SUCCESS in 41m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/04f3dfb71a484435998cf1653846db00 : SUCCESS in 38m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f2779afdef84343b5f0d35a876354ec : SUCCESS in 19m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca915b510ed84464a69c7462ea4c19b6 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05550c6001f14b38b1cf1e0bcd6352d0 : SUCCESS in 23m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcc08f74efad4ec192a40b656b3c0394 : SUCCESS in 45m 01s","accounts_in_message":[],"_revision_number":10},{"id":"bb3ea02e0b2f9ee780036b2afad87ce6dc8559f9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-26 17:01:40.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"ef1a23711c761a817b31082e4dab5bc0aa58a62e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-02-22 05:45:31.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"c90c36e509623d61e5d8cb6facb5de4f9b22d605","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 06:08:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26e5713ebe104488a80ed36bfca9a89a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/045b5251961b4916afb1578005c8357c : SUCCESS in 14m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df50bc119ff44937ad9b5f8323096739 : SUCCESS in 14m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/356c75edd28e48058116de28dba4b204 : SUCCESS in 21m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e04c8f4328f403a90f0941ed1753c13 : SUCCESS in 17m 37s","accounts_in_message":[],"_revision_number":11},{"id":"26e7431668fc288f922ed40cbc9700d1f86ac108","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 06:51:55.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a5f58ea18a64a55b63fb45017a4de36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/074d6ad45079467ab1a2bdf317fed172 : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/a9ba681b5dc8413aa44729aa036996b7 : SUCCESS in 53m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/48b05d7c3ec6425eb87ef7f36c27b4b6 : SUCCESS in 42m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6b885442d064f949bb653f883560e34 : SUCCESS in 1h 00m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/edf768db34a144a0934be8817d45e75c : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7f585472c114422a1e281ffa545139f : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2cf4e2cab7e3495e81a4161ead19ea44 : SUCCESS in 21m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4b753d0cb77c45a9abde5a66166e1b10 : SUCCESS in 8m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6432cbac702642c985b9ae66d279f78e : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/608b7f76b69f4706b521d649744a47da : SUCCESS in 20m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f2323b40e7334864a57a880da3e4d9c9 : SUCCESS in 17m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0fb9a82078df44f99eb5bf796681cb43 : SUCCESS in 15m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d859d951a143404cb334cc696d68e883 : SUCCESS in 16m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ce8532de598e472b938763a9d22192ef : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a3c0362cb2394e7b86667923fea6b529 : SUCCESS in 14m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a77476120f9f410d85f990f203d1bee0 : SUCCESS in 14m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9d439510e5e24d4c8c2f6dfec925b1af : SUCCESS in 13m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/269104f682b34ee59431dce8b181f0ff : SUCCESS in 11m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3529b0a9a0354f139822db3bfaf69d12 : SUCCESS in 14m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b7d9139ccf9e48c1b85c971ba81f3a03 : SUCCESS in 5m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e2e1eb4339a84671bee86441c413be9b : SUCCESS in 17m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fb24b23d957c41d1ad5e613c81db0a0f : SUCCESS in 6m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c02232a9601e496d97b1a1e1b2c8d19a : SUCCESS in 58m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e1af04243a84dff9020775479fd24d4 : SUCCESS in 56m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/64eb042c3cf6403a8fb552ec63280bf1 : SUCCESS in 52m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/64e8f7625de14a689f24a53de31226e7 : SUCCESS in 39m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff1abdca12674cd1ae2b7b44a1a63dbc : SUCCESS in 16m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ede8dfea8bc46e9b4aa4985f2ffdcbe : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d73e91c961e04433a5f962ed865539a1 : SUCCESS in 20m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76e08bfbccb941eb9bfa4c40520ebfc3 : SUCCESS in 46m 25s","accounts_in_message":[],"_revision_number":11},{"id":"3485bd975787d322fa23767fb6de8c4b4a186105","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-06-27 04:40:03.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"3ebb27b2e5142cd708e4292c7161d61f5574bb04","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-06-27 04:40:37.000000000","message":"Patch Set 11: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"ff610697e9cf7a47be4a3c5775646f1b01d30d69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 05:07:04.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b685780b23a74c0dbb5c256d8b608133\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b35c66add3cf4dcfa082a7c93b6c8ec0 : SUCCESS in 16m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/171774bfd3894f4b85f481d611034a03 : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/017a68d48fd5442ebdc4e264d4bfd64b : SUCCESS in 21m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fefea0dd3893422d9d395d0a0cc90ef0 : SUCCESS in 17m 59s","accounts_in_message":[],"_revision_number":12},{"id":"bdafc4d4383a4a817b4a661fa16f437e200e71a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 08:56:05.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcaf305ce34842269462c9e794faa959\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c249d1fc3346799382f96a8f3ae96e : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/a7bb34062ff4416a8a4cf707f1cc3110 : SUCCESS in 1h 13m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d94a7623fa8e466ca84cdfcd176b9de8 : SUCCESS in 59m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/492d0796974f43e59456efdc9fbff857 : SUCCESS in 45m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9981eb66aca64bf2865edb4d2bde11e8 : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dac8c79e19464859916fd76dada96a6d : SUCCESS in 15m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5fa93111a4c84cf7a6c7224970001933 : SUCCESS in 25m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f780bedb3c2a492db8722c375446c318 : SUCCESS in 8m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b3df4e7f503643fc96a799328b35a838 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/add14a14ed3a44c8a031f6a08eeb3964 : SUCCESS in 24m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/006ce189252d4140915dd49f07013f14 : SUCCESS in 19m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b8856b29351e4f1c956d5cc161574c24 : SUCCESS in 17m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e450e6daf61249c8a9064865ff4ea96c : SUCCESS in 17m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8a36662727374ad98a92f9daf4a700ec : SUCCESS in 24m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e038733bdfc9431d8894b5f0e49d8cb2 : SUCCESS in 20m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0fa81c81bc91457293418d45e9d8904c : SUCCESS in 23m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/411cb837ea514f7b92ecc49a8adf14ab : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/73f9722e88b3495897e2847340cf3c14 : SUCCESS in 22m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6eddaccfc69f4b43a89d3b03f6464482 : SUCCESS in 24m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ca4b788c82a343069f92dde65e9cbc9f : SUCCESS in 7m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7c85e72580ef4f97bf985ba51d47b1fd : SUCCESS in 21m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cbf2f05a02e244978791aaa24435157b : SUCCESS in 9m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2065c5b070d84046a69acf2756484500 : SUCCESS in 1h 00m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/844993f10602488e924c8b605d7d74a0 : SUCCESS in 1h 23m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e579009aeef84c23896de54037d2ada5 : SUCCESS in 50m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/484dd67c3b3d4dee80dc3faa321c0de6 : SUCCESS in 1h 08m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8399fd5d5dd843a2b5fac26cc9940ff8 : SUCCESS in 17m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e028519507774b5398d6b2240805a387 : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/054ee40ee30a4742b75aafbf0168c9e9 : SUCCESS in 21m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c42a62d5c6f24efcac417d5b18eea2ad : SUCCESS in 1h 09m 09s","accounts_in_message":[],"_revision_number":12},{"id":"0627d229f02c8164fca7f11dfdc8e48d58ac03eb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-27 18:17:50.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"c540228aea8d468ac8613ff1a6110e1f722e1f25","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-08-17 16:15:03.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"29b63b159246ab118af377df11fc2db5f8af00f3","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-01-26 00:49:21.000000000","message":"Patch Set 12: Code-Review-1\n\n(32 comments)","accounts_in_message":[],"_revision_number":12},{"id":"e8b774570d39d8192bb0295663034a9e307ee13a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-27 00:24:21.000000000","message":"Patch Set 12:\n\n(13 comments)","accounts_in_message":[],"_revision_number":12},{"id":"7335c1e63783c6e8a72f701e72a5ea641b24ded4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-29 18:59:58.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"4c9a7b0a7c696ae874df3e548cc56867f507996d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 20:12:39.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5efba074673041d4a51e116d6ab3961d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b40a8705ce6f4acf9a7d1fcc41f32c53 : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/db277c8bc76649988918f24683c1aabb : SUCCESS in 1h 06m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3b4a92d03572426380464abfcae1f296 : SUCCESS in 1h 02m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a25e0c38007045d086c0513be6549f01 : SUCCESS in 51m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b79b7eaa4424e3aad32135693a43c3a : SUCCESS in 28m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/24cf5324b446439d93dd15d26851d9b4 : SUCCESS in 18m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6015d29f22e849228c1348244e4cc4e9 : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9e4e05f10704459e9a08fb6150d82d8e : SUCCESS in 19m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/13cf1980286e4f5683bd3f8ff3902335 : SUCCESS in 11m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ca6f93ed1e9442fe8ab0d62ed965fa69 : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ab73768b172945d09507acb951a393be : SUCCESS in 19m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bdab07026c1346a1a7afcf05634ed6d7 : SUCCESS in 17m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/37bb84f2e0b54a3e9a4519d7bef781c0 : SUCCESS in 16m 11s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6db5b591f1944e3a91bdd776c907aa6f : SUCCESS in 16m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/50c0b161239f4f078de1e92cdf62809d : SUCCESS in 18m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/df22bd5fe4374a6eb10bbc69630d3d9a : SUCCESS in 20m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/64e982f4691f49bbac39fbfa633fd978 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/970b9bdab7a34f71953ae9db3a82fb4a : SUCCESS in 18m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4c24e5b8230648ddb512540f2718a03e : SUCCESS in 12m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b6a5404bee8d46b6b2875c4b3c1c7f5e : SUCCESS in 12m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1f2591f63bb64b58a94a1d0517136143 : SUCCESS in 14m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b7a8a6bc9b614019afc81ad36d8db4fa : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/36a0ca3504934084bb8fe2b0912273db : SUCCESS in 15m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b301082672f24dd3aa1d4332da411856 : SUCCESS in 7m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/78b8c782f9ce4de89a458587524b57f7 : SUCCESS in 48m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f423ba45aa04652997496038b49e950 : SUCCESS in 52m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f4cd6cbc8a8c4a56b105c12a71ca8ebc : SUCCESS in 55m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fe2e89e2f93f41929f1424d95b2dd9c3 : SUCCESS in 55m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a847829dea645d381596a87493e3045 : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2776c2cb50e24b30b0b2f76535680846 : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/36d5a4809b444aa3be5b8b2c81e8586f : SUCCESS in 20m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2372b8706eb146c999730b922767bd0c : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":13},{"id":"88a8969af3ded1026f1e6bed4752fbb9a9e073f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 20:32:42.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbbc63e11ba44bea8627980131331c2c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b27e82bfcfc4b40a81289e473f3fb43 : SUCCESS in 15m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d80483941ae64403a9914b4de092a9bf : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c93f6364c03428cbd7a97efcfc4462b : SUCCESS in 22m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/54e2e7661fbe4550a6bf5d83f964a6ad : TIMED_OUT in 1h 31m 03s","accounts_in_message":[],"_revision_number":13},{"id":"f1a7ad347737435f37fb03a9b94c3f313aef252c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-01-29 20:37:15.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c551215d3143c73023445e054625725b74f14d36","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-02-06 15:38:48.000000000","message":"Patch Set 13: Code-Review-1\n\n(36 comments)","accounts_in_message":[],"_revision_number":13},{"id":"7199bd058e93ab0d1adcfcd8a35d991048e3553e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-07 07:16:02.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"6e2c2c9361f2096b69ed2360b130aab798959235","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-02-07 07:33:47.000000000","message":"Patch Set 14:\n\n(12 comments)","accounts_in_message":[],"_revision_number":14},{"id":"4515ec8c7d08b5a27cc60d1ced89c8f81d04c9f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 08:35:42.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba98816f30784334b5aaf942d14c9e00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a21bdcb10b04934b0867f1430d9e70e : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/57a23b2f434c4c84bbe4914aece1785e : SUCCESS in 1h 03m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5a78736ac4c44cc2b87cf0145377e8a6 : SUCCESS in 50m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba2b5861653048f29d1d6083a4197f1c : SUCCESS in 46m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b6f1cf4d67d4251a496980ebee28160 : SUCCESS in 25m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9c5803bb30e4e2aa21d45dd563007e2 : SUCCESS in 21m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0da6e7488074260b206eb1975dad5ae : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2943fac572444041a8585bfea376f11a : SUCCESS in 22m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dbef92286f7a4942afecd1070b9cf0cd : SUCCESS in 11m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ae0ec4ea9060403f82f378ef06cf0be6 : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d2df8bd3f9ce4830914a446e729d42ef : SUCCESS in 23m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3e27b588e06e44119e348f094f58ae54 : SUCCESS in 25m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1aca5498e32d4e86ba348b7a699d26b0 : SUCCESS in 26m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c0ffe5f34d1049fca9a78eb9001a1338 : SUCCESS in 22m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/24ef92b9465243079683d05950571d01 : SUCCESS in 25m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/40043f8673ee4e93849bc2b9ea6fcb7a : SUCCESS in 22m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/22d0c5fe985c4559acec34c25dc9b176 : SUCCESS in 23m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e2a4dbbc7a3e4f75ac2135c74d24a487 : SUCCESS in 19m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3c70dc55762645a7a540146f0b187345 : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6fb7238fb5bd4f58b728ab19cc546d21 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/895a47d728cd454ca1d7e882a9bbd2c7 : SUCCESS in 18m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6f47747fb42447e5bea91a3dbbcd4d0e : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61d2cff3fd9844c8a17eb3005aa7a512 : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/24b4cad912fc40fbb1524e02cdd1d1a7 : SUCCESS in 7m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4a4a02f6d49a40818122f29cf49702d8 : SUCCESS in 1h 09m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/378f637730224ae69a7210e65ef3310c : SUCCESS in 1h 05m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b10de053c81e48a2833efd6d555cc9df : SUCCESS in 55m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/918e63e2a76242fa92de6e90f9f57ce7 : SUCCESS in 53m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dcf3e5fde954a4db04f10220a063b08 : SUCCESS in 18m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42eab517c9f049f98d5a69d007696053 : FAILURE in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f2ef8fc76004a4cbf76495550bba8f6 : SUCCESS in 19m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/173a9d2f2d56434dbdb522584159d2a1 : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":14},{"id":"640d48ac29ff1d13766f3410c09858cd8779a16c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 08:48:51.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7017e832b9ce4bc1969365037aa778ca\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53503d7db78642ceac43e8276ecad835 : TIMED_OUT in 1h 30m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e01bc88486ec4c74beb4e91956dceaee : SUCCESS in 51m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7ec11179ed54b9d872e8c66e0b4aa5d : SUCCESS in 19m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/922dccb66ec14d93974314443348abc3 : SUCCESS in 1h 15m 46s","accounts_in_message":[],"_revision_number":14},{"id":"a5235bd5e5fe82d4bf77e08f2074021f81f2def0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-02-07 16:56:50.000000000","message":"Patch Set 14: Code-Review-1\n\n(40 comments)","accounts_in_message":[],"_revision_number":14},{"id":"3fb0c01a3d45b920384171f61f0b74ed00a097eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-21 03:51:45.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"81839176b033aba338d5c873760de3a5ca08a8ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 05:00:52.000000000","message":"Patch Set 15: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7439e8d85e734f47afcdb35f4f7900f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fee0a0712874e899830a12f073c4e89 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/fa2c144921af40128d0917ed53efdda3 : SUCCESS in 56m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/998abeb6a9934f44864b6f99afdd1f13 : SUCCESS in 38m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67f52746d95e49e9b8a3f47b6d236280 : SUCCESS in 58m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f11cf24f7ee74b53af3e15b33e06282f : SUCCESS in 20m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8608840338da446997d2ef9c6f24a3f4 : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dff7d7b7ca864ced9f183d5211d2eb92 : SUCCESS in 18m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8812e4f18e054102889f8a883e07c732 : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/acbe966d932046abb23ba708549f9ca6 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bf5bf432f68745db8a9aa6d4d14b2e15 : SUCCESS in 19m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5d7d15ad157643ebb5a39df07075c352 : SUCCESS in 19m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9793f0b7a26e4658afa8f45308b4119f : SUCCESS in 19m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ffc52f2fd83a4d0695f539446d038119 : SUCCESS in 17m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9804a9ab19064b57a1e3410a6caea3a8 : SUCCESS in 23m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a5664f50750d42b88f2b2da2e9249f3d : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f8d1d97fdd20471aaa4a9f41bc7ffdd9 : SUCCESS in 19m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/00569e411a7f4a9a80612907a42bf226 : SUCCESS in 18m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/504bd3ee8f254f698d5eec318cf38555 : SUCCESS in 27m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9b0e2129ae264f70b48177035a6a9b09 : SUCCESS in 20m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8aaac54a747845f28938f7b36d0fac31 : SUCCESS in 17m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/18bbf636af8d47849d2033821a9a92bb : SUCCESS in 6m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a89747cb9fd94fb19a2c52f14cbefc23 : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c33b52f34a994e2c81c2c82011a50ab1 : SUCCESS in 8m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/df4e06c0b50c4717a1aedaf938e02f01 : SUCCESS in 1h 00m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b4d678c2105342fbbf21388602afe4b6 : SUCCESS in 48m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b417106f480412fa7d6c9a86d661dba : SUCCESS in 54m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ac79133e6a44cf0a0b90b5037c0ed8e : SUCCESS in 17m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdf513f93d8c4fd2a573144a69ee5cab : FAILURE in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8967fb9dd4c04e5babcc4e2ae2ad0158 : SUCCESS in 21m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05f593ab51464f5c977aaef444179451 : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":15},{"id":"a35d0e47b05928ec1af05377d7003ac6d0b89fbc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 05:24:53.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3467de5180b14b41b7dd298a51c4071a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6471e3182c4542758ad84665bf179c75 : SUCCESS in 17m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7213a25bd274414caee59a9f534a1747 : SUCCESS in 49m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f6f79596d3c4d0393940d0bf2e95378 : SUCCESS in 26m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8aac9346416f4bc4856a454d77e6784f : TIMED_OUT in 1h 31m 03s","accounts_in_message":[],"_revision_number":15},{"id":"6064d6192b1fe81bf9c588d3e01435219de21de1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-21 15:04:55.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"5c6ab1a7cebd3359234324bfce36fddef729adc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 15:56:01.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dae449c228b54bf68460f3f8aeab874d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f1c7146e6d74fb9944df97b8d7bded2 : SUCCESS in 16m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad26c9318a334176ad41dab507459d6e : SUCCESS in 23m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1952de35377248269c81640b8445ed83 : SUCCESS in 20m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f42353d74b094e06a9c5620c2bd4a3d5 : SUCCESS in 45m 40s","accounts_in_message":[],"_revision_number":16},{"id":"9249971df5c66c18d2e5c4583775d7ea6b978125","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 16:24:30.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21d93c4e0c4c46bbb175c44d1f8d5c6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84e19dcd46b342e19d4840312dba723a : SUCCESS in 7m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/7f866d6171cd4e1f963c2854d0084b63 : SUCCESS in 1h 07m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/550ed729603148beacf0f9399bf12a74 : SUCCESS in 55m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8151e9c94d704f679e1a3e01d18bd168 : SUCCESS in 52m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3cfcfc79559143178463dfd3bdf0ab74 : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e251cf1ef84d48fd9d0f1129e2ab635d : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78fa602130264d9eb3eb507ee0d138ad : SUCCESS in 22m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ea19668e8c042c8a235110e09451afa : SUCCESS in 10m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/eb1dc2fba27f445d9f47773a6d950ed7 : SUCCESS in 7m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1b8682bc73ab4bce95465e6da46f8a16 : SUCCESS in 18m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/363cc6b950874960957930d24ce2fa67 : SUCCESS in 21m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/082f5e9dc5524a9ca7471a6a5735e74a : SUCCESS in 18m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f7e46f2ed32b44a9ba948da84e30c1a1 : SUCCESS in 18m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8b2a4d31560448f8812a6de4ee260a0e : SUCCESS in 19m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/398be6f592bc44c1957837d95a60f1be : SUCCESS in 20m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/afa4143afa5a4b2fb79fd3e3e0fc098c : SUCCESS in 18m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4ed438d94ce04683a696b7a5d38b187e : SUCCESS in 19m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/71dbd511f4194e86b6571fc863061b89 : SUCCESS in 19m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7929fda3eda9414ca95b2cfd8398059b : SUCCESS in 17m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f530e7d95c524d04b331909e22b5c801 : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f05aa266530e4c9c960343cfe77e7563 : SUCCESS in 5m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5034f0296dbe46f38eda39dcc6d3b091 : SUCCESS in 17m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c5fb90ed00314ef9996df96503fa0863 : SUCCESS in 7m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b7b31384bda4a80af86996a7a2c2587 : SUCCESS in 1h 00m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dcfe5dbecd884df6ab8ff57fcb8f1790 : SUCCESS in 46m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2384d705be294257a757d46cd4521629 : SUCCESS in 40m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a26dda281124069af581966c29fa0e8 : SUCCESS in 18m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f423115e917e4a63aaf93520f7f6067d : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d41613e058c6493a91fcde7518eacb15 : SUCCESS in 18m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b1b05a9205a48608698f59ee36abcf1 : SUCCESS in 58m 55s","accounts_in_message":[],"_revision_number":16},{"id":"46bb647007de55ac40975ad41a09df3efa4780ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-05 21:20:49.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"ca2c7e34fd8eae2c3bfbd27d129151bb1716189a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 22:23:05.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25b0018c278b4a7ba3e18cba3267651e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a0eaaa56b347c2bc36277fba0e2201 : SUCCESS in 7m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/5ee1774aa826475eac280aeb246487bc : SUCCESS in 44m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/48861fc197754f38bd481fa9cab939e4 : SUCCESS in 35m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bd1724bfc1b4822ae181edb5d9fe5a8 : SUCCESS in 47m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/885da59ee44345889f02fb50cca2ea17 : SUCCESS in 16m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4afbcfc0caa49689cfed04f9ef4d4e5 : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c948c48f141f4f08af2c85c08617e88a : SUCCESS in 19m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d93e26b42a774d6facc7b1303f3c629a : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f4e51973dc7142e2baa40dc8631f73c3 : SUCCESS in 16m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b9a9ee2942ae40b4858ceb778532ab17 : SUCCESS in 19m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d2cdc7a1d21e4e67a27e839e15d4a619 : SUCCESS in 16m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c78dfdd6d2c34599b1f898fe8cd638df : SUCCESS in 23m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a1ef433b2ff148c6a4983f5cdf417af0 : SUCCESS in 28m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/764207096a0345bd84e875adf782587c : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/407e83dea1324d6995fe0156cedc45eb : SUCCESS in 17m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0a2938eea530453e8ec7d0c5fc8c05a4 : SUCCESS in 15m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/13a0d185e8d84adea1ac03ecb98c569c : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4742ebd52a4640339eb0de6a8269c95a : SUCCESS in 16m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/45cc6ba713244593a9877ca303eabce4 : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ae38f83669f94a95b81796a03018fc72 : SUCCESS in 5m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2e13ae7caf2743269440bd52f4cf6a24 : SUCCESS in 13m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d30ac4b887be43d69c7b91af6bcc5f8c : SUCCESS in 5m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5a41d3f77a9486dab2e4ee42fdf25e7 : SUCCESS in 46m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4fedd70885342a5bfbd39be81815fe2 : SUCCESS in 40m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/826b7362bdc442468320e2bb3c579187 : SUCCESS in 55m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e0f6dd8b61f467aa7d124ba0e8ac1af : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16644835c37948d9a9a3d5eaba67b93a : SUCCESS in 3m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf2cfa09d1594696b9c0ce32a36e4372 : SUCCESS in 16m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81fde5ae5f2c4bf68ba79df4a1eeccbb : SUCCESS in 1h 01m 31s","accounts_in_message":[],"_revision_number":17},{"id":"a1af7330f6af8fdc07500d7fc9d06f973c88e919","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 22:24:42.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b20fc5e45d014b608c9cc9f4c194d57b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5b7c25ef88f4d5ea6d7a3fb27e93adc : SUCCESS in 49m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2b124c0ece6477bade167199d868b77 : SUCCESS in 55m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e225444805bc433ea9acdc0395e7202c : SUCCESS in 58m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ebb1671cf824a8a85965576017bb19c : SUCCESS in 1h 02m 03s","accounts_in_message":[],"_revision_number":17},{"id":"48e457f70501fe4ea15393c01f968c0c6b252486","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-28 19:52:39.000000000","message":"Patch Set 17:\n\n(7 comments)","accounts_in_message":[],"_revision_number":17},{"id":"d3b0b3e7c7faa486d83ea35410c5a3b4d8ad0519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-21 14:44:52.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"6db796f248a9a61f7087dfa87917e713b1ac2daf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 16:17:56.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d47747347464467b8db8262db5216f3e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8f8f78bd94f145658aa156346f454bef : SUCCESS in 1h 30m 09s","accounts_in_message":[],"_revision_number":18},{"id":"62b60fba24d5930620a73e0911e4b1bcd4062e15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 16:19:22.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98972052459f42f4a5a4d85e06aacc41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36a780cbc6a4536ae64ede7455ae2bd : SUCCESS in 6m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/bc28129d521b4aba82ae0484a2b31b09 : SUCCESS in 44m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3b2bda5f87fc462693cd7d499c00bc13 : SUCCESS in 1h 22m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b4fede5c980454b872000b1e843e563 : SUCCESS in 49m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d94dc19600748c4918d6e0477a5fc69 : SUCCESS in 49m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff1ba04e814740f8b18cbdcb59f4ef5a : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77fcfcc7e6524476b707e67c3fecacdb : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94c04af5a37c43dea1f45a36c7b035e4 : SUCCESS in 19m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/920df4a3608b46a5b7d5f47da787ade2 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/96be61cd43ee4086a5d962344d650840 : SUCCESS in 17m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/81e4d70c3c8a4f9b8712734a0d727c6b : SUCCESS in 15m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/198f4a2db94042348c8b5582952ac40c : SUCCESS in 17m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a9137246df6544d4b9824e21641a9d94 : SUCCESS in 10m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/73faab3e943047b59bfe24db3b0cbf78 : SUCCESS in 23m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d1e9de65357d4291850af4206ae79c3f : SUCCESS in 10m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/85a86b1a2b744ef1925d1e10603a11ed : SUCCESS in 18m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bb55eec5298443d99b0e71ce95d9d49f : SUCCESS in 16m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/213bb7db7caa4313b8708e4147e26af8 : SUCCESS in 12m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b1ab22422b534d6d976701fa085b1124 : SUCCESS in 8m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/466517d4fb6b4bdb93aa69a20c78d12d : SUCCESS in 17m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c6cd3b1c801e4f1a9f0c26783a00fc96 : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/37f12108038943858528e1e7dabefb89 : SUCCESS in 18m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/08e20826ad334e7f93a6ed7f7f0ade25 : SUCCESS in 4m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d350b14c3994f459a8969527847836f : SUCCESS in 57m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b9c39f8f6b194d96b78c4e85c5340d32 : SUCCESS in 40m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/735293580b90422f92472d7d4ae8888b : SUCCESS in 48m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e586be87701d4010a0dd835a145cf9f2 : SUCCESS in 18m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba5231e3134a41c0bf755c60c2d8e3d5 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa29990cf2914186b46052281c2eec9b : SUCCESS in 17m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5f6f1c733604f80ab0adc4b369ffb3e : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":18},{"id":"a14d8d66e040a4416394e6d0448c227d1f10176a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-19 00:34:43.000000000","message":"Patch Set 19: Patch Set 18 was rebased\n\nThe following files contain Git conflicts:\n* swift/obj/replicator.py\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"d4d303a7c610bef8f5d9a33ef64e1310417f5f77","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-11-19 00:40:58.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"34a510614012ce3bab79b1a2b336b82cd873dc33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 01:47:33.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec8940074c424891adfc28033a746e28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec3b7e1cdeb41b3b5ada2fc7cfd1b33 : SUCCESS in 7m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/c90bfa6a71994961bb8d7a1ef99e294d : SUCCESS in 41m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f0383cfb04c143ceae02a38c2f3d3844 : SUCCESS in 1h 05m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f6dbda3f8fa46beaf050985927d781d : SUCCESS in 38m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/676be46baf8448598a7b8879c03bd238 : SUCCESS in 46m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9a4c5347c684cf5be586d1538e96f9e : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74b21b83d74846d0bc93ad5539cca245 : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/660241598a2c4c159f97fac59edcea0c : SUCCESS in 21m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b01562eaebc9420cba98db5f31534899 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c34d5a6c0cff4dd9b7b01d50a6d86cb3 : SUCCESS in 20m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/74b3e58798e54e698338ef6f0bb5dbb0 : SUCCESS in 16m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8bde69144ec344eb81d308b43ee3e958 : SUCCESS in 17m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d411286553e5453ab87c7b5892582e30 : SUCCESS in 34m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2174e7d8128440e5be9922c09ed6acbf : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a30653d2af4644e19e7fb9f184dbb2ad : SUCCESS in 16m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/65451c9915d84ee092945b684398bfa0 : SUCCESS in 18m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/952195ad68224261b601d174bfece67c : SUCCESS in 14m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a8427b8e73514f699f5ad7bb8c109302 : SUCCESS in 14m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/150b3109c7e342b4aba91be1523f142f : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b9adb39c51a422ca461f296efcb42cd : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b2476496b1804329bd6afb2491b3fd55 : SUCCESS in 14m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/83e4955807d3465e8ddae4270e72db7c : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/49a9285d5bd146ea9341b09f4d075982 : SUCCESS in 47m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f9e7f6da505046f29130d13e084354fe : SUCCESS in 35m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/360e8827a8a043c5979e895ec25e0362 : SUCCESS in 32m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c361428eb6d64160975c0546aa40ef3c : SUCCESS in 14m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8702009cb3a472995067194521c1e3d : SUCCESS in 3m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/78ed86e82289450e8b453a042a3ea4b8 : SUCCESS in 14m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab5c8e39d9794f3c9076b119929df1fc : SUCCESS in 40m 14s","accounts_in_message":[],"_revision_number":20},{"id":"932715df154694caa65a2710177e4a0793fbd11e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 02:48:00.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e18dab38b7d4906a25d5503bd830b77\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a6149fc8f1e47789ad2896878dfeebd : SUCCESS in 2h 04m 42s","accounts_in_message":[],"_revision_number":20},{"id":"0ee50597aafe1fb431ebbd01036920ef54535c17","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-11 00:10:52.000000000","message":"Patch Set 21: Patch Set 20 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"7241c36ed1b0db388342c4a2bf072c0ac24975ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 00:51:43.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eaeab97978742229eb247b8450bdbc6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1747513f8a844b5aaef6c59cee451abb : SUCCESS in 38m 37s","accounts_in_message":[],"_revision_number":21},{"id":"cf565ed6da95e720e624c55576f34db866bad5d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 01:14:33.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/182fd51e92fa4f14a1fe9efff6cd9cc6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/610441a7e68148d98d098d42ac3e5dd6 : SUCCESS in 4m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/84b75f8a8ae34414b36aaedae03129f5 : SUCCESS in 54m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/86d90c20066f4fb99707a8fd286a3d0c : SUCCESS in 41m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/159ae662fef244af927eda4fb3e9d868 : SUCCESS in 51m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8737c7cba2e649279a5368f2f43ed62a : SUCCESS in 58m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ff00f0bc35d41639fa4c589593d0d6d : SUCCESS in 15m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8c597938eeb494d8c92c5fdc81f886c : SUCCESS in 15m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd052520352a4419a28d98a7675a42a8 : SUCCESS in 20m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/20ea632fa1254e9683aae29b462d99d4 : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d69f1039da224ed6ae059e2e3358f380 : SUCCESS in 16m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/33bc76f272404a47b3950cb760c063ba : SUCCESS in 17m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8836ca1560844c96b56732006e330600 : SUCCESS in 16m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cf7ab26c228c4a34894b1633fc828d6b : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b7ee4848d926413d8fc55c2b8fc1f998 : SUCCESS in 9m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/283d38333cda41a495b1533fc31aab2e : SUCCESS in 15m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d6c5ec062b8640eab82ff58bdf6a1fe0 : SUCCESS in 7m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/23482d162ad8414f951785f5271c54a5 : SUCCESS in 16m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6abc16a2df3c4cdba9eabc5871511a42 : SUCCESS in 6m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/06b55a47895a4518b930e8afebe8979d : SUCCESS in 55m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85c981e8d9634500a36ad9c40c8b2f32 : SUCCESS in 39m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/997d8b901ec54d9e949d94a2957e0a41 : SUCCESS in 36m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffe0ceb949694c179d0c56ffe227c079 : SUCCESS in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f51d0b2089a24e72aa5fffcf89011691 : SUCCESS in 3m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c5bbeacd3c144cc0af5aef98cc6052a4 : SUCCESS in 19m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e48a774363d49289fb0c4f44f38aa14 : SUCCESS in 29m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1133e6afee84499ace38b9077fc1b41 : SUCCESS in 51m 54s","accounts_in_message":[],"_revision_number":21},{"id":"2c3a85d93b1b229f5260e20eaf09b2872ff93eef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-19 23:36:32.000000000","message":"Patch Set 22: Patch Set 21 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"96996f8ac4565e0320b6d03ba6383d7054b5af73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 00:16:24.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01d40270167745e489b4e707556b918d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9e60ce9a1ec849b981781df434746438 : SUCCESS in 38m 11s","accounts_in_message":[],"_revision_number":22},{"id":"3ae9c9d54c94615bb6c758ddfcc91e50b6eb9f0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 02:39:18.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/695b743840ed49179df10dba96fbf245\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22bf4ec07ba14ed3baf25161e2a943de : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/d32c246d2f1d4613b061d31925a37794 : SUCCESS in 49m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/15170f7a6ca842019f03073e2e1102d9 : TIMED_OUT in 3h 01m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/47fce2ca02ec4e919ad410e7a7f72408 : SUCCESS in 41m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/875503c1666842c8a7951c3f4b4cb3f5 : SUCCESS in 31m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa38520e57ff498fb08b9b4db545d93b : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d035c8d931de4492bbf5323a3aa25885 : SUCCESS in 10m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d35f56c12f4d4a9981291995be48451c : SUCCESS in 23m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/04fe61ee286d4250a8d9b3567a4dc71b : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0c76d274e4c5426f90973c8ddfbea146 : SUCCESS in 8m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c4429adacac847dfa34747f54710b55a : SUCCESS in 20m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9ce2617b471941f0b6c82d472abb80eb : SUCCESS in 10m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/72c812dfe1b943e2b7dada1d73208d03 : SUCCESS in 30m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cf5de15212454c9e814b1c50d2a42c2e : SUCCESS in 9m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/35e80c34a82849ce9af866058040b68e : SUCCESS in 13m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e2ed1a561a81435e89fbcf0641a2d0ed : SUCCESS in 14m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/501df24dcd154856b0fba44baa2e67d4 : SUCCESS in 5m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd4fe7c06f3a4303a9f01935e1103c43 : SUCCESS in 14m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9b22eb58612d4f8295c17cbdde322213 : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/adb99487c9f1454d96d95dc85d7f51e8 : SUCCESS in 55m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/304b2ae3a9274f5aae71ed490d1fb6ac : SUCCESS in 43m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0f292ca308da4169abad65ab538607e3 : SUCCESS in 43m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3650290a80f74544a1155c50f3b1e70c : SUCCESS in 20m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/166c0dcbdbc340468a16143a50c8856b : SUCCESS in 3m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/809b73beb6a347bb8920ad94d4921b9e : SUCCESS in 17m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27efefeed3c8454ab37dbe01b855c31d : SUCCESS in 44m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bea0b0e529f431bb70916a13e36eb9e : SUCCESS in 56m 13s","accounts_in_message":[],"_revision_number":22},{"id":"622a3a78777ca7b2d75820598cde9554b5cefc21","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-20 03:24:55.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"af0bcfe219a92fe2c0841930aca2a68cc4a40863","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 04:14:45.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a67ef8b8419476199087b0e43806967\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9cc82bfc4b7e4b4e88212b920a994d2d : SUCCESS in 48m 23s","accounts_in_message":[],"_revision_number":22},{"id":"436758c94c4b088551710363aacf61308551f85e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 04:22:46.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b669f5b150ee413fb3859e8bda3d2374\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5218257ad5a140e29776b5d61583619c : SUCCESS in 6m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/86e4207b665f4b968ff9b8d0ac180dbe : SUCCESS in 47m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/43d1d1ed92a74f139a8955d033d635cf : SUCCESS in 42m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cb93b75a58584f50887ffb51e625cf54 : SUCCESS in 52m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0b166d942b344ceb0ffe6c60f13165e : SUCCESS in 50m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f5ebe3f1f8d475991d58d93432dab39 : SUCCESS in 12m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9e233be29124fdeb8773e27fb92c2c2 : SUCCESS in 17m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b7656b0083b4487bfc51a3571dd176a : SUCCESS in 17m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f0df924792d748ce95979c43427b872a : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/507d65d216a34f52a893cda22089d70a : SUCCESS in 18m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7feadfd49c6d458195d23f3fa203dba6 : SUCCESS in 17m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0a536d1d5a02436fb95426d6cf4be64d : SUCCESS in 16m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3d20d9a155b942a083b44bdc70ddb4ab : SUCCESS in 21m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ffdce1182f4d49a5bed2d84f074fbda3 : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/09579b4099ae4e2e9614e5400394dccf : SUCCESS in 13m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ec614a61bfdb455c948e100e05db5b21 : SUCCESS in 14m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/75724e49823848b1b9957223a543a33d : SUCCESS in 5m 53s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/46722c17229c4d9d98c6eb90db285956 : SUCCESS in 14m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/eebdec1e96c5440f81c06cc74669db7a : SUCCESS in 7m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d1dff464dbe4cf4927fbdc558f94645 : SUCCESS in 33m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/367eadc6f4564df89ed8ddf30cb09464 : SUCCESS in 46m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d0f2890455f3423794afb23b28588ac1 : SUCCESS in 18m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb238e5e6ce64ed7a5528a887b88cb3f : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/183e1147eec947d395df9d340ee7b6f2 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/190209fdde45430784f2c1c43c9aae38 : SUCCESS in 8m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78dee383563a4b2c9aaf6263d56d8b5a : SUCCESS in 49m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8f8de17da0d472a94b74bac3482a109 : SUCCESS in 26m 56s","accounts_in_message":[],"_revision_number":22},{"id":"758539d8b92e61b57e1d8f9c1a6687d20390847c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-20 14:31:37.000000000","message":"Patch Set 23: Patch Set 22 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"302bc6530b981953739ffe1c7c73a84a53ae0634","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-20 15:56:54.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"73897fea82cb8fbe5d749cf836b7ac6c99a62bd8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-20 16:59:34.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"8d885478e085eb08ec1aa606fff9ce8793419424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 18:32:41.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d68eab2bc7e4763b8eda7bf35902619\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8b23c4c3cc142228382159e63cacc45 : SUCCESS in 14m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/7da587cc55524c778ba568ea2868b051 : SUCCESS in 1h 02m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c081c65692224e1298fb8d99c1ea2369 : SUCCESS in 53m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9a124e984a4c4640a6df122f266d689b : SUCCESS in 47m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a81c7bcb2b934839a35c51a8d3344ba3 : SUCCESS in 1h 03m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3758571e9ea462383c233e960e8f933 : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7d64ee5b3214256a7397c949625181b : SUCCESS in 21m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2021f5b8d8d44c4d8d31a04344cf571b : SUCCESS in 19m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d8933394159d41268f754342143c5ea3 : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6e03fbcc4d4240a2b08ce2cac39631df : SUCCESS in 23m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9e4e6525901a43fbb62777f8fdfb5993 : SUCCESS in 8m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/41d60acd36a147d6ad7f969488eac38a : SUCCESS in 23m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/391a5d0537724632b8376da4f77de70f : SUCCESS in 26m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/81293dbfe2cd465cbd0c691c1e49b714 : SUCCESS in 13m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/50d0bae7c1984585b0b3153fba2da257 : SUCCESS in 14m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/947f02cef3ad4069b32480387c09d3d2 : SUCCESS in 15m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1887c7ad9b884229a3eb31ff84a8a8c0 : SUCCESS in 8m 14s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c81fcb7860a94e7eab87da1a4aed1325 : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/11c342f537b54cbaa8b5f75fbd4cae39 : SUCCESS in 9m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/83bc09437a28438d9b28c5eabea4f82e : SUCCESS in 52m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/45cc78398e5d40aebf59b62b3fb105ed : SUCCESS in 44m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9ed5c12737024e2fafeb324cef2162e9 : SUCCESS in 44m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e72951fa3924f1cb43f51b3727f6d51 : SUCCESS in 18m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9ebe77dd2d648439d179d2fe86eb71f : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b310dce7f73b414e81c216a34dfdb1e1 : SUCCESS in 26m 03s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bde50ea47fc941ff9594109f4a4c7d2d : SUCCESS in 48m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee4e702c65984c73b78aba178af1b755 : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":25},{"id":"d60a01a076ff40ed5e323599e2fa6405b9d1d992","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 20:18:11.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0642e3be978d451bb78a4af013024924\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/13d0a61f396449c7a600a7f6ffaf9700 : SUCCESS in 1h 08m 32s","accounts_in_message":[],"_revision_number":25},{"id":"643263023f4df352229befb7dfb35fedbd995db0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-05-21 15:57:32.000000000","message":"Patch Set 25:\n\n(4 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":25},{"id":"ecc53c7494e9f15920f6e4d6551d93a1ea8c35ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-05-29 01:09:42.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"54c1c1b3bce772653b357d28e6bca35aa60e016a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 01:52:50.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d4e86dbbc50494ea9ac88cea64fe44d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e458b75551f4d1ba44143de2e13f92b : SUCCESS in 40m 54s","accounts_in_message":[],"_revision_number":26},{"id":"cd8785a20db5d2690a832a579feee8a40adbca09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 02:33:50.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cad89b0ea08b4d12aea8d0da02b36373\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56eff784b4234529817e1f69d4e90212 : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/135aec19c6144928993d1dc27a429768 : SUCCESS in 1h 00m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9a4a7349072d4a9c857484c0e34fd003 : SUCCESS in 1h 04m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/29bc5d8fcc434e378133471cf30daf1e : SUCCESS in 37m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6492436e8914658a8f5f74ee484bbf2 : SUCCESS in 1h 07m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95dbb346567048e38bc8c9d57260d2cb : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33744485ba9c4b48b97f0f8b4446892e : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/989384b6f75c48038af2bc7453647463 : SUCCESS in 21m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4ee301098d3943bdac6966032fa8496a : SUCCESS in 4m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/898c116e5934442b9b131f4f55c03c2c : SUCCESS in 13m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/065528a5ecb644e9a644be4d48456f22 : SUCCESS in 17m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5f6f9cb1c92845d8a62d796eb7f2c7c6 : SUCCESS in 17m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/200827d4af2f4c8787eef1e6f36bfa38 : SUCCESS in 21m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3ff2090c56b94329be6f803d6b861b14 : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d973c49183b74d238afa3f34e14e8b70 : SUCCESS in 14m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/87d96fb5794a46bfb4973f8bf8930828 : SUCCESS in 15m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e57054b2f0fe41b8b3489a3b0f4b817c : SUCCESS in 5m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8eb78af40c794a9eb2b75d2683748d36 : SUCCESS in 14m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1ea38b26c26440b3beca5ca91d26e6d4 : SUCCESS in 7m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e52a50afb711411186df90105e6019c2 : SUCCESS in 49m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f5b26336d3e4f37adfb9431359375e8 : SUCCESS in 39m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/79ec851fecf240b9b94d61e89daed71a : SUCCESS in 36m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0ff510e557a4f8f9dfb3ec4836215e2 : SUCCESS in 21m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b226e000439149d98b9aa96d1000b9ac : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88e6cce10bc3460984773850b7d848c9 : SUCCESS in 18m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d6427d2a8d14e6e934ee906653be9bf : SUCCESS in 49m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00e7220c3d814ef287a5c7b174328edb : SUCCESS in 57m 41s","accounts_in_message":[],"_revision_number":26},{"id":"263d05f5c6073e0aff71d39e632553ac29e1cd1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-07-30 18:44:13.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"57e1a5d9c69bf0d162dbca765118461cbc9ba32e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 19:44:16.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8acf0168cca24cf4a74a91cd09a9a185\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/32d50395848d4248ba7f3ff079dcf603 : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":27},{"id":"8e32200dcbd15a404ad95ade2546b87d102f5b9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 19:50:25.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8e104284a624f69bd93c4522353b508\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2457dfc306c448f69873d364cd26797f : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/45452177248145718f282612e98a8547 : SUCCESS in 49m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cd43f7cf94f74a379fb27c1ce2664bb8 : SUCCESS in 59m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/05e44f4f0fe34091a25a372c710fcc50 : SUCCESS in 56m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9bb6c8c2624441794e7781464f9267b : SUCCESS in 58m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86e108f270f14aa683cdc4e71c616918 : SUCCESS in 21m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf516e5d3f7a488cba867d98ca51c855 : SUCCESS in 10m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb0faa84ffa74378a99e80385ad9caf6 : SUCCESS in 17m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b1a658e2b2294ff399b03c01ad60a5d7 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c9a156e9d07f40e5aee0edea77f60072 : SUCCESS in 21m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/98133be4010d44b69bcaa926ff7b5726 : SUCCESS in 22m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5dc5afda607e4886ab90370444a07f66 : SUCCESS in 16m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a5296594f07d476793e52328d20fd069 : SUCCESS in 29m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6bc99164722147f48c7bfecc22b8c5f2 : SUCCESS in 12m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9ed85e54603348c294c47aba8d55ef73 : SUCCESS in 12m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d34a4d085e6b48e0b24346f85d79b975 : SUCCESS in 15m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4c817ea70e9a4531aa0d7b9d68b9916b : SUCCESS in 5m 55s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7081d3a276014809a062bb5ea28b9ab0 : SUCCESS in 14m 14s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5654e91205f5460db7e59455dfefb03c : SUCCESS in 8m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6267fd5a286d467eabc7edc897d5fc8c : SUCCESS in 52m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5ac54e997b0e4e6e95d54e66759916bc : SUCCESS in 39m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/56d1a407b66a4a729cfa1fe564b54655 : SUCCESS in 50m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07a4de13227d4370b538b0d24859f6aa : SUCCESS in 18m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0e283c6d2a749c6913422519aafb799 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/719a805de26e4357a4c9f4929a969d71 : SUCCESS in 17m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93036aa695004db2b9e46d5fec573968 : SUCCESS in 35m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c632c429b6b341e9b3d9f75ac1907685 : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":27},{"id":"861222add770b9480ac251feee4bca459233d504","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"real_author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-10-14 17:53:30.000000000","message":"Patch Set 28: Patch Set 27 was rebased on behalf of \u003cGERRIT_ACCOUNT_34892\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"_revision_number":28},{"id":"96512b617d4e8e6f139a58fe9a78caf87e125f73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 19:02:28.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7de7ab7353424f499e9c92e582d133cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e4ccd6dc36846bd90d2b2fd79510739 : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/9fb12d83da3c4f598878334eaf2366ae : SUCCESS in 1h 03m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6461e4c9da354663914f54bda6f45080 : SUCCESS in 49m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/46cfc975fd5a40d08964133d5b14a14f : SUCCESS in 52m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fb97907e3a948f2a40dd27d013222c0 : SUCCESS in 59m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9596146935874859a204bf3fc385daf2 : SUCCESS in 19m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02d18a21c02f460db3e4267310caee0f : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3540d9eda7f34b2c8a805a8da2da603b : SUCCESS in 20m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7eb3572a2f64998ad14683cbd1dec4f : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9960e38fcabd4be7aa53eca14ac8b3db : SUCCESS in 26m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1ddc154ba89046b09bf502210bffd248 : SUCCESS in 21m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6a6282589d664f8b934155b66a40f5fa : SUCCESS in 36m 47s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8819a77dca5c41c094bb5749db0de577 : SUCCESS in 29m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/854251a4112b4930a66eff12e808b720 : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d2c3741bcc8b47589728207326838cac : SUCCESS in 13m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0444f437db2d49a1bf17bfa7ea8e4f4a : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b797347510d94495b360fac5c789464e : SUCCESS in 6m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/23abeca235df4465a7f33de0a1efca87 : SUCCESS in 15m 19s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/95e285b1735e4efa96f1b049236ce39c : SUCCESS in 8m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4efbcad14dfc47ba80201924cba10e03 : SUCCESS in 59m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/98e0180e7d594bba883e78876e0eef94 : SUCCESS in 54m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4aaedca01aaa4e14a0f19b787ee5c006 : SUCCESS in 25m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c81f3c96defe4821997f95cecabb94f5 : SUCCESS in 29m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/177f3c50f2334e1982497f93fc3e0f06 : SUCCESS in 3m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d4bbfea9f4b04c7fbc6e8ffe85bc1f50 : SUCCESS in 16m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9377d5c1bd664f3caf26a4841870eb25 : SUCCESS in 49m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c95c1e63ece6470ea6eed69107db0667 : SUCCESS in 40m 03s","accounts_in_message":[],"_revision_number":28},{"id":"7168694b17b587809d9a440f7ef35cb64c04df33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 19:21:06.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00eb587d93f34cc7bd2d35ff03a1709d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0cc804163e8140fe896938eb3ad4c03e : SUCCESS in 1h 19m 38s","accounts_in_message":[],"_revision_number":28},{"id":"675a41717bf856e3ebfbf9ba858ccf4dfbdcf01b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-03-05 00:44:13.000000000","message":"Patch Set 28:\n\n(3 comments)","accounts_in_message":[],"_revision_number":28},{"id":"3ab7a8febaec574718e937890b4f12d104d5f19d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-05-11 05:45:50.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"98941f166a2f2e3f14f82bd43010b5b26f498fc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2026-06-02 14:35:58.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":29},{"id":"6c7eb0af12869bd784b09d6e4cb38aca513f6a2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 15:53:31.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adf4d7640a1049a5a96a418eb30547e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1608b3ea50449f4a70a8eb854428e2a : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d905d5ba5f8b49b4954788ee461fdd3b : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/569e496c9f4c4caa81f9edb92da311bf : SUCCESS in 1h 12m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ac23b71c9e0b4af3b8d10fe9ccb8b099 : SUCCESS in 43m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db8bf013887e4e80bce4a81135faec50 : SUCCESS in 1h 02m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a495298c9ce482ebdf42febded05b7d : SUCCESS in 42m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4391af85b06049038e0585f293096cdd : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b798fd2d334748859b743c000ee5519e : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9fe6253ff4e49ecb7048841e95c01e2 : SUCCESS in 22m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/156a21b6abec488d87ef93d62312173e : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c7400e866d104fcd8e5002274d9ec5ee : SUCCESS in 15m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/19b4315a750f41598425b677e7d4c89f : SUCCESS in 23m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/32d193446d0e4be89955efcc8147dae9 : SUCCESS in 29m 42s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1f25a4631edb4277bf1aaa02b20aca0c : SUCCESS in 15m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6635fb3bd0ee419d833d782e5c270bf1 : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/222f8e2a8ca44d35af007ae814bd6280 : SUCCESS in 12m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4f5da27c9d0f43969926deec7242c21d : SUCCESS in 17m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fdb356b3240d44a6bfc5bb7591f02e69 : SUCCESS in 6m 58s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/09f5251111b14ec5b00e2bca96a26058 : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8dc557b5f963451496e21b7b21db3550 : SUCCESS in 9m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/53322aaced9543e69db2956d45599f52 : SUCCESS in 1h 00m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6f9329f7507b44dfaf5196a59b73b934 : SUCCESS in 51m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f86f27b27bad49d8aefac633c8c93b35 : SUCCESS in 54m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3745af25ebf4386b47166a0562e8e0c : SUCCESS in 30m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28ac8350ef2f4a5686211199a3473615 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fee04348095846468548e4a9cde7cdf2 : SUCCESS in 21m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ae2bdb2809b4dd583c6b9f4ee160cd2 : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":29},{"id":"32d333b1ed07ae52467328d33ba9ea0429b6540a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 18:53:13.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f37ac4e90a174a76b731aa82b308f360\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a3fe57d457ac45a3a78f108aac583bd5 : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":29},{"id":"f528b437cd2bc73d8f700c84a93c2aa1060ad6ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-11 18:25:51.000000000","message":"Uploaded patch set 30.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":30},{"id":"77051c149da3520b17cf1def726c2388ff861d0c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-11 18:26:14.000000000","message":"Patch Set 30:\n\n(2 comments)","accounts_in_message":[],"_revision_number":30},{"id":"216fde46afaf13faeeb52c97a67646e36bed8d8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 19:36:51.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1105fdeb58a7457c912a5fd7c66f46ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28dce387136b4afba5aaec92a93692f1 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66df8cfd5a694ff09f013a33f6c23804 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/7c5df2435e644ff8af2a8b4a672de16c : SUCCESS in 46m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9a935bb7aaaa44f89abf1eecdddecef2 : SUCCESS in 44m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/02f49a1fc8f342c6a244dff2a11973af : SUCCESS in 45m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3c6eb3688a9419480a14aa0af8d18f2 : SUCCESS in 49m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/666416fbcb784feca0711e222bc05d64 : SUCCESS in 16m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce8cbe532b144347914e5198f8281977 : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97ee4009854e4fba9d2e308cc20d50ed : SUCCESS in 20m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d12a6f0c40ae4c778f50808a502c3249 : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3e083e0f58ca4df68732e6b0100f61d9 : SUCCESS in 26m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d72d46a1f7f743999dc1047e078adfe9 : SUCCESS in 22m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/16a58d84fb5349f59ceb0aedf3049734 : SUCCESS in 35m 01s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/1e2eb19a114d4a129e8a19a3f519e9af : SUCCESS in 21m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/98143fb7ad0e44a7a90e46c273ddd7bb : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2920172e589b4e1fa937a7311ba2f262 : SUCCESS in 15m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a4cbdbbebed44bc19eb793938b86019d : SUCCESS in 15m 29s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/507d0e2f53b448bf8445d13b49bd795c : SUCCESS in 18m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cf120a2aa21b45278fa11732ed0515ca : SUCCESS in 6m 16s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3f515e3c760749dba64c61f4d7e890cc : FAILURE in 7m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ccc46dfec08248c3b000558cbca80053 : SUCCESS in 5m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f2e1ea6bed34f6bb55619653a1ef40e : SUCCESS in 1h 04m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/61938b27e45c45009014e87603e2b23c : SUCCESS in 38m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/26bbacfded114f70b330a18ac8bcf676 : SUCCESS in 52m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1e82282549e483ba0976495e721b4b8 : SUCCESS in 21m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34a047fa571e4455a4025d5cb5828bb5 : SUCCESS in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/047194d2f30342a8aa37a17bd462e49d : SUCCESS in 18m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4f5a0ef693246c2bd5c84ff58098805 : SUCCESS in 33m 16s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":30},{"id":"32ea906560279b0f0bc5312bcd9d1ae343b4ce3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-11 19:48:02.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c2cc4be3a544d1d85c5e3772e054fcb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5b63d53832814303b44c531ff79d0888 : SUCCESS in 1h 19m 01s","accounts_in_message":[],"_revision_number":30},{"id":"878f3165836f23d48cb910e4b1a4fa3d9484c29c","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-08-20 11:50:34.000000000","message":"Patch Set 30: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"8c755accb4c408c78e6f1af0ee30eefaa6eb0aca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-20 22:12:43.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":31},{"id":"0716fe4206119fffe16d81a0b205eee7fa453607","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-20 22:13:01.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"f03b661fcbe2dbee665e4561137adf38b7005b06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 23:16:56.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdce1849d16a49e69780550f066a8732\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef36ae53bf8a4897ac570738a09914c6 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1803162c75a4e409778f20450232977 : SUCCESS in 3m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/394942ecf2124a828a83805e0ef3f0b2 : SUCCESS in 35m 07s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/597d99376e5742d6ab8b04398b0de592 : SUCCESS in 29m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/39ce17e43a984083b6dfc97d3783dea5 : SUCCESS in 30m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/460e03cb077644c2958691aa02d9a882 : SUCCESS in 56m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89c7fcf686784b47b3a34fa698b53349 : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60065d6f151f43c8951209a55852555d : SUCCESS in 10m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6aa2f58b9234ddfa6e5de92031bd3c0 : SUCCESS in 12m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c8f3a9061e742b49fdb24dcb1581cf2 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9f70269d81994e9c821db09ec73f8ad4 : SUCCESS in 10m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/849788cced314445850b251700f3fbb6 : SUCCESS in 9m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f5a678477c98457aa07606b47a828366 : SUCCESS in 32m 07s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/edd5e983608943eb8424880996e7da55 : SUCCESS in 9m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3403e9eeb0ea488ab87d13b282798d57 : SUCCESS in 14m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7c9c03b087d24c1eaab00dd00121fcde : SUCCESS in 8m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0c0e94e2137747eebda1866a6b5a8c85 : SUCCESS in 15m 57s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/045c14ea8e514e70bdaf5304e7b91599 : SUCCESS in 9m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b2b864486774ea781efa522c8e26ea1 : SUCCESS in 6m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b38134b35175473e9f3ca4535aa287cc : FAILURE in 6m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/aadfb7bf9eeb4e028783559943606102 : SUCCESS in 4m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aecc49f2e5184ceaae663385a4221ca7 : SUCCESS in 36m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7e800290226c49bdae86596707e9a51a : SUCCESS in 36m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b41e9cffddcb4a50b57222e5b38812f3 : SUCCESS in 22m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e54058b60b034222baf9bf94fbf8f336 : SUCCESS in 28m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a81d453f2b924e65a127fa8c8ed9bb4b : SUCCESS in 2m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/987d689e767444978adf4098fedfc83e : SUCCESS in 10m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b4f69d1b060459da4217098f61e64da : SUCCESS in 57m 41s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":31},{"id":"46f36e64463d8ea88057a66c0417742d62b7cc1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 23:34:43.000000000","message":"Patch Set 31:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/752c9c6c86e64717a77b404c02c3b09e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b47da7e8b9274d26906e6f91fc04cf37 : POST_FAILURE in 1h 16m 15s","accounts_in_message":[],"_revision_number":31},{"id":"6c61b6c8679bddf50228aa26198d4539fa9230a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-08-28 18:35:28.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":32},{"id":"75c741f97352e595b959012c2cfaa4d98b8f0583","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 19:13:54.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe0dd3f6920645bfa458c97fab3ab720\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ce4ce36e3ef240fda8fc3463f8381443 : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":32},{"id":"8a123aa8d6c29fabf01c6a8f9ad1be8c17f578b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-28 19:46:49.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c20b10b0b2948398d0f3d7afe9d8b4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0fd73ba71704ef69870825aba9709e7 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdf7aa0b48744987a15e90ee8c013c21 : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/79f7d1f88c094029b5ddda4d9c7a5a19 : SUCCESS in 1h 03m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/63f78f0676d44645ad17827cada84c51 : SUCCESS in 49m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4bcbe52fb447442aab4178dfa6c065cc : SUCCESS in 51m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd54f06101b740dda2bc5d3ccf6ee482 : SUCCESS in 1h 03m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84379dc39e6c47abb1113b0bdb60a29b : SUCCESS in 18m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9bd2c2c35c1d4e27a493cab3df342b6b : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b498f2f6354e4e11abf6abbd258604b4 : SUCCESS in 25m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c478a54c9fca45a986d7142018f7a0ba : SUCCESS in 9m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/00b4207fcb3a48f2a20f4c14c0277170 : SUCCESS in 12m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/df759e05d7df4371ac8b093825871aee : SUCCESS in 20m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/661878026283467681722d78a01d2a54 : SUCCESS in 30m 10s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/99af7cef1e8649c7bbd27380634c31ec : SUCCESS in 19m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bce95f63210d4e3faf796e636d1a876c : SUCCESS in 14m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9238243ed981494fbd940b84366810e2 : SUCCESS in 11m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0909bd8be8a1401bae1d8cf4d300ad08 : SUCCESS in 16m 09s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/06e7d0564dc64585b34718f019f88c68 : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8b354970ffe34c3d95e29b38a93a9058 : SUCCESS in 5m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/62e1aaf6a05147f4bd003411cbc7a354 : FAILURE in 6m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f9b8d2a3af5946de90079e5454daf6ec : SUCCESS in 8m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b115661d4fcc46de909d2f092fd9d8de : SUCCESS in 49m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a606de308d1c4916aa6db3925850071c : SUCCESS in 57m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5253c9f04e8447a5ade22d6328968339 : SUCCESS in 1h 00m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb160509fefc4ead996b656f3482bf94 : SUCCESS in 25m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7df2274bbe0647089eb55b3bbb19ac0d : SUCCESS in 3m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ab6b146af1c46bc9f393932dd32da64 : SUCCESS in 17m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/330df516387e45868e75de1c5b7925d6 : SUCCESS in 56m 40s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":32},{"id":"5e74c22c628a795292cedfecf2a948927242b123","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-09-03 16:44:57.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":33},{"id":"7644d13b430e8dd4c09bd2cc707cc014cdee149e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 17:52:14.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e1670e42eb64e218ad8b98d53519d14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16df892c21584fa8bbe3216ecea56949 : SUCCESS in 10m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fabf0dbc1f944a985b70c2aea8149f3 : SUCCESS in 4m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/cb412614fd524a068c3aadad4d802478 : SUCCESS in 56m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5cddbe3529d04a54817ba27ff029bbf9 : SUCCESS in 33m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eb8016df4adc4f23bd03cc913a20be07 : SUCCESS in 28m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0edebe79a45a4e1185e21e212e184e62 : SUCCESS in 52m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f4ef922321f743ec8889947b9640f2d0 : SUCCESS in 20m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6a9a995551342d19e2b8d17d405f594 : SUCCESS in 21m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31178070669a466aa6eac13f6f1575eb : SUCCESS in 23m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5125303508de4c1394440a4d7e209bb6 : SUCCESS in 7m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9dbff2c096924657bc40e20c31bc36e7 : SUCCESS in 12m 21s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/249e7908752949a4b8c3737aeb36d571 : SUCCESS in 10m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/32e3e27625574ec3bf0fb7143c955972 : SUCCESS in 27m 37s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b69eae92b39a4c3c9b9a72e7e683f48a : SUCCESS in 10m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6c853ad2be914ff99876a16e56736def : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0db8083e76264070a6d5e47f255871ab : SUCCESS in 16m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fa4529c87b6540aeab919fbb86c634cd : SUCCESS in 15m 05s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa4f279301f5469db22264236c3b330e : SUCCESS in 17m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3c3c0d1b78b74cbb985dc7b966e557b4 : SUCCESS in 7m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e7ef48c68ef54ea483bd44155792472d : FAILURE in 32m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/62207df9050e422aa9108ef6323d5e56 : SUCCESS in 9m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ae7b80e88684ce1aba123de19b3705e : SUCCESS in 58m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9fda04e5c8e54f75abc5753571e6aad9 : SUCCESS in 38m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fcde94d39f9e4ef492d251783463490c : SUCCESS in 1h 03m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/904404ed5c4840558e43c8ae9bdb2cd8 : SUCCESS in 29m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7669c602e1984a02bd632d24a8518c93 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9c0546fefcb143e5baf494434026e02d : SUCCESS in 17m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d26a4b2dc9fc41c9a89fecfdb45771ac : SUCCESS in 27m 37s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":33},{"id":"bf4be5c9e1516c8397518c93318e5035b7acd380","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 18:05:38.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8547b40baeed4ad180b24d10963a8911\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1cf04e55de394c9d955e296b8e5632e1 : SUCCESS in 1h 17m 08s","accounts_in_message":[],"_revision_number":33}],"current_revision_number":33,"current_revision":"377a20108e9f875336e5f3da9b3ac18bb2e1aae3","revisions":{"f657bd0fd5c73c5721255419ae97faefef8a057c":{"kind":"REWORK","_number":1,"created":"2022-04-27 23:28:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/1"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:27:47.000000000","tz":-420},"subject":"replicator: Add rsync_max_suffixes_per_revert option","message":"replicator: Add rsync_max_suffixes_per_revert option\n\nStill try to revert all handoff data in the first pass -- just be\nwilling to break it up into (randomized) batches so we can show progress\n(and delete data!) more frequently. If any of the batches fail,\nimmediately bomb out.\n\nThis reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f657bd0fd5c73c5721255419ae97faefef8a057c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f657bd0fd5c73c5721255419ae97faefef8a057c"}]},"branch":"refs/heads/master"},"c006f2120d4cf7a2f057af4b7d1e1825063bcee9":{"kind":"REWORK","_number":2,"created":"2022-05-03 18:03:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/2"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-05-03 18:03:49.000000000","tz":-420},"subject":"replicator: Add rsync_max_suffixes_per_revert option","message":"replicator: Add rsync_max_suffixes_per_revert option\n\nStill try to revert all handoff data in the first pass -- just be\nwilling to break it up into (randomized) batches so we can show progress\n(and delete data!) more frequently. If any of the batches fail,\nimmediately bomb out.\n\nThis reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c006f2120d4cf7a2f057af4b7d1e1825063bcee9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c006f2120d4cf7a2f057af4b7d1e1825063bcee9"}]},"branch":"refs/heads/master"},"06efd74d8374b6160f4161ead0a525f7ef1caa79":{"kind":"REWORK","_number":3,"created":"2022-05-28 06:36:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/3"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-05-28 05:44:47.000000000","tz":-420},"subject":"replicator: Add rsync_max_suffixes_per_revert option","message":"replicator: Add rsync_max_suffixes_per_revert option\n\nStill try to revert all handoff data in the first pass -- just be\nwilling to break it up into (randomized) batches so we can show progress\n(and delete data!) more frequently. If any of the batches fail,\nimmediately bomb out.\n\nThis reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06efd74d8374b6160f4161ead0a525f7ef1caa79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/06efd74d8374b6160f4161ead0a525f7ef1caa79"}]},"branch":"refs/heads/master"},"2ef38df5781400407646cf4fac974ff7dc102583":{"kind":"REWORK","_number":4,"created":"2022-05-31 15:24:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/4"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-05-31 15:23:31.000000000","tz":-420},"subject":"replicator: Add rsync_max_suffixes_per_revert option","message":"replicator: Add rsync_max_suffixes_per_revert option\n\nStill try to revert all handoff data in the first pass -- just be\nwilling to break it up into (randomized) batches so we can show progress\n(and delete data!) more frequently. If any of the batches fail,\nimmediately bomb out.\n\nThis reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ef38df5781400407646cf4fac974ff7dc102583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ef38df5781400407646cf4fac974ff7dc102583"}]},"branch":"refs/heads/master"},"ad0774e5b5c48ee4813a608958ce123766d0374b":{"kind":"REWORK","_number":5,"created":"2022-06-24 19:17:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/5"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-24 19:15:17.000000000","tz":-420},"subject":"replicator: Add rsync_batches_per_revert option","message":"replicator: Add rsync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be\nwilling to break it up into (randomized) batches so we can show progress\n(and delete data!) more frequently. If any of the batches fail,\nimmediately bomb out.\n\nThis reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad0774e5b5c48ee4813a608958ce123766d0374b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad0774e5b5c48ee4813a608958ce123766d0374b"}]},"branch":"refs/heads/master"},"8493f1d81a999f5642fe2381008520aa6048b797":{"kind":"REWORK","_number":6,"created":"2022-06-24 20:44:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/6"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-24 20:43:24.000000000","tz":-420},"subject":"replicator: Add rsync_batches_per_revert option","message":"replicator: Add rsync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be\nwilling to break it up into (randomized) batches so we can show progress\n(and delete data!) more frequently. If any of the batches fail,\nimmediately bomb out.\n\nThis reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8493f1d81a999f5642fe2381008520aa6048b797"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8493f1d81a999f5642fe2381008520aa6048b797"}]},"branch":"refs/heads/master"},"0bac190a800baad6d18c772c968dcc8643de0812":{"kind":"REWORK","_number":7,"created":"2022-06-28 22:28:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/7"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-28 22:21:26.000000000","tz":-420},"subject":"replicator: Add rsync_batches_per_revert option","message":"replicator: Add rsync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be\nwilling to break it up into (randomized) batches so we can show progress\n(and delete data!) more frequently. If any of the batches fail,\ncontinue to the next batch but don\u0027t delete the partition.\n\nThis reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bac190a800baad6d18c772c968dcc8643de0812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bac190a800baad6d18c772c968dcc8643de0812"}]},"branch":"refs/heads/master"},"e4738b70f324095331068754bb1b1833661f6451":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2022-06-28 23:11:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/8"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2022-06-28 23:11:40.000000000","tz":0},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be\nwilling to break it up into (randomized) batches so we can show progress\n(and delete data!) more frequently. If any of the batches fail,\ncontinue to the next batch but don\u0027t delete the partition.\n\nThis reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4738b70f324095331068754bb1b1833661f6451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4738b70f324095331068754bb1b1833661f6451"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ce55760a18940d67f580770602069af3776d5ca5":{"kind":"REWORK","_number":9,"created":"2022-08-22 20:06:42.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/9"}}},"commit":{"parents":[{"commit":"f6196b0a221eb17e1d5649d732b997dc200b2fe7","subject":"AUTHORS/CHANGELOG for 2.30.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6196b0a221eb17e1d5649d732b997dc200b2fe7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-22 19:57:33.000000000","tz":-420},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be\nwilling to break it up into (randomized) batches so we can show progress\n(and delete data!) more frequently. If any of the batches fail,\ncontinue to the next batch but don\u0027t delete the partition.\n\nThis reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nAdd a new interface to the hashes.invalid API -- if the line starts with\n\u0027!\u0027, remove the suffix during consolidation rather than just setting it\nto None.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce55760a18940d67f580770602069af3776d5ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce55760a18940d67f580770602069af3776d5ca5"}]},"branch":"refs/heads/master"},"5bb0417186e26b578f95efbf053818d673883c84":{"kind":"REWORK","_number":10,"created":"2022-10-24 22:08:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/10"}}},"commit":{"parents":[{"commit":"8d541fed4ee12d8472c15a62049de6c14bc0b02a","subject":"Merge \"Clean up some replication error messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d541fed4ee12d8472c15a62049de6c14bc0b02a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-24 22:07:57.000000000","tz":-420},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be\nwilling to break it up into (randomized) batches so we can show progress\n(and delete data!) more frequently. If any of the batches fail,\ncontinue to the next batch but don\u0027t delete the partition.\n\nThis reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nAdd a new interface to the hashes.invalid API -- if the line starts with\n\u0027!\u0027, remove the suffix during consolidation rather than just setting it\nto None.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bb0417186e26b578f95efbf053818d673883c84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bb0417186e26b578f95efbf053818d673883c84"}]},"branch":"refs/heads/master"},"c88b68c80090425a80f5166919ace02212c611aa":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-02-22 05:45:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/11"}}},"commit":{"parents":[{"commit":"63e618b232c5f1a50e16baa0e685f4e25716f737","subject":"Test x-backend-replication header with POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63e618b232c5f1a50e16baa0e685f4e25716f737"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-02-22 05:44:54.000000000","tz":-480},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be\nwilling to break it up into (randomized) batches so we can show progress\n(and delete data!) more frequently. If any of the batches fail,\ncontinue to the next batch but don\u0027t delete the partition.\n\nThis reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nAdd a new interface to the hashes.invalid API -- if the line starts with\n\u0027!\u0027, remove the suffix during consolidation rather than just setting it\nto None.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c88b68c80090425a80f5166919ace02212c611aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c88b68c80090425a80f5166919ace02212c611aa"}]},"branch":"refs/heads/master"},"7e3f0563854cabc839e93e8716dddb37f0b41979":{"kind":"REWORK","_number":12,"created":"2023-06-27 04:40:03.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/49/839649/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/12"}}},"commit":{"parents":[{"commit":"7b1f7ee857b0d7019b3e6e17a75d38ebbf650c15","subject":"Merge \"py3: Quiet RemoteDisconnected tracebacks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b1f7ee857b0d7019b3e6e17a75d38ebbf650c15"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2023-06-27 04:39:55.000000000","tz":-420},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be\nwilling to break it up into (randomized) batches so we can show progress\n(and delete data!) more frequently. If any of the batches fail,\ncontinue to the next batch but don\u0027t delete the partition.\n\nThis reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nAdd a new interface to the hashes.invalid API -- if the line starts with\n\u0027!\u0027, remove the suffix during consolidation rather than just setting it\nto None.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e3f0563854cabc839e93e8716dddb37f0b41979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e3f0563854cabc839e93e8716dddb37f0b41979"}]},"branch":"refs/heads/master"},"df279b2ef18b61e010909bdc4ccd311495feeef4":{"kind":"REWORK","_number":13,"created":"2024-01-29 18:59:58.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/13"}}},"commit":{"parents":[{"commit":"7b1f7ee857b0d7019b3e6e17a75d38ebbf650c15","subject":"Merge \"py3: Quiet RemoteDisconnected tracebacks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b1f7ee857b0d7019b3e6e17a75d38ebbf650c15"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-01-29 18:58:48.000000000","tz":-480},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be\nwilling to break it up into (randomized) batches so we can show progress\n(and delete data!) more frequently. If any of the batches fail,\ncontinue to the next batch but don\u0027t delete the partition.\n\nThis reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nAdd a new interface to the hashes.invalid API -- if the line starts with\n\u0027!\u0027, remove the suffix during consolidation rather than just setting it\nto None.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df279b2ef18b61e010909bdc4ccd311495feeef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df279b2ef18b61e010909bdc4ccd311495feeef4"}]},"branch":"refs/heads/master"},"4353dc6e513fff9853879d12be88ca01e693610d":{"kind":"REWORK","_number":14,"created":"2024-02-07 07:16:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/14"}}},"commit":{"parents":[{"commit":"c5d743347c0f8e9aa0c1f64c26231ae0ce45462e","subject":"diskfile: Ignore invalid suffixes in invalidations file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5d743347c0f8e9aa0c1f64c26231ae0ce45462e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-02-07 07:15:47.000000000","tz":-480},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4353dc6e513fff9853879d12be88ca01e693610d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4353dc6e513fff9853879d12be88ca01e693610d"}]},"branch":"refs/heads/master"},"8238165554eafe6ca7ee680c05fc205967b47ca6":{"kind":"REWORK","_number":15,"created":"2024-05-21 03:51:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/15"}}},"commit":{"parents":[{"commit":"337079f21f66c4f288c5ffa736a177f446472e5e","subject":"Merge \"replicator: Ensure handoffs can clear with large handoff_delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/337079f21f66c4f288c5ffa736a177f446472e5e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-21 03:51:09.000000000","tz":-420},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8238165554eafe6ca7ee680c05fc205967b47ca6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8238165554eafe6ca7ee680c05fc205967b47ca6"}]},"branch":"refs/heads/master"},"119e8637432d323b4a92e5408878199109f5ecac":{"kind":"REWORK","_number":16,"created":"2024-05-21 15:04:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/16"}}},"commit":{"parents":[{"commit":"337079f21f66c4f288c5ffa736a177f446472e5e","subject":"Merge \"replicator: Ensure handoffs can clear with large handoff_delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/337079f21f66c4f288c5ffa736a177f446472e5e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-21 15:04:24.000000000","tz":-420},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/119e8637432d323b4a92e5408878199109f5ecac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/119e8637432d323b4a92e5408878199109f5ecac"}]},"branch":"refs/heads/master"},"0555d3eba874ad35324d230c3878e19936a11e53":{"kind":"REWORK","_number":17,"created":"2024-08-05 21:20:49.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/49/839649/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/17"}}},"commit":{"parents":[{"commit":"b400a1fdb3ae609d5f86d4d0f1b590640337b8f0","subject":"Expirer: add delete task iteration related metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b400a1fdb3ae609d5f86d4d0f1b590640337b8f0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-05 21:20:05.000000000","tz":-300},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0555d3eba874ad35324d230c3878e19936a11e53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0555d3eba874ad35324d230c3878e19936a11e53"}]},"branch":"refs/heads/master"},"8ef87396dab9be4e0e43ec65608f029ae2573b1c":{"kind":"REWORK","_number":18,"created":"2024-10-21 14:44:52.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/49/839649/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/18"}}},"commit":{"parents":[{"commit":"4f69ab3c5dfaa8dcd321bdabb74dc079f7973d82","subject":"fix x-open-expired 404 on HEAD?part-number reqs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f69ab3c5dfaa8dcd321bdabb74dc079f7973d82"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-21 14:44:42.000000000","tz":-300},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ef87396dab9be4e0e43ec65608f029ae2573b1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ef87396dab9be4e0e43ec65608f029ae2573b1c"}]},"branch":"refs/heads/master"},"0864ab1254555766c4e085474c6a3f25cd09054b":{"kind":"REWORK","_number":19,"created":"2024-11-19 00:34:43.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/49/839649/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/19"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-11-19 00:34:43.000000000","tz":0},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0864ab1254555766c4e085474c6a3f25cd09054b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0864ab1254555766c4e085474c6a3f25cd09054b"}]},"branch":"refs/heads/master","description":"Rebase"},"0a01f21a85632fd0fb49e75ea45c410d4ac7d6bf":{"kind":"REWORK","_number":20,"created":"2024-11-19 00:40:58.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/49/839649/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/20"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-11-19 00:40:46.000000000","tz":-480},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a01f21a85632fd0fb49e75ea45c410d4ac7d6bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a01f21a85632fd0fb49e75ea45c410d4ac7d6bf"}]},"branch":"refs/heads/master"},"c1f0124a3adacff30e5e32e44d1d96f8b9cbdf8b":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2025-04-11 00:10:52.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/49/839649/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/21"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-04-11 00:10:52.000000000","tz":0},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1f0124a3adacff30e5e32e44d1d96f8b9cbdf8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c1f0124a3adacff30e5e32e44d1d96f8b9cbdf8b"}]},"branch":"refs/heads/master","description":"Rebase"},"4d6063219e703162ed99cdf8769ba98ced041275":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2025-05-19 23:36:32.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/49/839649/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/22"}}},"commit":{"parents":[{"commit":"84a70769b1c56cc376a1485391f81569cd7ce494","subject":"Merge \"s3api: Allow PUT with `if-none-match: *`\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84a70769b1c56cc376a1485391f81569cd7ce494"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-05-19 23:36:32.000000000","tz":0},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d6063219e703162ed99cdf8769ba98ced041275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d6063219e703162ed99cdf8769ba98ced041275"}]},"branch":"refs/heads/master","description":"Rebase"},"4980521d147f3d0b56ffbcbcee9e75babe394a2d":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2025-05-20 14:31:37.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/49/839649/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/23"}}},"commit":{"parents":[{"commit":"b735b3d034b11e1472f7b471919649d165c447e0","subject":"object-server: return 503 not 404 if meta or data file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b735b3d034b11e1472f7b471919649d165c447e0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-05-20 14:31:37.000000000","tz":0},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4980521d147f3d0b56ffbcbcee9e75babe394a2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4980521d147f3d0b56ffbcbcee9e75babe394a2d"}]},"branch":"refs/heads/master","description":"Rebase"},"2fa085a5dd55cba89d8601a9881a101ba6e40842":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2025-05-20 15:56:54.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/49/839649/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/24"}}},"commit":{"parents":[{"commit":"84a70769b1c56cc376a1485391f81569cd7ce494","subject":"Merge \"s3api: Allow PUT with `if-none-match: *`\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/84a70769b1c56cc376a1485391f81569cd7ce494"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-05-20 14:53:23.000000000","tz":-420},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fa085a5dd55cba89d8601a9881a101ba6e40842"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fa085a5dd55cba89d8601a9881a101ba6e40842"}]},"branch":"refs/heads/master"},"26cc99b57a94db52c5422d3d8269468353309c23":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2025-05-20 16:59:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/25"}}},"commit":{"parents":[{"commit":"b735b3d034b11e1472f7b471919649d165c447e0","subject":"object-server: return 503 not 404 if meta or data file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b735b3d034b11e1472f7b471919649d165c447e0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-20 16:58:10.000000000","tz":-420},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26cc99b57a94db52c5422d3d8269468353309c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26cc99b57a94db52c5422d3d8269468353309c23"}]},"branch":"refs/heads/master"},"c04061894f81ad52e6c210ba2f21bf773c23cec3":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2025-05-29 01:09:42.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/49/839649/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/26"}}},"commit":{"parents":[{"commit":"d0e97650f3d1e832135c3b220d0928f5cd986501","subject":"Merge \"CI: run S3 cross-compat tests as part of DSVM jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0e97650f3d1e832135c3b220d0928f5cd986501"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-05-29 01:00:34.000000000","tz":-420},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c04061894f81ad52e6c210ba2f21bf773c23cec3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c04061894f81ad52e6c210ba2f21bf773c23cec3"}]},"branch":"refs/heads/master"},"e5cdd0ff4813867c647d947659bebfc673069b59":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":27,"created":"2025-07-30 18:44:13.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/49/839649/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/27"}}},"commit":{"parents":[{"commit":"f39133055fff953f69f6b7ea62c47127d5b5c9a5","subject":"Merge \"sharder: make gap and overlap warning logs shorter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f39133055fff953f69f6b7ea62c47127d5b5c9a5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-07-30 18:43:29.000000000","tz":-420},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\nSigned-off-by: indianwhocodes \u003cnairashwin952013@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5cdd0ff4813867c647d947659bebfc673069b59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5cdd0ff4813867c647d947659bebfc673069b59"}]},"branch":"refs/heads/master"},"e2d46f1fe2e8d19f3cd269d6682fb102678f9eb8":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2025-10-14 17:53:30.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"real_uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/49/839649/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/28"}}},"commit":{"parents":[{"commit":"48a5d5e42f017ab1438fcb257cfc53d3818f1b02","subject":"Merge \"test-db-replicator (trivial): just one tmpdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48a5d5e42f017ab1438fcb257cfc53d3818f1b02"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-10-14 17:53:30.000000000","tz":0},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\nSigned-off-by: indianwhocodes \u003cnairashwin952013@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2d46f1fe2e8d19f3cd269d6682fb102678f9eb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2d46f1fe2e8d19f3cd269d6682fb102678f9eb8"}]},"branch":"refs/heads/master","description":"Rebase"},"d9cf904be261b033a610c5aeed52a04be1540c9d":{"kind":"REWORK","_number":29,"created":"2026-06-02 14:35:58.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/49/839649/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/29"}}},"commit":{"parents":[{"commit":"f52988bb3f3bca035ea913643d71114d43ef43ac","subject":"Merge \"Reject oversized chunked S3 XML bodies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f52988bb3f3bca035ea913643d71114d43ef43ac"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2026-06-02 14:33:58.000000000","tz":-240},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\nSigned-off-by: indianwhocodes \u003cnairashwin952013@gmail.com\u003e\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9cf904be261b033a610c5aeed52a04be1540c9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9cf904be261b033a610c5aeed52a04be1540c9d"}]},"branch":"refs/heads/master"},"26a0a37f4a8390e3a88e72e4ef6598ca4998befc":{"kind":"REWORK","_number":30,"created":"2026-08-11 18:25:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/30"}}},"commit":{"parents":[{"commit":"482a93529b5822e28e707bccd94037acf4c16d8d","subject":"Merge \"versioning: more tests for OPTIONS requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/482a93529b5822e28e707bccd94037acf4c16d8d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-11 18:24:37.000000000","tz":-420},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\nSigned-off-by: indianwhocodes \u003cnairashwin952013@gmail.com\u003e\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26a0a37f4a8390e3a88e72e4ef6598ca4998befc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26a0a37f4a8390e3a88e72e4ef6598ca4998befc"}]},"branch":"refs/heads/master"},"f9bcd43140f8ca2c350081cbbb9059b9874704ab":{"kind":"REWORK","_number":31,"created":"2026-08-20 22:12:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/31"}}},"commit":{"parents":[{"commit":"482a93529b5822e28e707bccd94037acf4c16d8d","subject":"Merge \"versioning: more tests for OPTIONS requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/482a93529b5822e28e707bccd94037acf4c16d8d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-20 22:12:28.000000000","tz":-420},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\nSigned-off-by: indianwhocodes \u003cnairashwin952013@gmail.com\u003e\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9bcd43140f8ca2c350081cbbb9059b9874704ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9bcd43140f8ca2c350081cbbb9059b9874704ab"}]},"branch":"refs/heads/master"},"85ed99e836962d1ca46a9dad389693a46f2fe1d3":{"kind":"REWORK","_number":32,"created":"2026-08-28 18:35:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/32"}}},"commit":{"parents":[{"commit":"482a93529b5822e28e707bccd94037acf4c16d8d","subject":"Merge \"versioning: more tests for OPTIONS requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/482a93529b5822e28e707bccd94037acf4c16d8d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-08-28 18:09:48.000000000","tz":-420},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\nSigned-off-by: indianwhocodes \u003cnairashwin952013@gmail.com\u003e\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85ed99e836962d1ca46a9dad389693a46f2fe1d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85ed99e836962d1ca46a9dad389693a46f2fe1d3"}]},"branch":"refs/heads/master"},"377a20108e9f875336e5f3da9b3ac18bb2e1aae3":{"kind":"REWORK","_number":33,"created":"2026-09-03 16:44:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/49/839649/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/839649/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/839649/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/839649/33"}}},"commit":{"parents":[{"commit":"5cab258d88fd56fc341ec8768682f96be1e738ab","subject":"Merge \"trivial: Remove some py2 comments and validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cab258d88fd56fc341ec8768682f96be1e738ab"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-27 23:01:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2026-09-03 16:44:13.000000000","tz":-420},"subject":"replicator: Add sync_batches_per_revert option","message":"replicator: Add sync_batches_per_revert option\n\nDefault to a single batch to match prior behavior; this default may\nchange as we gain more operational experience.\n\nStill try to revert all handoff data in the first pass -- just be able\nto break it up into a configurable number of (randomized) batches so we\ncan show progress (and delete data!) more frequently. If any of the\nbatches fail, continue to the next batch but don\u0027t delete the partition.\n\nThis has two main applications:\n\n- When a cluster is nearing full, the progressive delete provides relief\n  to full disks ASAP, without waiting for the complete partition to be\n  sync\u0027ed.\n\n- When old hardware is being decomissioned, drive errors may cause some\n  suffixes not only to cause rsync errors during replication, but also\n  to prevent quarantining during auditing. In such a case, the entire\n  partition would previously hang around indefinitely; with batching,\n  the partition would gradually drain, leaving behind only the bad\n  suffixes.\n\nIt also reduces the likelihood of timeouts (either rsync_timeout or\nhttp_timeout during REPLICATE calls). These timeouts now apply\nper-batch.\n\nAdditionally, it gives periodic opportunities for other replicators to\nsneak in and grab an rsync connection slot.\n\nSee also: https://review.opendev.org/c/openstack/swift/+/818017\n\nChange-Id: Ie12dcb4e524ee3af3ddc2b70c027fb62f68720fa\nSigned-off-by: indianwhocodes \u003cnairashwin952013@gmail.com\u003e\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nSigned-off-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/377a20108e9f875336e5f3da9b3ac18bb2e1aae3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/377a20108e9f875336e5f3da9b3ac18bb2e1aae3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
