)]}'
{"id":"openstack%2Fswift~839674","triplet_id":"openstack%2Fswift~master~Ieb52c187ee16bd843c72efa97a6ccda3a166c49f","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-10-17 04:06:39.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-10-17 04:06:39.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"hashtags":[],"change_id":"Ieb52c187ee16bd843c72efa97a6ccda3a166c49f","subject":"ringv2: Add _dispersion_graph to v2","status":"NEW","created":"2022-04-28 01:13:55.000000000","updated":"2022-10-17 04:06:39.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"abfeb04c76aa3cb2eaa5283b400f39eafc46a23e","_number":839674,"virtual_id_number":839674,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-28 05:25:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 01:14:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-28 01:14:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-28 20:34:03.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"047a936154ed6345f447399a8c08205a2fb3f2d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-04-28 01:13:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9b2cfeed2c56feaf97e215a06cb8940e43fe949","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 01:14:33.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 834423,3","accounts_in_message":[],"_revision_number":1},{"id":"612a575b40531d690fad73c2040238bf5a67e220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 01:14:33.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 834423,3","accounts_in_message":[],"_revision_number":1},{"id":"a2effdf08b3d418a66a8a080c707c4440b82867f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-04-28 01:17:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e47998cb38ac4f2337d838ccb47dd7108ba5ee7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-04-28 04:20:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b32a7d8ce8dd82e46cad21267ed361cfec9b5c18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 05:25:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c59b157c33ec4dd8b87f2c93c226278a : SUCCESS in 8m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/68056763ed4145caab6c8be23b617f15 : SUCCESS in 46m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bd6d29e4b0274f4dba542b617a1c27ec : FAILURE in 20m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c35d4dba0ca3499288b57530569529bf : POST_FAILURE in 24m 18s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b997161ea3aa43e08abb6ea28912634c : SUCCESS in 26m 01s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0ecc8b93a620485f9acff5df53293166 : SUCCESS in 23m 06s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3c65e731a0084a3b8c8aadfeff1fb1f9 : SUCCESS in 24m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e6a64278691c4cb99fafb10790df29e4 : SUCCESS in 7m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8bb88a62f99945f6b1eaff6041f5ef19 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c135aaba14243c08262b940f0adbb34 : FAILURE in 19m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1f41fedcb6a44d7d83eb64b80a81598f : FAILURE in 20m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1f84c9c6a2f8454da08262cf62a744cd : FAILURE in 16m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5e8fdfc3f9a9458c9a1c8857685b8e88 : SUCCESS in 20m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/215f8209a7d5474dbfcc83cf913e99ed : SUCCESS in 19m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/73ea9291617c402492ec3375d3ebf41c : SUCCESS in 19m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cb1e6f4d70034f55b3baea0a70160052 : SUCCESS in 20m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4a02989aa07c408a8a776b92950d1cf5 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f1902d4139c74ffb838a8bea81393825 : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4f2861a275e04cf784b1189aca8708ed : RETRY_LIMIT in 4m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9841233cf43c4f309d1bb7dd12946926 : RETRY_LIMIT in 5m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c188816cc3bc4648be112d9a11592ea1 : RETRY_LIMIT in 4m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d08c5b7e1995422aa7773f80c0727d5f : RETRY_LIMIT in 5m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/99224ddd3752417891151ebbb66ac4fa : RETRY_LIMIT in 12m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f9ae4cf7e254080a9864325ad0596bf : RETRY_LIMIT in 12m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91f987d7fd974bf787c594849c79f7f6 : FAILURE in 18m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a68536c6a6e5431e93e22683439cedcb : FAILURE in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d5269798282d4ab681189ab18b8cee15 : SUCCESS in 22m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29e73f454bd14f108e4f8dd25c06e727 : FAILURE in 29m 23s\n\nWarning:\n  Comments left for invalid file swift/common/ring/ring.py\n  Comments left for invalid file test/unit/common/ring/test_builder.py","accounts_in_message":[],"_revision_number":2},{"id":"7068a8038b331108ba47c0395a83e00611f8d167","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 07:06:54.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24fb2ca9cc3b4ba4af7c54c1ae425531 : FAILURE in 47m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/30814fe7fa3749cea348fe1d67dcc8aa : FAILURE in 39m 20s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/7032c3af36f64ea68e700b8954005912 : RETRY_LIMIT in 6m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f41aae48eebb4527be6b01147186b5ac : SUCCESS in 38m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7b98bd87eb54728a1632d211ce5b395 : FAILURE in 49m 49s","accounts_in_message":[],"_revision_number":2},{"id":"b530e5225963c7094b8655f1113aa0824c92ad5a","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-04-28 23:49:27.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0d5d34e255024e7fb0831db77643a87b7dcae10d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-09-28 20:34:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"abfeb04c76aa3cb2eaa5283b400f39eafc46a23e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-10-17 04:06:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e21ef726fb8a85a08ebe9ef451c36336b8f5a2ce","revisions":{"913fb402c5969bcc102004657f663fc31d833d40":{"kind":"REWORK","_number":1,"created":"2022-04-28 01:13:55.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/74/839674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/839674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/839674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/839674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/839674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/839674/1"}}},"commit":{"parents":[{"commit":"46a19145e5c7e72d968e97aa83d9f4a2c481905a","subject":"WIP: Merge ring builder into ring v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46a19145e5c7e72d968e97aa83d9f4a2c481905a"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-28 01:10:50.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-28 01:10:50.000000000","tz":600},"subject":"ringv2: Add _dispersion_graph to v2","message":"ringv2: Add _dispersion_graph to v2\n\nThe dispersion graph can\u0027t be json-ified as it\u0027s keys are tuples. So\na separate serializer was required.\n\nThis serializer json-ifies each tuple key into a string and then the\nwhole graph can be serialised as json. Deserialization just goes back\nthe opposite way.\n\nChange-Id: Ieb52c187ee16bd843c72efa97a6ccda3a166c49f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/913fb402c5969bcc102004657f663fc31d833d40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/913fb402c5969bcc102004657f663fc31d833d40"}]},"branch":"refs/heads/master"},"e21ef726fb8a85a08ebe9ef451c36336b8f5a2ce":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-28 04:20:33.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/74/839674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/839674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/839674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/839674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/839674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/839674/2"}}},"commit":{"parents":[{"commit":"bd27b04c3c9f6c23599d0640e85e53ce84449384","subject":"WIP: Merge ring builder into ring v2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd27b04c3c9f6c23599d0640e85e53ce84449384"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-28 01:10:50.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-04-28 04:20:14.000000000","tz":600},"subject":"ringv2: Add _dispersion_graph to v2","message":"ringv2: Add _dispersion_graph to v2\n\nThe dispersion graph can\u0027t be json-ified as it\u0027s keys are tuples. So\na separate serializer was required.\n\nThis serializer json-ifies each tuple key into a string and then the\nwhole graph can be serialised as json. Deserialization just goes back\nthe opposite way.\n\nChange-Id: Ieb52c187ee16bd843c72efa97a6ccda3a166c49f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e21ef726fb8a85a08ebe9ef451c36336b8f5a2ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e21ef726fb8a85a08ebe9ef451c36336b8f5a2ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
