)]}'
{"id":"openstack%2Fswift~839785","triplet_id":"openstack%2Fswift~master~I31efa3fef673a5541505e4855e52bfd0c9668ffd","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-04-29 15:45:49.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I31efa3fef673a5541505e4855e52bfd0c9668ffd","subject":"Drop arm64 probe test job","status":"MERGED","created":"2022-04-28 16:02:57.000000000","updated":"2022-04-29 19:06:15.000000000","submitted":"2022-04-29 19:02:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839785","meta_rev_id":"23be48d36a10c204a0789fa652855a18544fc728","_number":839785,"virtual_id_number":839785,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 19:02:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-29 00:48:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-29 00:48:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 17:23:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-28 18:06:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6690c680c890f338dc8fe28beac76436ec29c4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-28 16:02:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ca5ec1a9ef7fddf798d474ecfdd4c02da6b2c5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 17:23:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6bf099ba456431bb6948f9a69130af4 : SUCCESS in 55m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f3e3eb9e8b6402288e39179a4ce24a9 : SUCCESS in 51m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/15e692dfe3864b2d8b6c18ace80b0245 : SUCCESS in 1h 14m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f320ae52a071472e9490b8763fb97845 : SUCCESS in 1h 16m 05s","accounts_in_message":[],"_revision_number":1},{"id":"469d429a0228e66ed387fd15546181842042813b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 18:06:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d468133910b472983d041fbb329c6f3 : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/3241d0f16eba4c9bb2dc5577003c7b41 : SUCCESS in 54m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3c70cbb3e24f421c84b38c26fe98b400 : SUCCESS in 44m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e84c8a9e79b047a9b69f119e19637e47 : SUCCESS in 47m 09s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c6371458458f4690986f6dfad79de14f : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e2429fed2a1140dba688491b841f4641 : SUCCESS in 23m 13s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/356c200430024db59a4c760987baf811 : SUCCESS in 23m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f19edf5c468c40e19e2e3ea60db3c102 : SUCCESS in 7m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/75788adb1ded41b8a6421d7408a743aa : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e49533edf29646739aac66763cbec9c9 : SUCCESS in 16m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bf36d40aa6004746b94fa0026fe10d84 : SUCCESS in 16m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cad39e1acbb54deda920d879de0a9891 : SUCCESS in 16m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f5910abd9a2642b5abf455f839c14cf0 : SUCCESS in 28m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/935d293675b046c3869b1182c4a7e0e1 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aa229fbd12a248378e031b6b936b02d0 : SUCCESS in 17m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f8e319a2dd5542d58dc24cacf4eeec27 : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6450615d47d6496497ba8c972b7f6873 : SUCCESS in 16m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5c4db11d1cab4943a983199cc372c3c0 : SUCCESS in 15m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0d095bd05a864d4185d8b0a2d43aaf0d : SUCCESS in 8m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4881eab659b14576925aa15e87a3bb8a : SUCCESS in 26m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/52fe04beb05d46b287a7ea64502fa4cb : SUCCESS in 47m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/68bb2e231fae4453b955df16c0233cfd : TIMED_OUT in 2h 01m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b3c7147f86445929820d016d11c8e8c : SUCCESS in 55m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/653aebb4c2434e44b252f2926ae13429 : SUCCESS in 57m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39db3229fced41e0a31509752df94daa : SUCCESS in 27m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de7b8c3f891449528b1eaf0a419dbff0 : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/80365b10710040759752696013e4f1a5 : SUCCESS in 19m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d6f574272494544964254d9271ed212 : SUCCESS in 56m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f900540c40ffb4af6f6e2c3910591c08ebc40cae","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-29 00:48:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01c2e4b7b302676adcd9ad2d5bfc8384b969a345","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 02:22:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ab5dd5b329346baa81e4033cf17a19b : TIMED_OUT in 1h 31m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2ef2c1593fa4a038aff178f23076c60 : SUCCESS in 35m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c5a2ec1b8134570bc44bc126eaf3681 : SUCCESS in 42m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e06be05a1fed45a4a68faae2310ffc1c : SUCCESS in 35m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c81840d88656dd709e0f5913018b18c3e5d66ace","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 02:54:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4b7718cb21c487da6d10788438689ec : SUCCESS in 7m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/b087cb4c4e59473ca9db59a73cf346a2 : SUCCESS in 41m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/110055ff194e4803b85e5705ef7ca142 : SUCCESS in 34m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/116b05ab9a0b4534826f7c095dd989f1 : SUCCESS in 37m 11s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/dcb936b902ea4ef7a99f6cfdc8dbb4b4 : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1acf4f0c53e147d484e4811d914e01b6 : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e5c0137161594193bd68cccd5391c9a8 : SUCCESS in 21m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8c1f783d32c54455aa4668db5fb8b39c : SUCCESS in 6m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/564395861ad54e108a2bb669be46fc0d : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/16dcad325f944d9d8ce375b2b0007dd3 : SUCCESS in 18m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d540ea017f994e158feb1c39d797af07 : SUCCESS in 15m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cd1b6207d5bd452ca24706425cb467a4 : SUCCESS in 15m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7d7787d8a8534f858cfddcaddb77be78 : SUCCESS in 18m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6c0c7abd8ad344b18d8515ca38d2193a : SUCCESS in 17m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aa36baa01cc348e798243fa4e0fde1d6 : SUCCESS in 17m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9a137ee6d3574e98a9780508f2311443 : SUCCESS in 20m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/15cedd1b5912475ab8e411d7bf745132 : SUCCESS in 15m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8018f738aa0c4fc8aca939a135db768c : SUCCESS in 13m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/58a3952279f54fc3a6affcbb80c620a8 : SUCCESS in 6m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b0a1cbbab5584b04a3e9612ce4111355 : SUCCESS in 16m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e4e9af34806b46e6bf61b566b10dc903 : SUCCESS in 54m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba6f127cb8a84eaeabf5b390e85b385a : TIMED_OUT in 2h 01m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/13fe570bccac4a0588b44e2bc29b3ea8 : SUCCESS in 49m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c126eb0e089c4801add4f118be10cb4c : SUCCESS in 29m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6a1c0d98198486d95773fa283559ae3 : SUCCESS in 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/750ac8d48e7c49cf8febaabebd9182ef : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/41955efba9fd47bfa8f9ec11252516e9 : SUCCESS in 20m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa99c0d748874217bdddc3f490b6d447 : SUCCESS in 37m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d2b5d295ccc01e90fb0486fa50066eef5578b53c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-29 15:45:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87b291d79fe4c32340fcb68404d0a1f0ae7ee758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:51:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/364196d8c3d343cc8222ddb8da54cd72 : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/6471c42dbf004a929bf1dc3dede0bdf2 : SUCCESS in 54m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cf9b36127c9f4400adde26f9ab7f9cd6 : SUCCESS in 52m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/122cd9d1d20c49969e5df00fbe1f3490 : SUCCESS in 46m 02s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8abb4c8f1dc642369b1fdcd74b695546 : SUCCESS in 21m 26s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/28f145b4b4a84e41bb4367858d8a2628 : SUCCESS in 23m 13s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cbe967e6e05b4207b59e5d6adac23b52 : SUCCESS in 22m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5870ee2f2662481ab4403fd6b876b212 : SUCCESS in 8m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d4264479556942a2914f6d1d95e58d68 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c681614d78484a59ab88bd3a14dd16f9 : SUCCESS in 16m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/be7bfe639cb043fbba78611013f2d510 : SUCCESS in 16m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/840575a4c21043cabc6e4907d43607dd : SUCCESS in 13m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fd2aa0afe4d54d69828d62826893bfc8 : SUCCESS in 20m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/79c2bae4819b48d19962aec1c6eda1f5 : SUCCESS in 16m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3428416fce6243eebc189781d4b01e48 : SUCCESS in 41m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/99a40fc2b7f5480cad533693bd761c82 : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/07de3b78e3914907a202f03d64d2dc69 : SUCCESS in 14m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/876298479c0444e98fdc22daf7b18e1a : SUCCESS in 39m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2417f0f7b52b435187bddeee5405090e : SUCCESS in 7m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6433aff90e634a3ca60cc319bbec350f : SUCCESS in 19m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8a96fa201688471188bacc7d28f63c49 : SUCCESS in 52m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d32dad99606949dd97f1b64c04ef92bc : SUCCESS in 55m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/361c138a054c4c1c8bf27c3f28e4dd78 : SUCCESS in 39m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8ecd485d151a4e439c6505c2fd72d3b9 : SUCCESS in 41m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e53964c2d2b4f13ab14bf4120b02540 : SUCCESS in 35m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e76bbb022294c4592ada4d24af48294 : SUCCESS in 6m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/549213a027d7441da0f802cce9e771b7 : SUCCESS in 20m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09a1820bd2984060b4fed6a9a6e87fd5 : SUCCESS in 1h 02m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1dc2cdb90d35cb0f2b9c83eac0a6362cc0001f3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:53:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f12da525e9267ba3797b1c21c94dc9fc763bb4ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:59:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ee24be51423435c95bf95bd2345d86c : SUCCESS in 35m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf5ab1b6678e4d54b682e13e36cdab88 : SUCCESS in 1h 11m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a180142d5a044a79f8b3acb355f1d20 : SUCCESS in 40m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/23d47f5406434cfaa62de4ec387f455f : SUCCESS in 44m 15s","accounts_in_message":[],"_revision_number":1},{"id":"0f688b120adcc1e1f6c93e45ff82fa01874b68d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 19:02:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bd9bb0766244f95a1f5b1e465a0659e : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/03b37fec02904d05a54bb6f8cb781fe1 : SUCCESS in 56m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ecb6175a1664718b6f08c7d29a29362 : SUCCESS in 50m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecffdcc751614e7cbb1cc153c8d35b0a : SUCCESS in 55m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6b17c6c079af48acb8fbebcefda06522 : SUCCESS in 9m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/753722d052b54feca79d240760aa68ff : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3f048f0b0cc84c64ae577933e4d2c3f1 : SUCCESS in 17m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/524c83ef72e04324af5abaa042858739 : SUCCESS in 17m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6e2f8d90806c419ba0e23b79e733af6a : SUCCESS in 15m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/91bbc31e60944d2cae74aca962128341 : SUCCESS in 18m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4166a6ca65374db3bc38501b674495b0 : SUCCESS in 18m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f341a6b01871446ab120bb0605640098 : SUCCESS in 17m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b8f0ed05081a4a8b82d2cd083e9e5196 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2da3b45b0bb44188ad1983102b7f827f : SUCCESS in 15m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eff5bc6405564bc993c7a4b4183bcc01 : SUCCESS in 16m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/89e4a56e95c342f99906613eb8a2e726 : SUCCESS in 8m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4e00243b30eb4845a6f19f12def2cde0 : SUCCESS in 59m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a8b946278f049fb8781d17edfa05a1a : SUCCESS in 52m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8678620e8b0c47d594d3e2f6b73e9499 : SUCCESS in 46m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f525d80874f44dd4bcd1e1ea9f9cc529 : SUCCESS in 41m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d8c86c0bef04163989ebfa01503ef95 : SUCCESS in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ca57d1d53042da94f296526903f2ce : SUCCESS in 6m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f35b95c4a6e482cba91d354aeb4da86 : SUCCESS in 19m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b1d899dd4ae4d1ea55982e20403a351 : SUCCESS in 50m 51s","accounts_in_message":[],"_revision_number":1},{"id":"04cd36b3d0d408c0c0990f8158018f84ed43ee17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 19:02:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"23be48d36a10c204a0789fa652855a18544fc728","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 19:06:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a02587e0ff494cfd8574a5ac583b76db : SUCCESS in 1m 17s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e5590c3bd2984ef8bbaa6929b163baa0 : SUCCESS in 1m 02s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b698ea5de1e24108874c9cfaacb11cd2 : SUCCESS in 1m 15s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0708edecdf6286d4a8295d91e0d2b4cd95e66211","revisions":{"0708edecdf6286d4a8295d91e0d2b4cd95e66211":{"kind":"REWORK","_number":1,"created":"2022-04-28 16:02:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/839785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/839785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/839785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/839785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/839785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/839785/1"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 15:53:06.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 15:53:11.000000000","tz":-420},"subject":"Drop arm64 probe test job","message":"Drop arm64 probe test job\n\nMost of the time, we\u0027re so resource constrained that we trip the\nthree-hour (!!) job timeout, while the x86_64 job can typically complete\nwithin an hour. When the arm64 probe tests *do* fail, they\u0027ve often\nfailed due to eventual-consistency problems; things like object-updaters\npopping 10s timeouts and not getting asyncs off disk.\n\nThe job last passed back in January.\n\nChange-Id: I31efa3fef673a5541505e4855e52bfd0c9668ffd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0708edecdf6286d4a8295d91e0d2b4cd95e66211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0708edecdf6286d4a8295d91e0d2b4cd95e66211"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
