)]}'
{"id":"openstack%2Fswift~839797","triplet_id":"openstack%2Fswift~master~I5e4a270add2a625e6d5cb0ae9468313ddc88a81b","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-04-29 20:19:04.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I5e4a270add2a625e6d5cb0ae9468313ddc88a81b","subject":"Clarify that rsync_io_timeout is also used for contimeout","status":"MERGED","created":"2022-04-28 17:07:56.000000000","updated":"2022-04-29 22:31:41.000000000","submitted":"2022-04-29 22:29:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839797","meta_rev_id":"bfe5b1e9aaba129eb599f37af7ac4cd052a8c991","_number":839797,"virtual_id_number":839797,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 22:29:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2022-04-28 19:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2022-04-28 19:22:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 19:11:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-28 19:22:42.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"c2be815e4d8afac43b4d6249aac25cbbab578855","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-28 17:07:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"930e27b02932043ae69f350892ee097676342d19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 19:11:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0b59bb57edf44cca52a5aef9c7e6df7 : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/1c1084f5d37f4cddab9aa748d178877b : SUCCESS in 1h 04m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/51134466087f4599bd1b5f9bcd50a8c4 : SUCCESS in 54m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a0a72c75d7d4874bded47e973c41cd2 : SUCCESS in 43m 24s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ce9950f6a16c412a91f15d86a977bdbe : SUCCESS in 23m 11s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/95dbe9551bae4df69e90b2d91a883163 : SUCCESS in 22m 38s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a669a6d4997d470690a785e875ce4d48 : SUCCESS in 18m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87e175e84b8a43058dd1254e108db1bd : SUCCESS in 7m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f35e086ae39640d990962b130a1dd1c2 : SUCCESS in 6m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8e3e13a5c92a4650bdb137fc64e15081 : SUCCESS in 18m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/57d91090d87640a88501a7ac4559645f : SUCCESS in 21m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/262effc34e9048aca69bb4566e5fbd15 : SUCCESS in 19m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2740d4e79ceb4593a2bbfac39c4b2488 : SUCCESS in 23m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1295f251b2ef4f9ab941ea66ee9829f3 : SUCCESS in 20m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a9cb90c938a14c75a4846231c8eabdf4 : SUCCESS in 23m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7ee9b0dc098b465ab79a07d11eaf2636 : SUCCESS in 22m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6f1f9aa3790d4ba58f81b3e798df463d : SUCCESS in 16m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3eb5fb832752402cbe4a18648edb228e : SUCCESS in 23m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ae1538dcde0b4af18441d14072d93c21 : SUCCESS in 7m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1296c0b493a646d6a67b9300b5cbd6a2 : SUCCESS in 20m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bfc875cc207d465d8a94cae6353b0e42 : SUCCESS in 49m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3855b614978f409db7f3469b2a8233d6 : TIMED_OUT in 2h 01m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a92a08af803e4032952bb7d2236383cf : SUCCESS in 37m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2fbb24841a8343f19eabe2abd1421bc9 : SUCCESS in 37m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d8cdcebd42e41ac9e5ee68b599f3ff7 : SUCCESS in 22m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ed30a803f1c4d0dabaa6eef095a969a : SUCCESS in 26m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee4c391534ef4bb6b3ab495d7410b7fb : SUCCESS in 56m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0185e8ec5ad47a789300d16ca4bb9cbea1e4f2cd","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-04-28 19:22:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91feed8f260241fc642b9c29d93280cecb01cd84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 20:20:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff7033f6230746f19b1cbd595587b847 : TIMED_OUT in 1h 31m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd463bab976b4aa98affd939a6d2b2e4 : SUCCESS in 43m 45s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/c44f0763521148dd8a743e6851dda260 : TIMED_OUT in 3h 01m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/40a895fa364a429ea5bc2d5b75099b06 : FAILURE in 56m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c05ad1e2355420ab385a95aeb37830c : SUCCESS in 47m 22s","accounts_in_message":[],"_revision_number":1},{"id":"63b3787eb6d9839ce538ac29a14f630727780689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 21:32:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18dd6f62237243598015d448dca7d6f7 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/7265be5d36a347cab4ed6b2997f53df9 : SUCCESS in 45m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2111c3bc3988409eb108f7aa0ac83f68 : SUCCESS in 43m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23c05303a63c4ba49586c0deb4c80822 : SUCCESS in 49m 46s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2a2757558efe4672aae499c8eda4da14 : SUCCESS in 31m 38s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9f96d3264fe940b7b9d7ad58c7d0a4c2 : SUCCESS in 31m 32s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/aa77a44997d24ecdab73909bc6e886ea : SUCCESS in 35m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/871c18d2373d4d929c80891155783fe6 : SUCCESS in 7m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2ba5d3569228476dad5af6de3694de1d : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cd9f13e7b58d4aa795441b83f9419b4f : SUCCESS in 28m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/278e37f7f28047bca56e3374fe5b94f1 : SUCCESS in 12m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7f682488bd994a22b0d5538cb8a073b0 : SUCCESS in 27m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b867d7762bd840eda6c6fc9db6b8d052 : SUCCESS in 29m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d4ab222c816b45f2ac7936e1cde862df : SUCCESS in 28m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5829fadcc65645b481726c09d3f2ce65 : SUCCESS in 30m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8ae7e145104a499a86caccd7395dbff1 : SUCCESS in 28m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0ae7877246814b26b63250f7b20781c7 : SUCCESS in 30m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/78dec099b4c7493c8d3ba2508f538c67 : SUCCESS in 29m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d21705bf505b48f4a619034404cf59a5 : SUCCESS in 7m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aaa4a0647c394ceea510472b02b923df : SUCCESS in 28m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a66e4e3f5b7f44999ba137118c8e25d2 : SUCCESS in 1h 21m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f25e9bef2294b21a9787e52715990d0 : TIMED_OUT in 2h 01m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/488c5e4b7aa24a86a271bc5fbeaaa1e0 : SUCCESS in 56m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8458f1b4d6624ccba71526addbca2a58 : SUCCESS in 32m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b969aa831a94fed895d4a689f30f27e : SUCCESS in 13m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/891b02658e4146a297e174c75072d7a1 : SUCCESS in 18m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb74dfcd7cd94fce8d29652ffe3c36f9 : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e0221714e35f0226ac49c4b2d13448f47eac1ae4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-29 15:47:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ff374028a2a3a039009d6fbc9cf8a4b4c8c9677","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:57:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0e8b04235ca4eddaa37797c50c26f2c : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/14f160f798d74225b5d288d4a800581e : SUCCESS in 1h 04m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/973af891b55043728a1c02bf4c6dc731 : SUCCESS in 55m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8c19863d18f41b4a9035be073cea15a : SUCCESS in 1h 00m 17s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/31cbbbc264504e47968f5700f46ba873 : SUCCESS in 21m 38s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/70e7272dae62466eb42d99c157f048f8 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/36a85e430071435497f1d02806a676bf : SUCCESS in 25m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/827abd88cc8341469e2f244a6189316a : SUCCESS in 7m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ec95d72c1a6042edadc0332dc8143fce : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aeb3a951bcda4b79993c7a11fdba2b99 : SUCCESS in 14m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1cc492690d5a4e1996079224bdcb65dc : SUCCESS in 15m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/764689a439874d32815a4fc1eb984397 : SUCCESS in 19m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/570b207de87f407a91651676da9f6b16 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fe08c246bd1e4c0894d157b651e9310f : SUCCESS in 16m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/169eb5eea41f43c8ba066c76a159aba0 : SUCCESS in 18m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/35360ed058464e3bb68b06ab4fa3a682 : SUCCESS in 13m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3d7d281017814bc5b3d51c67042f7f26 : SUCCESS in 19m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a73b18eb92df46c198c3fe190ed327b0 : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cfa52fb2df5f45cb8b5a3da1d3d4c808 : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e5f2209a9a246a5a8a76a44a033583a : FAILURE in 14m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2fc7d2b2739e44a8b5426f93e4f7fddf : SUCCESS in 56m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b2295b88a79453d80863037f09f92c7 : SUCCESS in 55m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f306587e1655431bb0d5819d7717584b : SUCCESS in 31m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ef09e0045a0f4012800eb32902c22b97 : SUCCESS in 1h 02m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/545dca800ede457ab6f6f00bf30b2643 : SUCCESS in 16m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/09045f2222a14ae7a734ace9129d524e : SUCCESS in 23m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5b88213e8bc4e29a1aff72485c997b0 : SUCCESS in 42m 05s","accounts_in_message":[],"_revision_number":1},{"id":"09a5e718b5679d374c9b4cbb9ffe50e7d264c7ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 16:58:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b2bb5f7e5fbb2a734525425c432e15fc467de30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 18:57:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a2c760e48b44a76b98d0af86a896911 : SUCCESS in 37m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c001c16cc9364f488be3fcde2eb4b790 : SUCCESS in 39m 51s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/6ff775a0cad04affb230f50913197e0c : TIMED_OUT in 3h 01m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/818170a217f343b198dc2bbe1dd92011 : SUCCESS in 53m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3ae60b884a7471cae674b378e6e74a5 : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5d08f1b2cff5d194412af02207117c070b1f4d81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 19:02:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68305ee5992b4ecfae4f6fc4323296bd : SUCCESS in 10m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/1f89058ae8de4db9ad6b8ddee0f9fe55 : SUCCESS in 1h 04m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b0566b8fac044774bec3be854b481290 : SUCCESS in 57m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85833ad013b549bb9ec81deb3a33809a : SUCCESS in 1h 00m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/aa14e573aa094c0c9a94f14016bdacdd : SUCCESS in 8m 35s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/64ea8d512e0e48efb82221209564894b : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e1287b3eb04446bb9626417daf1bdeba : SUCCESS in 16m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/11e6610ac55c46f4b11171958e866b53 : SUCCESS in 16m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e38608be0c764ffbb65718d03054a1ef : SUCCESS in 15m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f97cf2f556df43bdb92000f070f108c7 : SUCCESS in 18m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b2e10824a7cf442fa2687125a76ac9a6 : SUCCESS in 16m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/358538ab48824325a1799524292805be : SUCCESS in 17m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c5c57cae21084791b0eac249a67d5b6c : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c7ec4c0a97724d53834f7ca192bc0e2c : SUCCESS in 15m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/16ba5bc281b44c1a90f988611ecd0b91 : SUCCESS in 17m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/28345242842a4929acd6eccf9e644428 : SUCCESS in 7m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c67aa3d2cbc246d0a4425cf47c26211c : FAILURE in 45m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5f5cc541b4c4248b22172a6f329f242 : SUCCESS in 1h 01m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a95695e58674342a67d28edaf5ae29c : SUCCESS in 33m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/752b570ff2144d8e88c971293f9283f6 : SUCCESS in 40m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a620df595f04f9ba630d7e12928c1cc : SUCCESS in 16m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/52572c5f39cd4ba297b30285e56f612c : SUCCESS in 20m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/138bea9c8eee40f089dcc6cafcb5c889 : FAILURE in 40m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f5f21b29e6d3a42580b5984527784c9eb5409aee","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-29 20:19:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3b871345fcc05bc07dcd9ca89a58b61b8615602b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 21:14:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3451039d2bab4b06b504a8dc788c5f5b : SUCCESS in 53m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d402e9beb15c4626b6ca288a8ad84081 : SUCCESS in 40m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7fefa9aa5db740dc8c577b737007a0e0 : SUCCESS in 42m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/008f0a5e5d4a4aabad15d451152b060a : SUCCESS in 41m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4380e0137ccdbd6e3b3830f1b5ca233d34056fb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 21:26:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/451924670d6442a7a23b2128c46423bf : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/d7e3839bf1364ef0b3cc2013608787b5 : SUCCESS in 37m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/74f339e3be6b4eb49c48f605ffc837ab : SUCCESS in 52m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5db22d88e080451a9eb57f0d7e8f42df : SUCCESS in 46m 56s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/be71860c1fc0441490197bb05f95f216 : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7c1afc31bbdf404eaa42a40dcca26f29 : SUCCESS in 22m 10s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/62def8f35f194b2bad3d24395733ed72 : SUCCESS in 21m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5176c8f0b167490d93938a40eafdf1e9 : SUCCESS in 8m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/02d341f905e446418767733e47e2faa2 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a0e29462322f4c3685d8c54273da4b6c : SUCCESS in 17m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5d6aefdd90e54944aaad4acb0de6447c : SUCCESS in 16m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/886919bdde514e1eb5c4a989b242021d : SUCCESS in 16m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/df01ad2eaf9f46eba7ca854c4393db9a : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0a4d9bbadd40411f8a0902c76dce0fd9 : SUCCESS in 18m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b181f28e39041b6bc2ffbf0f9005e4b : SUCCESS in 19m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5eaf32f7f8584fdeb73929db3f869498 : SUCCESS in 13m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5853bdbbe2454b4686d6c22c83a45e91 : SUCCESS in 16m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5735bd2e5f5c4cf89ef80081103b878d : SUCCESS in 19m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ef0cf9ac845345ccab34128f7cf705e7 : SUCCESS in 7m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0b9b0392468746a9a7749742c81cf529 : SUCCESS in 14m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9ea5174e9ed4447db54a9b8e6e67899f : SUCCESS in 52m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c2f4440df124e0ead28277cf0051b94 : SUCCESS in 1h 00m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/91129f53f68f4a22af1074c7ed423b52 : SUCCESS in 46m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/82ed8857dee24bc98d643c6859ae44e5 : SUCCESS in 41m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0c016e8cb504cb2b4dc13b40d0c2008 : SUCCESS in 21m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a8fd0faeb4a347cca2f7ee1ee1e0c48e : SUCCESS in 20m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1b9451f8bcd469e8ffea22dd4890f6d : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5f2b146d92c4661745bf4883676d12f51aa81db7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 21:27:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1883d4aa395ceb2be0f4c5116c82f54d98ec119a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 22:29:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1b256eba5f41aeaa1b7cdf3d26c153 : SUCCESS in 11m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/a5a2d473969743acb18f0a08b9b6db07 : SUCCESS in 1h 00m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a25de4b7b46d4d718e08bf74fd1c6c22 : SUCCESS in 48m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ccd67ca666d4afba3be10a65f7a8a43 : SUCCESS in 55m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e1cefc180f1048eea2249a58b44fe6dc : SUCCESS in 10m 19s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/6c7e4147ca46440d9f6029569f3905be : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2fb1a395d0ef4300885cdf2da6d5e28d : SUCCESS in 14m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1c1edded6bbb4bad917edd273b013017 : SUCCESS in 13m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c69de5e567b48bdbed72fcd9f4da570 : SUCCESS in 13m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6e4f366e41864463b333fd6fb6d3b400 : SUCCESS in 28m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef940ebdef244748976da35f732a48be : SUCCESS in 15m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4ad3f0e5c20141a1b3e3f9eac132edeb : SUCCESS in 15m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d5f9a137ff73491aa35c49b39d3bfc69 : SUCCESS in 13m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/73885161bcc248e19d9c913de71f1138 : SUCCESS in 18m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f39cbdfdfc1b45af8569ff348ba4436e : SUCCESS in 17m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ebd958cd88024150b9728c5c146847aa : SUCCESS in 7m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/44654dcdb6f548799d5430e09f02e3e2 : SUCCESS in 59m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fbee71e4f2549d59461397d46fc9441 : SUCCESS in 56m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/49b8b05b3dbe4e32923a20dbe510cc4a : SUCCESS in 43m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/55308de21a0e4e35a55a7f42c8823c83 : SUCCESS in 40m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b64825fecfc34314971fa2ff5c0786c5 : SUCCESS in 21m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/580900fa59c249af9047a5af8b99748d : SUCCESS in 26m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3639f2d7ca64c1083988411d3624f63 : SUCCESS in 47m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c792e6b6cec33af1af746cbe695eab6f4a2859f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 22:29:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bfe5b1e9aaba129eb599f37af7ac4cd052a8c991","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 22:31:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2122a3d8d7a14bb68f2b68123205c103 : SUCCESS in 1m 27s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b940e6cb76774efaad000f1fd78f5b8c : SUCCESS in 1m 01s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/7a315479f3f643b18c7257a597a2d018 : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"043e0163edea35f75ddc2263feb59c32a8e8eaa6","revisions":{"043e0163edea35f75ddc2263feb59c32a8e8eaa6":{"kind":"REWORK","_number":1,"created":"2022-04-28 17:07:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/97/839797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/839797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/839797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/839797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/839797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/839797/1"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 17:07:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 17:07:50.000000000","tz":-420},"subject":"Clarify that rsync_io_timeout is also used for contimeout","message":"Clarify that rsync_io_timeout is also used for contimeout\n\nChange-Id: I5e4a270add2a625e6d5cb0ae9468313ddc88a81b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/043e0163edea35f75ddc2263feb59c32a8e8eaa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/043e0163edea35f75ddc2263feb59c32a8e8eaa6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
