)]}'
{"id":"openstack%2Fswift~839822","triplet_id":"openstack%2Fswift~master~Ibe01f54d157d6ea540c36ba4e77cc816e864a81a","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-07-10 21:34:31.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ibe01f54d157d6ea540c36ba4e77cc816e864a81a","subject":"memcached: Log command name when logging exceptions","status":"ABANDONED","created":"2022-04-28 20:59:41.000000000","updated":"2024-07-10 21:34:31.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9a55422ec69c532c3ce794e677b959da64210a48","_number":839822,"virtual_id_number":839822,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-28 23:02:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 23:02:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-04 19:51:59.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"26ce49d5e75cb5ca5bb29a4a660004bb24d15639","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-28 20:59:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b7d3e91617c2c677f4e5e253594a095423fed1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 23:02:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a0964de8cd54508b9823363a852ab5b : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/5d6d5e05e1f549bda5a8364948869d18 : SUCCESS in 52m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/179966cfa050424f80736bd49205fb8e : SUCCESS in 44m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5c04dc26f1d402dad610f7d2de35130 : SUCCESS in 47m 35s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/23f79ef74c064d4e9868b0e426540841 : SUCCESS in 24m 06s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/02421327963b43268021d793e7bea166 : SUCCESS in 24m 30s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/486f7bbf478c4538890edcac83917b68 : SUCCESS in 25m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7c17e154a0a14e99a931ea30f6a68b88 : SUCCESS in 7m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/39546ae080824594bbf1dd86bf405946 : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7f85d6a877bf48dba469bdc149f8be6b : SUCCESS in 22m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e2cd9238f4324d838ef046102c44ef8c : SUCCESS in 21m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/38db2f5754d848c69822d74a17622ade : SUCCESS in 20m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e6a04607c7f642c98fa6cd0bc48893c2 : SUCCESS in 23m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/caa85f67ccb34ae99d2d67261dd5b47b : SUCCESS in 23m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4bdfb026e2c046d9b91ef68e1bab1b34 : SUCCESS in 19m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a286a1929fe740a18355225221e65206 : SUCCESS in 21m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f1e8c2ce7b03427e93e268b810912521 : SUCCESS in 21m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/493000da45cd4816a34a1343c2cb054a : SUCCESS in 23m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2873c086ca104bf0b73d425b0b7aef40 : SUCCESS in 9m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e260c402369d4854860fb340fff84b7f : SUCCESS in 21m 04s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/73de2c45eb3149c69e4b5fcc0ae0d009 : SUCCESS in 51m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c207880505242dfaa804ab675b9f074 : TIMED_OUT in 2h 01m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2106a77a24404fc6b9e771a932010e97 : SUCCESS in 40m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/827242c6183244d597e68571c054e5cc : SUCCESS in 44m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e89104c9e8b64cb29400c3147997be44 : SUCCESS in 17m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ff0aaa0b30c4a75b311e6e0b022d8df : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ee6507bda99742a7b1deb746878c818e : SUCCESS in 27m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23b6927a39ba490fa9cc9a93a2f2b0c3 : SUCCESS in 44m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3203220ab7b3cbf4a3b7027612fc9b23ed1c333b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 00:03:32.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e23fd7ec7f9e44e3aaf4ad0713884f09 : TIMED_OUT in 1h 31m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/47abe78bd0354f999552e231ada4e9f6 : SUCCESS in 48m 01s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/168fbab5df2a4314845c283be2e8d2e7 : TIMED_OUT in 3h 01m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f85cb517e7914702bba94833e36de2d4 : SUCCESS in 45m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f1f7c05d8404fc6a10daa4642b519dd : SUCCESS in 51m 51s","accounts_in_message":[],"_revision_number":1},{"id":"fcb3dabf5c23238058165a86c9768bf2592106ca","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-04 19:51:59.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9a55422ec69c532c3ce794e677b959da64210a48","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-10 21:34:31.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c9b9fa105f5003c2e8c040d429b9bd2220da730","revisions":{"0c9b9fa105f5003c2e8c040d429b9bd2220da730":{"kind":"REWORK","_number":1,"created":"2022-04-28 20:59:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/22/839822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/22/839822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/839822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/839822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/22/839822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/22/839822/1"}}},"commit":{"parents":[{"commit":"9bed525bfbe3ec05cb45c9835333fe45f8e04b40","subject":"memcached: Give callers the option to accept errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bed525bfbe3ec05cb45c9835333fe45f8e04b40"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 20:49:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 20:49:14.000000000","tz":-420},"subject":"memcached: Log command name when logging exceptions","message":"memcached: Log command name when logging exceptions\n\nChange-Id: Ibe01f54d157d6ea540c36ba4e77cc816e864a81a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c9b9fa105f5003c2e8c040d429b9bd2220da730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c9b9fa105f5003c2e8c040d429b9bd2220da730"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
