)]}'
{"id":"openstack%2Fswift~839823","triplet_id":"openstack%2Fswift~master~Ie6bae96eee8e32d5e336ef42ad76d7fae9e42a31","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-04-29 15:49:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie6bae96eee8e32d5e336ef42ad76d7fae9e42a31","subject":"daemon: Log pid when logging about SIGTERM","status":"ABANDONED","created":"2022-04-28 21:03:45.000000000","updated":"2022-04-29 15:49:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ee42908a6697b5d7508758c5d4a008472deffaaa","_number":839823,"virtual_id_number":839823,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-28 23:09:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 23:09:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"525f785eb9223364ce0c0cff4f63a1497d00f7e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-28 21:03:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f5ee0dddbe3f4618e2cb4fb70b669302cfa4b28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 23:09:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c13fd9ae13664fcb9194e91b2de40765 : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/ce4898525bfe4701b4c52283b471484f : SUCCESS in 57m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3841e52559ed48f59b8a406fde0f3078 : SUCCESS in 53m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f93b8a73c9945c98bbd672330d82cba : SUCCESS in 51m 00s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e3ad4e3a542844ae8b3ee7faa37f38f7 : SUCCESS in 18m 05s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/daf41f9bd5e6474ab193203fb532f883 : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/25fbf43d1c74471b85fdcfafc33a53e8 : SUCCESS in 19m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e6abc83e2f994e53891a8128c0c9c352 : SUCCESS in 8m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f0e74dc3056341c98adf4042c05d1684 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/526eaa47f9624124a5a71ad4ccbdaf4a : FAILURE in 17m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/de271b6c76f9455c89401424ce1ee088 : FAILURE in 16m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/67ea5008cc7643b9921e295f6650b6c2 : FAILURE in 16m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b5990cdc87004fb994ce509c15f4e0b1 : SUCCESS in 18m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9f7f825d471d45bb9dc0c5ebbf85d640 : SUCCESS in 18m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a87fc82fdb89435b89825206158e319d : SUCCESS in 23m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c6bd314b3acd4a95bc3e0900a387fb8e : SUCCESS in 15m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7949f8a9091a447bb151641616416268 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f2b513c768c148f4b70d632c7d6eb338 : SUCCESS in 17m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e03cc52d9de14d119d484f1b17c45d48 : SUCCESS in 7m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0d52e16492f74d65965b328b44f59209 : SUCCESS in 24m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e837318f2b0846b6b09885dc2aa49334 : SUCCESS in 52m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fa25e2b215549689a3acdd2a85d6560 : TIMED_OUT in 2h 01m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f2db215f45d54db2b2fad6cf02eb68ab : SUCCESS in 44m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e528c3d43d8b493bbfeb402cfd62c59b : SUCCESS in 32m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/087bce2766334586afd560a3808a4cad : FAILURE in 15m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/944bbbd926424b449be7eeb5d4a4d4e1 : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d14cbd99949546af852ef88a0e1cfb63 : SUCCESS in 27m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af772e5cf79d45f9ade31404d4b1df9b : SUCCESS in 45m 00s","accounts_in_message":[],"_revision_number":1},{"id":"01233f2a1fcaccd0b317e45aaf0781ff94d1e2a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 00:23:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81559e403bca446e8a02ab74a533885a : FAILURE in 44m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a69a1e102384afeaf77efb12cbd2465 : FAILURE in 33m 53s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/67567086521547078771db703a3de623 : TIMED_OUT in 3h 02m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/59fd85283de441a29d2e356e427cadf1 : SUCCESS in 37m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7417b043f17a4c57ace8b222eb74947c : SUCCESS in 42m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ee42908a6697b5d7508758c5d4a008472deffaaa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-29 15:49:39.000000000","message":"Abandoned\n\nSquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c18cd6ebb27811d36f42249405967777afa81f9","revisions":{"6c18cd6ebb27811d36f42249405967777afa81f9":{"kind":"REWORK","_number":1,"created":"2022-04-28 21:03:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/23/839823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/23/839823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/839823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/839823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/23/839823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/23/839823/1"}}},"commit":{"parents":[{"commit":"73e1e8c8761b6761aa01d48484a368f37598084e","subject":"Log signal handling at notice","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73e1e8c8761b6761aa01d48484a368f37598084e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 21:03:29.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 21:03:29.000000000","tz":-420},"subject":"daemon: Log pid when logging about SIGTERM","message":"daemon: Log pid when logging about SIGTERM\n\nChange-Id: Ie6bae96eee8e32d5e336ef42ad76d7fae9e42a31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c18cd6ebb27811d36f42249405967777afa81f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c18cd6ebb27811d36f42249405967777afa81f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
