)]}'
{"id":"openstack%2Fswift~839828","triplet_id":"openstack%2Fswift~master~If9403c2176f868158370b3f9e1195d2f7304583b","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"If9403c2176f868158370b3f9e1195d2f7304583b","subject":"Ignore py36 deprecation warnings","status":"MERGED","created":"2022-04-28 22:48:37.000000000","updated":"2022-04-29 04:01:46.000000000","submitted":"2022-04-29 04:00:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839828","meta_rev_id":"a90bd019ded3a95697968717b805ab5f8744a9d6","_number":839828,"virtual_id_number":839828,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 04:00:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-29 02:52:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-29 02:52:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 23:57:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca5fd21637f5b87ce13053346181c884dabf9dc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-28 22:48:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"989bc88f3416a22128081d7bb368dd7bdcbd4dd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 23:57:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff289be2ccd42bcb29becd2639092d7 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/569a8c62f05643909a6a0a2b9a7f3fa0 : SUCCESS in 54m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/43e707e73b2d4cba9de40bcdc8fe72b7 : SUCCESS in 42m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46f0f122389f4ed9b180ab2bdd92f8df : SUCCESS in 49m 30s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2c4f363855eb42fdb9dc53930c996160 : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f260258412dc408c96b6a52a2cb6f406 : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6b61a76d13734a5c8b3b511aa4c91e07 : SUCCESS in 23m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8c82db709d804802bdc4f37f3f5dd818 : SUCCESS in 8m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/139129fbd2684e96bd78902a24580d71 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c63598f876a347a9933847ab86cde3c6 : SUCCESS in 16m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/788205390eda46aa9d63560042869eb1 : SUCCESS in 16m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/189ce5f8fa13402fa74631bd86d014cd : SUCCESS in 20m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ded004438fca4942a245adfe6762e3d4 : SUCCESS in 31m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/87d2bb98530e42b694ccb6000b38e47c : SUCCESS in 22m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1fb90d3be2ce43268548e952d977ff11 : SUCCESS in 18m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e450dcdd722f437d99802331a8d2d82b : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ca95e46c5f3540f4b152ba499802df1e : SUCCESS in 17m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e72f703b1cb44758aaa52de1317f482f : SUCCESS in 18m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a06f162b67104d90a94317af548694a4 : SUCCESS in 7m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1e81e7d5974f4b3baae8eac47e8804be : SUCCESS in 18m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2c7dbf77d13b4e30ac23a959638a725c : SUCCESS in 55m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fcd471bde6d14010a7db384426a3b763 : SUCCESS in 1h 06m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/05f687a9495f4024b4d04fc49046a796 : SUCCESS in 47m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a0a3f37e76e24855acfd4c1cb71ce372 : SUCCESS in 41m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5467b482b6204c13ac1c02c891eee8d9 : SUCCESS in 18m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6900e84f2741fca3e5e885a83cb510 : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b95fcba5542d4e4fb586ea9b6553ffc0 : SUCCESS in 21m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1ea49752d3d4151b20176133a5b3467 : SUCCESS in 45m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ac60f43329811f6eabdaa032701ce5e522a675ad","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-29 01:04:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dbfcf37d0019f90bdd62a71511c9c49d9aa2d243","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 01:41:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/75f89cff9ea14f1783fe54410c88dead : SUCCESS in 35m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/479106bcd0ee4836902fa50ec9b79d6b : SUCCESS in 36m 39s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/9a142a4e9698484aba4537f4096bb598 : FAILURE in 2h 50m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6858dc6c0be4debaed819c5560772f8 : SUCCESS in 39m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3b614045bc74430b243f6d32db87d98 : SUCCESS in 47m 56s","accounts_in_message":[],"_revision_number":1},{"id":"95f7544a9f5e35f2d794b2890bdb115ca5412ae5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 02:14:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf7ee73323945658b0e7aab5f44b5bd : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/a6c6eb3a7ece48219428866e635c2d6d : SUCCESS in 1h 03m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c0bd8caa10c3464a9194c73161d81aa2 : SUCCESS in 56m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10215b77ee994e619eabfb962a4fe4b4 : SUCCESS in 48m 22s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d353988b3bad42cfa76e177b76b8b08a : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fa972af037c240f8b7e6642d194ceed3 : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5f9168d419284c7aa6d6c5948046f62b : SUCCESS in 20m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/29bc880aecab4ce3b1646377f3049d9c : SUCCESS in 7m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e61670b7dd704feeae0167ae18e01efe : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b682dd8b5334177945b6dab6823e9e1 : SUCCESS in 17m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/de47348c3c1546cb8f16d7192bba3d41 : SUCCESS in 14m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e16b2bfcf35743d38cbe06e38bc8e8cd : SUCCESS in 14m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/646a8ae62d5b4209879afd397aa643b0 : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fcd1dfcd9f214ae0b4c611eb3d064535 : SUCCESS in 17m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/542364cce2ec483488fbe27e43a12282 : SUCCESS in 18m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cc8cbf5203144eeca368722c527f25e8 : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4695560a351f417388e97a1757f08376 : SUCCESS in 20m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c4617c26da56472f945ea3ebe68bec76 : SUCCESS in 16m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/71948f167fa54d6ca0081ac82a718646 : SUCCESS in 6m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b0f3465d86c45b3be3833d89fef98d0 : SUCCESS in 16m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2f25ee71d0184abc9e86ddb18ab43f2d : SUCCESS in 49m 54s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a0cbfefaff448f8af7104fa3fe86baa : SUCCESS in 54m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2a960fc7ad644807889aaa09a7382788 : SUCCESS in 40m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/549eee42bde64a8c9391be06d2249ce5 : SUCCESS in 52m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1796e3f98e945718175f6c84d59530d : SUCCESS in 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eedc9c91a87148d599805fe7bdc32e3a : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7f2e1c3ba29d49208bbbc98fde0764cc : SUCCESS in 20m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17a16edb652244f19a1a9d415b11786d : SUCCESS in 55m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7818d599d0ebcc6df403852ae0f273d39aebbcf9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-29 02:52:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81f62c4448ad13443122ac6969b3d713c28c2751","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 02:53:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd49e77c9069169f05a726902e58e770dc62b5ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 04:00:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/858e804973394d0e86f208246e1b6ec4 : SUCCESS in 11m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/ce08569ba82e49f7be411937c3e71182 : SUCCESS in 1h 01m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/72d1be1a5dba409d9183fee6bcea2d4c : SUCCESS in 39m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/689101fc98994dceb8e794ba2c87190d : SUCCESS in 45m 09s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bb70060272234e8595445feeb516a7d8 : SUCCESS in 9m 21s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e46538fa773248dd8cbe78f2447acfb5 : SUCCESS in 7m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b04322aae8724117a72da1102b2ed1fd : SUCCESS in 36m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/737c69914a8c4d78b29ad32d907d0d36 : SUCCESS in 16m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/214cf5ee4a6341ab8cc9b9c6e54e5e18 : SUCCESS in 17m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/18dd36108f144dee9f747807430ae740 : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8c4414e8b4fc48a8a283a6d380c091f4 : SUCCESS in 20m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5b221ade66bc42c88247f4f4b4c91caa : SUCCESS in 21m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9aec9bc247a64680a8a57c7a5cca52d1 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/14fc363d001847789d6005259a11bc18 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/96364a7f83304d2987834880660b1ccd : SUCCESS in 17m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d94bb2e7a9ce4669b29b48eb646768a5 : SUCCESS in 7m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/93791b6fc11349119f3076c0ee36292d : SUCCESS in 1h 04m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0b4b14bce654f2ca75b912021593c57 : SUCCESS in 55m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/45dab3e970b84f508447c55143d08fbd : SUCCESS in 43m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a3b6215a14ce42f78b35368862b1b2ac : SUCCESS in 44m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/026987bf23b94d248473b0e944e046fa : SUCCESS in 18m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/760f4fdcd9814142a1b178b019e50ba6 : SUCCESS in 7m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca36080b038342adacde8faf2656d14c : SUCCESS in 21m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4195df650709484f9165666690c17fc6 : SUCCESS in 58m 34s","accounts_in_message":[],"_revision_number":1},{"id":"4b462ec8c895a434233eae628fdbdcadb0025741","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 04:00:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a90bd019ded3a95697968717b805ab5f8744a9d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 04:01:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fa09d4b53b64672814dbb908301ba04 : SUCCESS in 1m 04s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b40d6d961c7e48a5bc814524710359d3 : SUCCESS in 58s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/3a00b9d180c44ef3b89ffbb0450d6cbe : SUCCESS in 55s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7298038ed90fe5824c250975045a6f0a3eb39d04","revisions":{"7298038ed90fe5824c250975045a6f0a3eb39d04":{"kind":"REWORK","_number":1,"created":"2022-04-28 22:48:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/28/839828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/839828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/839828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/839828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/839828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/839828/1"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 22:48:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 22:48:13.000000000","tz":-420},"subject":"Ignore py36 deprecation warnings","message":"Ignore py36 deprecation warnings\n\nChange-Id: If9403c2176f868158370b3f9e1195d2f7304583b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7298038ed90fe5824c250975045a6f0a3eb39d04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7298038ed90fe5824c250975045a6f0a3eb39d04"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
