)]}'
{"id":"openstack%2Fswift~839831","triplet_id":"openstack%2Fswift~master~I34f98fe897ee2d15fc360216c88c1f67aecd8081","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I34f98fe897ee2d15fc360216c88c1f67aecd8081","subject":"probe: Increase job timeout","status":"ABANDONED","created":"2022-04-28 23:19:28.000000000","updated":"2022-04-29 20:41:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9aba725fd5594bf1ff10a27e2d61dc5102b88958","_number":839831,"virtual_id_number":839831,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-29 02:06:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 02:06:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"778bcc2b23e323de7d4edfb4a38ee1331eb197d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-28 23:19:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d09c16033901865787ba3f07751821241fe3b9b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 02:06:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e1b9e9e11a4f508eec19f691424abc : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/35d4f9b6e29946fca67f1f895ea1d0ce : SUCCESS in 59m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a376ebdb80c543ee8cd9dd6b184b9e76 : SUCCESS in 52m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3479848f6c1744c1a4e7cfb3f887065d : SUCCESS in 41m 04s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/95fd5994ec064678ae73391a93ac838e : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/01e405706faf42aebecd344757444918 : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9fe2a494fba64f20a8c0364ecd5a971a : SUCCESS in 16m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ccd88dd2c03147fa94b239197a1f9ecb : SUCCESS in 7m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0866254229e049208e49f830a1c1d012 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5d9f50c829a14bec9fa0d599b6631d09 : SUCCESS in 17m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fedb22554bf140d9afb3dd070caa123c : SUCCESS in 11m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5b07cfdf91e449bfbae1fd1eb96c564a : SUCCESS in 13m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a5fdc1050dd247c989a5a947e04e86f9 : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/558d0d4620e64424ae0b4b2df5dfcec8 : SUCCESS in 17m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ba5054207c564ddeb016385b70c00445 : SUCCESS in 18m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/00fafbd7d90a4f078aff94ccfd3ac300 : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3e6ad20b77b343a1aa4414499c94fe93 : SUCCESS in 14m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b69de056707e4370a9e85dcc8a088e57 : SUCCESS in 13m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f32cfa36331046e39b448f7f62d3cfb1 : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/de9aae923150419ba94446b9bb11c392 : SUCCESS in 13m 50s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3b3d821dcd054f13b0f7aae591496a01 : SUCCESS in 59m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7caf87b0a0f64f2caa45493a07d15591 : SUCCESS in 2h 39m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cc08b8a7f7a24742afbf06cb5814c6ea : SUCCESS in 41m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a4f18b2e7333449ea6591eb020c733de : SUCCESS in 26m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c01c28965cf45e9b2f7b8b30c502311 : SUCCESS in 16m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10040268ade34c9a896ad6835f880d52 : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dc0302425d61498ba54d966a00da9696 : SUCCESS in 22m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78096edb6c84480986d6d29589b415a1 : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":1},{"id":"450a875ada18fd7dab46d464ecbc5c3f657b8e53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 02:30:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76b86cb6fd3049a4a97b7d6843f8e369 : SUCCESS in 45m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed0e89dd2a41498b8525d259e723dfb5 : SUCCESS in 45m 22s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/5f6b9ed1e1b147b2b7dd03a3bcf8ddbf : TIMED_OUT in 3h 01m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c655f494dc59414bad24fde9e1e199cb : SUCCESS in 51m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/eea2d67f85424366bfabf4f8c662a9c8 : SUCCESS in 49m 21s","accounts_in_message":[],"_revision_number":1},{"id":"9aba725fd5594bf1ff10a27e2d61dc5102b88958","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-29 20:41:29.000000000","message":"Abandoned\n\nLooks like https://review.opendev.org/c/openstack/swift/+/839828 addressed the job slowness.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee3370ef05e886d01893c58c4b387f7db09d9af7","revisions":{"ee3370ef05e886d01893c58c4b387f7db09d9af7":{"kind":"REWORK","_number":1,"created":"2022-04-28 23:19:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/839831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/839831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/839831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/839831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/839831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/839831/1"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 23:18:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 23:19:22.000000000","tz":-420},"subject":"probe: Increase job timeout","message":"probe: Increase job timeout\n\nThe last couple days, we\u0027ve seen a very large increase in probe test\ntimeouts. Bump it up, see if they\u0027ll eventually pass.\n\nChange-Id: I34f98fe897ee2d15fc360216c88c1f67aecd8081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee3370ef05e886d01893c58c4b387f7db09d9af7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee3370ef05e886d01893c58c4b387f7db09d9af7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
