)]}'
{"id":"openstack%2Fswift~840135","triplet_id":"openstack%2Fswift~master~I25ce3bbaa107ddf51dfd2b69b3ab5423b3d24377","project":"openstack/swift","branch":"master","topic":"drop-l-c","hashtags":[],"change_id":"I25ce3bbaa107ddf51dfd2b69b3ab5423b3d24377","subject":"Drop lower-constraints.txt and its testing","status":"ABANDONED","created":"2022-05-01 18:13:17.000000000","updated":"2024-12-19 16:46:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"00aca93b2563c5f2f12383c55ba63b0c435d2ffa","_number":840135,"virtual_id_number":840135,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-01 19:21:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-01 19:13:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-01 19:21:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"998bbe3fbd8caf42b544117a3f74c45ecb996f89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-01 18:13:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5da2e25b310ebab4946958ad8f52fb797f81e609","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 19:13:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3239f29e84754117b095cae549fe3600 : SUCCESS in 47m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00da566299b54b2e9086d2b9a5fd80d1 : SUCCESS in 43m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e283262ce6743eba9fe7b66a83aaa46 : SUCCESS in 58m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/402f4042468d4ae3814451beeb6f7d66 : SUCCESS in 38m 54s","accounts_in_message":[],"_revision_number":1},{"id":"981a09f618e37e310804fe75e473d15709b124d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 19:21:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/59e67663e0c84f3bbfaf6b9ddf419fc6 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01db534aee944ba8828bfde9fbf0bfa8 : SUCCESS in 8m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8177c39d857a4c999957464583d46e5b : SUCCESS in 2m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb57dc1adf544c3f9b39672deb33aa9a : SUCCESS in 4m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/899ad50b37214ecab036136ee6bdce5a : SUCCESS in 1h 01m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/efa265cc0b334bdba6ef20fb1df16c92 : SUCCESS in 37m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73aa293fac5541e1a758803ea26ca475 : SUCCESS in 38m 10s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ee687a8c063547369d4c369a5d664fd9 : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/bb7b868356a04f1397136d44883c1f53 : SUCCESS in 22m 25s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/deee123b4cef4ef7a15fdb38c28076b8 : SUCCESS in 21m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/902345a8bedd433c86bc062c78502d32 : SUCCESS in 6m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/74c5ac4500834be8a09434b61ce4cc36 : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e27b16b5c4ce4a69a54f6f5429c81ec8 : SUCCESS in 17m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4686fb8425444aa880d16761ece5a62d : SUCCESS in 16m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cc00c185ec88467c8f703c703f760e10 : SUCCESS in 17m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d4ca8f0f9b3d40daacd09df4ae5adc61 : SUCCESS in 19m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/875e7a9ffce94af3aacf254b6ea88350 : SUCCESS in 20m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/61d28a6d973441d69451d94cc94f9906 : SUCCESS in 21m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ae09f4b72e0f4596ab99e24f736ce67d : SUCCESS in 20m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/68f206b502da46a1b365f14a61af42c0 : SUCCESS in 19m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dc4de6de1c7a432090f5de32a2e1af06 : SUCCESS in 20m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5c4353e6c957403fa5c4e5bd7f83ffbd : SUCCESS in 5m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f8b569fa631a404e98f06266b07656be : SUCCESS in 17m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eb30e76c7dee4db1b197a18039e5cc93 : SUCCESS in 45m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/73fc628844b048059f0ea431febdf8a3 : SUCCESS in 49m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fa8c9ee653e44234833c826e560b385b : SUCCESS in 34m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8f99cca78e654a46a5ed5cac663fc882 : SUCCESS in 34m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aeea82d679e40c8a5c2a2dc1c865637 : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b8a7e1f23b24f11834932c3e1b79b1f : SUCCESS in 23m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbf30bbf11e44ef9a3d9f86edc1db69a : SUCCESS in 38m 32s","accounts_in_message":[],"_revision_number":1},{"id":"00aca93b2563c5f2f12383c55ba63b0c435d2ffa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-19 16:46:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92d8f31a3160e4f1e248253d138fcadea2a4e034","revisions":{"92d8f31a3160e4f1e248253d138fcadea2a4e034":{"kind":"REWORK","_number":1,"created":"2022-05-01 18:13:17.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/35/840135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/840135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/840135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/840135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/840135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/840135/1"}}},"commit":{"parents":[{"commit":"66270beb392d2d5addc156edf33e3d5449ea8b34","subject":"Merge \"Clarify that rsync_io_timeout is also used for contimeout\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66270beb392d2d5addc156edf33e3d5449ea8b34"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-01 18:12:20.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-01 18:13:16.000000000","tz":-300},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: I25ce3bbaa107ddf51dfd2b69b3ab5423b3d24377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92d8f31a3160e4f1e248253d138fcadea2a4e034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92d8f31a3160e4f1e248253d138fcadea2a4e034"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
