)]}'
{"id":"openstack%2Fswift~840154","triplet_id":"openstack%2Fswift~master~I87a392a6fd11a68d19444aaffbf8f4e55838c51c","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-05-04 06:48:00.000000000","reason":"Matthew Oliver replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I87a392a6fd11a68d19444aaffbf8f4e55838c51c","subject":"Rework signal handling in background daemons","status":"NEW","created":"2022-05-02 06:18:45.000000000","updated":"2022-05-04 06:48:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":10,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"ccf8e2b2b6d060d49867fa9747238ef4eb6bd603","_number":840154,"virtual_id_number":840154,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-02 08:28:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2022-05-04 06:48:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-02 07:19:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-02 08:28:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-04 06:48:00.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"216b7fee8eebbf3ca0b413bd835b241634d212d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-02 06:18:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8857c591a09eb734e1171d8aa2b3effd88b8188","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-02 06:24:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe4b5ac8b0b8e848cd1f86064cdf9dffa7743083","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 07:19:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0208fefd300413aa8f3452f2fbbacef : SUCCESS in 53m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2935f8786cfc4382ae672863e56b792d : SUCCESS in 45m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/72e438a088d3477fb595a30b1bf78f24 : SUCCESS in 51m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/31a7d7c1b46a462780835235e6931889 : SUCCESS in 53m 08s","accounts_in_message":[],"_revision_number":1},{"id":"909d1670127676c634525cccaed98b2d3b2b28c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 08:28:22.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03523eb126934b1db708ab0acff7e00e : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/4a33440071074e6ebe9014efe7a10cf5 : SUCCESS in 47m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/60127c377df741a29429479d407a31f1 : SUCCESS in 39m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f826bdf4a1174a82beb7dc465aa7f926 : SUCCESS in 50m 55s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/10bbc986eb93423b9554e3ce269cb631 : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f4ab08eed1aa4582a0420ebd15675efa : SUCCESS in 22m 30s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d32c330b3b4645b590538e857cdeb9be : SUCCESS in 13m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/19ff518a95c4433bb0c594a82cbfd3e8 : SUCCESS in 6m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7d00faa53a10444fbfed73993c939aad : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d93b2f1525574b73948c2c1e48cecaa9 : SUCCESS in 12m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ae6681c3816149ec8197651d3891190f : SUCCESS in 12m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5108f648ab664bc2af2556a8d42b28c4 : SUCCESS in 19m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0697fd4d9876472192ad43835d208bb4 : SUCCESS in 13m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d8bcc5802d11485b950b113587d4e844 : SUCCESS in 17m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/af2ef9bbcc874c12afc4c4844487939b : SUCCESS in 13m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5a28f1971fe7410b80af5f2097cc7a80 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6452f8e0e6a84c4d896eddf9183387ad : SUCCESS in 12m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7619ebabd26f482396420844cd0247da : SUCCESS in 19m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a95e839b25c45d393d314b415cbf81e : SUCCESS in 5m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4972bf5e232b43cf961773ece24578f5 : SUCCESS in 18m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d7f7cdbbbc9f43a8b158d0dbd827b7c5 : TIMED_OUT in 2h 01m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c89eb2b4570c420ab3a9bf927976b1e4 : FAILURE in 1h 05m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c04d403d3cb944378ab3e26bf87bacbd : SUCCESS in 27m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c608fcb1d88348a2b589c648213a10fd : SUCCESS in 38m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39f22b8eff044441ac352a24a5789d21 : FAILURE in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0012dfb658f54dd89ae1769e77a14c2a : FAILURE in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5ae3e319e7e54cdbb12dee3a52624155 : SUCCESS in 15m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7321fe74cb294ea4a906cec4a9dc2a98 : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":1},{"id":"4d49b51694c88325c6d072e1f642d08bd5b0f68c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-03 17:02:37.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ccf8e2b2b6d060d49867fa9747238ef4eb6bd603","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-04 06:48:00.000000000","message":"Patch Set 1: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"299ab0ce63c362954bea5c21d34b7c20b22d9b65","revisions":{"299ab0ce63c362954bea5c21d34b7c20b22d9b65":{"kind":"REWORK","_number":1,"created":"2022-05-02 06:18:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/54/840154/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/54/840154/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/840154/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/840154/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/54/840154/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/54/840154/1"}}},"commit":{"parents":[{"commit":"ab612dd26d2c0e63fb99c9842aaf2ebaf19d8068","subject":"tests: Save ourselves 20s of sleep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab612dd26d2c0e63fb99c9842aaf2ebaf19d8068"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-05-02 06:17:05.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-05-02 06:17:10.000000000","tz":-420},"subject":"Rework signal handling in background daemons","message":"Rework signal handling in background daemons\n\n* Stop logging in the signal handler. Logging may cause us to block on\n  the shared handler lock; if whoever was holding it was blocked waiting\n  on another lock held by the main thread, we deadlock.\n* In cleanup, signal all children at approximately the same time with\n  `os.kill(p, signal.SIGTERM)` instead of looping through children\n  individually. This has also ensures that grand-child processes (such\n  as rsync) get killed.\n* During cleanup, wait up to 10s for the SIGTERM to have an effect. If\n  there are any remaining children, hit them harder using SIGKILL.\n\nChange-Id: I87a392a6fd11a68d19444aaffbf8f4e55838c51c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/299ab0ce63c362954bea5c21d34b7c20b22d9b65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/299ab0ce63c362954bea5c21d34b7c20b22d9b65"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
