)]}'
{"id":"openstack%2Fswift~840249","triplet_id":"openstack%2Fswift~master~I59488f3ecdc348027f35968e4a4e2d53d3f15e56","project":"openstack/swift","branch":"master","topic":"fix-logging","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-05-03 00:08:41.000000000","reason":"Clay Gerrard replied on the change"}},"hashtags":[],"change_id":"I59488f3ecdc348027f35968e4a4e2d53d3f15e56","subject":"patch pyeclib logger lock too","status":"NEW","created":"2022-05-02 21:51:51.000000000","updated":"2022-05-03 00:08:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"8475699ecf313eea930ff15b9f996e5f2dd7178a","_number":840249,"virtual_id_number":840249,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-02 23:54:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2022-05-02 21:59:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-02 22:46:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-02 23:54:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9235bfad3709732da7a01e51397e1d34ee94910","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-02 21:51:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a246a11b56e588a6d1ad511fb6545f9a5f80fdcf","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-02 21:59:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95dc11999a3650bf29e669f2aa3f627e2f5a31e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 22:46:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e33b077ea0d471082ea41603a4f1e75 : FAILURE in 35m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc603bc0fa5441fcac91c5cfb02ed18b : FAILURE in 39m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/90cdb9b901574f969f94f23d3c19b8bb : SUCCESS in 52m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a292214c23444930b717d2f4fbc88591 : SUCCESS in 46m 39s","accounts_in_message":[],"_revision_number":1},{"id":"4066997f74079ebb987f4f21fe0793c1800abc7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 23:54:12.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6211d46bbc04d999bd4d674805c7357 : SUCCESS in 9m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/cf794deacf9443689b76da0eecc926a6 : FAILURE in 49m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a96ce05ca65e44ddb8cfe9034045000d : FAILURE in 30m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/333e48d42d5d428a96da93b7568fdc99 : POST_FAILURE in 32m 14s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d77457f7c0c44ce49431505be7bd6dfa : SUCCESS in 21m 25s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3e4dc7c2cc0d484a9d1ca7314cb41a61 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/845888b9b80c413b8ce84561b4940192 : SUCCESS in 20m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/40f8f52951b2436da30168e5eedf5905 : SUCCESS in 7m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ad43131d55214149a954f9115e298ecb : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7eba43714b00490886bae3a589374be4 : FAILURE in 18m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/17cab211bfff4eeba8bb7750d2f3eec2 : FAILURE in 15m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dceaa87a93144d1a9a8b796f946b573b : FAILURE in 16m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/104ac192f685472aa23ffb85c7e9b411 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d30d4fc7294480796dbc89cf2082982 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4d7ce57530b649b58300564bfd15ee77 : SUCCESS in 19m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fed201babf02434ebd9a9db6ccca048f : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2471b67b2fb24ddc88b4da30039cb880 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f396311bab64492c9692d8aaaaba5471 : SUCCESS in 14m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/29ccf27feef545cabbc20fbb00e0ec8c : FAILURE in 5m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a3ee5c5494fd4a9ebe206092abf895d3 : FAILURE in 4m 41s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9695e078202c49e4966534414b8da0da : TIMED_OUT in 2h 01m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/035f9250e08f40a492372de407e8689c : TIMED_OUT in 2h 00m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/70a7cb02387840a382a461506ba13cc5 : RETRY_LIMIT in 18m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6483bbed347242cabb6710ac1a08bcac : RETRY_LIMIT in 20m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d209dbf9fca4be6aa0ecf6163f9c393 : FAILURE in 18m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5706d7830c2944689de4a24f7068254f : FAILURE in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6ea9b55dfaee49fba72061e7ff8afb72 : TIMED_OUT in 1h 31m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d445749695af4257861fb4bed3a199e8 : FAILURE in 17m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8475699ecf313eea930ff15b9f996e5f2dd7178a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-03 00:08:41.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4eb2316e6cd41dde9a84a2da0e5c2c72202d03b4","revisions":{"4eb2316e6cd41dde9a84a2da0e5c2c72202d03b4":{"kind":"REWORK","_number":1,"created":"2022-05-02 21:51:51.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/49/840249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/840249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/840249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/840249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/840249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/840249/1"}}},"commit":{"parents":[{"commit":"476986654eb3248b25303ab4c2d9fda1bdeee75d","subject":"Fix py3 logging._lock patching","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/476986654eb3248b25303ab4c2d9fda1bdeee75d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-05-02 21:51:42.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-05-02 21:51:42.000000000","tz":-300},"subject":"patch pyeclib logger lock too","message":"patch pyeclib logger lock too\n\nChange-Id: I59488f3ecdc348027f35968e4a4e2d53d3f15e56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4eb2316e6cd41dde9a84a2da0e5c2c72202d03b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4eb2316e6cd41dde9a84a2da0e5c2c72202d03b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
