)]}'
{"id":"openstack%2Fswift~840250","triplet_id":"openstack%2Fswift~master~I33a672f49dc09886f26b58f4fef6916d11f69547","project":"openstack/swift","branch":"master","topic":"fix-tempauth","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-05-18 22:33:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I33a672f49dc09886f26b58f4fef6916d11f69547","subject":"Don\u0027t give clients made up tokens","status":"MERGED","created":"2022-05-02 22:29:27.000000000","updated":"2022-05-18 22:35:09.000000000","submitted":"2022-05-18 22:33:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"840250-fix-tempauth","meta_rev_id":"cd0f3c7c52ac0c0e4e2ff7458a9ce2956d7e1159","_number":840250,"virtual_id_number":840250,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-18 22:33:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-03 01:03:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-18 21:21:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-02 23:16:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-02 23:42:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-03 01:03:31.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"96ab6f426ef329fec73c815b6b2528453a87939f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-02 22:29:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49f163920c39e4ffd947f5ae1d5f2d217d3ea9b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 23:16:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d3752f48ff44f5e98ea2aab255f7d0f : SUCCESS in 44m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca9960d57ab2452c88b58c8c579c345f : SUCCESS in 39m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5de3b4817924fc5b8cd2c3659a9ecbb : SUCCESS in 42m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a0ece4d452f43da8e4e56d6f9dc36be : SUCCESS in 39m 28s","accounts_in_message":[],"_revision_number":1},{"id":"2cb09ca9172b5223b20fcdc4843bc9909d018460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 23:42:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9eac02740b3418e96f6741611c8c529 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/a0d2b085cf3942b88ef654641fb91adb : SUCCESS in 1h 06m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b0fbba7804574becb1d0171f754c31a8 : SUCCESS in 56m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/652eb08c2a234fed922a1362cd1e06dc : SUCCESS in 55m 17s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/bbf77e575d0b4d9e930762bed0826e63 : SUCCESS in 18m 41s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8ecb5d3e76404055b53a894ba1d80137 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/94ccf52cca1740b78424a97c4e316903 : SUCCESS in 20m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2a4d3353727f4241a6cc904706112fb3 : SUCCESS in 6m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c55d9850e1ba4a1cb484b1ee1c44be80 : SUCCESS in 4m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2f51e224199b4c54939cbf94fbfe376c : SUCCESS in 16m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3acb63dbf6c54b129ffee543b9b698da : SUCCESS in 28m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/623d33e10a744c179e016d72ea10f3d9 : SUCCESS in 17m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6340d572bf8046038a7880583e18f188 : SUCCESS in 29m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3614c13074354b27828ec6212d9dd9b1 : SUCCESS in 17m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/006151623e474c738b89659da894c7b5 : SUCCESS in 19m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c1556d82a61647d1aa16a087c679e2b0 : SUCCESS in 29m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2e51421d9ca442e7ac4034ed491beb22 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/be1e4a1e57364e1892f65f62bbae065b : SUCCESS in 27m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/549d98b6b73c431694fbe01a726d568c : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7bb554e0052a4d9fac2666698c6826c6 : SUCCESS in 23m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d48a3010780d4ad5a0c4d8cd1695b2c6 : SUCCESS in 54m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/02e0f7229c91493189f9fedc282a2885 : SUCCESS in 52m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/89453b1ffb814d6399b45ed8441a8c72 : SUCCESS in 42m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bbc01f321417412f9ebae5d17f8d5332 : SUCCESS in 43m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37f8a82ea20142598a88eb494b0bf323 : SUCCESS in 24m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ea85d24cd2b479890b5582b7961764a : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7508179f7ee4511b6eb39161cdc2af4 : SUCCESS in 20m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/692f1fbf92fc4ffabbc84d6f0be75a4a : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":1},{"id":"04b039a5b17371a06cee79e7153a894532378062","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-03 01:03:31.000000000","message":"Patch Set 1: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"03e6da142790be8e774e16877a09f4c34aa82ac8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-18 21:21:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"769204948b900de9eb6eb30bfeb71b61f55c5513","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 21:22:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"061fd2fbe63f42d3aa8695e25f4887d2117da4ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 22:33:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/095580fd5f434cbbb893f188fe16f92e : SUCCESS in 9m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/39755f46a43a4a5b9ceee9fac179c7e4 : SUCCESS in 43m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4652ffe6de014b858c8eeef24475ce90 : SUCCESS in 47m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3fc3c0e62944872bdd5f238d4759e37 : SUCCESS in 48m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/91e89e596b20462a90e3cbefef6b12ad : SUCCESS in 7m 33s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/368f66f10e1146b09f04d41e6bd71789 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f63d58d40fe44f42909a5b8cd1b31722 : SUCCESS in 17m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e7ababbeb266482eba2264f49dcfc1d8 : SUCCESS in 19m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0466afcc91ec4e04a14efa92f1cd4932 : SUCCESS in 14m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6e2cd8e1a648463dac01efc4c0cd1a89 : SUCCESS in 26m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/02863c83a49a431bb0b8ffaa3c4b38fd : SUCCESS in 20m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a21e4d089154864a7cb8ea909b2f701 : SUCCESS in 28m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/46c563fc7b83469eab82472b895142e2 : SUCCESS in 27m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/afe3371aa54e41ceac72da472d8f0d66 : SUCCESS in 17m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/770adb505c414a6e9a70334e638e3ca8 : SUCCESS in 28m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/29ce03a9c8094aa98ef101178ebcd429 : SUCCESS in 6m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f9e4b42b9d73450986fd6bc1fd2721d2 : SUCCESS in 51m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9ce60f350954ed9977a9fcb9e77e3f1 : SUCCESS in 1h 05m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ca40e636c0342e69e60aa2f7394c913 : SUCCESS in 40m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0d980822be40470da1a0ecb5d6d520e4 : SUCCESS in 43m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a129508826444ed287275c11ff6f97a1 : SUCCESS in 19m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e13c56713896431abb20eb0a6111ebef : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/24e707e4fbd04b6ea20ebb6450288829 : SUCCESS in 26m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3aa091eeb8044a985879e6c90c97a51 : SUCCESS in 46m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f7af98f337c4a40b1cadb98a06c73442109eb081","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 22:33:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cd0f3c7c52ac0c0e4e2ff7458a9ce2956d7e1159","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 22:35:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7401887c67346628a5e18aea991e886 : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c86a2b22aa7147ef9c8da4542c36c4ba : SUCCESS in 45s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b3eef45194e04d4e923e3afb34034cd6 : SUCCESS in 53s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"94226bdd39a64415e456057740add0ee6190d033","revisions":{"94226bdd39a64415e456057740add0ee6190d033":{"kind":"REWORK","_number":1,"created":"2022-05-02 22:29:27.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/50/840250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/840250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/840250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/840250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/840250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/840250/1"}}},"commit":{"parents":[{"commit":"9bed525bfbe3ec05cb45c9835333fe45f8e04b40","subject":"memcached: Give callers the option to accept errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bed525bfbe3ec05cb45c9835333fe45f8e04b40"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-05-02 22:28:03.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-05-02 22:28:03.000000000","tz":-300},"subject":"Don\u0027t give clients made up tokens","message":"Don\u0027t give clients made up tokens\n\nIf we can\u0027t set the token in memcache return the client a service\nunavailable error.\n\nChange-Id: I33a672f49dc09886f26b58f4fef6916d11f69547\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94226bdd39a64415e456057740add0ee6190d033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94226bdd39a64415e456057740add0ee6190d033"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
