)]}'
{"id":"openstack%2Fswift~840531","triplet_id":"openstack%2Fswift~master~I6554e37c6ab5b0a260f99b54169cb90ab5718f81","project":"openstack/swift","branch":"master","topic":"p-backend-ratelimit-reload-log","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-05-11 20:26:40.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-05-11 20:26:40.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-05-10 23:06:08.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-05-11 18:08:49.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I6554e37c6ab5b0a260f99b54169cb90ab5718f81","subject":"backend ratelimit: support reloadable config file","status":"MERGED","created":"2022-05-04 18:35:27.000000000","updated":"2024-05-11 20:27:50.000000000","submitted":"2024-05-11 20:26:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":75,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"840531-p-backend-ratelimit-reload-log","meta_rev_id":"3de4fa1867a0264f85979d45788792cf968384a7","_number":840531,"virtual_id_number":840531,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-11 20:26:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-05-10 23:06:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-05-10 23:06:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-04 19:40:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-19 20:19:50.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-05-02 00:43:59.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2023-05-23 23:07:30.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"5efe60a06c0e2a61757c4caf81b13c3ca2f65e76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-04 18:35:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a20704130606d99bd54faa8dad351575a39762c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 19:40:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4bb016a9dd14dd0845a021c1c44c58a : SUCCESS in 9m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/3eb749a2c4dc43e486d098d41c34a7a5 : SUCCESS in 47m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/93f48afc8b3840ee850e7dd9554b2b5a : SUCCESS in 46m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d942118600f94cafa875a741e36e407b : SUCCESS in 44m 33s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/34401155ec8d4efc8c0a87680ebea3e6 : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fc016167f3924abb91c5e5d7e68abf81 : SUCCESS in 14m 26s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/64d763ede92848c0b2dbd35831814042 : SUCCESS in 25m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c57b804ad87244e1a1a183d91a0ce3c3 : SUCCESS in 6m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e9d97484b5e443b699397aba2af39559 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c7600c51054f4ce99961e736ab40f4ec : SUCCESS in 12m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/63e392148b9b4b2cace2ac4086f453e2 : SUCCESS in 16m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8a3f4fe08ccf446c92ea680d57d2f6ec : SUCCESS in 11m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0019ea7bd9bb4d34b57b0ecb8cc582a4 : SUCCESS in 20m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0a6407a068dc4509afdae3495723fc29 : SUCCESS in 13m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/72babf1275f84440a5c93bc5c2d30884 : SUCCESS in 21m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0b3d3688322b41acb1b8648e551b270e : SUCCESS in 19m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/008c61c970c749b7850eb5a4fb635cec : SUCCESS in 20m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d7ae12f46b2148d697e06edc753c5f0d : SUCCESS in 13m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e359d88489214779b469e5bfe2925ded : SUCCESS in 6m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/75c1cf7c81a749c7ae92f1f0a6f14023 : SUCCESS in 14m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8551a517378a4664a5b1b5868327da1a : SUCCESS in 37m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e7fc6fbcba64be2b17f5e37a44c88a6 : SUCCESS in 1h 03m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4cab28c320e84d3da1e19129992f9a3a : SUCCESS in 28m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/13baac4cf43643b4b74a08c1a71e20ff : SUCCESS in 41m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2db8afd78e2e4754aabad33c57601487 : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80c7af95c851472dae1a865b009954ec : SUCCESS in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5bf1e0794f1542cfa795a774b66cf0da : SUCCESS in 17m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca7f601f16ff4e289337abfbd90fe87a : SUCCESS in 43m 56s","accounts_in_message":[],"_revision_number":1},{"id":"08d26c503eae30a2d2e160497ed11bed148af485","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 20:01:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3216f8d528ce4467881bfa3d58f2d430 : SUCCESS in 1h 00m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4874cdf86444dba875f1cc5f62a2d04 : SUCCESS in 1h 02m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/306940a555364906b7000ef31d65c4c0 : SUCCESS in 1h 19m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e8724c5093a44abb97a2d6717fa915b : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":1},{"id":"78520609dffdd7f123d0d2599587176447976e74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-05 19:49:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a9d0cc6ae9184ea762f72e282af858935fdb7a98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 20:39:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2792248640f4b0bb2db6ed7094561e0 : SUCCESS in 42m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a66ee5c9e24486d88add6d00aa9ef8b : SUCCESS in 35m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bac6bf23b3b4556873a6e14c9ed4207 : SUCCESS in 37m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/31747843e5194d04b7ba2e9ead418b12 : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":2},{"id":"48ab13ba323381ba363b8601889561a431959fa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 21:02:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9daef4a72fc64d6e936a20ed62d25f1c : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/a0017cb9b16e422a824c8c811c0ff306 : SUCCESS in 1h 07m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/81fea7fbec854fc4b73b083a53659ec6 : SUCCESS in 55m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d18711c49d5740998f9fe5485358e24b : SUCCESS in 53m 03s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2a66e9dacc9c493e9ccafed1170f7082 : SUCCESS in 24m 12s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9c316b71967a4d87a92bc308297fd646 : SUCCESS in 25m 18s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e72f675c253e489cb89f34265850e47b : SUCCESS in 24m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/23a16bad5efe4be2aa37f40962cdbe40 : SUCCESS in 7m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c248e6b6d17c4ef69e06305f4532842c : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2a0e4355cb154fcaa59b8eb24f101262 : FAILURE in 16m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7bb521147c624180a775875bee224d49 : FAILURE in 20m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/84ba7ce9584443feacba97056cbe722b : SUCCESS in 19m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e45be561c51c42b999884a01bffe57ac : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fb9489ed12664274b86857a0a8fbba8d : SUCCESS in 17m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cc2fe1ec137f4116bf2c32e6f8497ab8 : SUCCESS in 17m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b0d108ce85e443d2971c76bfdc83ea11 : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7db73ab0213a4012a540fd65a66a90d0 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/658bdc766bc24605832036b0805550a7 : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7ab94d5fba534c78aa4c8bd1c8d8aa71 : SUCCESS in 12m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4be6a55f3e644f6eb5fedeacbec1776c : SUCCESS in 14m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/360104b72efd4acfbed58bf397e35d42 : SUCCESS in 46m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8549221d617f48d184a4c1881001e5fd : SUCCESS in 52m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/99c447e1c07c4f5d8ff86198794926a8 : SUCCESS in 43m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/87c6a18afb6542c5b3721c87324736e9 : SUCCESS in 42m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09010b67d83b458baab5b3cf3039d1e0 : FAILURE in 15m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20fb2b90c87f418786355cdefcf0bf43 : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/65bde2cf9c6f4fa499dd7b738cf5aee2 : SUCCESS in 19m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5dab154c8914187951120abec099bc0 : SUCCESS in 41m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f0f1e3fd3b6085ce0852b70497cda96195f13470","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-06 13:05:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b7fcfd8535c8f850a51c8ad1d2c74dcacbf24a47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 14:05:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4d9f958c7154dfd97f44e3d7c0903df : SUCCESS in 48m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3ccdc836ef37415396eca147448a0a35 : SUCCESS in 47m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/10a069ea6c0645d686cd3f56737abdcc : SUCCESS in 56m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2fa447a87c0645979f09797f2ad8e02f : SUCCESS in 51m 42s","accounts_in_message":[],"_revision_number":3},{"id":"5633bfe0382cfd1e242b91cd5ef7a219a81a146a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 14:09:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4fff4220d0147ec92d2e4884d4719de : SUCCESS in 15m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/93fd7fe437e64c96b5ce1fe326672b89 : SUCCESS in 57m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/27b3c72b3ddc46009d22843052d353d0 : SUCCESS in 51m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aeb48a98be0c4137b4355770f114c1a9 : SUCCESS in 50m 35s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/94ba28fdcacf4c6f8c6d40c2357c0fba : SUCCESS in 22m 57s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/aad3e869397346ec9de63f2cedad32bf : SUCCESS in 22m 00s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c7aa381d0c50479fadd6686abe542f1a : SUCCESS in 24m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f5358a8ead5444b994cd53c9f7a21f6c : SUCCESS in 12m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/49927573527948bbaee18c6097dddb24 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a7999aa064a04213806ccaa74218c080 : SUCCESS in 17m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fc19341b02164321a05820ad68ff0289 : SUCCESS in 17m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/11c30441a8cf4bf4a46f0cd495fcb334 : SUCCESS in 18m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e40867e808294ad3b00a578260f7c417 : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8241bd385d1c4be19be9d621c22d3df3 : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e9f7e0727923401d8081df89907aa128 : SUCCESS in 17m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3619115841c64fb0a95490644107faa4 : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2ae1d1e481a64f3ea03c834e75103dc3 : SUCCESS in 16m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6c3ef3d7788541a9a62af6870c4a26a5 : SUCCESS in 16m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d2b8e5420f043148f7805e28de47abd : SUCCESS in 7m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ffa05941d03a406ca9adba1fb1144c89 : SUCCESS in 19m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d221ddb123374bcb9ddcb9e9e7ae4415 : SUCCESS in 51m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1513b65b350a40e6b52b2e56e0e80fba : SUCCESS in 59m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b861eb384e354973a396ff82e0864bf1 : SUCCESS in 46m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2c1c99cd0d054651a8c478436de006f4 : SUCCESS in 30m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11ecc1ee03ac4bc5b223027bd005e9ba : SUCCESS in 20m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab47f530716e4b08a179244a81003fc7 : SUCCESS in 6m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b54d4b2f7bc34f4bace0567691663673 : SUCCESS in 19m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6e0d3bd14544bf4a0469ccac921969d : SUCCESS in 46m 35s","accounts_in_message":[],"_revision_number":3},{"id":"40d2d1d42e771cd986f77aa0f788a6d7c496dd5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-19 13:35:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"c5a2d5d62f9827d7c02200ccf77bc480ec7ee500","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 15:18:24.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76c02d104bc14e8c92471b7306b10f96 : SUCCESS in 49m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf8d9d17cbb04f95be822db005374592 : TIMED_OUT in 1h 32m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/95bf160ab2d14ad48ebaa58e915aa314 : SUCCESS in 58m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/39c0bff8ad994436ae7aaad236ace7e6 : SUCCESS in 58m 08s","accounts_in_message":[],"_revision_number":4},{"id":"2e95dc37d62ee28113a12de8d01f4a16f283bb14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 15:40:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ef51cfeffcf4e2b958050f93cb98f7a : SUCCESS in 9m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/6a77296ed0044409a4410747e4ab338f : SUCCESS in 56m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/baff7234d792492098553a0737ba79e2 : SUCCESS in 46m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/294bc96de8b34a10bc4558cdf0b349b4 : SUCCESS in 49m 58s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a671044ac5a84376842e7cfd4dfa7c90 : SUCCESS in 22m 44s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/88723fd530db427890e81f38b0b0023c : SUCCESS in 22m 43s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/442c71854fc846b2afc3d1c21fce7842 : SUCCESS in 24m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4bd4f1d315b748a9b1797d6e4c49c016 : SUCCESS in 7m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d40ac8fd09dc40e0a5a503059003f05b : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/213ffd0407444b86993ddb21cba028d8 : SUCCESS in 17m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fffc42c0218744cc90a2b38604413d48 : SUCCESS in 28m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ef35e45b1bb642a58f3bfa74a9cb500d : SUCCESS in 16m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b63ed2a4bdbf46fe9a4c553ed87637ba : SUCCESS in 22m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f17fa2175d5e4394a10339ddf51ebee7 : SUCCESS in 22m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9688d2745de94ba1a89192d6d3d538ca : SUCCESS in 17m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a6592a12aa6a445f859abfd6d8b87b64 : SUCCESS in 19m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e25feef90047473e9cce3d82ed296e48 : SUCCESS in 15m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4ad67f20c4504eeab3b347f1e7e2e7a8 : SUCCESS in 17m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b5e8de4d97e14d0c999fa0d116ea1ea3 : SUCCESS in 6m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/03d057c9ec1743a7937248e9c94b800d : FAILURE in 16m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/687facb9236e4eec9768e9ba180c8603 : TIMED_OUT in 2h 01m 10s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/418eda0460114c3e8c16b537b02c385d : SUCCESS in 1h 20m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/414da47e3a984f05a8f40b87b578f7bc : SUCCESS in 43m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1366b112688043358e59194be6831393 : SUCCESS in 54m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/038ed9716f0841ea8efabaf4636fc728 : SUCCESS in 30m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a4684985064a409710e1d990c06e05 : SUCCESS in 6m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e5c503e3c46f46da968724076fa907b7 : SUCCESS in 21m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d213a9fceacb4626905997de3030b9b9 : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":4},{"id":"49069d90ddc634baf568ac6df21a14a022c384bc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-19 20:19:50.000000000","message":"Patch Set 4: Code-Review-1\n\n(21 comments)","accounts_in_message":[],"_revision_number":4},{"id":"97abc5dfaca045c8ab43d09eb7895b6d40817af5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-20 13:25:28.000000000","message":"Patch Set 4:\n\n(19 comments)","accounts_in_message":[],"_revision_number":4},{"id":"538a69d70141d86a30c7f739c2ec15cc99c08443","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-20 13:40:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"067965ac638c9372da07228e0de200f967c59837","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 14:32:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/921ba1a86a6b4beb8671f1a332ba089b : SUCCESS in 39m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/447998b7fb7546a08f32a3cc418f0375 : SUCCESS in 44m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8bbb2c704ecd470db549f3b0ee2bede4 : SUCCESS in 49m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e208e39cbbb34d68b84cf097ffb71550 : SUCCESS in 49m 55s","accounts_in_message":[],"_revision_number":5},{"id":"0ba2f8723195282de18b6b304e7231ce270b74db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 14:33:42.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a759ab6d29e46ef8c0027ecec01b7c4 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/044ffa539c8741debc4c01c624c402a3 : SUCCESS in 49m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4294dd069f5e433d8d2326fd62aebd3e : SUCCESS in 34m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd5ccfbc989d40c7baedf15cd72dae88 : SUCCESS in 47m 36s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/af0736c3e34c419e8af767598f3220dc : FAILURE in 7m 32s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/47f24f4d985b4954a91cca11429544a9 : FAILURE in 6m 18s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/35f5d1b7723648da90a6912b6b754828 : FAILURE in 6m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/28961d901b754da789120c61d5483906 : SUCCESS in 6m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2e8f0c2fa6a840c7983446c292cc86e2 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/787002fb4b8b409e96f42d9c519cae8f : SUCCESS in 16m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/88aace6cfc5244268cf1ee7679f9bc6c : FAILURE in 4m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a0a939785dbe4095b17278b5564f11e8 : SUCCESS in 12m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f26317589b1545dbbba3e75014d7cffa : SUCCESS in 13m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/94e48baed76f47df9a031848c09d5e02 : SUCCESS in 17m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/61bb35d44514486d930b70a996edc035 : SUCCESS in 15m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c3cf39356d0140d2a44b5865fe4c86e6 : SUCCESS in 12m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c27fd2e0b33047289527efcea849263b : SUCCESS in 16m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ac4e1003230446d297ad586b13cf1ef5 : SUCCESS in 22m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9e4687dbca494699b4aa2745f4acc3b5 : SUCCESS in 6m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/44e460ff8f5246fd85772d8c236247fb : SUCCESS in 17m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0fbbca5219474ccd8241d426b07f7fbb : SUCCESS in 49m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/488bb7d664bc440e9dda3060bdbcc8f5 : SUCCESS in 42m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2859603c0c8d464abb8285d3dec72acd : SUCCESS in 27m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/39038020ba5647a8878f3d393c4f685f : SUCCESS in 39m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af3b61355fd54948aad168c9e78da66d : SUCCESS in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e58fdd9af17b420abfed20d8f5663719 : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b914b8fe10944fe6a48429e255f758d1 : FAILURE in 9m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce39737bf50041ff93bd132fd717b60f : SUCCESS in 43m 34s","accounts_in_message":[],"_revision_number":5},{"id":"cc970dbfad04d38c5def239b98e2ad4b54b5b75a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-20 16:37:02.000000000","message":"Patch Set 5: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f32df1cf7983fedf61f8f1432ff779118bd27efb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-20 17:42:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"eafec995803e37198fa3bd58d0812d455527490e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-16 14:10:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ead0005687b9545c681989b096a95c01980632a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 14:34:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d6fc8581ed04810a0f649f6d91dbfb9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/42ba1277f742444c8fdfe95c15367fc9 : SUCCESS in 15m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4db4b9ac079e480d9da260dc2addfe2a : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f53b9ff592144fee98d62f6603d57146 : SUCCESS in 19m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7a976474e0942a2b55cb3dba5321da5 : SUCCESS in 22m 48s","accounts_in_message":[],"_revision_number":6},{"id":"97634e2394659c7cd547d5842710140d1a0d73ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 15:20:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd1f7941c01344538e7b4c555286137f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d66f96e751074e06902103eb02ddaaa6 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/d45fb0f051ef4ddf899ab794d42ac3dc : SUCCESS in 1h 08m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6bc84d108e504643ba1308eb2be32861 : SUCCESS in 49m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89c14e4d863e44ef8cf1eeb46ff37238 : SUCCESS in 59m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2882b494ac144915956b981c24e963a1 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4210751cc89a4c73901142da4c9f7238 : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc6ae01fe8744b928ca21b82c19fcc51 : SUCCESS in 20m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1f0461aa1c24e5cbd4b70a5840e7eb2 : SUCCESS in 9m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0f3987eb96684b34a214e33a43b4bd00 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/548261aea492473e90adb1f1fe2224c0 : SUCCESS in 17m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/005f446ca96f46fba1d9146e7c464090 : SUCCESS in 16m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5db796d6e19143bbaa19c7a6a89617df : SUCCESS in 14m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e22a28c9357b4973848e716446bf26e4 : SUCCESS in 17m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e316dd2a90ee4b1eb5781397e11a86eb : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f96a9e1c2e2246a1b518bbd37d93b9f4 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e76e14d56e7744e191fef71dbfea1e8b : SUCCESS in 19m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/01b2bda9cb2d49ddbac621c386bd4da1 : SUCCESS in 13m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4d0ef240145146d2b8d09f2b79cb4f7d : SUCCESS in 13m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b267f46cb9df41e6b01145f3ef74760c : SUCCESS in 14m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9ee9bd9f16cd49b7ac1d387938bbc423 : SUCCESS in 8m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/978d3d3680434f28afef7262d67c93f6 : SUCCESS in 15m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3cee5bf6f5e14c19be183a1e8f6b7d39 : SUCCESS in 9m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2d1808ae371a4272aca9adb18dc77f7f : SUCCESS in 57m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d24f9e65e31b4352a5318d91ae563fa2 : SUCCESS in 58m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/546678902f794a2dad121c604ea6ebce : SUCCESS in 49m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/adb1ccaf523344a58f421d794d795509 : SUCCESS in 46m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30efdd883b514b9aa68f4cb2c5ce48f9 : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bd6d8749b864e7ba696e95e40592e60 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c700b575d36743ebae269c58789bea6e : SUCCESS in 16m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38921d50d93046d88d736563ddb73eea : SUCCESS in 51m 51s","accounts_in_message":[],"_revision_number":6},{"id":"1fd6217dff24a352114de5901dbc105d49cae878","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-02 00:43:42.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"06f3e6006231011b929cc3b21b726ed032503fce","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-05-02 00:43:59.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e61202b694da015a4bc779ea7bd13c9bffef6aa6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 01:05:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7162372d5ec4095bc0eb0276c5b4b82\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/55ad5d926b764ba084653d4849635cda : SUCCESS in 15m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebba0071dea04e8b853adcd2cf38c013 : SUCCESS in 14m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/61b62564ff7c4f538a52897279c65c1e : SUCCESS in 20m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/49ac467fdf69492bb9b1f403cb588d76 : SUCCESS in 21m 11s","accounts_in_message":[],"_revision_number":7},{"id":"efe4f0bba08a39686ce52e2561c303926822b7d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 01:49:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b03fbe5dc19a4ff5acb8aad8e7d02b50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78fd97249924bf7888ce83ad6a90d87 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/443b345d5b7f4cf59f40459d895a4b8f : FAILURE in 34m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0d507f02aa104dc6811b2bdf513c9c2e : SUCCESS in 33m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42339b9a4b8c42fe97b49ae4b471ac56 : SUCCESS in 1h 00m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e921a16b1634f339d05d43e48a3dfc2 : RETRY_LIMIT in 1m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f44540cfa36f42bd8ff991418e40cd66 : RETRY_LIMIT in 2m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a2bc16b849a400da050c6274e9d271c : RETRY_LIMIT in 1m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bef078e48f254bf48cda5c2b47529c66 : SUCCESS in 8m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/35b1dc458a774796bdede6f6f43fa3f8 : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9ea5e1e46d7b43b3bde3ec80bccf918d : SUCCESS in 20m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0854df91783044df8ae8032ac479ed82 : SUCCESS in 16m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7f4a1e946bf549019eb43c4c52f456e2 : SUCCESS in 18m 12s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c958492d4aae449dbe7128346dfe6090 : SUCCESS in 19m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/07b39d554cc049338a8f8fb664f33bee : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/98b525ef5f094f15b4e8067dc0202939 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fb3469553e5a4cefa05eaedb1000039c : SUCCESS in 21m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1a26af819b5a415c9e1c60452712f853 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/daca7e15eaaf4cfc95428afa36eda630 : SUCCESS in 14m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9ec7950df16e4d9c9c168beddc6e551f : SUCCESS in 19m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/73308849cf874836898290b305d0a162 : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/97224f2c3f0448cb9543c86a4b1e939e : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e4e4eea41ed44629869a2158fd497c94 : SUCCESS in 7m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/df0b884c8a584ecfb28868473af20a28 : SUCCESS in 39m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/336b707ccac540b490aa5aff228b13ac : SUCCESS in 47m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d22b40cd58484111911619588269ec95 : SUCCESS in 33m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f097b644537742a18934d62c79511870 : SUCCESS in 32m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af80f201f6de419185810becaf94de48 : SUCCESS in 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdafe106d4504f81abc01c81cacdb9f7 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ba5ecc97b2404c569426ba9c6b109797 : SUCCESS in 16m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01d3dfc1c4aa451d829a6c943b7b6364 : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":7},{"id":"fd9eb7a6de9aedfb533b360baa8b6ffacc8323cc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-23 23:07:30.000000000","message":"Patch Set 7: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"0a82e33c191ac0c00e36878ee436ec316d0c95a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-22 18:26:23.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"2211738d6c569596cf1c83ca416ed9606dd89537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 19:40:49.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ea60a4543484bfc8a74a8a4e1b799ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbb409b8239e47dda8941e1edd3dd256 : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/7468de499cd94f96949075c3e3b8bc8e : SUCCESS in 1h 02m 32s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6d3cb3b13c2f43849a4383d1c36846c1 : SUCCESS in 55m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f9b590e9b6654cab812931f6e5718f6e : SUCCESS in 51m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7ed3ae7afac42d2894965fa58bb83ac : SUCCESS in 55m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9436311205d40369225690368e703a4 : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2584dc7b6adf4b5591147c920f4f5786 : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba5d4e96e0c5450e8affbaaf6659a180 : SUCCESS in 20m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/218c11b6495142fca5b04367761b3037 : SUCCESS in 11m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e4f86c96484c49f7b5b92c93327f7554 : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d7525db1a23643e99d194dfb1a4a3ef3 : SUCCESS in 16m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a37c4cfc5574dfb9ebcc8b64b6bc459 : SUCCESS in 19m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cebb980ac8dc49a587d8377dfc6a7bc7 : SUCCESS in 22m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/035c73bd4eeb44de9e7d6f70cd760418 : SUCCESS in 23m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bbaec6e86dff4ee485e941ac11322573 : SUCCESS in 23m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1e1cde366d28412694bc3dc3febd24e1 : SUCCESS in 24m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/50fec45ef8704a578791f0ea53efcf21 : SUCCESS in 23m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/49680649a722429ba2873e8392c0dbce : SUCCESS in 23m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/302ce8c77e684fea86ed30867f1f8e34 : SUCCESS in 21m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f496c0ea116145048409e6532f59a7ef : SUCCESS in 23m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/490d494622e44ba9921259af20182dfd : SUCCESS in 24m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0e0705a100fd4142a165dd3e0f16485a : SUCCESS in 8m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/447f92aa36ed490e84c46c364f75eae8 : SUCCESS in 14m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a82a3973b50742e189f497f871d50afe : SUCCESS in 9m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f16007e3b81f4798ad6fbdab1702a26e : SUCCESS in 45m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6418d1f7a9c9413e897ab53086e2cac6 : SUCCESS in 1h 06m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/22a0a0cf103a4b0abdda82d5f44f6465 : SUCCESS in 1h 02m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6d9b9e2902434e918ac1e8b62bf59659 : SUCCESS in 55m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8b952bd720345a595b4e6e0958c2788 : SUCCESS in 21m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02adb3240bc34c7fb4400090a28c01f4 : SUCCESS in 6m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d1c7edc5a15442a2bf47a0fea9677c18 : SUCCESS in 24m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac91d672c51045cea1b9ee699b4ab99c : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":8},{"id":"95c3bfc3d7c08bfce5c4e5cce0bc61e1213e48ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 20:13:30.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/984ae9704214408e947b15f00bcdcebc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/395da9b0117e4e18a968322cf7fddf50 : TIMED_OUT in 1h 30m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/39e74d18436f418db2cb88fb8bb8109c : SUCCESS in 1h 06m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/12d1fbdb7957493b8e402a33f36bd84e : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f97f66d6015b4389b97b2dbb6208c50e : SUCCESS in 44m 56s","accounts_in_message":[],"_revision_number":8},{"id":"f42c0358c8f04398776a8b43b728e07d3d5285fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-23 16:42:49.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"1603233540b0f88b38c5af794662a52afe057d67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 18:09:46.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d59ca2ae9a640539021c2d33685eb0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d0a48fd077144709a27f5e1aeed43be : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/c8d5f3f33d6944ad8d6c06c7c2743392 : SUCCESS in 1h 04m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a08214d2fa9f4c6bb106989f56cbc978 : SUCCESS in 42m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/030286dd1bbf4c0e88953f51249ea81a : SUCCESS in 45m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f770225dd92546f49bcf711f9769c26c : SUCCESS in 55m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5a0bb489f3a45b29a35be69d4ac9524 : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6ac6ce426354c92a4b520efa600db28 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffa5d810d4874f15aeecca250e7e9c16 : SUCCESS in 17m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a568eae6a284cf5a21b1374924514e7 : SUCCESS in 11m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ca33fca68d6f419da6bb86c18ef517a8 : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/54d28d9cf7244f3c8186e92438136f0e : SUCCESS in 19m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/19485dc1e3274b91ac303142cda374de : SUCCESS in 16m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/517c89ec0a25483b8d6b1c0fd100d1cf : SUCCESS in 34m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/00466914bb634cc194a6f3e5680d4297 : SUCCESS in 17m 07s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/edc1c28489c2421c8989bf20fd984bc7 : SUCCESS in 17m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/49d1e172cf2445a38a5eeac99b9469bd : SUCCESS in 17m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/43711a6ca4ad43d4979883b74258a535 : SUCCESS in 18m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/821a4882f71e4ae08850e2b193eab732 : SUCCESS in 15m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3cbfa73beda649ed837b995c805cc057 : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e2b3ce1196dc4c68b164e6913e2dfbab : SUCCESS in 16m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/20af5ab56bb24b0b99856d651dc8df5f : SUCCESS in 14m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/57096fab45bb4f6d9e279b8e2451ccfb : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/00e5a50434a94c4e89417f88817f42bf : SUCCESS in 21m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b487d8abd1f04bafb5ace3ee4ada0f06 : SUCCESS in 7m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3c35610f6d2e41caa1b4e9c76d0ffeeb : SUCCESS in 1h 04m 58s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/47622e45650e41ad8990cafa9af306c9 : SUCCESS in 1h 01m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8c51995cc1ae4949b36a191fe42c7025 : SUCCESS in 48m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/550452ed38c74d2e99514c0242c807fc : SUCCESS in 51m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/302eae941fc347b78053733e8ef44bef : SUCCESS in 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d35fd0b9659a47839f1e9f90c867f6bd : SUCCESS in 4m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d6dfa220159a43a3a074612ceb3b1478 : SUCCESS in 20m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f73c186c17054264aba412204f9e1f10 : SUCCESS in 1h 03m 53s","accounts_in_message":[],"_revision_number":9},{"id":"f19ded4c81cdcca3eaddfe21fb7d3406425bf3f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 18:19:44.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c6943f842a44cb7bc01eabd00329a8d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae5670a9896c4834b3168b790fe2034d : TIMED_OUT in 1h 30m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2aac1d3703f449cbbfecf6bbb412da2d : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c86e3d4e86344f5cbb2523a4f43eeaf7 : SUCCESS in 1h 24m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/779bc078fafd4f0b9dad7d498fe93e1e : SUCCESS in 29m 12s","accounts_in_message":[],"_revision_number":9},{"id":"0c752d101a67f89f080372675eee430db088e0b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-11 18:14:24.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"b78cd1f8fae24a9b0f9669514add2558bb765925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 19:32:44.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64057cd828234d5d9f3f18d0fa56706d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd99559201ae43fea6d8b68d58539677 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/f67844a0c317481a88780190d6dd389e : SUCCESS in 57m 02s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/21eae2824eef48ba9b470a65d090b903 : SUCCESS in 1h 02m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/591e3c1a5e0d4f2eab63cb3e5ca70bf2 : SUCCESS in 53m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41478e3ac5004c958c23196b4192bdac : SUCCESS in 58m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8fbeb260549144a5b95449d9859a4b9a : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b4d289244174e038ad074ec424b4a4e : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59498899481645709f67608b03757fa6 : SUCCESS in 42m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/659ccc36dcac471fa59468f6f3becf57 : SUCCESS in 10m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2cc81a2be0d844c1a80e92da5233910f : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/91f1885c04b54cf3922559415fa538fd : SUCCESS in 38m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/43d18bc945484742ba33e564379bcf7c : SUCCESS in 14m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/070625bae97b41b7b2ecfc672ffafbfe : SUCCESS in 14m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7bffc588f89b413388a06c53f0ffa389 : SUCCESS in 18m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/12ed4f821f1d45a8b6966fe979272e38 : SUCCESS in 14m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/27fd4c34b519419cb13bc34f99c87b78 : SUCCESS in 21m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e42e37387b774bd795641154a56ba12b : SUCCESS in 17m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f8375f3aa0e148fc9dc71c1480559ecb : SUCCESS in 20m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fbb4585fc71e4361a7529eea71e067db : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7a25735b4ec84555b1e89855f661ed13 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/92cd96563c1043af8d8ff51750aa244d : SUCCESS in 15m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b7808568374c4366b219f177b8556a42 : SUCCESS in 8m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8dff847e489b4986a3e3d3b9a6a292b7 : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/99269e3084914a85ab8ee6745c11e84a : SUCCESS in 9m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5569b27d391544afad83d0911943c420 : SUCCESS in 1h 08m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6311e041c4d54b70b045e5424ac48b1e : SUCCESS in 45m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f0cf10b91bdc4536b0d469fa8f118a68 : SUCCESS in 49m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/838b5883ef1f4d71943b7755d1e84349 : SUCCESS in 22m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1026134cca4d4f33a253f882d9b3c212 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/046652f74c214d24a9ed21310e48a3c3 : SUCCESS in 20m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ef9b449c242402b8c6a9a5c69d3b20e : SUCCESS in 1h 02m 06s","accounts_in_message":[],"_revision_number":10},{"id":"5a136b74ce3c1ac18197bb8bc549d73583cc535b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 19:41:16.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/503d7c8bc32341bf93cb57c98026edf3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e8a184fb60d43a783ed833b50b5c30f : SUCCESS in 1h 24m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27b930f6e5c349c684c0924bac3148e3 : SUCCESS in 1h 15m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f518ffc145934d829d049df85a6c9786 : SUCCESS in 1h 17m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8dce5ee21011461cb40baa68a3c4e00e : SUCCESS in 30m 09s","accounts_in_message":[],"_revision_number":10},{"id":"f78b83311d77c127dbcd3433ead94a549c148c44","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-14 10:15:37.000000000","message":"Patch Set 10:\n\n(10 comments)","accounts_in_message":[],"_revision_number":10},{"id":"fedcd9d064936a925e67b70488e190019c8c1e50","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-04 15:14:29.000000000","message":"Patch Set 10:\n\n(6 comments)","accounts_in_message":[],"_revision_number":10},{"id":"f89e5f9ae6fe143544a34ff88156e4ad9c4f7a10","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-10 23:06:08.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"2bbbc26361bd800a19eb278f523387477d7c4da5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 23:06:41.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"3e60068affb970a988d93485b83555117226c5ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 00:29:24.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/65831f76e93a4e39bc6486285910c522\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/745de963b33642bab6521119db07a9a9 : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/2914ea11a0114e6994dff342a6acc5e5 : SUCCESS in 51m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/32239d0943cc42dda4161eaee6efc28b : SUCCESS in 43m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99dad981f7034a54a5eed91184e525c5 : SUCCESS in 45m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/aadcdbf776bb465dae10e0e5d21ee712 : SUCCESS in 9m 26s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/cb6c3bdf72214c3b915f72b8ea469f90 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/31ebc7d1fc014ac6a060773ca6ac3c5e : SUCCESS in 17m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e7c005eceea047288ca158a7a09dd184 : SUCCESS in 18m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/89766c41d502425ba02b8a9170ddab2b : SUCCESS in 19m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5f9ab706b17e418e9a7010e9113058af : SUCCESS in 17m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/72d03ed6c2094f8f982b1cd68b3acd42 : POST_FAILURE in 33m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/37ddfe3c36d04f1fb7766ec3a08b8629 : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dc38eff7df3f499f857c2d0f4ae3cc18 : SUCCESS in 19m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d6962d36bdf340f5b8d0ea6036875fa7 : SUCCESS in 19m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/17d6802471584912998c6feff75a50ea : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/54e65aff0fff4e69b631d4a4aaf4c223 : SUCCESS in 15m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/29b4cb0ba2374f09877b7eaadede7165 : SUCCESS in 16m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8d9465e523a24ceba1082f2ac31fb689 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e3ac2ba163cc401f8c33c0d3a571539f : SUCCESS in 7m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a416ecd4ea644e578c688a8d6fac5916 : SUCCESS in 58m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/136b7033d70c4a78a8b5dbe61120afef : SUCCESS in 50m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f2a5715377314e03b75f720edf012b25 : SUCCESS in 34m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e076af9ef9f54555b01091567c3393af : SUCCESS in 17m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3393afaabf8e4857bc7a0e697a9f2632 : SUCCESS in 3m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2d718c9494e41e984de558c5417baa8 : SUCCESS in 48m 25s","accounts_in_message":[],"_revision_number":10},{"id":"3364601fdd90d2d7dafab6a938f393d8b5b14262","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-11 18:08:49.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"95ab90a7899a33e3d3bb6e6cb3024031d4c98d1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 18:48:03.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15ffaf462ec244d4a17217caa58713a9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee031614cf764ef3b958a2ef8d0c5be4 : SUCCESS in 14m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f01a88f92e51483ba68a2298bb0a9476 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2bc129c565fa4f7eaf68858755ad58a5 : SUCCESS in 37m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6909ce98e15f40d4abb4d91ad6088552 : SUCCESS in 20m 15s","accounts_in_message":[],"_revision_number":10},{"id":"a5290ac2fba731cb2e573a45ac03d8a2c1ffe3d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 19:17:22.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a70f7740d05412eb6d2cff2d8630050\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae5c5cc570d140e4954c0251ac3bd559 : SUCCESS in 6m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/b6ff8970394c40f28ed6cbf79fe120ba : SUCCESS in 39m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cc9a511f67894862b978f3dee4149236 : SUCCESS in 32m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddf69c9031cc41048c95702080ac9f31 : SUCCESS in 1h 02m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31925574ec764ce0b10c3bac28789b9b : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee09fd3e6786403dbd67c83f985d8769 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5763aa1e948c4c6d943cd73b2e33dd81 : SUCCESS in 19m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/07d14bd60238446a9a1301b38588cecd : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6d9b7a446dff4c8a8ce206a4d4d5d370 : SUCCESS in 4m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a23982a9e9fc483b858d19bf85c04d59 : SUCCESS in 18m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a06248dc893a4b288959a7ee4cc39b3a : SUCCESS in 18m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9c58d417508840da95276b4fd084f9d8 : SUCCESS in 16m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0a3bb111a1074a1fb20b8a4616fa9453 : SUCCESS in 18m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f6542d9aea0e40398e75b129b80c7f4f : SUCCESS in 14m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8979e306e2cd45cab43150e3cb13d5df : SUCCESS in 19m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aa4b9d27e949467384e5782a0fc90c3f : SUCCESS in 18m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6ac59c0e98524b4da7b59ae079e22c22 : SUCCESS in 19m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7ddad47e1d844600bd3a76a9b5330b95 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f9dc524e2515480c9251932569d93d7c : SUCCESS in 16m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0c8b92dcbfaa43f697a42a9a48e897d6 : SUCCESS in 16m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d52223f565d14589859d9fa7e76793f2 : SUCCESS in 5m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3eb6ffd6b6594b01bea8217b8833a443 : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/95e516b275504b7ea832da7307de9d30 : SUCCESS in 8m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8c2c8563bdc466d9058b682514377ea : SUCCESS in 49m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a5f97cbb789e41f08b0b0338e2eaeff2 : SUCCESS in 35m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/86a3d393bfa14715930df3c816fd2b01 : SUCCESS in 36m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c18243f19757487f9986b18bd490a3e0 : SUCCESS in 14m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e2ac77161934243a8e8b15b51ed401c : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7b651c42ad604ffe8525cd064c1c9a5a : SUCCESS in 15m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56f8eb34d6e74038b55c2bfc71f061c3 : SUCCESS in 38m 09s","accounts_in_message":[],"_revision_number":10},{"id":"6e88b3f3f3539ffb9aed727f622e4a4f16a0ea2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 19:17:54.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"4b7af40acdb6c86b757cea46f10faec3bb54960c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 20:26:39.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5b92e2695d54aebafb4d76e8168cb7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f01126f932504ac0b0e5e18c5920bf86 : SUCCESS in 6m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/4e6c8cedebbd47308a3df5257ede3b7c : SUCCESS in 39m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89c8619286eb48759cc4bd7203e6749e : SUCCESS in 33m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ff7cc5e22d0440c9977656296b312d8 : SUCCESS in 1h 03m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8b8f8a139e4e4df2a03e9eaa212d0aa2 : SUCCESS in 10m 47s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e817153ba1e24691bc178fa29c03773c : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c7b6cbe3972e43e58e71cb81c7494d17 : SUCCESS in 20m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d4a70fbba7de41f59e90c20ad1746ccf : SUCCESS in 14m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/266a11c34da14978bcd34e395d07e0e4 : SUCCESS in 19m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4a631f94158e42b9853aabfc029a5e36 : SUCCESS in 18m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2f4d9a60a89e4a7b8f94d75f8e09c096 : SUCCESS in 23m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8cd0aab740e843b4b469e28351e14c98 : SUCCESS in 18m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ea047df00b0e4faea46953bb087dd4e0 : SUCCESS in 18m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a005e8d0f044c0f8712def96e1178cf : SUCCESS in 21m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/138df9f2dd454a95b5502d5ada52df1a : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9e264cbb22c148ff972e8f71876a0c2c : SUCCESS in 16m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eba3d53ec0fc4d1a8ec706ebb706ecaf : SUCCESS in 17m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f5c9e86d3f6148a48fe7b4b4d10e4539 : SUCCESS in 5m 12s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/201acc392d8946babe17a5db15f1582c : SUCCESS in 7m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/03c9d0964a374afe84e48bf13818d79b : SUCCESS in 45m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/79ae78bb236e442d9bf90324fd189df1 : SUCCESS in 32m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f9b826dfee91458b8a9ecea46b0d77ff : SUCCESS in 33m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7045a8580daa455198e13a64ddff2e2e : SUCCESS in 14m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6824404082e4724856fb72f834b88f6 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3070a50202f442bf8c926294f8f3d2e1 : SUCCESS in 15m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d7f4858d3f44a8aa3252b70cec2faf7 : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":10},{"id":"73b1a205f722074c39abdebd1b4dc9761ef3c2be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 20:26:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"3de4fa1867a0264f85979d45788792cf968384a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 20:27:50.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e30dbcd8f55743d38efd67b709be9876\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89028e739bf544a0a2e69af4ea23f08e : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ba08ec762a364e5abdd0082e201d0f02 : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/14695a096e534acaaa1c6b2c664a68cf : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e9abfd76ee5ef94a51b2dbe5a758432cfeeab8e9","revisions":{"3adb7da6e23695522f6952593cdf9ca07500234e":{"kind":"REWORK","_number":1,"created":"2022-05-04 18:35:27.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/840531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/840531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/840531/1"}}},"commit":{"parents":[{"commit":"7d1707db84ffff594bbbc0b9c01ca47890fe9b61","subject":"proxy: don\u0027t error limit ratelimiting backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d1707db84ffff594bbbc0b9c01ca47890fe9b61"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 18:32:10.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 18:32:10.000000000","tz":60},"subject":"backend ratelimit middleware: support separate config file","message":"backend ratelimit middleware: support separate config file\n\nAdd support for a backend_ratelimit_conf_path option in the\n[filter:backend_ratelimit] config. If specified then the middleware\nwill load config options exclusively from that file, otherwise it will\nload config options from the [filter:backend_ratelimit] section.\n\nChange-Id: I6554e37c6ab5b0a260f99b54169cb90ab5718f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3adb7da6e23695522f6952593cdf9ca07500234e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3adb7da6e23695522f6952593cdf9ca07500234e"}]},"branch":"refs/heads/master"},"e1ef3dcdbb6aef97f75c454beacb93498ce508c2":{"kind":"REWORK","_number":2,"created":"2022-05-05 19:49:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/840531/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/840531/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/840531/2"}}},"commit":{"parents":[{"commit":"7d1707db84ffff594bbbc0b9c01ca47890fe9b61","subject":"proxy: don\u0027t error limit ratelimiting backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d1707db84ffff594bbbc0b9c01ca47890fe9b61"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 18:32:10.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-05 19:44:40.000000000","tz":60},"subject":"backend ratelimit: support reloadable config file","message":"backend ratelimit: support reloadable config file\n\nAdd support for a backend_ratelimit_conf_path option in the\n[filter:backend_ratelimit] config. If specified then the middleware\nwill give precedence to config options from that file over config\noptions from the [filter:backend_ratelimit] section.\n\nThe path defaults to /etc/swift/backend-ratelimit.conf.\n\nThe config file is periodically reloaded and any changed options are\napplied.\n\nChange-Id: I6554e37c6ab5b0a260f99b54169cb90ab5718f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1ef3dcdbb6aef97f75c454beacb93498ce508c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1ef3dcdbb6aef97f75c454beacb93498ce508c2"}]},"branch":"refs/heads/master"},"26032c9ff2bd355c376ab53bb07ce2db292c9aed":{"kind":"REWORK","_number":3,"created":"2022-05-06 13:05:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/840531/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/840531/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/840531/3"}}},"commit":{"parents":[{"commit":"7d1707db84ffff594bbbc0b9c01ca47890fe9b61","subject":"proxy: don\u0027t error limit ratelimiting backends","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d1707db84ffff594bbbc0b9c01ca47890fe9b61"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 18:32:10.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-06 12:26:58.000000000","tz":60},"subject":"backend ratelimit: support reloadable config file","message":"backend ratelimit: support reloadable config file\n\nAdd support for a backend_ratelimit_conf_path option in the\n[filter:backend_ratelimit] config. If specified then the middleware\nwill give precedence to config options from that file over config\noptions from the [filter:backend_ratelimit] section.\n\nThe path defaults to /etc/swift/backend-ratelimit.conf.\n\nThe config file is periodically reloaded and any changed options are\napplied.\n\nChange-Id: I6554e37c6ab5b0a260f99b54169cb90ab5718f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26032c9ff2bd355c376ab53bb07ce2db292c9aed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26032c9ff2bd355c376ab53bb07ce2db292c9aed"}]},"branch":"refs/heads/master"},"897e8d1741960aa3c924d72b7e84a7357d22c1d3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-05-19 13:35:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/840531/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/840531/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/840531/4"}}},"commit":{"parents":[{"commit":"f84b8475ab63e85a74f2cefcb51cc47e88f0a8ac","subject":"Add backend rate limiting middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f84b8475ab63e85a74f2cefcb51cc47e88f0a8ac"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 18:32:10.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-19 13:30:57.000000000","tz":60},"subject":"backend ratelimit: support reloadable config file","message":"backend ratelimit: support reloadable config file\n\nAdd support for a backend_ratelimit_conf_path option in the\n[filter:backend_ratelimit] config. If specified then the middleware\nwill give precedence to config options from that file over config\noptions from the [filter:backend_ratelimit] section.\n\nThe path defaults to /etc/swift/backend-ratelimit.conf.\n\nThe config file is periodically reloaded and any changed options are\napplied.\n\nChange-Id: I6554e37c6ab5b0a260f99b54169cb90ab5718f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/897e8d1741960aa3c924d72b7e84a7357d22c1d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/897e8d1741960aa3c924d72b7e84a7357d22c1d3"}]},"branch":"refs/heads/master"},"7730d4806c4dc4bb9a5a630f2f84e05229a6d726":{"kind":"REWORK","_number":5,"created":"2022-05-20 13:40:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/840531/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/840531/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/840531/5"}}},"commit":{"parents":[{"commit":"ccaf49a00cea1990027eb699bff8f10c664b2c9a","subject":"Add backend rate limiting middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccaf49a00cea1990027eb699bff8f10c664b2c9a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 18:32:10.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-20 13:40:10.000000000","tz":60},"subject":"backend ratelimit: support reloadable config file","message":"backend ratelimit: support reloadable config file\n\nAdd support for a backend_ratelimit_conf_path option in the\n[filter:backend_ratelimit] config. If specified then the middleware\nwill give precedence to config options from that file over config\noptions from the [filter:backend_ratelimit] section.\n\nThe path defaults to /etc/swift/backend-ratelimit.conf.\n\nThe config file is periodically reloaded and any changed options are\napplied.\n\nChange-Id: I6554e37c6ab5b0a260f99b54169cb90ab5718f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7730d4806c4dc4bb9a5a630f2f84e05229a6d726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7730d4806c4dc4bb9a5a630f2f84e05229a6d726"}]},"branch":"refs/heads/master"},"ae44a6dbec1eda2ec8b31b3ae496fe585c3bc64f":{"kind":"REWORK","_number":6,"created":"2023-03-16 14:10:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/840531/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/840531/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/840531/6"}}},"commit":{"parents":[{"commit":"ccaf49a00cea1990027eb699bff8f10c664b2c9a","subject":"Add backend rate limiting middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccaf49a00cea1990027eb699bff8f10c664b2c9a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 18:32:10.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-03-16 14:09:59.000000000","tz":0},"subject":"backend ratelimit: support reloadable config file","message":"backend ratelimit: support reloadable config file\n\nAdd support for a backend_ratelimit_conf_path option in the\n[filter:backend_ratelimit] config. If specified then the middleware\nwill give precedence to config options from that file over config\noptions from the [filter:backend_ratelimit] section.\n\nThe path defaults to /etc/swift/backend-ratelimit.conf.\n\nThe config file is periodically reloaded and any changed options are\napplied.\n\nChange-Id: I6554e37c6ab5b0a260f99b54169cb90ab5718f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae44a6dbec1eda2ec8b31b3ae496fe585c3bc64f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae44a6dbec1eda2ec8b31b3ae496fe585c3bc64f"}]},"branch":"refs/heads/master"},"08ab90e0d1dabd6c738ee80abf8cb54c86a2f13b":{"kind":"REWORK","_number":7,"created":"2023-05-02 00:43:42.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/31/840531/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/840531/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/840531/7"}}},"commit":{"parents":[{"commit":"f99a6e5762896c7789d168bc49d8cdcb47903264","subject":"Merge \"Log (Watchdog\u0027s) Timeouts with duration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f99a6e5762896c7789d168bc49d8cdcb47903264"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 18:32:10.000000000","tz":60},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2023-05-02 00:42:43.000000000","tz":600},"subject":"backend ratelimit: support reloadable config file","message":"backend ratelimit: support reloadable config file\n\nAdd support for a backend_ratelimit_conf_path option in the\n[filter:backend_ratelimit] config. If specified then the middleware\nwill give precedence to config options from that file over config\noptions from the [filter:backend_ratelimit] section.\n\nThe path defaults to /etc/swift/backend-ratelimit.conf.\n\nThe config file is periodically reloaded and any changed options are\napplied.\n\nChange-Id: I6554e37c6ab5b0a260f99b54169cb90ab5718f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08ab90e0d1dabd6c738ee80abf8cb54c86a2f13b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08ab90e0d1dabd6c738ee80abf8cb54c86a2f13b"}]},"branch":"refs/heads/master"},"9bdaaa1d8ae2adf1ad962efb441451363feb4d8d":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-02-22 18:26:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/840531/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/840531/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/840531/8"}}},"commit":{"parents":[{"commit":"50336c509877e7c325823498cd67bf6cf4f4bef3","subject":"Merge \"test: all primary error limit is error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50336c509877e7c325823498cd67bf6cf4f4bef3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 18:32:10.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-22 18:24:30.000000000","tz":0},"subject":"backend ratelimit: support reloadable config file","message":"backend ratelimit: support reloadable config file\n\nAdd support for a backend_ratelimit_conf_path option in the\n[filter:backend_ratelimit] config. If specified then the middleware\nwill give precedence to config options from that file over config\noptions from the [filter:backend_ratelimit] section.\n\nThe path defaults to /etc/swift/backend-ratelimit.conf.\n\nThe config file is periodically reloaded and any changed options are\napplied.\n\nChange-Id: I6554e37c6ab5b0a260f99b54169cb90ab5718f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bdaaa1d8ae2adf1ad962efb441451363feb4d8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bdaaa1d8ae2adf1ad962efb441451363feb4d8d"}]},"branch":"refs/heads/master"},"4788fba71bc564649cf23a444b69f1d80f29d555":{"kind":"REWORK","_number":9,"created":"2024-02-23 16:42:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/840531/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/840531/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/840531/9"}}},"commit":{"parents":[{"commit":"50336c509877e7c325823498cd67bf6cf4f4bef3","subject":"Merge \"test: all primary error limit is error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50336c509877e7c325823498cd67bf6cf4f4bef3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 18:32:10.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-23 16:38:38.000000000","tz":0},"subject":"backend ratelimit: support reloadable config file","message":"backend ratelimit: support reloadable config file\n\nAdd support for a backend_ratelimit_conf_path option in the\n[filter:backend_ratelimit] config. If specified then the middleware\nwill give precedence to config options from that file over config\noptions from the [filter:backend_ratelimit] section.\n\nThe path defaults to /etc/swift/backend-ratelimit.conf.\n\nThe config file is periodically reloaded and any changed options are\napplied.\n\nChange-Id: I6554e37c6ab5b0a260f99b54169cb90ab5718f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4788fba71bc564649cf23a444b69f1d80f29d555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4788fba71bc564649cf23a444b69f1d80f29d555"}]},"branch":"refs/heads/master"},"e9abfd76ee5ef94a51b2dbe5a758432cfeeab8e9":{"kind":"REWORK","_number":10,"created":"2024-03-11 18:14:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/840531/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/840531/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/840531/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/840531/10"}}},"commit":{"parents":[{"commit":"50336c509877e7c325823498cd67bf6cf4f4bef3","subject":"Merge \"test: all primary error limit is error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50336c509877e7c325823498cd67bf6cf4f4bef3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 18:32:10.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-11 18:10:24.000000000","tz":0},"subject":"backend ratelimit: support reloadable config file","message":"backend ratelimit: support reloadable config file\n\nAdd support for a backend_ratelimit_conf_path option in the\n[filter:backend_ratelimit] config. If specified then the middleware\nwill give precedence to config options from that file over config\noptions from the [filter:backend_ratelimit] section.\n\nThe path defaults to /etc/swift/backend-ratelimit.conf.\n\nThe config file is periodically reloaded and any changed options are\napplied. The middleware will log a warning the first time it fails to\nload a config file that had previously been successfully loaded. The\nmiddleware also logs at info level when it first successfully loads a\nconfig file that had previously failed to be loaded. Otherwise, the\nmiddleware will log when a config file is loaded that results in the\nconfig being changed.\n\nChange-Id: I6554e37c6ab5b0a260f99b54169cb90ab5718f81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9abfd76ee5ef94a51b2dbe5a758432cfeeab8e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9abfd76ee5ef94a51b2dbe5a758432cfeeab8e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
