)]}'
{"id":"openstack%2Fswift~840542","triplet_id":"openstack%2Fswift~master~I2936f799b6112155ff01dcd8e1f985849a1af178","project":"openstack/swift","branch":"master","topic":"p-ratelimit","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-05-13 16:11:19.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-05-13 14:43:16.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"I2936f799b6112155ff01dcd8e1f985849a1af178","subject":"backend ratelimit: support per-method rate limits","status":"MERGED","created":"2022-05-04 21:18:56.000000000","updated":"2024-05-13 16:12:40.000000000","submitted":"2024-05-13 16:11:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":63,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"840542-p-ratelimit","meta_rev_id":"0b36a37b5c18c19cddb62fa3bb879b98f2fa5bee","_number":840542,"virtual_id_number":840542,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-13 16:11:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-05-13 14:43:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-10 23:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-05-13 14:43:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-04 22:31:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-04 22:40:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-08 23:30:15.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2024-05-10 23:52:45.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-05-13 14:43:16.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"e44141c5119b86dcc22014d158079a1a74334020","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-04 21:18:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21b2088b3f3b999721b58914205af867f87c9f18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 22:31:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0da49939559347ec8f7517598422764e : SUCCESS in 55m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/58fe5304854645f49ec58d9e955b9fc7 : SUCCESS in 1h 09m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3dd0a2bf0ca0499fad29053ed449ea09 : SUCCESS in 1h 00m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d01753c01374c9285240f373929d1ec : SUCCESS in 59m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f41cd0853a2fb2a44bf6dff3cb2c1628fe2b53f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 22:40:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee5e3a6291f84abdab41288c3c05fc89 : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/b2c7c4105d4346af95083dd9ce9c3b49 : SUCCESS in 1h 11m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/862f19d59c9c46179567adf865819f2c : SUCCESS in 1h 05m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74a5384786b6452083e8d025ac727521 : SUCCESS in 50m 19s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/71a0715747e7460aa508e363a9588d4e : SUCCESS in 30m 05s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/16c26cf881464f7db84034c4f5a43ffe : SUCCESS in 24m 22s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ebd6e2629d36439e8b58bd69b2c18c48 : SUCCESS in 19m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fd961d2ba48948c192e6afe23f18466a : SUCCESS in 7m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8277757c4755454fbbcc1f907ff77cf4 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/da49ca96ba2d4a39982e0b406d60f5c3 : SUCCESS in 15m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f3b6df67607d403b9478db7135f829ba : SUCCESS in 14m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4fdb4e2b02b6468aa55125ecbd47faff : SUCCESS in 15m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1d249434570944e99f282feaf3a7adac : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7d2afcce196c452aacc9991549165389 : SUCCESS in 25m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/565be55fa18e437382cb2bc4e2f0e194 : SUCCESS in 16m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/50e7af25f0474f3480a4ca28da935c21 : SUCCESS in 14m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0e125ead656e4509a4a7b17f0f483a58 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/daed775b0e8f4bc8b9e069b71ee83b27 : SUCCESS in 14m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a4bbf3655ac34cc3b7bd70aa11b794f4 : SUCCESS in 6m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/06786426a34948399850ecdf9608c606 : SUCCESS in 16m 41s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6077c330da9145d18f66f26619486b0d : SUCCESS in 48m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/36a295517e104aa5a600763f0afce1c3 : SUCCESS in 51m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/647305de61424f56a28bc09df7d1bf0a : SUCCESS in 46m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/35a68d8753a547fd8a1ef589b9ee7ce6 : SUCCESS in 39m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/082f0d2574d0459088e0636ea5305d92 : SUCCESS in 16m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b416b94c8f54cf8a1d4c09ced892f9d : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d2d83f45d49043d18593fbb59a7e9d27 : SUCCESS in 20m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b83b54bfc15466d9192eb11534dc961 : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c5bbe9c27e8af459b6b2b9b2d5d2e850aaa42f58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-05 19:49:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8708ec5b2e1093c5969a77a21e44960707aa513e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 21:01:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f62480f5bd4f4e35b08409a9f80c8633 : SUCCESS in 12m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/7eeb59b75e474c019725108251555ed8 : SUCCESS in 53m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/66bfabab67a24863be82a81979a0d3b4 : SUCCESS in 58m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c7d1d5ae97247a0be4d480d8def2165 : SUCCESS in 41m 58s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/116be57795714c848a2c2a5e9d8d8a37 : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ed8844d7a7204aa48fe80cccfb03930d : SUCCESS in 23m 17s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/04e1c3b514cf422dae46f444d0e23d3a : SUCCESS in 19m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dd72c55a145842c5984534d170a49668 : SUCCESS in 8m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d59f55ff80e643ea81a475d1f30f816c : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d81c3b71c5eb4fbaa53ab400397e5085 : FAILURE in 16m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/92fbe06a22204294b47937e40ce45d59 : FAILURE in 20m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0e9ae77c0dfd41d996032ffc29744591 : SUCCESS in 14m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cc3063651adc4d6d991634360d70079c : SUCCESS in 20m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e62fae278d3f44eabc6119db21b23d40 : SUCCESS in 18m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a7913e1c64d6418080993eeb63e0fdc5 : SUCCESS in 21m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/92052e618be3495f91bea9ad70e3a25f : SUCCESS in 18m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6e46c9fd0c7048369bc7e7d4b7246661 : SUCCESS in 17m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c72b5ad03b694ab0a4a992510dc85c3d : SUCCESS in 17m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0f7fefd488874d348b1b0425c5b0a340 : SUCCESS in 9m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b7d8f85a87fc4f3287edfb1682a92833 : SUCCESS in 16m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/220218192adf440ba9be21be2f80199b : SUCCESS in 50m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0221fbe146064aac9c258b15351f805f : SUCCESS in 51m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ea19233097424c5a967442615c43f647 : SUCCESS in 40m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/70eda6b4eecd4708b3905e3a25cde3a8 : SUCCESS in 36m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd2222d606d84bbda77db6afbd12d48c : FAILURE in 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6421c0148960492eb30b54c9c947b971 : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5a96fe19e98744dc823f2b340fb437e5 : SUCCESS in 23m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81d660e1882c4694891f7d0cc4afe1bd : SUCCESS in 48m 28s","accounts_in_message":[],"_revision_number":2},{"id":"13cea3ba9ce3c8f8d7ee87c367a726d935d5ef99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 21:23:21.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa5482c8b87244baa48e6264a4fc09cc : TIMED_OUT in 1h 31m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3af987d6925e4e6fb8f463fb2e199940 : TIMED_OUT in 1h 31m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f13a97ba7204a54aba06e318a675c1a : SUCCESS in 35m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b09e0743d2264134ad6d824d831e51cb : SUCCESS in 37m 56s","accounts_in_message":[],"_revision_number":2},{"id":"0ad0d291af494340e36110f3b9843769db77626e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-06 13:05:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"17f1972e4a3899922926f25369a8b4cb26bb45ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 14:14:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cdd7425272d4ebfb3d64e812754bbce : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/2f37b6f51a1f4ca781f7c7a07dcc9a6d : SUCCESS in 1h 03m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4477140f46634bb094809e7b35211046 : SUCCESS in 54m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/363f4683827d466aa08d421d7af068b0 : SUCCESS in 47m 48s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/decc3185fe5a49699092852332888986 : SUCCESS in 22m 01s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b25e1cc5e95646f5b0e404f9de87d4f3 : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8c431849dcc440e18c736d0c0dbf0d4a : SUCCESS in 24m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a3fe2d15e2a14730a38bb32535e98131 : SUCCESS in 8m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9edeb2fe00ee4a449ce6264bf64b822b : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a6a3b85bd6f34d6cb10229e7093af85e : SUCCESS in 18m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3de65cde35994209bd8f6ac15f8d2bd1 : SUCCESS in 15m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8897d516abb8445e8403203b42f45710 : SUCCESS in 14m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aa54ab1ac9c148bca4e5b95fdf4b89b7 : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7f83995c91dc475eb84db94936ec3b92 : SUCCESS in 20m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1559768f408a415bb32bd7848b581929 : SUCCESS in 18m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aaa021d4cb1d47cc99a595b50ec911f3 : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/68eba07bf62749aab740af71faa5b9a6 : SUCCESS in 18m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2473e68d5d004b88ab56f8efb2b28b74 : SUCCESS in 15m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ac8c3dd5124142ccad93947d68b6e1e2 : SUCCESS in 6m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6ee8269f1b1443bb97b9b2d1e24c2442 : SUCCESS in 16m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7e3599a40a3f4cf38e5d57f5beca5cea : SUCCESS in 59m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7078db2b39ae48bf88beaca5a8966c16 : SUCCESS in 55m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9c6124e15cc9447ab075f9956920629a : SUCCESS in 39m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dbd06a2a7d60405aa47166377ac67947 : SUCCESS in 43m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30edc872dc61403cbc52fb72c77d1e02 : SUCCESS in 17m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b71bb9ab6b0f4ec7918529504d54159a : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0b73d68b9b7d4585aad8c963279d244d : SUCCESS in 27m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65805e167a0e4218827298785e01109b : SUCCESS in 45m 05s","accounts_in_message":[],"_revision_number":3},{"id":"f7ba967fff7c6bbbe5c0e8ba26f653954957aec1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 15:04:19.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77864906c1b84b889ec8d70a086a34b9 : TIMED_OUT in 1h 31m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f127db3e88ee474c94b4d8a605ef778f : SUCCESS in 44m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b378b57705f64dd3a2be908250e98698 : SUCCESS in 56m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/29a3491e89fc4d2e808506e359a2a7f0 : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":3},{"id":"81e76db17a5c2078cafc4f797ce52ee9911f7cc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-19 13:35:45.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"ffe46e4d748a7360719fc38b252abb632a68455d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 14:50:18.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e541ef97a3b47439652f50ea686f6ef : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/63815eef8bb348459e6f376fb266e64e : RETRY_LIMIT in 3m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dfde72c4e07645878649bcf5eaee73c6 : SUCCESS in 48m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2e6c975ba888479fa6171878489b5c66 : SUCCESS in 47m 37s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0176b11e4f6e4d5da78907f8ca42f31e : SUCCESS in 21m 50s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7ed045c2320f494a965559252fef8eae : RETRY_LIMIT in 2m 15s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3aad2bc8754d4411ba170705309f103b : SUCCESS in 19m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2ee487059e77483c9b01faa3af90a1b7 : SUCCESS in 7m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1a4b925a27604e36874ed4be1bd09a7e : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1413468146674b62a0efc4512cb6be47 : SUCCESS in 15m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ba1b1ee5e66d4e959d75f9f6019e16eb : SUCCESS in 15m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/45ecb0cb057843268f340596252bdd14 : SUCCESS in 16m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/22f108d0ec034d9082f2f9c68b8f8b41 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/018c4bf02963486abd2194cfda3a1537 : SUCCESS in 17m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f492306c02b44c088fdfb2be848e69aa : SUCCESS in 19m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/95b6ad2e39494c8883ca349d94f4eacd : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0444266a28414f1f9bbd1b17817b27f4 : SUCCESS in 30m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f0f7125fbf69470b81ac1b11ccf8a936 : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c2a52af5e39f4199a56262a409568081 : SUCCESS in 5m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/69f62e3c401243e2b8f28a8d00b3dd1d : FAILURE in 14m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eb2b1d12fc6b4252a1a945a0d6c3463d : SUCCESS in 58m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/97576467d4864b02836fd4fcaeb8ea78 : SUCCESS in 57m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/890bcc9dec2f439ca0cf55ed50d45f0c : SUCCESS in 51m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1168aecd76b64282b22d770142a19a7f : SUCCESS in 40m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de9227ec6c97400f9dfc2e61ef14271b : SUCCESS in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdfd587ce60241dcb0bd4a4baa6d505f : SUCCESS in 7m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/951231ac44064d00abc7973380714274 : SUCCESS in 25m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e89aa9808e84a89a4983f73fc622f4c : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":4},{"id":"4e5b9f7beec21283111bc70812d016f7dc61882b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 15:10:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d43001a3951b4ba5a3e1fc59785ae6dd : SUCCESS in 1h 03m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dcaeefabea72499c8e0fd391d0b0a738 : SUCCESS in 43m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca64856b0eb649ea941af5fa1d99eef3 : SUCCESS in 1h 01m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4775cb96016c4a0b9c75f70b8d4a8f16 : SUCCESS in 46m 01s","accounts_in_message":[],"_revision_number":4},{"id":"984367e3ef3b5ce3caf8887a299b86d0677b9c1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-22 18:26:23.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"1ae10fe900e0330ba403b85a9afd39814b89566a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 19:10:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/069f162d91564b5dbbd5321addddc8d5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7467e6373f8f4da7bbe3a885f4115aea : SUCCESS in 39m 37s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e9de58fab7f74fb7aae9c798187309f8 : SUCCESS in 38m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/780fb651b36e4b66b12e714a843fdf33 : SUCCESS in 41m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/aedf34c020df483797d7d2a64ade0246 : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":5},{"id":"120061a3f7cdfc969d877c18cda9c9e620e5cd32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 19:47:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cf5f6ea9c0446028eddd7222779e4d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a9d6a82bec14529956945d7e71c4e0d : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/e576ba75783746dc851ed33becb58909 : SUCCESS in 52m 48s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/879efc7565234bd79fadb620b2398cac : SUCCESS in 56m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5b01e6f903d94813b3817393e9aeaea6 : SUCCESS in 46m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b8b397a4d8243958507cc6c8c726e9e : SUCCESS in 59m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e4e898d2bfd414c989fd27718315df3 : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb0999c249f241028c68e5052a041227 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/72960f23e3984f37be9d153a6455a89c : SUCCESS in 20m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b9b5221223a4275835c25f7fec7e7ae : SUCCESS in 10m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7375d4cae93e4381ab309affe0096fc7 : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c4f8eaabe8604024869778aa94153f1f : SUCCESS in 19m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5fb6ef7ad8f14b08bdb2b66ff85c64ec : SUCCESS in 17m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bf902e137dd648c7aed32747631f9b5e : SUCCESS in 19m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/40bb8d26d5e942918fb0af2fbe579646 : SUCCESS in 14m 48s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2c5621b0ed914543b76e33bd766b99bc : SUCCESS in 24m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71f64980c68c4640a7d0e6a407711d34 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/76fc12c842494f45878e12603d3cd957 : SUCCESS in 18m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/031dddb6df3e40aa96f57e156029f421 : SUCCESS in 17m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/06340609fd094e9c9133d35107f74de0 : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/85591d1d829849359cb3a4d28c34c5d3 : SUCCESS in 14m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/71b85534146c4b65824a780ccca2df6f : SUCCESS in 15m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ad5c3bf107ea432b97fbb41fe929e0a7 : SUCCESS in 8m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/19a55900111e4b19b40c2aaae1101974 : SUCCESS in 15m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/849728d06df842af883c0187c0645c32 : SUCCESS in 9m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bd4d68e9246d45eea8c6c12b854d3565 : SUCCESS in 1h 00m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/39ffc58346434d41854903dbde8a9b52 : SUCCESS in 1h 10m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c23286dd52424a5385c06fbdd5c595f1 : SUCCESS in 52m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bca679b0361343aebd5a5752a5ee190f : SUCCESS in 48m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82655e288b00433d9adbc69fd9ca25dc : SUCCESS in 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/339404a79d194b6288457fe2b0a42de5 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/427301caad074274b51456656b59d767 : SUCCESS in 19m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebf5d6f3e75e4f98aded4a9dfe0705af : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":5},{"id":"fe31e14c935ba47e4afbbab932ebf998c5d501e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-02-23 16:43:34.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"b90b80b3674a1044d96684d68dd7222f8ee35bbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 17:25:20.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e86c70e46e7343cb891cf61ab8170f32\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ceb7e7e61b974962b5032db2386ad40d : SUCCESS in 20m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d98767b779142788ef5bd7545c9a7f4 : SUCCESS in 21m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1cfea0c419754789854b7e0f68b95d61 : SUCCESS in 24m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae30113e14c741c68dc75f8ced0a3cb6 : SUCCESS in 24m 46s","accounts_in_message":[],"_revision_number":6},{"id":"575e298fa4b87cae95eeaf3d0728d6c0ca530929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 18:19:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd63582f70ad4df2a6a89f63402adcd3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af36d92ca25d449f9e890d15c6b43f18 : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/8fe7d5ea6507449a9408a14994f7bad9 : SUCCESS in 1h 17m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/236d489266184131805536823683ad55 : SUCCESS in 1h 00m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53084851acdd43649511b41b33817c9c : SUCCESS in 46m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95b888c4686241aa9a190ef5d12c0794 : SUCCESS in 1h 14m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b9f2f7b71184aaf864e6faa88543d36 : SUCCESS in 30m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf8bac35bb8f40ae863e1d1515aedcb1 : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34e6d3967e9a47659660019ce323e498 : SUCCESS in 19m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1b6469ea15d4eeaaf8507f9a59404ab : SUCCESS in 10m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/df4246d11eb44e62a2303d6e69658466 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae71a1dcda44470b9252c23a2b7173c8 : SUCCESS in 20m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a1a8c6de7b5455b9e3c679c4a97245c : SUCCESS in 30m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/902f4e21388244a7850be8ebd514d0a3 : SUCCESS in 17m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e9b2c60b6ff64b6a9aaf210738595362 : SUCCESS in 19m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5dbd7688577b4949a871c0e7695749c8 : SUCCESS in 20m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e03e1044a3ba42f79b3aae7df387db16 : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/70826e374ba34e2099aa8f86011b978e : SUCCESS in 19m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/588dcd913ae94a34a3494d35e6810dc1 : SUCCESS in 18m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/70f4dcbebe114ce4a6fa89cceefb9008 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ab675d041398457fa21a0bba37098bea : SUCCESS in 16m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/55cdbdbd258f4baca649027fb999c1e0 : SUCCESS in 15m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d8c56e7c5bb74b9690a85152555e6773 : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7e9543f66b074c0fa4d1ac624f88534b : SUCCESS in 16m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4afd3cc57b194ce1a0fcccb0cb595425 : SUCCESS in 7m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/02bbd06e465c48c681bd7c839c36371b : SUCCESS in 51m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/021bbd79a6df4962a375eed26940028d : SUCCESS in 1h 05m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/082b366f91c044349328659422c807fc : SUCCESS in 49m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7eeec70e33b34f10bc8c8b29e0eea38c : SUCCESS in 59m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cad47321e28c47b9a28121d7f49892d5 : SUCCESS in 22m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbf92bc748794665b705aaf49e7ffd99 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/809e78a04a004039a40abb714f6e2789 : SUCCESS in 19m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea4e926f3fec4da5a06c8e2790198f03 : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":6},{"id":"3623036dfbbfb7c29ab87b1eeb0534b59bbf8dab","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-03-08 23:30:15.000000000","message":"Patch Set 6:\n\n(24 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ad44c9f84a2afb9ba6337aaebd41270c859439f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-11 11:36:53.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"34214f378bc15b2b7af22290bf52af6aea8884fc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-11 11:36:58.000000000","message":"Patch Set 6:\n\n(20 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7b166c4d7cdea07410b8ca50f484f9900c825cec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 13:00:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1dce2617e944c83bd8c925514feede1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f3c560395b842fca8305de86849eda2 : SUCCESS in 9m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/30330b6ea92246dba2574ac797f2ced2 : SUCCESS in 1h 10m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/256f13012dd145c9a4b51f893df55810 : SUCCESS in 59m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2621ebbe34754879b1fe0d28d2844a28 : SUCCESS in 52m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/659276bf7ef84c77befcd94aa2928556 : SUCCESS in 59m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38ada9aa5d80451b91d29bddb19aea08 : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2be6028edebe4ace9998d057b3896bbb : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b9d9ebf2371493eb54fb74201ccee85 : SUCCESS in 29m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4a2302b410ca49aa86469ccd21726a27 : SUCCESS in 10m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e9fe2d0693934b5a95c6837e3e836a27 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/04ac01caf68a4329961682404b6b33ba : SUCCESS in 18m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a06f0027354f43c186c111da33c0e379 : SUCCESS in 16m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cb98c7e0d9974679a33c2505091caf3c : SUCCESS in 16m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/183c0fad37d742f1a1e340f1939d22bb : SUCCESS in 16m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/293b3af3469346dfb5ebffaf75e46d04 : SUCCESS in 17m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/461f6714db9d423ab75d564a54a2ea96 : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8f25a4cd4c884508888663cc1e2269f8 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eb1ec16847c04784a545c43fb2e59493 : SUCCESS in 17m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0e8a143ba96e490db9313b6849bd55a9 : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/875631a2dae243cd85c86fa4f569c221 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8e0ab906d49f4e1b917ac7a13874fcc1 : SUCCESS in 15m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b67c4b6523694a89884151a7f72f2fcf : SUCCESS in 7m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df181a542e06455b860f3876fa3afc3f : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a7e2e98c86cd4ef6922f28dbf04e71d1 : SUCCESS in 6m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e94ba5a727894ab095bb130486bbfac4 : SUCCESS in 1h 06m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0fe9eba785bb497aa35619cde9a92af8 : SUCCESS in 1h 00m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/10a2331c1d4c458593b6a79378c8e37a : SUCCESS in 1h 16m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d54930486034026857d704b04c10eab : SUCCESS in 28m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba353983e28c44cf9031ba3408d7d375 : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e9250ede40594fb196f277fa98296bf4 : SUCCESS in 18m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ff263c67fe84542903b664109ce8dde : SUCCESS in 46m 17s","accounts_in_message":[],"_revision_number":7},{"id":"e0adb1dc139396f513cf91f757e5a9b6ebf2f0fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-11 13:31:22.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76dd44eaef6c4928bff6e428405aa65b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6416dd0bfc5b440e983255703dd8d835 : SUCCESS in 36m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/233b1a4b35794348af09caba9efc630f : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c33d0c03741c4fc9bb85f4863ebece5a : SUCCESS in 47m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c63f2f6a7ce04d63a27b3f4c8745a043 : SUCCESS in 47m 42s","accounts_in_message":[],"_revision_number":7},{"id":"077cd62c3dfe1ba5b0dabd8e3efae8d84747300b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-12 10:42:43.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"a0d610ca7e939f8c154e2ceaa050f764c56b2670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 12:12:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d1475afc2a24863b317d52154ea9aa7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a5a0b8fa7e24b11b2b5f36211391878 : SUCCESS in 7m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/168805602b2a4f63b787352bfe5a918b : SUCCESS in 45m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/13b21882d6cd4f088fc2e06927fafd41 : SUCCESS in 56m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7e01613186de42bda68fae8d8be0e759 : SUCCESS in 57m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b01acb4ae79e4bb789fab18266252cbe : SUCCESS in 46m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/278bbffed9384e01ab21a214d19226cb : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81ed92ff93864b0baf43073f42adc51d : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2c303ee5041418ba3e914f97bc50d12 : SUCCESS in 18m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c8121c67c03c49ac86d931a86c6d4ab3 : SUCCESS in 9m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/51d08f3a15c84f03b8417e573856e482 : SUCCESS in 6m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a124031ce577457693383d7c5320777f : FAILURE in 19m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/190537653b70430eb0e30152ceaa2d94 : FAILURE in 20m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b6340527203b47fb9734d2fb4eeaabbc : FAILURE in 18m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c3f6740891264471a5954dffb6c8d3e4 : FAILURE in 17m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f5acfba8bdba4b9585b897121888c8e1 : FAILURE in 18m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c5c05dcf4c124a8fb7e8cc166591e120 : SUCCESS in 22m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/87e96f573f7f423aa200c8cc20538c3b : SUCCESS in 21m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/60e53e8cd85c482681373e1db5622048 : SUCCESS in 19m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f9687cbff2b54af5b438fccfe6a1cc77 : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1e21d9f94f344b6b8ffb1b1f3e851955 : SUCCESS in 16m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bd6a2f84946f45b3812183b948f48978 : SUCCESS in 24m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/462f12429487478ba264db286df3f455 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4e6ad91972ea472a8b655b43b8b16d08 : SUCCESS in 15m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cf46cfd24a2946289d2bb2d928e28113 : SUCCESS in 8m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/67bd11bdeb074ee9a05dd7bc37ca6738 : SUCCESS in 1h 05m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/293c71cd47b246fca4200fc96cc3c22c : SUCCESS in 54m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/83f522e58e414a8fac5b0e0ce339f3c5 : SUCCESS in 51m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d99fb30499144ef684c23f27b893b319 : FAILURE in 17m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe5736b22a524ab5b8cb2eb9645f2950 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e585df1a315342289b432812c8e6f200 : SUCCESS in 17m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64f7a730beed4eb5a568fa0b8281bffa : SUCCESS in 1h 02m 59s","accounts_in_message":[],"_revision_number":8},{"id":"1e0dfe4c9954bc653e06f941f59602f1cfbbccb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 12:16:36.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/508590c383a54c38b005dc4a879ab731\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d351a6cecab4017bc3507f37fe69779 : FAILURE in 56m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a6e4f2e13f940ef9d67a5320eef2bd2 : FAILURE in 29m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/14dca8c200224e83afe591c3a5e588d9 : SUCCESS in 33m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a226a1061a2a43d1960b79b98e0f3b3b : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":8},{"id":"b0607452a9bbc30b0df43e4b439c044ab47bb34b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-03-12 15:55:07.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"ce36c77b6fb5ddea841c7d36999debc168011180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:14:46.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aacb880d051141f6aa4a824fac462726\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5fede9df7d449deb2a792604d11d0cf : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/8eadcf00d4ad4449b42b6b9cbb792c87 : SUCCESS in 46m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/02426e21e33f4e6eb9c6449d1f23e8ae : SUCCESS in 55m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/23f4976a73c74aeda938f2bb03f5e7ef : SUCCESS in 51m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1bb6012bec9b44edb02dbed06ee1fd1d : SUCCESS in 45m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f752fb866c2e4947b02adbf66285eff5 : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c60e63bea5894ad784e67f78928591eb : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79b3e61c59e643bbb455411a904c3ca0 : SUCCESS in 19m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/12ef91870c2949fa9e49b8ad63012115 : SUCCESS in 9m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a62afe8cff9340009fb40452f6f9b13d : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6b0c0bf6cb3a467c8ff1d6147bff789e : SUCCESS in 17m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/998fcf38f3d241d695ee6d0a8879e149 : SUCCESS in 19m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3c6014b30e2a41dea3f5c030628e49e4 : SUCCESS in 17m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/023bbb58d97c48789108b6dd30eb7ea9 : SUCCESS in 19m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e242e5b770804350a712d3ffa0732a29 : SUCCESS in 16m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b7e666ca50184b0d8ce2bbb58f3a0e2d : SUCCESS in 18m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9cb121965b014cbc8d9861089c549074 : SUCCESS in 16m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e0feae8b7d9a4f109b1afbeba25d8ed7 : SUCCESS in 18m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c5d51902c64f4957a70ea54f0708e393 : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fce79b3ea40c4eb2a7972b1bc21bd66b : SUCCESS in 16m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f247c4efe2ed48a790db2f51f0c2021c : SUCCESS in 14m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/977a36d2a8e148e8bb53f1e64b22367e : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bfaacf5c61304cf1a4669d9096bfc2fb : SUCCESS in 16m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/218814fc658546ccaec99b117829a845 : SUCCESS in 6m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/48b9d9de330c4331a3b880e459777572 : SUCCESS in 1h 01m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08ddee3459e948a992f531a4834e0646 : SUCCESS in 47m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9c14414afce8468d84a7553b8b64f958 : SUCCESS in 48m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3bfae769e894500a5a1fea2119dab73 : SUCCESS in 21m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70a7036645c14c9396c08f2a64405237 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a476228b8204cf28ba0e7d7e8a7f41d : SUCCESS in 24m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5021d9b9a1d45dca0c34ecfa099c67f : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":9},{"id":"8e22ba03daa5b2e95b9232d1fab7ae4548ba78c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 17:27:51.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/544662b57764453bbc278370e3b1aff0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5fd36c1bb8244e4a9f5c9f472b8f62a : SUCCESS in 44m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/078fcb9a3e5841c2b5bd06a9b6ed5f74 : SUCCESS in 1h 15m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b626ada6054d4d638cddc681680ce87b : TIMED_OUT in 1h 30m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/821c306f5ad84385bb50d040bfb07828 : SUCCESS in 23m 56s","accounts_in_message":[],"_revision_number":9},{"id":"544f69c8aba709b411e9c39107d54b6c3ad1459e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-10 23:52:45.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7d10b986ea3778bdc3d731434f003286e174a5ac","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-13 08:20:47.000000000","message":"Patch Set 9:\n\n(9 comments)","accounts_in_message":[],"_revision_number":9},{"id":"5ece86edf0b1db1bd0cfa924a777664b6c18ba0e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-13 14:43:16.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(9 comments)","accounts_in_message":[],"_revision_number":9},{"id":"94f35f4b11a32520b8c45c280cc37899852a3975","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 14:43:50.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"2f74a9a53291d96c2912e870c3cfdcb7d6628c78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 16:11:19.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a18bcfdc45fc442bab47bc3fcae184e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47c7736d5bea465292e969ba82b2cab7 : SUCCESS in 7m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/be9385c055bd42bfb0f131222bfe44e7 : SUCCESS in 51m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ac9d8e41979d4e538e6a7fc2b30aa867 : SUCCESS in 54m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5381f969fb34f4e98c79fbe07514ee5 : SUCCESS in 1h 05m 33s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bd3d8eb55ad243e3948a73b725464476 : SUCCESS in 11m 35s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c0c1f3593493407387908bc0b68db4bc : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9bbfec91062546668be3fac802cfd323 : SUCCESS in 18m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/44d25fa6c45b4340ae8f7686a03fe12a : SUCCESS in 18m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/70b12fdbc75c4c92b04a01bdafc83339 : SUCCESS in 14m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/03378cb0464547e39102d00e419c867d : SUCCESS in 19m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cd228ceb82d341f08d92cdd7e952a2a3 : SUCCESS in 16m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e65ecec11c1a40208392db873437028c : SUCCESS in 23m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/783fe15cdb8a4e639f17397bbf70c33a : SUCCESS in 22m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6741fec5aa234089adc7d17c0d517594 : SUCCESS in 19m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/23c095feb6e249ebba77167a1aa9209e : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4366f8549eda4498962694adb0525a36 : SUCCESS in 19m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b82b2977b49d4738966556d541054e88 : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aea157c9ad1f4a139d01bce477a4bf1b : SUCCESS in 5m 51s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/abc4f149abb645639da2b76f99a77ed1 : SUCCESS in 7m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1373aaa800c4bd58b63d381343628d2 : SUCCESS in 56m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a5a11e5b46e4a3f89ca62d3bf3f1cde : SUCCESS in 37m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8696c647d44346d3b386cff0ea3a7920 : SUCCESS in 54m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62cd80585083431cb27c2d391f55ccf6 : SUCCESS in 21m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73d8fb1dc0274d64b12d463c5db32871 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1e12d990be294b24b0b943aa97adbfaf : SUCCESS in 18m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e72d186fb93d47b580d08f62d70e581c : SUCCESS in 54m 25s","accounts_in_message":[],"_revision_number":9},{"id":"f103f1fba34c8e6df0e0e5dfe3aa685578bfd3f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 16:11:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"0b36a37b5c18c19cddb62fa3bb879b98f2fa5bee","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 16:12:40.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5d7f12d22b9476a86de78bd10fcbbbe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9826746adb64fdd91b1c8295fe98502 : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/38527ad0ed534412b38da795eaedc40e : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6dc0768093b443f6ab1a25c36ff1a9c3 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"3517ca453e0f1994cdbde69b2c4e1764c91586f2","revisions":{"e7505677947fd74e5047c5afeaf057e02ee21e44":{"kind":"REWORK","_number":1,"created":"2022-05-04 21:18:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/42/840542/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/840542/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/840542/1"}}},"commit":{"parents":[{"commit":"3adb7da6e23695522f6952593cdf9ca07500234e","subject":"backend ratelimit middleware: support separate config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3adb7da6e23695522f6952593cdf9ca07500234e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 21:13:52.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 21:17:34.000000000","tz":60},"subject":"backend ratelimit: support per-method rate limits","message":"backend ratelimit: support per-method rate limits\n\nChange-Id: I2936f799b6112155ff01dcd8e1f985849a1af178\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7505677947fd74e5047c5afeaf057e02ee21e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7505677947fd74e5047c5afeaf057e02ee21e44"}]},"branch":"refs/heads/master"},"3af4cb839cf228c31f23cfce29d5ce139bede2de":{"kind":"REWORK","_number":2,"created":"2022-05-05 19:49:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/42/840542/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/840542/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/840542/2"}}},"commit":{"parents":[{"commit":"e1ef3dcdbb6aef97f75c454beacb93498ce508c2","subject":"backend ratelimit: support reloadable config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e1ef3dcdbb6aef97f75c454beacb93498ce508c2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 21:13:52.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-05 19:46:51.000000000","tz":60},"subject":"backend ratelimit: support per-method rate limits","message":"backend ratelimit: support per-method rate limits\n\nChange-Id: I2936f799b6112155ff01dcd8e1f985849a1af178\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3af4cb839cf228c31f23cfce29d5ce139bede2de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3af4cb839cf228c31f23cfce29d5ce139bede2de"}]},"branch":"refs/heads/master"},"39b9d17e5834817deeb54f53064f4af8974ea0ce":{"kind":"REWORK","_number":3,"created":"2022-05-06 13:05:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/42/840542/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/840542/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/840542/3"}}},"commit":{"parents":[{"commit":"26032c9ff2bd355c376ab53bb07ce2db292c9aed","subject":"backend ratelimit: support reloadable config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26032c9ff2bd355c376ab53bb07ce2db292c9aed"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 21:13:52.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-06 12:36:12.000000000","tz":60},"subject":"backend ratelimit: support per-method rate limits","message":"backend ratelimit: support per-method rate limits\n\nChange-Id: I2936f799b6112155ff01dcd8e1f985849a1af178\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39b9d17e5834817deeb54f53064f4af8974ea0ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39b9d17e5834817deeb54f53064f4af8974ea0ce"}]},"branch":"refs/heads/master"},"63854e8f719ee6597dbe656acf03bb50a8ce1ded":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-05-19 13:35:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/42/840542/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/840542/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/840542/4"}}},"commit":{"parents":[{"commit":"897e8d1741960aa3c924d72b7e84a7357d22c1d3","subject":"backend ratelimit: support reloadable config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/897e8d1741960aa3c924d72b7e84a7357d22c1d3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 21:13:52.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-19 13:30:57.000000000","tz":60},"subject":"backend ratelimit: support per-method rate limits","message":"backend ratelimit: support per-method rate limits\n\nChange-Id: I2936f799b6112155ff01dcd8e1f985849a1af178\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63854e8f719ee6597dbe656acf03bb50a8ce1ded"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63854e8f719ee6597dbe656acf03bb50a8ce1ded"}]},"branch":"refs/heads/master"},"e7d34d03204465eadcc054560c7901804c12401d":{"kind":"REWORK","_number":5,"created":"2024-02-22 18:26:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/42/840542/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/840542/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/840542/5"}}},"commit":{"parents":[{"commit":"9bdaaa1d8ae2adf1ad962efb441451363feb4d8d","subject":"backend ratelimit: support reloadable config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9bdaaa1d8ae2adf1ad962efb441451363feb4d8d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 21:13:52.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-22 18:26:03.000000000","tz":0},"subject":"backend ratelimit: support per-method rate limits","message":"backend ratelimit: support per-method rate limits\n\nAdd support for config options such as:\n\n  head_requests_per_device_per_second \u003d 100\n\nChange-Id: I2936f799b6112155ff01dcd8e1f985849a1af178\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7d34d03204465eadcc054560c7901804c12401d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7d34d03204465eadcc054560c7901804c12401d"}]},"branch":"refs/heads/master"},"e2873285fed45aafdad7cdd6778159d49079d4de":{"kind":"REWORK","_number":6,"created":"2024-02-23 16:43:34.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/42/840542/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/840542/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/840542/6"}}},"commit":{"parents":[{"commit":"4788fba71bc564649cf23a444b69f1d80f29d555","subject":"backend ratelimit: support reloadable config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4788fba71bc564649cf23a444b69f1d80f29d555"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 21:13:52.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-02-23 16:41:15.000000000","tz":0},"subject":"backend ratelimit: support per-method rate limits","message":"backend ratelimit: support per-method rate limits\n\nAdd support for config options such as:\n\n  head_requests_per_device_per_second \u003d 100\n\nChange-Id: I2936f799b6112155ff01dcd8e1f985849a1af178\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2873285fed45aafdad7cdd6778159d49079d4de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2873285fed45aafdad7cdd6778159d49079d4de"}]},"branch":"refs/heads/master"},"d6b46aa353766c13546c695d627e76f53fe9d87d":{"kind":"REWORK","_number":7,"created":"2024-03-11 11:36:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/42/840542/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/840542/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/840542/7"}}},"commit":{"parents":[{"commit":"4788fba71bc564649cf23a444b69f1d80f29d555","subject":"backend ratelimit: support reloadable config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4788fba71bc564649cf23a444b69f1d80f29d555"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 21:13:52.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-11 11:35:57.000000000","tz":0},"subject":"backend ratelimit: support per-method rate limits","message":"backend ratelimit: support per-method rate limits\n\nAdd support for config options such as:\n\n  head_requests_per_device_per_second \u003d 100\n\nChange-Id: I2936f799b6112155ff01dcd8e1f985849a1af178\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6b46aa353766c13546c695d627e76f53fe9d87d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6b46aa353766c13546c695d627e76f53fe9d87d"}]},"branch":"refs/heads/master"},"51cbd120dfff3be469a0e3b7acfa1d23962de4da":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-03-12 10:42:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/42/840542/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/840542/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/840542/8"}}},"commit":{"parents":[{"commit":"e9abfd76ee5ef94a51b2dbe5a758432cfeeab8e9","subject":"backend ratelimit: support reloadable config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9abfd76ee5ef94a51b2dbe5a758432cfeeab8e9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 21:13:52.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-12 10:42:18.000000000","tz":0},"subject":"backend ratelimit: support per-method rate limits","message":"backend ratelimit: support per-method rate limits\n\nAdd support for config options such as:\n\n  head_requests_per_device_per_second \u003d 100\n\nChange-Id: I2936f799b6112155ff01dcd8e1f985849a1af178\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51cbd120dfff3be469a0e3b7acfa1d23962de4da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51cbd120dfff3be469a0e3b7acfa1d23962de4da"}]},"branch":"refs/heads/master"},"3517ca453e0f1994cdbde69b2c4e1764c91586f2":{"kind":"REWORK","_number":9,"created":"2024-03-12 15:55:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/42/840542/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/840542/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/840542/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/840542/9"}}},"commit":{"parents":[{"commit":"e9abfd76ee5ef94a51b2dbe5a758432cfeeab8e9","subject":"backend ratelimit: support reloadable config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9abfd76ee5ef94a51b2dbe5a758432cfeeab8e9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-04 21:13:52.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-03-12 15:54:54.000000000","tz":0},"subject":"backend ratelimit: support per-method rate limits","message":"backend ratelimit: support per-method rate limits\n\nAdd support for config options such as:\n\n  head_requests_per_device_per_second \u003d 100\n\nChange-Id: I2936f799b6112155ff01dcd8e1f985849a1af178\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3517ca453e0f1994cdbde69b2c4e1764c91586f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3517ca453e0f1994cdbde69b2c4e1764c91586f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
