)]}'
{"id":"openstack%2Fswift~841143","triplet_id":"openstack%2Fswift~master~I8883b63be315a7891c4bfbec662c81f218d1f263","project":"openstack/swift","branch":"master","topic":"p-repair-gaps","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-05-23 10:02:15.000000000","reason":"Matthew Oliver replied on the change"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-05-31 20:23:55.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-05-31 20:23:55.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-05-31 17:52:45.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I8883b63be315a7891c4bfbec662c81f218d1f263","subject":"manage-shard-ranges: add gap repair option","status":"MERGED","created":"2022-05-09 16:20:01.000000000","updated":"2022-05-31 20:25:39.000000000","submitted":"2022-05-31 20:23:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":47,"unresolved_comment_count":19,"has_review_started":true,"submission_id":"841143-p-repair-gaps","meta_rev_id":"816b02544aff53c93e394204331913975edc17ea","_number":841143,"virtual_id_number":841143,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-31 20:23:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2022-05-23 10:02:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-31 17:52:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-31 17:52:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-09 18:29:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-10 00:35:13.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2022-05-10 10:28:31.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2022-05-11 19:29:41.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-05-23 10:02:15.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"b5cadbea4e7d54b320652c3ca0d3551549b11e23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-09 16:20:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"103dbffee0a28cd501a1a5b685489909c345e029","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-09 17:31:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d26873284ad9d301055fc2a90d2d5b670d4f4a12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 18:29:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb955f864765438f843780c1a8367bb5 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/84284e15ad504f329c6df71db9235082 : SUCCESS in 54m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/506a2a294c744b29afc1480ac8c741a7 : SUCCESS in 45m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f72c866d98345aca99283a5a3d50b3b : SUCCESS in 48m 42s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3f71231ea17042fd8b0534511fd85c2a : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/37e55faae5684de1a7933f8d928a32c4 : SUCCESS in 22m 20s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e1891eb522374e5884c036287085f1dc : SUCCESS in 25m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9932bb9bc9494135b37d816b94cc0096 : SUCCESS in 7m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ad8f24b7b5ae429995b6ba8cd24f9bfc : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/85469aa6e9f54164a310c57b1541c121 : SUCCESS in 20m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b095570fa6f24b39b5925f7016f0cf97 : SUCCESS in 20m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2388a6c07771439f97b1b34a68e729fc : SUCCESS in 17m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c97deb3a30d74e32b4a746fa1760392a : SUCCESS in 21m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a9f9dacd5c9e4ee0be735e126e20a4ce : SUCCESS in 22m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f343f99ea8fa46b4a8043847953f1882 : SUCCESS in 22m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6fd34f2d37d14070b4147ec15f9a2e00 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/89d416d7e7474a5c99d91728a71a4962 : SUCCESS in 17m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d8c43d52e0d74aa59ccd3563e55d0abc : SUCCESS in 19m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e89487b6d70044a5a84d83263ff9820a : SUCCESS in 12m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c20ad1b6c240467cadf7f896978a3b66 : SUCCESS in 19m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/31bb0eea63ca43809406050a9d9fc887 : SUCCESS in 43m 04s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e9aefbedd20422883c370dabe1adf92 : SUCCESS in 54m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/275934f68c5b4b6bba0b76e8f8b90201 : SUCCESS in 52m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e4b60579b4b54652ae30c2eb3368fcba : SUCCESS in 42m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/240b7003b4814d9b8f2fa49b1856c650 : SUCCESS in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08d9efa98cb042c489a296f377189b98 : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0a33dcefc74549e99a0e4ec7d8e0500e : SUCCESS in 20m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3eda82c8a32941f0af82c239e503c781 : SUCCESS in 47m 56s","accounts_in_message":[],"_revision_number":2},{"id":"aad106a845e9667a1104214f7775bce57698bde3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 19:14:39.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0037c30bf9c84acdae6bf02388ceb85e : SUCCESS in 1h 04m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa2a592a65a8486b8d5004011ea00d73 : SUCCESS in 1h 04m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/80774ff8da554f2f82b0af9ce02188dd : TIMED_OUT in 1h 31m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb7057b35b4b4c188facc12d915c44ce : SUCCESS in 1h 08m 28s","accounts_in_message":[],"_revision_number":2},{"id":"38e4da1f03c89fec7bed191d86d1499b7fabd5c2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-10 00:35:13.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7a4fb5908c8889b262796370aac49ce001a9ff30","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-10 10:28:31.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"91738c46c5ab474c9440d6875fd06901db82615f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-10 14:03:55.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3adc5670f9435688f29fc34f2f89d322854fc28c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-10 17:02:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d5e6a9624bdfcc364294fd289aff4cefe28f0bb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 18:36:30.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35b0da8973684f7f86732f7b1ab0c143 : TIMED_OUT in 1h 31m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85f354cb52074e78baf6955ef45a9481 : FAILURE in 45m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/38821268dfbc4ed98bf71b8092877751 : SUCCESS in 54m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e1f99bb62d74327b7f7d0de867fab4f : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":3},{"id":"a1f61fc575219fb3b1f7d48354ed66338f7410e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 18:39:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6665c59a8f34e6db705114edcb70c46 : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/c55e6b6940504cf19dffd6321241ce8f : SUCCESS in 1h 09m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/60c2b7816e3f45f19fa2bb946cca4690 : SUCCESS in 50m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd1dcf5ea32647478f3b359690595fcf : SUCCESS in 51m 53s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c3180ec92c8346c1ab6001cbaca40ab2 : SUCCESS in 25m 10s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d54f4c7055ac429192451d5f9d645a1f : SUCCESS in 25m 20s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3e01dcbaabd342f89cfa0635b811444b : SUCCESS in 24m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3040b0be4c2746c499219c552f4143f9 : SUCCESS in 7m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b40048ea96dd4a089dccef33964174ee : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1cd1f239717f4cefacbdbda8f5f0be05 : FAILURE in 14m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aeff8057c746444f9004022033e8f3e0 : TIMED_OUT in 1h 31m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/56b2e5bc471c42f398680b1e4201dee4 : TIMED_OUT in 1h 31m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/63e933be10f4491ca4e0a9a41cfd4d0b : SUCCESS in 23m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/82688a9cd1614cc0b7645848011b6803 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9a860fb5cc6a4bc9b37ae7067f59874a : SUCCESS in 18m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/011ba536ad2f4ef2ba6a54dfad39fac2 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1d39c9ded4e4403ca468aac3690d437e : SUCCESS in 22m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4448298496e14f0db4cf74e0117c7b49 : SUCCESS in 14m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a55919484433423483970ac6bd3937af : SUCCESS in 9m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a42fb87738ec46679b31170de7b19f5d : SUCCESS in 21m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/970db139f9d3462a836a74624e94e92f : SUCCESS in 52m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8398781f76a54dafabe048abcb77f0e0 : SUCCESS in 54m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d291bd91bb9c44ccb89020f21d8f13ce : SUCCESS in 42m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d93b600f95b42e5bb2e544cce6054a9 : SUCCESS in 34m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a0f58085abd4725a22db4d6e48a2a57 : FAILURE in 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f524ce4240949178d19e1c48fbfdc49 : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2a5133477fa84e2691723b7c5befd399 : SUCCESS in 25m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1dcc52d9ad6844d3869b3db8cee3cc33 : SUCCESS in 47m 33s","accounts_in_message":[],"_revision_number":3},{"id":"1b970ae0faa94f137bb7686e7b57e60771e6e7b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-10 21:05:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"66aa5d9be4567e2d3dc88108945e8ab3492b73d7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-10 21:06:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"952d55825da50fa1a1dd5d4204c3bd179c65c348","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 22:38:46.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24132f09ec044378939c556847062153 : TIMED_OUT in 1h 31m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/79eb10be74b04be08a6e5d7773456514 : SUCCESS in 43m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/730837f6990e4c0285701c6140fe69f1 : SUCCESS in 41m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbe0174614094af6b4b554e80e2e88b9 : SUCCESS in 43m 52s","accounts_in_message":[],"_revision_number":4},{"id":"c5e1c69c4c027e7a25a121f6c947ade047afcbd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 22:43:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfaf425bd80d4d7eafcfc2afcc043002 : SUCCESS in 10m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/4d77a207b1e2485395ddfd74ee60c4ab : SUCCESS in 52m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/18e84c2c706c4f3fb9db3b657091adf3 : SUCCESS in 56m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dc3230d6fca4515bec3979f31656ad4 : SUCCESS in 57m 27s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0682430d4c68484ab3ccd5cac47df311 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d1e3693ce06a433a95427910380b232c : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/26e139ab14804120985d74ea00909013 : SUCCESS in 21m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a72d86a208546afa466bd9daa870291 : SUCCESS in 6m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/908bb92b26cf4c45a03b435768c6a47c : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ba07202c167f40acbeb66627dca88c3f : SUCCESS in 16m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0ed21aab878b4cef9af9703b03c596ed : SUCCESS in 15m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cf317950fe2644b8ac93d161c557639a : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bee5ee80e67744378836469b893fb82a : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/731e1d4047604a9998e4e89abd46508e : SUCCESS in 17m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fb2531cf3c0b438185442ae1d8136f47 : SUCCESS in 18m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c44ffcadda8646b3ac61bc66f3a8d7b2 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3d568b3250a24620886e20f8d2a0eee5 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dd10c7931d7b45eca0e86ffb2fe345cb : SUCCESS in 16m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/caf4269ca471498fb14addc2ab010577 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/755619f7f4b84f548e5da1687f6fb4bd : FAILURE in 15m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/655800348e774c0a9cc4f51d85a4de8b : SUCCESS in 49m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/732303d8e4794e4aa1e8f89ddec3a885 : SUCCESS in 56m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8de6d94a4a2b49a2b3420708094c102f : SUCCESS in 46m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7a12099590f141589fecb9733b56b646 : SUCCESS in 41m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d61a0d4a5c614556a44940dd74e16367 : SUCCESS in 16m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc9bab3b7f3d42baa99a0443f16d3223 : SUCCESS in 7m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/464524c31aa74ef28a8352994096cda4 : SUCCESS in 26m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49dfa11520b947a2bb8c3aca45cf7fa5 : SUCCESS in 49m 18s","accounts_in_message":[],"_revision_number":4},{"id":"a7d09aa17028d5f0da33ee3198bb4fd7f5290b4e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-11 01:26:46.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"dd8a05aec3a86f009b4beddfe646dc8faea11d81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-11 14:31:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1de541d53a074ae4ef8172de2454ec10d547803d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 15:38:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6a575456c4e4717ae5b41ae2965662a : SUCCESS in 1h 04m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57f7b563ac2043aaa328d41174f2e140 : SUCCESS in 49m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bef1b9320a4b4b0e962e697ca4e3f7ae : SUCCESS in 47m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f0e34af6a8d46b0b54e22df2dec056a : SUCCESS in 1h 05m 01s","accounts_in_message":[],"_revision_number":5},{"id":"768f3fc7e6763ca51657138fd8d3016538c5ed4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 15:41:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52c81934c65e4263993597c8de50f115 : SUCCESS in 7m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/1c021694d91e44bb880f4b275ea5cbf6 : SUCCESS in 45m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/58379ac9d93d4a3f962ec39a13b20791 : SUCCESS in 36m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b2127228bbc4e59b8490dcfcce56af9 : SUCCESS in 40m 19s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/da447845cc05458bb87e81f786c3efa3 : SUCCESS in 17m 51s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/60206fd1e2864882ac4e7705f9e08279 : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d1d092e36786464daa47fa2cc5557b7b : SUCCESS in 18m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ce014ac6033e4354870ab28af33e2bee : SUCCESS in 6m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a51b9b0eb61e4a94af7f7924442ba521 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8febe5be98d042118aac1c0aa04cf4c2 : SUCCESS in 14m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d42d490d0a83472692dc24d07890dbd6 : SUCCESS in 13m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a4807f64467d4f16a399690f681096ea : SUCCESS in 14m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e78d17c902794835ac6213b5738c0944 : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/67767eaca9674804a37e4bf31d4798dd : SUCCESS in 15m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4a028064138423c923a75292d8f25d9 : SUCCESS in 15m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ef3a3ce622b447dca9033a7a6ce36995 : SUCCESS in 13m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2abfeb22f2284ef7b4427052e5d37b00 : SUCCESS in 14m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ee8362feb2f34bc48bf2d2b19b332fe8 : SUCCESS in 11m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f6893bb22546492e9787ae42cde7c175 : SUCCESS in 6m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8e6c976ca8e844a2a05b8b05c98ed834 : FAILURE in 14m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e24e163baa824c65b24f16a876d0093c : SUCCESS in 46m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/219a7cc3153d4fd68c7a60bc1fcf411c : FAILURE in 48m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a15cc79f1cff4c8db5ba30848a6487db : SUCCESS in 32m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ddb6a96d416f4d7facefca3b0d0278db : SUCCESS in 35m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b87f3de0ed6401380ff2b05c526046e : SUCCESS in 13m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4992c4c1ee254ade979a0a5a91ca3ca2 : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30383a84832b4911a216f48b1969c120 : SUCCESS in 21m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36d80ff66ab44c16bce1de30163dda35 : SUCCESS in 1h 03m 32s","accounts_in_message":[],"_revision_number":5},{"id":"81e01ff94a112b427679a35aa455d08ef446eaae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-11 15:51:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a676bbb4082c721cc7cc83ab895105b13ef73451","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 16:55:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8f06b5a95b44a89a745bec185f89e23 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/84cbaba14c3e4a93b9463788e6d2e2e2 : SUCCESS in 1h 02m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d3a3292e59a46a2a191ad7aec827028 : SUCCESS in 56m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9aa17e10c1154801927297c8342285ce : SUCCESS in 54m 35s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0c8af7d62a314582ab03d3c4dff84a92 : SUCCESS in 22m 19s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e5d7273eefcd40a59f65f866df00a71c : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b4ae3711d35742fda373101b0d77db1d : SUCCESS in 25m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/973b244c90d74969995df41d4ffa2ae0 : SUCCESS in 7m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c834ebad53324774a794b40593f806b6 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/064d6c1b72964e44b55edb7a23c21e04 : SUCCESS in 15m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b741fb5cbc054fe69986ed0ce9faf339 : SUCCESS in 15m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/956e6d1870fe4147af04be6f14efed07 : SUCCESS in 15m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ed49daaa93964f29a85455a05823fb01 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cde5600ee9a543f5819f1bb21f953697 : SUCCESS in 19m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/95aded76a6994af0a5849c33f74b1e26 : SUCCESS in 33m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9b9e226be719471d96ba2bf3f944ae94 : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8bac9d2c9c494868a24845b22e37dcf6 : SUCCESS in 15m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a711ba371876422babb866ea637e7bee : SUCCESS in 17m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fcc980023aa14c62ab1fcc5ad4539d1b : SUCCESS in 6m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ab7d9b69ad0b4210b0041b962afcd5ac : FAILURE in 24m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/96fc1f132020475e99095bfc0a19ad64 : SUCCESS in 50m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/41eee52da508477a8b7388cf07fe82b3 : SUCCESS in 51m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/93b1b4aff6b9431c98692856c52f947d : SUCCESS in 40m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77ed1b2a3fc340399c38521bfe1a0652 : SUCCESS in 41m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecf1d5e022324ef1bdabeb4fb38767ec : SUCCESS in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/946e2ef841ab4d97bd52c3b409e8d830 : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30451ab715074bdfbb59246bf93e1bde : SUCCESS in 19m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8cc1e0976f744ba997d21c20a3f3365 : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":6},{"id":"aa1b7a69877b52a9f0c45d01583537956dc76b3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 17:24:51.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82484fac376b40e297853e45f0dec3ad : SUCCESS in 1h 05m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ea6248dcc6843c0bc7d5953398bc941 : SUCCESS in 1h 09m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f34b610a6874104acbb7922e949e93c : SUCCESS in 1h 16m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fec8a751427045ebb7743104eaf9d16c : TIMED_OUT in 1h 31m 28s","accounts_in_message":[],"_revision_number":6},{"id":"19c7546d6fc1677269703e6fca7646ddaee40c26","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-11 19:29:41.000000000","message":"Patch Set 6: Code-Review+2\n\n(12 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2d7143df9f36809acc44d4a8d8292a7836fbfb96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-11 19:49:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"6dd0c8a20329e43cf7b8c1f491b69cc1e9982741","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-11 20:05:36.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"9a869056ffa62e969853b6c04f7fa305f350df01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 20:50:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b0f4b85245f4d12961484e9ef4d4b72 : SUCCESS in 56m 21s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f3fc556a89e4c2e8dadef9956c00a79 : SUCCESS in 46m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/12612fb72e944f8ea38dbb929e774339 : SUCCESS in 54m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d0d739668fa4737a6d5da348a984b8a : SUCCESS in 59m 08s","accounts_in_message":[],"_revision_number":7},{"id":"fbfe4c7011b019b33d648d0256b900a7d2caede7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 20:55:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc56e928cf654a8f8f061c7520a0f9fc : SUCCESS in 7m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/99d1e31df18a49a2852978d4f7cc099c : SUCCESS in 55m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8aeb0d5f098b47169165c846b301c842 : SUCCESS in 39m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cd259f3d1944e05985b00c2a254edbd : SUCCESS in 48m 32s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0f2247cdec1e44249089c60d68296545 : SUCCESS in 16m 54s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/91e792141a7a4854a5212ae1cd999d78 : SUCCESS in 25m 39s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/acf556703f8a4c749d939a8e0a470c5f : SUCCESS in 15m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a9c05271c7a4136b392a3d70c9d16a6 : SUCCESS in 7m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/63f546614ca34a779ede14e5572986f3 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a056d8fdab3444d2b4d7c98b0088d9e3 : SUCCESS in 12m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ca66509d257e4a9e8341a3da2e577ce5 : SUCCESS in 21m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d13b9740c7d342b298193f2272e32d9d : SUCCESS in 21m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/306629b67a9041e2bad01443be47c8eb : SUCCESS in 20m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8f17c5acbf1e4ca28464ddde38addac2 : SUCCESS in 20m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b1bd3e44bfc04c1cae72a5244b4f396e : SUCCESS in 22m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/87783f139bb64734a3f164bf23be4343 : SUCCESS in 14m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/eb788b7177ca4bfd82d22894e62af745 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1a5f8c52520e4a1584d0846ee05f490a : SUCCESS in 20m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/534f5b346cca47a183386c9268d47e49 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a88129f151494b79bbd3f86ad78481be : FAILURE in 13m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9039add366e54ba6bb9a7fb1fc17e17a : SUCCESS in 1h 01m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bfc8fac26c194d0f833683f12c171fd9 : SUCCESS in 1h 03m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4b42f1431064953b41ea1f38640b3ce : SUCCESS in 52m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/46f8ae8cf5154dd6af9f478bdb50c73c : SUCCESS in 29m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1210157c72aa4b8fbefecb419d7fd025 : SUCCESS in 13m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3273188aee24477094e676c65c09d9f0 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/71b1d83deffc45b08a931efdeb2a0abe : SUCCESS in 17m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54a8d1f2fcea4dc2970d17c018ca0860 : SUCCESS in 48m 33s","accounts_in_message":[],"_revision_number":7},{"id":"83db78f1cca60873784f873d3523775f2f5df0e5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-11 21:10:49.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"aa047c0db77941c6c5c35f1e27dfdfd9104fed73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-19 18:17:05.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ed0635f3b017b30b41214146db7ea7932706bf4a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-19 18:21:37.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"a667668d5a4011d2e84c87e044585b491ce494f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 19:22:58.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9713d7a7e42343ac8e4c437fc3d8e291 : SUCCESS in 1h 03m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4cad29be061548a1bd9025473064b512 : SUCCESS in 41m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/054ce84af6d14b9fa9deb8636fdf6892 : SUCCESS in 52m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c976e8b916b41f5b205968f47678e44 : SUCCESS in 50m 14s","accounts_in_message":[],"_revision_number":8},{"id":"bb806862dac2628afb20668f06085dca32692ee9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 19:30:29.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d4b649432d94ad3b3f80b9f7eed9dca : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/9314e3e65b0d42a4a9bbda399efef396 : SUCCESS in 1h 06m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b9388aefc2df4dceb74dfc35c4b6c9ad : SUCCESS in 54m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7cda79cc69ad419494c28a89ce2e55c3 : SUCCESS in 52m 17s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1a8e2a39fbd44b0cab88d6c68fc66104 : FAILURE in 7m 07s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/dfbc8ff6f82c451d9bf05cc5eea07272 : FAILURE in 6m 38s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0a5b224643bb46549c046888540eab5d : FAILURE in 6m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d3006eaab68142eaa1935b71b77ef932 : SUCCESS in 8m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c8db72fab1e7487a80aa9cda7e6a76b9 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59170c75109344f084cb36fa2afd96c5 : SUCCESS in 19m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6c07d5ea9dca450fabc6a28635df111f : FAILURE in 4m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/964970af290f442da065f247edf67312 : SUCCESS in 16m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/495fb25f5e664ba086f917e8f7208fc3 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0c7b52fbf8f4f7385eec64af39abe60 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/437d81e3ed4943a5976fbd47fe3ff27f : SUCCESS in 16m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eb6309a181e3470f92d6f280b68c06b9 : SUCCESS in 24m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9670896fdc0948888dc78d3db0bd4300 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c9d55638bb534d56a3ef0d417191ba98 : SUCCESS in 13m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a833605ff91c410da6630bc657750a34 : SUCCESS in 7m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/630efae4e74c47169778131355f3551e : SUCCESS in 19m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f124992f47e9412d87cea9455ed7b436 : SUCCESS in 57m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/58fed4306e9c449f86d72f7352a26d44 : SUCCESS in 52m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7dd15253676f4d1ba257654d46d1c5e8 : SUCCESS in 41m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/56bb724e79a3421ea43fdf5f7dd100b5 : SUCCESS in 42m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bd53a9b39c14acbb5ac84417486d32d : SUCCESS in 16m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac9315cf289c44438d809d53b933988a : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/865059fb36fb484d810e714e1c026a43 : FAILURE in 8m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d288e4f6bbe942f889f6731d19a824dd : SUCCESS in 54m 38s","accounts_in_message":[],"_revision_number":8},{"id":"7381ccf880d0f6b3061641be934ec0e9282a8ddf","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-23 05:58:45.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"032ce39628a9584baa8d2bef4a9871ce9c0805b9","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-23 06:37:19.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4ee0f0fe48b44871dd7c350a90154aa4152e8eb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-23 08:44:24.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5eed57478c0b7e4a27985465cff4505ee4b324a0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-23 08:45:51.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fd2ad11a5cb2d4beea53f2d39ce30e2864cc51c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 09:56:16.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f94c4f8feb1549ebbe3861326c55dc24 : SUCCESS in 55m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/872dfe0d023841c2baac94e8504857ec : SUCCESS in 1h 01m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/327ccf4da6054604a14488c7f9313563 : SUCCESS in 1h 00m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7af248ab9ab9463ca211a6e0550bf3b9 : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":9},{"id":"f27fee3aa19a6d85e2141d9698af37ae519f43a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 10:01:20.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f999634de64794bf5eb90d613c5103 : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/5e861fced49d4f309c80f51057b0934e : SUCCESS in 53m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8d36dc19d259487794f7dcde3436b931 : SUCCESS in 54m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f5c507194194102b4b393688318293c : SUCCESS in 49m 50s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/60a65ceb0dad4792b00676ee2bad1670 : FAILURE in 6m 41s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fbe0b9aa85544f49ae41847e4e39a3d5 : FAILURE in 6m 12s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/705e1f1131164057bdecfde16a7de60b : FAILURE in 6m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0d40ad5ddab647d2ad3cf0b28c5a0189 : POST_FAILURE in 36m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/029ac7eb8add41d38a3dda265d1bf7a2 : POST_FAILURE in 35m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a5021a3e050947c69527e5db31e4ba72 : SUCCESS in 19m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/02373b5a613144f082e9261a98cec7ef : FAILURE in 4m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/109088afda844ad1a8aca661319bad9d : SUCCESS in 18m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/34ba71e51d354417a9bd57ffe8e284a2 : SUCCESS in 20m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/600675cb77e64542ada18e8fabbc94f3 : SUCCESS in 16m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2e2e646bc6f84744bd0e57f0768eca5d : SUCCESS in 22m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4da3326b039840e59d364c5a031bf7c4 : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6b621e0c9177485d99084e23d53d5ab7 : SUCCESS in 18m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dae09bd4a69f47eb9b50a081fd0c6491 : SUCCESS in 18m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/31a64812967e48d1b1a21a091c0b63b6 : SUCCESS in 7m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6f77eae6b34c4c95a5f304b8f7311415 : SUCCESS in 18m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/807d47fcb6b544d689d21de0f5cb07a0 : SUCCESS in 53m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e87790af877b441db0e9c1811d377d68 : SUCCESS in 1h 04m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d5d5a82c2b2d43deb5381659e427d6db : SUCCESS in 40m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/61e327a132c84eb5ba82549d9d7d250e : SUCCESS in 41m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef97243931dd488baa193eaa73b319ef : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/718cb619bf82460b9cdffdd14825606f : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e1953029a5f341698624e37007288209 : FAILURE in 8m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fe6e9aa4ecc464caf067584e98e2c72 : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":9},{"id":"3d3cb92c3befbfb9cdd1f9abc979beb7e879f5d8","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-23 10:02:15.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fb6b83c0d69fc980cdd9800fa95765ceb869720a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-31 17:52:45.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c7923088c37ae9eaf78146e8f7d1f4a37e0329fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 19:01:56.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1055b341389d4f0fb1ee0faa97c475e9 : SUCCESS in 44m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f81b0425fda4c8e9b9b7e73dcafc90a : SUCCESS in 29m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9e6a3c81f414b05abf2d79da2399534 : SUCCESS in 50m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/340cb98e2563415ab65a793c81b8cf90 : SUCCESS in 46m 57s","accounts_in_message":[],"_revision_number":9},{"id":"254a7bed940851fb42aa3b9dbe937ba7dca28c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 19:03:12.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e2d8a539b2740ee89b089dbc75d2011 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/e6057ac9dd4c4a89a23b8b0a5fe2c83d : SUCCESS in 54m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1fe585c93e2a4d1fb40113fe491185ac : SUCCESS in 45m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/032f15ddee7d49a49ea4392826dbc189 : SUCCESS in 48m 01s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6201095bde064b00aef18419075752dd : SUCCESS in 21m 28s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1541830c26a4488cae968331080e3c1e : SUCCESS in 30m 45s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c24e34b4b69446cfa369a1144a311a36 : SUCCESS in 22m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5cd892fd625482fa110f8575bcd154b : SUCCESS in 7m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/611c7abe52b948d0b433b6e99c1d5b2d : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8c396941786d4f2189a822d565bb075d : SUCCESS in 19m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/27fd7a346128413ea656f10dcf7d9bff : SUCCESS in 26m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8feaf5574a4d4c79b4490ea35d77436d : SUCCESS in 15m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8586d5b4e86c46a59d21426fd8df8047 : SUCCESS in 22m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/88e1f8c39aa94fe6b3786e4edba6fcde : SUCCESS in 24m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d3744dae75794a569c56f8ad3916101a : SUCCESS in 29m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0138e99be9f44a619cfe00f2eaba37a7 : SUCCESS in 19m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2c0507e58cbf455482fc4792499c3c2e : SUCCESS in 21m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b31f33412b8c4276b447f9f5b8e510f2 : SUCCESS in 29m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fa3c458218d94ce7b2a8c13ff631bd22 : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9291954819c44d599e3bc4cedc93ba1d : SUCCESS in 18m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a67905fe02304092b2af8904d3f9c429 : SUCCESS in 1h 08m 35s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/dea6915154c44a81ba54d8d37c508086 : SUCCESS in 55m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a81d59c029fc43e49e470748269cf349 : SUCCESS in 53m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30d9d81ec37b4e4380fb8b7f8d97cc51 : SUCCESS in 41m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72fb46149722434cb899dd60fa5ac10c : SUCCESS in 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10e0a1a0f52c45918ac86de52eb46fcd : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0fc98d972481449a98df9e0f55d363ed : SUCCESS in 26m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e656afdd22d442aae9dd78806a958d6 : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":9},{"id":"135f3c5eb8f426bdebc7725fedc73a4ba824812b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 19:04:12.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"427b34a13511b60f0a83df82bf160470cbcad481","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 20:23:55.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c2439690d67463a8f62f61dce96e62d : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/b39813d56df44260815cb778aabc7dc7 : SUCCESS in 58m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0aae7ae6dc9a47c2ab82fc9aa499479c : SUCCESS in 40m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7578514528a3457ab4b25990441fe0a1 : SUCCESS in 51m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/05b0770c187a4dad871faeb10d131af7 : SUCCESS in 9m 41s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2a2f99a36d0c412cb813d7b62f2b1285 : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6f9bca077f7f415fa6078ca3cfabc054 : SUCCESS in 19m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b089fb70166f4b3f87fc8b78c3a04ad2 : SUCCESS in 18m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2ee089acf8f345d492bf57011d953b51 : SUCCESS in 20m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ed14a47ebde742778c535cfce88a803c : SUCCESS in 21m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/79e288fb5be0418b8e9296e2ae3ee322 : SUCCESS in 22m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d379e0617d814aa3a92c2c41df7332a1 : SUCCESS in 22m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/804940b34ee34246b3c14b094e0866d6 : SUCCESS in 20m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/04801ff8dfb542a3944d6bdd883e5cd3 : SUCCESS in 20m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eb3fe9f19e9c4801a0d57d7739cd2a76 : SUCCESS in 22m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/31bb8052a99a46b79b3ce649f770e0ad : SUCCESS in 6m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0a92e41495964fbe9329c19d1eede856 : SUCCESS in 52m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/42736c7aa7404cb68ffda88546f47e0e : SUCCESS in 1h 05m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1dafcb938df34766999d28601dc4273a : SUCCESS in 37m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d91ca2dac46407aa056ff3f4348b313 : SUCCESS in 42m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97dbac45e35a4b9784742e672e4a298e : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/523318065edc42e3876ccb436a06e063 : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6488fc08db3b49af8f1b51c55e036908 : SUCCESS in 25m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dae6a8838d144159b8c621e611910fc6 : SUCCESS in 1h 14m 42s","accounts_in_message":[],"_revision_number":9},{"id":"b9149694f36b3c38be4024b8e35ada85913c2df8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 20:23:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"816b02544aff53c93e394204331913975edc17ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 20:25:39.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6544ddbe54a94dbe9d32637b172af6b5 : SUCCESS in 1m 18s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/392a660d31644265b68f607ce3b95472 : SUCCESS in 52s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/49262fa9efe647b3b068954595f10944 : SUCCESS in 45s (non-voting)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"b45b45fa729caa8c36f874658b69f92122dfbead","revisions":{"b931d4cb35426774b4720faf53de2356193e865d":{"kind":"REWORK","_number":1,"created":"2022-05-09 16:20:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/841143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/841143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/841143/1"}}},"commit":{"parents":[{"commit":"bff6e5f8fb9080eacab501a15510f627547e26df","subject":"Merge \"Rip out pickle support in our memcached client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bff6e5f8fb9080eacab501a15510f627547e26df"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-09 16:19:15.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-09 16:19:15.000000000","tz":60},"subject":"WIP manage-shard-ranges: add gap repair option","message":"WIP manage-shard-ranges: add gap repair option\n\nTODO:\n  - see inline TODOs\n  - write a unit test\n\nChange-Id: I8883b63be315a7891c4bfbec662c81f218d1f263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b931d4cb35426774b4720faf53de2356193e865d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b931d4cb35426774b4720faf53de2356193e865d"}]},"branch":"refs/heads/master"},"0a45bdf7f35a4c5128e02dec3faa47958ac09960":{"kind":"REWORK","_number":2,"created":"2022-05-09 17:31:21.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/841143/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/841143/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/841143/2"}}},"commit":{"parents":[{"commit":"bff6e5f8fb9080eacab501a15510f627547e26df","subject":"Merge \"Rip out pickle support in our memcached client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bff6e5f8fb9080eacab501a15510f627547e26df"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-09 16:19:15.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-09 17:30:12.000000000","tz":60},"subject":"WIP manage-shard-ranges: add gap repair option","message":"WIP manage-shard-ranges: add gap repair option\n\nTODO:\n  - see inline TODOs\n  - write a unit test\n  - functionally test: does the expanded range stick in the root db\n    and get pulled down by the stalled shrinking shard?\n\nChange-Id: I8883b63be315a7891c4bfbec662c81f218d1f263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a45bdf7f35a4c5128e02dec3faa47958ac09960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a45bdf7f35a4c5128e02dec3faa47958ac09960"}]},"branch":"refs/heads/master"},"b6390e2de90b9a1f2bb0a463577a7f593fd6a8e9":{"kind":"REWORK","_number":3,"created":"2022-05-10 17:02:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/841143/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/841143/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/841143/3"}}},"commit":{"parents":[{"commit":"bff6e5f8fb9080eacab501a15510f627547e26df","subject":"Merge \"Rip out pickle support in our memcached client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bff6e5f8fb9080eacab501a15510f627547e26df"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-09 16:19:15.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-10 17:02:37.000000000","tz":60},"subject":"WIP manage-shard-ranges: add gap repair option","message":"WIP manage-shard-ranges: add gap repair option\n\nTODO:\n  - functionally test: does the expanded range stick in the root db\n    and get pulled down by the stalled shrinking shard?\n\nChange-Id: I8883b63be315a7891c4bfbec662c81f218d1f263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6390e2de90b9a1f2bb0a463577a7f593fd6a8e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b6390e2de90b9a1f2bb0a463577a7f593fd6a8e9"}]},"branch":"refs/heads/master"},"892e9200ef78bbe17d0a06d9d988aa67d0a53165":{"kind":"REWORK","_number":4,"created":"2022-05-10 21:05:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/841143/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/841143/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/841143/4"}}},"commit":{"parents":[{"commit":"bff6e5f8fb9080eacab501a15510f627547e26df","subject":"Merge \"Rip out pickle support in our memcached client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bff6e5f8fb9080eacab501a15510f627547e26df"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-09 16:19:15.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-10 21:05:27.000000000","tz":60},"subject":"WIP manage-shard-ranges: add gap repair option","message":"WIP manage-shard-ranges: add gap repair option\n\nTODO:\n\n- functionally/probe test: does the expanded range stick in the root\n    db and get pulled down by the stalled shrinking shard?\n\nChange-Id: I8883b63be315a7891c4bfbec662c81f218d1f263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/892e9200ef78bbe17d0a06d9d988aa67d0a53165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/892e9200ef78bbe17d0a06d9d988aa67d0a53165"}]},"branch":"refs/heads/master"},"3a3e90232dc2f9857d192b9980de36055789d836":{"kind":"REWORK","_number":5,"created":"2022-05-11 14:31:07.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/841143/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/841143/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/841143/5"}}},"commit":{"parents":[{"commit":"bff6e5f8fb9080eacab501a15510f627547e26df","subject":"Merge \"Rip out pickle support in our memcached client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bff6e5f8fb9080eacab501a15510f627547e26df"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-09 16:19:15.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-11 14:30:18.000000000","tz":60},"subject":"manage-shard-ranges: add gap repair option","message":"manage-shard-ranges: add gap repair option\n\nChange-Id: I8883b63be315a7891c4bfbec662c81f218d1f263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a3e90232dc2f9857d192b9980de36055789d836"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a3e90232dc2f9857d192b9980de36055789d836"}]},"branch":"refs/heads/master"},"ceb0995adc34b5707912af323d2cc766deeddb0e":{"kind":"REWORK","_number":6,"created":"2022-05-11 15:51:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/841143/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/841143/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/841143/6"}}},"commit":{"parents":[{"commit":"bff6e5f8fb9080eacab501a15510f627547e26df","subject":"Merge \"Rip out pickle support in our memcached client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bff6e5f8fb9080eacab501a15510f627547e26df"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-09 16:19:15.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-11 15:51:08.000000000","tz":60},"subject":"manage-shard-ranges: add gap repair option","message":"manage-shard-ranges: add gap repair option\n\nChange-Id: I8883b63be315a7891c4bfbec662c81f218d1f263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceb0995adc34b5707912af323d2cc766deeddb0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ceb0995adc34b5707912af323d2cc766deeddb0e"}]},"branch":"refs/heads/master"},"35cd5f54e85f9757340fb32d121ebdf3ea701136":{"kind":"REWORK","_number":7,"created":"2022-05-11 19:49:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/43/841143/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/841143/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/841143/7"}}},"commit":{"parents":[{"commit":"bff6e5f8fb9080eacab501a15510f627547e26df","subject":"Merge \"Rip out pickle support in our memcached client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bff6e5f8fb9080eacab501a15510f627547e26df"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-09 16:19:15.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-05-11 19:48:42.000000000","tz":-420},"subject":"manage-shard-ranges: add gap repair option","message":"manage-shard-ranges: add gap repair option\n\nChange-Id: I8883b63be315a7891c4bfbec662c81f218d1f263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35cd5f54e85f9757340fb32d121ebdf3ea701136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35cd5f54e85f9757340fb32d121ebdf3ea701136"}]},"branch":"refs/heads/master"},"ae7dacac5cd17b48d523cc98a20d2fce813e51c9":{"kind":"REWORK","_number":8,"created":"2022-05-19 18:17:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/841143/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/841143/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/841143/8"}}},"commit":{"parents":[{"commit":"bff6e5f8fb9080eacab501a15510f627547e26df","subject":"Merge \"Rip out pickle support in our memcached client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bff6e5f8fb9080eacab501a15510f627547e26df"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-09 16:19:15.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-19 18:14:59.000000000","tz":60},"subject":"manage-shard-ranges: add gap repair option","message":"manage-shard-ranges: add gap repair option\n\nChange-Id: I8883b63be315a7891c4bfbec662c81f218d1f263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae7dacac5cd17b48d523cc98a20d2fce813e51c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae7dacac5cd17b48d523cc98a20d2fce813e51c9"}]},"branch":"refs/heads/master"},"b45b45fa729caa8c36f874658b69f92122dfbead":{"kind":"REWORK","_number":9,"created":"2022-05-23 08:44:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/43/841143/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/841143/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/841143/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/841143/9"}}},"commit":{"parents":[{"commit":"bff6e5f8fb9080eacab501a15510f627547e26df","subject":"Merge \"Rip out pickle support in our memcached client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bff6e5f8fb9080eacab501a15510f627547e26df"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-09 16:19:15.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-23 08:44:16.000000000","tz":60},"subject":"manage-shard-ranges: add gap repair option","message":"manage-shard-ranges: add gap repair option\n\nChange-Id: I8883b63be315a7891c4bfbec662c81f218d1f263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b45b45fa729caa8c36f874658b69f92122dfbead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b45b45fa729caa8c36f874658b69f92122dfbead"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
