)]}'
{"id":"openstack%2Fswift~841612","triplet_id":"openstack%2Fswift~master~Ie8404f0c7e84d3916f0e0fa62afc54f1f43a4d06","project":"openstack/swift","branch":"master","topic":"p-sharder-misplaced-deleted","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-05-19 00:49:14.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-05-18 22:31:12.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"Ie8404f0c7e84d3916f0e0fa62afc54f1f43a4d06","subject":"sharder: ensure that misplaced tombstone rows are moved","status":"MERGED","created":"2022-05-12 15:31:01.000000000","updated":"2022-05-19 00:50:53.000000000","submitted":"2022-05-19 00:49:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"841612-p-sharder-misplaced-deleted","meta_rev_id":"0437299b1e028a2e74a2e519b4ac13aa75a85002","_number":841612,"virtual_id_number":841612,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-19 00:49:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-18 22:31:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-18 22:31:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-12 15:34:19.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-05-12 16:32:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cc3900cf8e44a3d899c991ed61cc43fd45693cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-12 15:31:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66e7fb0b6c47cc851239260255929961fdca8f74","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-12 15:34:19.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"62dc799e72ac85e460f5d389105499a82b65a4d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 16:32:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/836520e055074aba9ae5ce02d6fba250 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/d84b795789234210a1abdebc96a97300 : SUCCESS in 53m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5d003c4797db4ebba100813830e1adf9 : SUCCESS in 55m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0831795e90ba4838a49a4a39e1c06cc5 : SUCCESS in 45m 02s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/266f7ff4480d4f89870dfae8428818d9 : SUCCESS in 17m 39s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/55c564314cde4e0f8acf8e2f475da579 : SUCCESS in 20m 32s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/94569f62a2ac43749a3a60f7f445144c : SUCCESS in 27m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0a7ec45f26734bd3af81c7b57602bd21 : SUCCESS in 8m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cfb25cd6f33d41f5bf335004d0c3bc82 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bd6b9d6ab35546838f03b37393edcfc4 : SUCCESS in 19m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/39bb919ebc2b413f8b5fa9506f2c48ba : SUCCESS in 17m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/72afaf5191e24d2983a893ee4545a6f7 : SUCCESS in 21m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/079b8a5233fe4d8ea1ff59dc566f4baf : SUCCESS in 19m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eaada8573a214377a470e1b175514a2e : SUCCESS in 24m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5cd03aa214bf4bc49fc4ec85b1a5bd51 : SUCCESS in 19m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8a71d16beb89411fb27bd98ce668f723 : SUCCESS in 13m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6b69e9c1e4bf474887a3ebcb928c1648 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/50213ca47942439dbc704ed9f44da1fa : SUCCESS in 15m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/669103da56114b088d4df1fbb2cdc59d : SUCCESS in 5m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1148256bfc6640b2b1f2e64819aac604 : FAILURE in 14m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2feed92d65ce487da9a95d43f03f02ed : SUCCESS in 43m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4f11fbd439242f8bfb9331c0f57b542 : SUCCESS in 47m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/195f2d2137694d27bdd5377d2836eaeb : SUCCESS in 32m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e6a5f8937afe4a8db84f17cece15ac12 : SUCCESS in 42m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c9a19f059c747b0a39240e742d29df5 : SUCCESS in 19m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add5ad0eb2974e898b6210c0cf4d4a19 : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5bfc8979d57b48af9186cf2715409212 : SUCCESS in 27m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e0585595a6a49f2bd8d79a09c926938 : SUCCESS in 42m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7f424a8388650240e901637a37c6ad838db266ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 17:04:54.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11342269843d46019b69e75188d2b82a : TIMED_OUT in 1h 31m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8273de3dfec34ba689e67f653c358def : SUCCESS in 42m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f42275747db24f4fa01aa0dd2ef6cb75 : SUCCESS in 49m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f11f178b1bf48edafac3f3cd9bd608c : SUCCESS in 51m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2fcdceee9c3893024053bfd25fc246bfc68b7f4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-13 07:38:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a51277981fd89ae4df8bd848bde1cdebd95f1c14","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-13 07:41:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e57dd3b8ceabca7c2670cb9e9dc341ad1e69b72c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 08:36:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5230ca31a91d4d988dac0c3c093c09fb : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/5b7bc2e8f17b46c987210dda17fcd98f : SUCCESS in 52m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7bb6a440f5814d8e8da47aa8d5dfb99e : FAILURE in 39m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94e8c41d862f484dab9a4840977fd988 : SUCCESS in 49m 19s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/053c556935fe4d11bde90b0b96c87765 : SUCCESS in 26m 01s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a11d9012133a4f2b9db78079e02177cd : SUCCESS in 24m 37s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b48d209fd790447a9545a261349177a8 : SUCCESS in 27m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c84431e488ea442994dc6df503032737 : SUCCESS in 7m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8c3f6100ba5d48de926a821e1858c585 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/81f263143f6a40d58f9d1653c00da24e : SUCCESS in 19m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bc2089934f3d483fa5bc0a5876e72000 : SUCCESS in 16m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8a1d728154e641fe9f04263dc0285f78 : SUCCESS in 19m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4bfcbafbbdff49bca6f22fc620172013 : SUCCESS in 23m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ddf70b7095bf48b9bf856f6e0fc9acee : SUCCESS in 20m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dee9e3c6999f41e0846a796ea4b82971 : SUCCESS in 20m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bd44e733216149c38cbecf605fa19a70 : SUCCESS in 22m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cb1d9e6de4354d20983f5733b8c396ef : SUCCESS in 16m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/69beb9930c3f401ba96a0f515bc6a224 : SUCCESS in 21m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/839449c2f35142119bff89c50bdff4ce : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0b2ce24065f14234a7ca73bcb5028249 : FAILURE in 19m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eedfdf241ecf494684b9986b055c82d0 : SUCCESS in 50m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/df6a8c4343de4aaba0401ebed360bd51 : SUCCESS in 54m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/46b95e6b70c5419191d435a7fd70f884 : SUCCESS in 40m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/02bcb8980384479e8f10b192b5fb2fc9 : SUCCESS in 40m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/663383e433de4fcc90312157dee26298 : SUCCESS in 16m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30927697599e42859199fe29031e76a6 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bd8d528c9d184c93a807fd9b02045adb : SUCCESS in 27m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5df94c0be42f499a8cd22b72917e7476 : SUCCESS in 44m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b758edcf50fa5342f5aaf3cce78dd0116658d0a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 09:12:24.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac7d04d6ff3a42559b00b93f36e52590 : TIMED_OUT in 1h 31m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/14f69eb157554ba29dedfb2739793394 : SUCCESS in 35m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5d4a730c95a4d5084cd72303cfdd41b : SUCCESS in 1h 01m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/477b7e3188a746979c091470a0253d44 : SUCCESS in 45m 15s","accounts_in_message":[],"_revision_number":2},{"id":"e9878a1768610f2cb5de70d2671a541d310a0cd6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-18 22:31:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e4c6966eb9d1b78ee2305b4c9cec3137cf916965","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 23:28:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e1f65b5f42b4cebbac9835cc24f6bbd : SUCCESS in 54m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f19bb49f6f04be38d72928d16545da9 : SUCCESS in 44m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7b43ffe45d14260be4518b1cf5221f2 : SUCCESS in 55m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/99e133afeb5f42c5a659e13d5bb1c956 : SUCCESS in 51m 45s","accounts_in_message":[],"_revision_number":2},{"id":"a4a419a5aa7556b259568a657ea2260aae4692e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 23:41:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8b3c1882f714829965785d22ed5ead5 : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/0a2d64b455b7480faf755920f568bf45 : SUCCESS in 55m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1fb3177a1151465fa7dc163d1e9d5b33 : SUCCESS in 45m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/827f0cf771784e518f9cc55d2d51fc7e : SUCCESS in 47m 09s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/db07393c156241ec9c41ff07a0ab03bd : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/828c0eafe50943fea9f4bd013cd226e4 : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5866de1e2a094fbc9549899f688865ae : SUCCESS in 24m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d18bd9d009ee4b8f931dff6fa8351fa5 : SUCCESS in 6m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8412a1214cec4df7a0bf7685fca53b01 : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a8a4f374bf8346499995e113e54989e1 : SUCCESS in 19m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a4ad44920b6b49b0843b003d3fc0dfdb : SUCCESS in 19m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/15d9f38f1bfa49488e79284cdabacca1 : SUCCESS in 19m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b027bf9b1a2b4c90845026f7e285d41b : SUCCESS in 18m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6b8ee0fc238b47e4a00894777bb5bb3e : SUCCESS in 20m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c6f04e26af9c4517aba7dac1b04ba99a : SUCCESS in 21m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7fa13e0ae93e4146860dfb59742075e2 : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/03ab99e6b4fd454188d3709e0996b529 : SUCCESS in 18m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e81f57c997d442839e6578b384e01388 : SUCCESS in 17m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/55ff9603b0ff46579e64b322bbe83da8 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fe9a99fa64824abc905d1386ad80188a : FAILURE in 18m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/681ddabff1aa4cf788a803d154e83999 : SUCCESS in 50m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fe016082eba2454580bc98e020ae3707 : SUCCESS in 1h 07m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/000a1c45dba04b7795e83da2cc36c5fd : SUCCESS in 38m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/250193dafd9644e4a355005dba4b4de2 : SUCCESS in 41m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50b33bec52654eff8d19844df6990131 : SUCCESS in 20m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca3f90516f1647b4915eda49ed5e21fe : SUCCESS in 5m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/170f68e2c54743cba041b531d3a27771 : SUCCESS in 26m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9898b8d17e1e47c1befa3d4e657dd139 : SUCCESS in 46m 04s","accounts_in_message":[],"_revision_number":2},{"id":"e9ca28d9c76890b9da8a1a1026c807a728ae2012","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 23:42:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"03d4b58b1e56d182d9f45b01c532837e85ad4859","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 00:49:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e692db176904b6f9b6bf242bc1a16f8 : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/dd427a54b45243a3a410f517260ca2ca : SUCCESS in 52m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cce060489ac544a28b2f909cd31d8678 : SUCCESS in 44m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/768c2a14a417407a9d23c07ead2c03e5 : SUCCESS in 48m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ad4f76cff4aa4e6d93de934ce30139f5 : SUCCESS in 8m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a77e1d2f01774b9490901c091478aa5a : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/75930412b9854739b181db81bb42209b : SUCCESS in 15m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/26cb1440f9bb471ba7ba61a2d200581e : SUCCESS in 14m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/07e2667d875f470483dc69289b27dccc : SUCCESS in 26m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/589e76d176154a5d9e1fee459af4c6d0 : SUCCESS in 27m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/58d8832542b74ff78a4a3e2c7105e532 : SUCCESS in 23m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bedc952b46844b0ab53371ab21d271ce : SUCCESS in 24m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a3dae5d8efaa461a93f1aefb791e0280 : SUCCESS in 27m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/42be2584f8fb4ce49566ff003fec096b : SUCCESS in 21m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/944be8f1da2b4380839dcbb89b1de5c8 : SUCCESS in 25m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/688af61c05fc4654aa4c2f9fd2e87055 : SUCCESS in 6m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5d9bd8209fc4432b8acd3cc4f3f97974 : SUCCESS in 55m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5349bf68ce64e15b4693bd5e4d8da95 : SUCCESS in 56m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6677590c6fd049a0a75b074e351662c4 : SUCCESS in 35m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d0e8075ab49e448195f7702ab41d6f02 : SUCCESS in 47m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/383e6b29611e4ab399c9eabc7cab504b : SUCCESS in 15m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d30bc1baf89448da8d0f8ead3c8fb27 : SUCCESS in 5m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e60a475aa5f445eb9edf7debe0b6c0b2 : SUCCESS in 18m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99cc74264b72456d80b6f36de04fdb4f : SUCCESS in 1h 02m 32s","accounts_in_message":[],"_revision_number":2},{"id":"3dc569fac78dd9a5ff6e1faed38f2d646653a0fc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 00:49:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0437299b1e028a2e74a2e519b4ac13aa75a85002","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 00:50:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d4e5f79428a4b28b4a030b3fda0c5db : SUCCESS in 1m 16s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/13baa7ce7e7948f98436a9d3a77e4ecc : SUCCESS in 41s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/3f1f92b4d4664a96b21df5da91876023 : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"019c955e1cff406a6a7c944d5e66a7bab61ab472","revisions":{"cd3b0d36b3a4862481d8a9306c1ee55c64cb3bb4":{"kind":"REWORK","_number":1,"created":"2022-05-12 15:31:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/12/841612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/841612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/841612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/841612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/841612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/841612/1"}}},"commit":{"parents":[{"commit":"75c5dbc29a34575ac6fab0fb0d150858157a6a2d","subject":"trivial: add comment re sharder misplaced found stat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75c5dbc29a34575ac6fab0fb0d150858157a6a2d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-12 14:55:36.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-12 15:27:20.000000000","tz":60},"subject":"sharder: ensure that misplaced tombstone rows are moved","message":"sharder: ensure that misplaced tombstone rows are moved\n\nThe sharder\u0027s misplaced objects handler should iterate through\nmisplaced objects in name order trying to place them in destination\nshard ranges. The destination shard ranges are consumed in order as\nmisplaced objects are placed in them, and once the destination ranges\nare exhausted no more misplaced objects are placed.\n\nHowever, previously, undeleted objects were yielded in name order\nfirst, followed by deleted objects in name order, meaning that the\ndestination shard ranges could be completely consumed placing\nundeleted objects. Misplaced deleted objects would then not be\nplaced. The same would repeat on every attempt until there were no\nundeleted objects, or placing undeleted objects did not consume all\nthe destination shard ranges. There is no guarantee this would ever\nhappen.\n\nThe fix is to yield objects in name order, as intended, with undeleted\nand deleted objects mingled in the same iterator.\n\nAlso, make debug message more accurate.\n\nChange-Id: Ie8404f0c7e84d3916f0e0fa62afc54f1f43a4d06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd3b0d36b3a4862481d8a9306c1ee55c64cb3bb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd3b0d36b3a4862481d8a9306c1ee55c64cb3bb4"}]},"branch":"refs/heads/master"},"019c955e1cff406a6a7c944d5e66a7bab61ab472":{"kind":"REWORK","_number":2,"created":"2022-05-13 07:38:27.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/12/841612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/841612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/841612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/841612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/841612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/841612/2"}}},"commit":{"parents":[{"commit":"75c5dbc29a34575ac6fab0fb0d150858157a6a2d","subject":"trivial: add comment re sharder misplaced found stat","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/75c5dbc29a34575ac6fab0fb0d150858157a6a2d"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-12 14:55:36.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-13 07:38:07.000000000","tz":60},"subject":"sharder: ensure that misplaced tombstone rows are moved","message":"sharder: ensure that misplaced tombstone rows are moved\n\nThe sharder\u0027s misplaced objects handler should iterate through\nmisplaced objects in name order trying to place them in destination\nshard ranges. The destination shard ranges are consumed in order as\nmisplaced objects are placed in them, and once the destination ranges\nare exhausted no more misplaced objects are placed.\n\nHowever, previously, undeleted objects were yielded in name order\nfirst, followed by deleted objects in name order, meaning that the\ndestination shard ranges could be completely consumed placing\nundeleted objects. Misplaced deleted objects would then not be\nplaced. The same would repeat on every attempt until there were no\nundeleted objects, or placing undeleted objects did not consume all\nthe destination shard ranges. There is no guarantee this would ever\nhappen.\n\nThe fix is to yield objects in name order, as intended, with undeleted\nand deleted objects mingled in the same iterator.\n\nAlso, make debug message more accurate.\n\nChange-Id: Ie8404f0c7e84d3916f0e0fa62afc54f1f43a4d06\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/019c955e1cff406a6a7c944d5e66a7bab61ab472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/019c955e1cff406a6a7c944d5e66a7bab61ab472"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
