)]}'
{"id":"openstack%2Fswift~841989","triplet_id":"openstack%2Fswift~master~I45051d9613c8a245db51e0578bdec3ed88c3b358","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-01-28 17:52:46.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-05-18 23:10:57.000000000","reason":"Tim Burke replied on the change"}},"hashtags":[],"change_id":"I45051d9613c8a245db51e0578bdec3ed88c3b358","subject":"Distinguish workers by their args","status":"ABANDONED","created":"2022-05-16 18:41:27.000000000","updated":"2025-01-28 17:52:46.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"da557ed0001b1239a93ba101c89d9a9898add788","_number":841989,"virtual_id_number":841989,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-16 20:14:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-16 19:24:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-16 20:14:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-17 14:41:17.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"b0ff9f2db77fd70fd49f340930ff21f5ca6c8bdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-16 18:41:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74ebabfc0195b0a8f2fb0501afac6448543d6b76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 19:24:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5de8bfef15cd4ae68a50efe76c962ec8 : SUCCESS in 32m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eaaa831050e4456ab423816a1b119f98 : SUCCESS in 32m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa621200b2bd4ff380e77bb773553197 : SUCCESS in 40m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c55038d67d44f7cb8b574394f59dc28 : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":1},{"id":"8c4984acf8fdabd65763891c2b5ad4fae96a9cb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 20:14:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8a3e6b89be7b4ec3a741f3b36428fe06 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/105161a20acc42efa1ca667cc81b73bf : SUCCESS in 10m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/b75bf72b2e1a44238c6f6aa0179c6f4f : SUCCESS in 1h 08m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1eacb90b62d94962b2c4bdb1d2b0bbe7 : SUCCESS in 44m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1516e2fdbc0146e1a16237872cf462e4 : SUCCESS in 48m 08s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b2c9839fb4f3440a9ebc27322951d47e : SUCCESS in 27m 25s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/174d2183b0004899bfe59ecc482beb3a : SUCCESS in 27m 25s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0585ea1844ee4c9ea8b306bc6f081563 : SUCCESS in 24m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/badbbf4acaba491491a4194417e4a050 : SUCCESS in 7m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3aa11b255714420a994a4ef2231c70f2 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/48f6556b5a9b47b4946d5cae76c6008a : FAILURE in 18m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1876114903ee44f6a3f0b4f0e00ca689 : TIMED_OUT in 1h 31m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ffae50e1f6dc4e8d9f59b6202644006b : SUCCESS in 16m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ed824f08ffcd4afdbbbe8ee699515c6e : SUCCESS in 23m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2f3ffd71e1934e12b8ffcc8efda90db7 : SUCCESS in 23m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fd7d801e2e594abcaf9e6f6dd38e9db0 : SUCCESS in 22m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/92a9e1a3eb1b43b584f4ca1e80b3505d : SUCCESS in 25m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1417a780d9684d83b9ef92f183457066 : SUCCESS in 23m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/63b66e02681d4a1b9e946cd733b61717 : SUCCESS in 22m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/926e4d4c01404b7791e3291aee9fc663 : SUCCESS in 7m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f21cf9e8fa9a4806a30c4e1329335bd3 : FAILURE in 28m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/992ff85349a340e8bef81290eaabd5f0 : SUCCESS in 51m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/dfef3122e391495da9432ef73beb5459 : SUCCESS in 46m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7ce3eaf1f1fb40f5b6e4f8b1adce96c0 : SUCCESS in 41m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2c02cbe4bf8d4a55901045b4e2285ec8 : SUCCESS in 1h 00m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e176063e2af24582a5b476b1becd0664 : FAILURE in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a13801c989794a76a4302496bd6c14bb : SUCCESS in 7m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f32d6557937491db80eaf041f0f3492 : SUCCESS in 26m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de684d49e55a46e0857169dde1049d44 : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":1},{"id":"de8b38dfafd79f3c6dff7347140f95769e792f6f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-17 14:41:17.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b3727b7b774d9a997c315891da4769505b008322","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-18 23:10:57.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"da557ed0001b1239a93ba101c89d9a9898add788","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-28 17:52:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30ae33e777ebcda378257411ff8aae90803c09b4","revisions":{"30ae33e777ebcda378257411ff8aae90803c09b4":{"kind":"REWORK","_number":1,"created":"2022-05-16 18:41:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/89/841989/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/841989/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/841989/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/841989/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/841989/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/841989/1"}}},"commit":{"parents":[{"commit":"eeb5533457871eb6ebe6f5529f936a605384425a","subject":"Merge \"memcached: Give callers the option to accept errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eeb5533457871eb6ebe6f5529f936a605384425a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-05-16 16:54:26.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-05-16 18:38:44.000000000","tz":-420},"subject":"Distinguish workers by their args","message":"Distinguish workers by their args\n\nAt least, if setproctitle is available. Note that not all daemons are\nupdated, as many manage their own forking.\n\nChange-Id: I45051d9613c8a245db51e0578bdec3ed88c3b358\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30ae33e777ebcda378257411ff8aae90803c09b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30ae33e777ebcda378257411ff8aae90803c09b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
