)]}'
{"id":"openstack%2Fswift~842040","triplet_id":"openstack%2Fswift~master~I7012e72952d83b4abc2a673c7c8a5e64de375b3d","project":"openstack/swift","branch":"master","topic":"ring-writer","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-05-27 03:22:20.000000000","reason":"Matthew Oliver replied on the change"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-06-08 23:29:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7012e72952d83b4abc2a673c7c8a5e64de375b3d","subject":"ring v2 serialization: more test coverage follow up","status":"ABANDONED","created":"2022-05-17 06:24:28.000000000","updated":"2022-06-08 23:29:12.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"13eec77980580bb95f24342bd97d3b53c528316d","_number":842040,"virtual_id_number":842040,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-17 07:22:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-17 07:22:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-25 21:25:32.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"aebdf21f54b66c1020e225c720ee0c3966a41b33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-17 06:24:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54f1080b9935a5b68514a04bac91b2b3447f3c93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 07:22:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dce2e33ba304b9696fb327dc14e039b : SUCCESS in 9m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/4f507834bb284bfe930edc6025746bb3 : SUCCESS in 54m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d73daa0daa84e3bad837638961ef3ef : SUCCESS in 47m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56a6e95707444b47acdd6039e3f37e32 : SUCCESS in 39m 47s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/07a3a7ae4c37417ebf14734e220cd385 : SUCCESS in 22m 45s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/72cd7dc7ecde4a1d86f5d93ce67a0d28 : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c64e56eec45e464586781c8849a29e21 : SUCCESS in 23m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5d6987f110534c829032329525f46baf : SUCCESS in 6m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a7883cb8919c45b39287b9311a07aef0 : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b333b8d64ccd49aea1d4dd1120bd09fe : SUCCESS in 20m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/22812ea6af1941d2a757ad1f401ff7a3 : SUCCESS in 11m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6f7a4a6025824897a610746ffe790164 : SUCCESS in 13m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/62c7bbd4999d42568054565452cd8456 : SUCCESS in 21m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ecee3ef9c90649fe8aaece2aad3c42d4 : SUCCESS in 17m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d789b392a49e45ecaabd463daf6405af : SUCCESS in 21m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e04a75fcc8e541ee8e4510bc18a2404a : SUCCESS in 22m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9603925481314c6e9bf7058594ecb1ba : SUCCESS in 15m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8ce8e3b63b124855bb73e167cdc1f215 : SUCCESS in 19m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/681e872234d9451984ee2b58cb290453 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0977c48ce81d48ecb3167d9fcf6fe2f5 : FAILURE in 14m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e4c19fdf19b249c19b0a43fe2c2f157d : SUCCESS in 52m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc58e49e26804dc7831ce545612b3bf6 : SUCCESS in 55m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f0e2b8031be044ffa345c9de6612b23c : SUCCESS in 31m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bf22663257034c249ca540e68b2fef84 : SUCCESS in 38m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc88134c62e1401bbca6822710835044 : SUCCESS in 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/583a106a79634bcdaecb5d450c4cf278 : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ca81d5301684a4e8dbe31a2186be2db : SUCCESS in 23m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ec2ccf8ec904c9fa06f06f70ac4113b : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7357bb6158be11550a24f75d71ac58e41885c198","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 07:58:01.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/030f27b87ac14d99baef8d807e2228ef : TIMED_OUT in 1h 31m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/779f43b708514886b4c30b25ed1c3ea3 : SUCCESS in 41m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd56aa9183ec484da4f577410e00f8b3 : SUCCESS in 55m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/56266c6747434c7295b28526a2cdd54f : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":1},{"id":"efa69e7a50f02a8aeb3d3f3e41b8f62c0b0b50a8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-25 21:25:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d61b73f5eaccdaf27314a5b021ebb4546728c8bf","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-27 03:22:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13eec77980580bb95f24342bd97d3b53c528316d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-08 23:29:12.000000000","message":"Abandoned\n\nSquashed into parent, minus the unknown checksum handling.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5517287fdb4ef51ca4ca9ea18c8a444604cc967e","revisions":{"5517287fdb4ef51ca4ca9ea18c8a444604cc967e":{"kind":"REWORK","_number":1,"created":"2022-05-17 06:24:28.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/40/842040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/842040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/842040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/842040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/842040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/842040/1"}}},"commit":{"parents":[{"commit":"b3da7c22b17209960b62ef15476651a9754a5cda","subject":"ring: Introduce a v2 ring format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3da7c22b17209960b62ef15476651a9754a5cda"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-05-17 06:23:22.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-05-17 06:23:22.000000000","tz":600},"subject":"ring v2 serialization: more test coverage follow up","message":"ring v2 serialization: more test coverage follow up\n\nFeel free to squash this into the parent\n\nChange-Id: I7012e72952d83b4abc2a673c7c8a5e64de375b3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5517287fdb4ef51ca4ca9ea18c8a444604cc967e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5517287fdb4ef51ca4ca9ea18c8a444604cc967e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
