)]}'
{"id":"openstack%2Fswift~842617","triplet_id":"openstack%2Fswift~master~I2248b43769deb5e79423369349f06dbf177304a3","project":"openstack/swift","branch":"master","topic":"rate_buffer","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-05-20 16:37:42.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2248b43769deb5e79423369349f06dbf177304a3","subject":"Reload backend ratelimit rate buffer per device","status":"ABANDONED","created":"2022-05-19 20:19:02.000000000","updated":"2022-05-20 16:37:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6ac0ebf6f58596beaa3957551a79f20043e6b032","_number":842617,"virtual_id_number":842617,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-05-19 23:48:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-19 21:25:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-19 23:48:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2d5b681a1b61d6f86e79aa2cd0967d9a954a12a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-19 20:19:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73ff0036494ba11f9ba86a13336e22cebc750f1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 21:25:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5b482830952c4b1b9ef13c3e5df7be7e : SUCCESS in 39m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5defec099424f319c0727891e5161dc : SUCCESS in 1h 04m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8bdfaa1b6c444fa0983d8a9138c567e1 : SUCCESS in 48m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/30d64a680dec47beb0d1c51fc34e3dd4 : SUCCESS in 47m 45s","accounts_in_message":[],"_revision_number":1},{"id":"77ec07c553f4cb7e2d79ad1496c8cae32d282123","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 23:48:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/695995953fdf43f9a1d2fd4238cc2948 : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/a1d162c3f85749d68778102fd66c8526 : SUCCESS in 1h 09m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/67bd84d9f316409dbba82f1cfaf87b3b : SUCCESS in 1h 00m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27f5776dd09948a5a1ef798e6cedcb14 : SUCCESS in 49m 50s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0373cdf17e5443cbbd256496948eae96 : FAILURE in 6m 24s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5adf1473637749038e6b31485c5382d3 : FAILURE in 5m 26s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d81b09d72d4a44528174a376ed59f7c1 : FAILURE in 6m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1ae40d1bba40444f99e721f10e454ad2 : SUCCESS in 7m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/de741b106a4f42b49f51e7bb98f45c66 : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1d8e8e137ca644efa5fd5534b6d8bc6b : SUCCESS in 13m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d28eebb6a23a4322af6ad1d778f66e8e : FAILURE in 3m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/57a26c89bcce4c7ba3de876bf780c28f : SUCCESS in 14m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71c78f671db943daa5e20abfde583de7 : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2517741073304de4ad1059f8f5c74af0 : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/44162927d7484f60921ecdc57c0d9a73 : SUCCESS in 19m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c4b590426eae46ea90b61d7312975b80 : SUCCESS in 12m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/011e0da7bd3d42ea933ac7c6b205b40c : SUCCESS in 15m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/18150f50a03549a58b373fcf31a7af39 : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f293eb4d46284defb1fef62611759d7c : SUCCESS in 6m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6e34685e3d7e4984928cd45631555653 : SUCCESS in 19m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/201da554ba1d43459dc0ba8ca77271bc : SUCCESS in 37m 15s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca85769f7bed4b54bbd82b485d952f11 : SUCCESS in 1h 03m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7155c7c883ff49b78ec263a264e2d746 : SUCCESS in 44m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8bc87ec5f2a04cf78a4682cb8dd39414 : SUCCESS in 28m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fba25dcf2cdb42dda5c262289e1a65f7 : SUCCESS in 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a549d513e0804af797699d1374a5b9dd : SUCCESS in 6m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8a69623c020344898ab4a17fd3a01649 : FAILURE in 8m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aaab5352e2fe411f918124fdace3638d : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6ac0ebf6f58596beaa3957551a79f20043e6b032","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-20 16:37:42.000000000","message":"Abandoned\n\nsquashed squashed https://review.opendev.org/c/openstack/swift/+/840531","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d56699ba6f3a28a49e7fdb041d4ed7f4ae695be7","revisions":{"d56699ba6f3a28a49e7fdb041d4ed7f4ae695be7":{"kind":"REWORK","_number":1,"created":"2022-05-19 20:19:02.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/17/842617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/842617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/842617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/842617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/842617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/842617/1"}}},"commit":{"parents":[{"commit":"897e8d1741960aa3c924d72b7e84a7357d22c1d3","subject":"backend ratelimit: support reloadable config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/897e8d1741960aa3c924d72b7e84a7357d22c1d3"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-05-19 20:06:02.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2022-05-19 20:17:48.000000000","tz":-300},"subject":"Reload backend ratelimit rate buffer per device","message":"Reload backend ratelimit rate buffer per device\n\nChange-Id: I2248b43769deb5e79423369349f06dbf177304a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d56699ba6f3a28a49e7fdb041d4ed7f4ae695be7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d56699ba6f3a28a49e7fdb041d4ed7f4ae695be7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
