)]}'
{"id":"openstack%2Fswift~842909","triplet_id":"openstack%2Fswift~master~I0582122ae7c50722c2ae8250148a334315560b44","project":"openstack/swift","branch":"master","topic":"ratelimit-test","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-05-24 05:16:23.000000000","reason":"Matthew Oliver replied on the change"}},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-05-24 05:16:23.000000000","reason":"Matthew Oliver replied on the change"}},"hashtags":[],"change_id":"I0582122ae7c50722c2ae8250148a334315560b44","subject":"SAIO: add backend ratelimit middleware and config file","status":"NEW","created":"2022-05-23 05:06:57.000000000","updated":"2022-05-24 06:50:02.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b2835edc67c5d030d7ea615962289140ccf13b21","_number":842909,"virtual_id_number":842909,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-24 06:50:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-23 06:04:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-23 08:58:56.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"d65fab12822c6d2cca26c99f7d09dd07c547a15f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-23 05:06:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34a0e9e7979a89542e57ab2ae767a39a8beec669","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 06:04:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d144486ce4e4ffebd4f0971a34bd417 : SUCCESS in 8m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8041b6a515304b138510ea8f6c409ed4 : SUCCESS in 5m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6bf937614af24da1ab46c41b5fad25a3 : SUCCESS in 14m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8426326979144e46902322ad7e2a6e95 : SUCCESS in 50m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6932dda0d18846fa87f48997da70b161 : SUCCESS in 53m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c762d80ff6c72ced17d347509491969daafb34ee","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-23 08:58:56.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1713279e148740c4ec7318e74c67948197b65b30","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-24 05:16:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42ac8c63ba4b87931d9e56006358d4ce29e79b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-24 05:52:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b2835edc67c5d030d7ea615962289140ccf13b21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 06:50:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843affd6207d4d64a3da24e26cca3231 : SUCCESS in 6m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6b50529482f54170a891a9514cf0a645 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c333883a8dc548fda14c303e78e45d1e : SUCCESS in 16m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e18e6ff965f249b1a697a321e9661f79 : SUCCESS in 50m 41s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cf47607d09e24032a90949c88356c14a : SUCCESS in 55m 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6881f9e86962cc5e57ecef55c4cfd27c40f253de","revisions":{"5407b826c3c948f83a57f3afd1b6bec68701cc68":{"kind":"REWORK","_number":1,"created":"2022-05-23 05:06:57.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/09/842909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/842909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/842909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/842909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/842909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/842909/1"}}},"commit":{"parents":[{"commit":"7730d4806c4dc4bb9a5a630f2f84e05229a6d726","subject":"backend ratelimit: support reloadable config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7730d4806c4dc4bb9a5a630f2f84e05229a6d726"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-05-23 05:04:43.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-05-23 05:04:43.000000000","tz":600},"subject":"SAIO: add backend ratelimit middleware and config file","message":"SAIO: add backend ratelimit middleware and config file\n\nThis patch added the backend_ratelimit middleware and the\nbackend-ratelimit.conf to the SAIO configuration.\n\nChange-Id: I0582122ae7c50722c2ae8250148a334315560b44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5407b826c3c948f83a57f3afd1b6bec68701cc68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5407b826c3c948f83a57f3afd1b6bec68701cc68"}]},"branch":"refs/heads/master"},"6881f9e86962cc5e57ecef55c4cfd27c40f253de":{"kind":"REWORK","_number":2,"created":"2022-05-24 05:52:50.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/09/842909/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/842909/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/842909/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/842909/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/842909/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/842909/2"}}},"commit":{"parents":[{"commit":"7730d4806c4dc4bb9a5a630f2f84e05229a6d726","subject":"backend ratelimit: support reloadable config file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7730d4806c4dc4bb9a5a630f2f84e05229a6d726"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-05-23 05:04:43.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-05-24 05:51:54.000000000","tz":600},"subject":"SAIO: add backend ratelimit middleware and config file","message":"SAIO: add backend ratelimit middleware and config file\n\nThis patch added the backend_ratelimit middleware and the\nbackend-ratelimit.conf to the SAIO configuration.\nWe use a different backend-ratelimit config for each \"node\".\n\nChange-Id: I0582122ae7c50722c2ae8250148a334315560b44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6881f9e86962cc5e57ecef55c4cfd27c40f253de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6881f9e86962cc5e57ecef55c4cfd27c40f253de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
