)]}'
{"id":"openstack%2Fswift~843100","triplet_id":"openstack%2Fswift~master~I308ebf72441e458dbd0da0aa22ccb6aaafd6f922","project":"openstack/swift","branch":"master","topic":"p-repair-gaps","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-05-24 05:33:34.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I308ebf72441e458dbd0da0aa22ccb6aaafd6f922","subject":"Probe: check gap repair acceptor in shrunk shardrange","status":"NEW","created":"2022-05-24 05:33:05.000000000","updated":"2022-05-31 05:44:04.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d064cc40bc3b87ca823dccbe5c7441b75e3a65f9","_number":843100,"virtual_id_number":843100,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-31 05:44:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-24 05:33:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-24 05:33:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"767b757391c95e05f90f9cb89936377e846e15ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-24 05:33:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81ee597c161d04098c3fda0119a09dea86e815d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 05:33:29.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 843100,1","accounts_in_message":[],"_revision_number":1},{"id":"55bb929592535a058fa9e6c396aad051cdcad3a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 05:33:34.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 843100,1","accounts_in_message":[],"_revision_number":1},{"id":"af8271d98d7b63ddc1b5c460b76cf3aaeee850ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-05-31 04:36:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"bee729bba4238e0aafe09b8e6d9d337dbfe0a5e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 05:36:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8534074a2b644549a510b5387b25cdab : SUCCESS in 28m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af1dbbfdc3da48f6a437d4e67a47fab7 : SUCCESS in 29m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbcc250344404405a600f84bb0cdaf9e : SUCCESS in 57m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c55a393b96b40f0bedf459eafdea840 : SUCCESS in 48m 47s","accounts_in_message":[],"_revision_number":2},{"id":"d064cc40bc3b87ca823dccbe5c7441b75e3a65f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 05:44:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/216b810641ad4b24ab1dee7b32254a3d : SUCCESS in 10m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c922dc4991944fb7b6bd65a2ac1c0d45 : SUCCESS in 52m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3004fc55d28a4f71937e8eb315cd3c48 : SUCCESS in 1h 00m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f08b92bab79435287f3902e6ef4cbc3 : SUCCESS in 1h 06m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a38063c59b4574b50f3bd5086cc090 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2d2ee8c5350f61cd70048e100601e89a1f5f328c","revisions":{"041756be23e4e81cd1f8216518e5d3cdb28d7372":{"kind":"REWORK","_number":1,"created":"2022-05-24 05:33:05.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/00/843100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/843100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/843100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/843100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/843100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/843100/1"}}},"commit":{"parents":[{"commit":"ae7dacac5cd17b48d523cc98a20d2fce813e51c9","subject":"manage-shard-ranges: add gap repair option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae7dacac5cd17b48d523cc98a20d2fce813e51c9"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-05-24 05:30:42.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-05-24 05:30:42.000000000","tz":600},"subject":"Probe: check gap repair acceptor in shrunk shardrange","message":"Probe: check gap repair acceptor in shrunk shardrange\n\nFor completeness sake, check the shrunk shard container at the end of\nthe existing managed gap probe test, that the expected acceptor was\npulled in and used.\n\nChange-Id: I308ebf72441e458dbd0da0aa22ccb6aaafd6f922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/041756be23e4e81cd1f8216518e5d3cdb28d7372"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/041756be23e4e81cd1f8216518e5d3cdb28d7372"}]},"branch":"refs/heads/master"},"2d2ee8c5350f61cd70048e100601e89a1f5f328c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-31 04:36:07.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/00/843100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/843100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/843100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/843100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/843100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/843100/2"}}},"commit":{"parents":[{"commit":"b45b45fa729caa8c36f874658b69f92122dfbead","subject":"manage-shard-ranges: add gap repair option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b45b45fa729caa8c36f874658b69f92122dfbead"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-05-24 05:30:42.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-05-31 04:36:07.000000000","tz":0},"subject":"Probe: check gap repair acceptor in shrunk shardrange","message":"Probe: check gap repair acceptor in shrunk shardrange\n\nFor completeness sake, check the shrunk shard container at the end of\nthe existing managed gap probe test, that the expected acceptor was\npulled in and used.\n\nChange-Id: I308ebf72441e458dbd0da0aa22ccb6aaafd6f922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d2ee8c5350f61cd70048e100601e89a1f5f328c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d2ee8c5350f61cd70048e100601e89a1f5f328c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
