)]}'
{"id":"openstack%2Fswift~843431","triplet_id":"openstack%2Fswift~master~Ie14658b39c48fba46da9ec2edbe80d0dedbd4a1b","project":"openstack/swift","branch":"master","topic":"p-error-limit-test","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-05-26 14:08:00.000000000","reason":"Clay Gerrard replied on the change"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-05-27 23:52:53.000000000","reason":"Tim Burke replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie14658b39c48fba46da9ec2edbe80d0dedbd4a1b","subject":"proxy-server: unit test error limiting reset","status":"NEW","created":"2022-05-26 11:31:02.000000000","updated":"2022-05-27 23:52:53.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"cd36c5422b0b9a3159d09b7fc6b352764e644112","_number":843431,"virtual_id_number":843431,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-26 13:14:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-26 12:19:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-26 12:19:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-26 14:08:00.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2022-05-27 23:52:53.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"8688895af4151587cfc3110d91d081eb843775eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-26 11:31:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b7c89ec4cc896f66892222d80059900e4dc3ddf","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-26 11:31:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe8ed2eb5083c620bd91722ad736bb1bb37753bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 12:19:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/20cc6aad506d4543aefb21dba86fe30a : SUCCESS in 29m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9535af257e8449e6a27953a986c1af53 : SUCCESS in 28m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0464dc91940940ac9c4ffc1ee1a2a615 : SUCCESS in 43m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/164d7e75e9a94290bc7521b491981d26 : SUCCESS in 45m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2afe8561f17e874b9b561bcc30c95c1f2ece5465","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 12:19:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb33d32082f442bdb4a4760974cad8ef : SUCCESS in 9m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f77ef58a4624487a3d1b9f5884faf15 : SUCCESS in 46m 37s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d45197d60b124aabad22c4d4ae8381fb : SUCCESS in 22m 55s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0546cc2f83fc42da8244e355cff318c6 : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2483810db85d437baaa10c75bb05745c : SUCCESS in 23m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cd1e80ca9bf94e589f5464eb25ab0816 : SUCCESS in 7m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/86e8ac1fb840452fba4f5c4be5554158 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/04bf98a081ef4474b8622c5966f139b6 : SUCCESS in 16m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d8a5c7c10a649eb8037be81c3c871e6 : SUCCESS in 18m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aecc6499d1cb4270864921e1aaba3206 : SUCCESS in 15m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6da26a85872b474fbda4a513de0d13ec : SUCCESS in 23m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8cb82af654f1441da4ffbc57649c9c9d : SUCCESS in 20m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1e63da49550149b181a12bb74a2e69d9 : SUCCESS in 21m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/864042939d574ab7a843be64a65daab2 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e91af21f48e1466f91ddd531229eaae3 : SUCCESS in 21m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6e71e25c18074a9696da69311413f9de : SUCCESS in 20m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d1177c0c61840369e8235bb5074bb59 : SUCCESS in 33m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c3a0787e0a2a42568cd590aa1570e5ce : SUCCESS in 41m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/593d6aede6e74024986535f8d056522f : SUCCESS in 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b004a23755244fbfa2ef668f1e099593 : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7c180fe72e0d404d98b61ed54a857d9d : SUCCESS in 26m 32s","accounts_in_message":[],"_revision_number":1},{"id":"df2bb7d5969a4c55b0cdbaba30f2c5a7ea073d47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-05-26 12:19:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5dc098acef8382cd66ba630d850fb0d9a3014a35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 13:14:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe4bc0f7a07d4100a53e789b1b50412f : SUCCESS in 9m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6a625776ef4442cb0ed9a714fd109a6 : SUCCESS in 51m 21s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/11c466293dea40479356cfe79b245287 : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/66b2467768d94406bfc2aa7a799092f2 : SUCCESS in 24m 22s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b39448c14e0e4b52b0b8693f95338394 : SUCCESS in 20m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9d6e8afc09294d32951da4dacceb4852 : SUCCESS in 8m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cdf1a21f69ec43068c4386f887e85f94 : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/611174826faf4a01a2d2300458a93f03 : SUCCESS in 15m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f1840e43219a423680b8bbb10eee54a5 : SUCCESS in 14m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cbc475c884b44638800cfcaa8dbae4fe : SUCCESS in 16m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5412d72ee1474bdb8956adcbccd0ae9a : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/880c4e37d4834d3084e258944643d71f : SUCCESS in 20m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/95532b0c50774fffa121b504c2aa1359 : SUCCESS in 22m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a39171f98a044b76af7cfb2c8e06d023 : SUCCESS in 18m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1ab5410871b643db90c4fcb517f274b5 : SUCCESS in 22m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/22921e86e2e64176b763798e7c54c418 : SUCCESS in 21m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/601ad4c610a24da6b1741acd34c6a4fb : SUCCESS in 40m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dadc17734bc2457f8fb0094f4649ccd2 : SUCCESS in 39m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9d014be44a64f66878d6cb6ba63783d : SUCCESS in 19m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13c575da66b34c1bbee4c1e42c24eaec : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f1fc806822434c369d94e0668d2f63c9 : SUCCESS in 23m 54s","accounts_in_message":[],"_revision_number":2},{"id":"0e280db75104dd9a224144516e9e99a60a4da948","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 13:21:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2754de779720443f912146ae6b1a65f3 : SUCCESS in 53m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/800f345a86d541278e968b14215b9392 : SUCCESS in 44m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf102770bfa9493a9fe6e4d98c1aa416 : SUCCESS in 56m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c0639e606d8459ca03cadcb03437c89 : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":2},{"id":"553f4b76f1ac2d3dc2c30f2a65dc7567af954bad","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-26 14:08:00.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cd36c5422b0b9a3159d09b7fc6b352764e644112","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-05-27 23:52:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cc89b8e3f32d1fa03acdf75b7499b547c747fa9e","revisions":{"744ae585315b4566e4cdc3f87f6adab2e4ce6554":{"kind":"REWORK","_number":1,"created":"2022-05-26 11:31:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/843431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/843431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/843431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/843431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/843431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/843431/1"}}},"commit":{"parents":[{"commit":"238dc035350610e6ca18b902f12890736fe0e095","subject":"CI: constrain py36 deps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/238dc035350610e6ca18b902f12890736fe0e095"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-26 11:24:49.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-26 11:24:49.000000000","tz":60},"subject":"proxy-server: unit test error limiting reset","message":"proxy-server: unit test error limiting reset\n\nAdd a unit test to cover the resetting of error limited node state\nafter the error_suppression_interval has passed.\n\nChange-Id: Ie14658b39c48fba46da9ec2edbe80d0dedbd4a1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/744ae585315b4566e4cdc3f87f6adab2e4ce6554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/744ae585315b4566e4cdc3f87f6adab2e4ce6554"}]},"branch":"refs/heads/master"},"cc89b8e3f32d1fa03acdf75b7499b547c747fa9e":{"kind":"REWORK","_number":2,"created":"2022-05-26 12:19:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/843431/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/843431/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/843431/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/843431/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/843431/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/843431/2"}}},"commit":{"parents":[{"commit":"238dc035350610e6ca18b902f12890736fe0e095","subject":"CI: constrain py36 deps","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/238dc035350610e6ca18b902f12890736fe0e095"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-26 11:24:49.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-05-26 12:19:40.000000000","tz":60},"subject":"proxy-server: unit test error limiting reset","message":"proxy-server: unit test error limiting reset\n\nAdd a unit test to cover the resetting of error limited node state\nafter the error_suppression_interval has passed.\n\nChange-Id: Ie14658b39c48fba46da9ec2edbe80d0dedbd4a1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc89b8e3f32d1fa03acdf75b7499b547c747fa9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc89b8e3f32d1fa03acdf75b7499b547c747fa9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
