)]}'
{"id":"openstack%2Fswift~844338","triplet_id":"openstack%2Fswift~master~I3de79e174834c7977c05a97ef0e4c26572dbfa98","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-06-02 17:36:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3de79e174834c7977c05a97ef0e4c26572dbfa98","subject":"Rename bind_ip to ring_ip in a bunch of places","status":"ABANDONED","created":"2022-06-01 20:53:41.000000000","updated":"2022-06-02 17:36:13.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"e7a489205bc08f08e2a0ba00ac444cc42bdd2218","_number":844338,"virtual_id_number":844338,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-01 22:28:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2022-06-02 17:16:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-01 22:28:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 17:16:17.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"bcf37110fb933f0cb40e6b66cea3deb87a57fdd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-01 20:53:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"824b7bde2b09a0c2c6d28c6f2f95f7016e579ff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 22:28:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f209ff8dafb455eafddd70ecd455397 : SUCCESS in 7m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/3a65e197a1c04dcb9486eb4921027b0b : SUCCESS in 45m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5be04c7929854bd9bd526bde26db016b : SUCCESS in 47m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00ea5bba24c64848b1b8681eeae155e7 : SUCCESS in 49m 08s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cc27ae75d5d649f1beb9032fdfd9cf85 : SUCCESS in 16m 44s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/bcbac0a283184ccdbe62efaa24f30f03 : SUCCESS in 17m 37s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/941d23374ff54cfc84edb89dbfe2d13c : SUCCESS in 16m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a89552c67a44e00b08d8bbedd1ccc74 : SUCCESS in 7m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dc993ec6dc4b45e193b22cd388dba482 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4e4a6ac85fec4ef6b7aceda9548ab847 : TIMED_OUT in 1h 31m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6a3a1f421fd543f48a48b4f60cd23ffa : TIMED_OUT in 1h 31m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/02aaeec6889a4a97a3f353cd70cec118 : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/082e3ea0cc974b12845996fba8c16518 : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/02a5a4a1f0204ed781ed88ab919b8ed9 : SUCCESS in 15m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e1a522f551b9450582afc4027a70f463 : SUCCESS in 22m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/40eb0d00214a4aa8820cd330033e204a : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/733af4afe38844d3b5554ca2e0f26756 : SUCCESS in 14m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/11d85b7435c94e18922444fe572c98db : SUCCESS in 17m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/81dc2a682dbe4f739d663bab3147a817 : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d6168a671fd94792a6a549044f88e68d : SUCCESS in 18m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/37676742b740403898fb556ad363ca0a : FAILURE in 45m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/66247cb1fb49473a9467d13ac4e3394b : FAILURE in 54m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2b4d9dbde9c640639ccf190547ecce8a : SUCCESS in 41m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/107c151400884c93b6326ad9b3d5efe0 : SUCCESS in 29m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad76a0b1709d4b6380ee381741ae4049 : TIMED_OUT in 41m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8e35e27f44347968a985b45d3948274 : SUCCESS in 5m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e51fb1949aa6444db66f8bcfb8dbc446 : SUCCESS in 17m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b872995efbd4bd4a3286e03f102ac38 : SUCCESS in 38m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d6ff5c151d335a2e0bd7624e2bcca0b2f993f4a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 22:29:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0161274612154114a66092b69c701bf9 : TIMED_OUT in 1h 31m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c56620695cf41329cc41905dcdf2ee4 : TIMED_OUT in 1h 31m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4a54cc5b27c403aaf2cb4a3882e9e09 : SUCCESS in 28m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5e6254b4f964f6cbb5d735d10429d3a : SUCCESS in 53m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6d492fff483c8ddc3468c5961d480da7de3ab5e0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-06-02 17:16:17.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e7a489205bc08f08e2a0ba00ac444cc42bdd2218","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-06-02 17:36:13.000000000","message":"Abandoned\n\nsquashed https://review.opendev.org/c/openstack/swift/+/840372","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3cb5d948bbea03401020358c06acde92c864abc9","revisions":{"3cb5d948bbea03401020358c06acde92c864abc9":{"kind":"REWORK","_number":1,"created":"2022-06-01 20:53:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/38/844338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/844338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/844338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/844338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/844338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/844338/1"}}},"commit":{"parents":[{"commit":"56f0479a90b5b70207721704fe27da71b7aa57c0","subject":"Add ring_ip option to object services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/56f0479a90b5b70207721704fe27da71b7aa57c0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-01 20:53:22.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-01 20:53:25.000000000","tz":-420},"subject":"Rename bind_ip to ring_ip in a bunch of places","message":"Rename bind_ip to ring_ip in a bunch of places\n\nBasically, wherever we\u0027re using the IP for ring lookups instead of\nsocket binding, we probably ought to use the appropriate name.\n\nChange-Id: I3de79e174834c7977c05a97ef0e4c26572dbfa98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cb5d948bbea03401020358c06acde92c864abc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cb5d948bbea03401020358c06acde92c864abc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
