)]}'
{"id":"openstack%2Fswift~845630","triplet_id":"openstack%2Fswift~master~I73598904fdd566c3b1d9e80e6d817a63a4ee1d82","project":"openstack/swift","branch":"master","topic":"test-enodata","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-06-13 20:55:19.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-06-13 20:55:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I73598904fdd566c3b1d9e80e6d817a63a4ee1d82","subject":"Feedback-I, add mock_utils_listdir_on_corrupt_data","status":"ABANDONED","created":"2022-06-13 19:57:45.000000000","updated":"2022-06-13 20:55:19.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dbb0ea533b6cd2298d1ea108392f9a78fda58690","_number":845630,"virtual_id_number":845630,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"CC":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-13 19:59:53.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-06-13 19:59:53.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-06-13 20:39:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"ebf16ec94fb64f781eb12f48900747b9de7c3aa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-06-13 19:57:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"766f3c9bd87f2703f70cc40702d59fd0ffca2bcf","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-06-13 19:59:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0717d06caff7263c6a66fdb378bb6be9e16d8d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 20:39:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8715ac768e1447398fd2b7f080a8e1c6 : SUCCESS in 29m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/421bc39cd5bb4e26ab6f6ec1f0535af9 : SUCCESS in 39m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/777c263b01474bbea569ef17c3afb68e : SUCCESS in 29m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/331f5afcf2a04e80996b6ff28df59a87 : SUCCESS in 29m 00s","accounts_in_message":[],"_revision_number":1},{"id":"dbb0ea533b6cd2298d1ea108392f9a78fda58690","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2022-06-13 20:55:19.000000000","message":"Abandoned\n\nreplicated in another PR https://review.opendev.org/c/openstack/swift/+/845056","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c7f0d6c70184b0c006359f8e0bb01c246b9aec1d","revisions":{"c7f0d6c70184b0c006359f8e0bb01c246b9aec1d":{"kind":"REWORK","_number":1,"created":"2022-06-13 19:57:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/30/845630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/845630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/845630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/845630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/845630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/845630/1"}}},"commit":{"parents":[{"commit":"b0ffaff34fa644c4256eba19e0dcc4ee32f2c711","subject":"Add replicator tests for ENOENT, ENOTEMPTY","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b0ffaff34fa644c4256eba19e0dcc4ee32f2c711"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-06-13 19:57:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2022-06-13 19:57:31.000000000","tz":-420},"subject":"Feedback-I, add mock_utils_listdir_on_corrupt_data","message":"Feedback-I, add mock_utils_listdir_on_corrupt_data\n\nChange-Id: I73598904fdd566c3b1d9e80e6d817a63a4ee1d82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f0d6c70184b0c006359f8e0bb01c246b9aec1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f0d6c70184b0c006359f8e0bb01c246b9aec1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
