)]}'
{"id":"openstack%2Fswift~846092","triplet_id":"openstack%2Fswift~master~I43e7a7ab7c14c57701b190b214a9987ae1abbb05","project":"openstack/swift","branch":"master","topic":"bug/1733634","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-06-16 16:56:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I43e7a7ab7c14c57701b190b214a9987ae1abbb05","subject":"formpost: info test assert for not supported digests","status":"ABANDONED","created":"2022-06-16 03:30:17.000000000","updated":"2022-06-16 16:56:56.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7098d9602a6b9d2da63f1b449e5f5aca0c5275e1","_number":846092,"virtual_id_number":846092,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-16 07:53:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-16 16:52:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-16 04:22:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-16 16:52:15.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"519667b13064c1e18315d1c9b976964eca1cf393","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-06-16 03:30:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e7b3a3fd07514fbca541352fc0e37c1fb9d8a45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 04:22:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27517e94209749679c6055dd6d24c1f1 : SUCCESS in 8m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f02de097b37e445b9cdb4df513ad5bb4 : SUCCESS in 46m 03s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c5431ffbae04445ea7b05ec42952928c : SUCCESS in 22m 48s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/48fd766c1f9e40778e7ca66d2999ee04 : SUCCESS in 24m 08s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f805a58486264fa8a51b605675d16df0 : SUCCESS in 24m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/01e09e5fa86945e291e4ab6c658e24f7 : SUCCESS in 6m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fdf5207a64a64c8ab151d245b5f97663 : SUCCESS in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/179c87a42deb4f038578e281e577053d : FAILURE in 15m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/df654d96d8ee4f969a94ed668089f5f9 : SUCCESS in 18m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/32ca6fad20104ed8a8307ecedb8d846c : SUCCESS in 16m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/04d319810ce04fd991d3a33af9daeb2b : SUCCESS in 24m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/465dbe247578465c91ef56f9f9664e46 : SUCCESS in 19m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cc8e7f3b1dec4322a0af0f1774a83695 : SUCCESS in 22m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1b35f6c2b9d24cfb9e580c28211b7d08 : SUCCESS in 13m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/428e33ed0d324fe9a6b250b2cfa15ffc : SUCCESS in 14m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/525aa60fa8e0458cbc1986b17e6424dd : SUCCESS in 14m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/42dee1b3887a4381848de3939cb672a8 : SUCCESS in 47m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3e943b804448471cbb3e43b713b057e4 : SUCCESS in 39m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65d2767678d7480f8224dc771f142a82 : FAILURE in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1bcaac5d66d4f44b055ab8f8f9790bb : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/560c0b050a86487abd1acd76d392fc67 : SUCCESS in 25m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ae60c127fee4665e84963fc9db9fd76633341463","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 04:39:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/159dba76895147b4819b7ad9179b337e : SUCCESS in 29m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4c057d90e474ab89d355a1373b3e5ae : SUCCESS in 1h 07m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9438f5a33ff342a993cf8dbe1e5afb30 : SUCCESS in 33m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/022a8648fb0a4e488d8e38eaf622d342 : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":1},{"id":"d4d803b5febc1bed8a44395d31589411cf675a70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-06-16 06:08:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0f34d5603df9fd80151de0164de354aea9296bd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 06:43:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3610b49123dd4926b9b6275edbdf7d24 : SUCCESS in 27m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41e1efd902494d1fa8186a69445d8696 : SUCCESS in 28m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/89690775eef74e6eae8b8f2e3b274301 : SUCCESS in 28m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/134ddd02802c4e9086dde69cb9485e15 : SUCCESS in 27m 27s","accounts_in_message":[],"_revision_number":2},{"id":"9fad8c4f4dfbe06c102f3065ebfe25700f4a2af5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 07:53:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4139b58e2ad944b186361048c28a0a44 : SUCCESS in 8m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/145cb7725abc42708de89a82f7f5d293 : SUCCESS in 45m 02s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4c009185761346a396adac54f50e87f4 : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/54cfc191196f4e3f80761dfac47983c3 : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4394768cd1c74210bb06898c0c29bcab : SUCCESS in 22m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3d8e722208974ed5a737a56a8efdd406 : SUCCESS in 6m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/00cba19fdf554da6a2e448addee23a06 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a9c705a2e78e4474a84e795c3bbfe4a5 : SUCCESS in 14m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/773b7828f08046c384edbef27975101a : SUCCESS in 16m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8e86ec8278b243b28cbba7a5a0ce19d4 : SUCCESS in 16m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d2a8f096b8f3494c953cd770142ebc98 : SUCCESS in 20m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/753f8c9e87384cd58662b5c7eaebf32d : SUCCESS in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dd0b6f49ab0549459f78ec22fdb47930 : SUCCESS in 22m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/46d96c0a2f4240f3975de134b3246966 : SUCCESS in 18m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/234b79c6da6145bda1cb4f1cf329b3b0 : SUCCESS in 20m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a8c4b0d5360d4b60b7a93f804209e32f : SUCCESS in 16m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/74954afae24345fcbecef3e5d38c91ef : SUCCESS in 30m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/910cda2f6ac64d21a93454435ea1851c : SUCCESS in 1h 13m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/975c1d466c0548f3a58c68a12ad6d6e6 : SUCCESS in 16m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a7d3a73cded42d6b52a18b12459bfc8 : SUCCESS in 6m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8c010389adbf498ab0a0fef233f8493d : SUCCESS in 24m 13s","accounts_in_message":[],"_revision_number":2},{"id":"d9b9b17545577e49e50cf09e73a43f3b091d004f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-16 16:52:15.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7098d9602a6b9d2da63f1b449e5f5aca0c5275e1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-16 16:56:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"27e1a1c78d68fef7eb76a7880df401bcec7650dc","revisions":{"3e1f9e9db11bad6f3d6c959be8ca5ce069206ad0":{"kind":"REWORK","_number":1,"created":"2022-06-16 03:30:17.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/92/846092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/846092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/846092/1"}}},"commit":{"parents":[{"commit":"f87968a5b73f3a8ddeaece0fc1e201ffc4fa48e8","subject":"formpost: Add support for sha256/512 signatures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f87968a5b73f3a8ddeaece0fc1e201ffc4fa48e8"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-06-16 03:27:07.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-06-16 03:27:07.000000000","tz":600},"subject":"formpost: info test assert for not supported digests","message":"formpost: info test assert for not supported digests\n\nWe have an info test, test_non_default_methods, where it adds non\nsupported digests. But we don\u0027t assert the log warnings.\n\nThis patch adds such an assertion\n\nChange-Id: I43e7a7ab7c14c57701b190b214a9987ae1abbb05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e1f9e9db11bad6f3d6c959be8ca5ce069206ad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3e1f9e9db11bad6f3d6c959be8ca5ce069206ad0"}]},"branch":"refs/heads/master"},"27e1a1c78d68fef7eb76a7880df401bcec7650dc":{"kind":"REWORK","_number":2,"created":"2022-06-16 06:08:15.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/92/846092/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/846092/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846092/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846092/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846092/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/846092/2"}}},"commit":{"parents":[{"commit":"f87968a5b73f3a8ddeaece0fc1e201ffc4fa48e8","subject":"formpost: Add support for sha256/512 signatures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f87968a5b73f3a8ddeaece0fc1e201ffc4fa48e8"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-06-16 03:27:07.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-06-16 06:07:50.000000000","tz":600},"subject":"formpost: info test assert for not supported digests","message":"formpost: info test assert for not supported digests\n\nWe have an info test, test_non_default_methods, where it adds non\nsupported digests. But we don\u0027t assert the log warnings.\n\nThis patch adds such an assertion\n\nChange-Id: I43e7a7ab7c14c57701b190b214a9987ae1abbb05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27e1a1c78d68fef7eb76a7880df401bcec7650dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27e1a1c78d68fef7eb76a7880df401bcec7650dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
