)]}'
{"id":"openstack%2Fswift~846792","triplet_id":"openstack%2Fswift~master~Id441688aac1088041e243b8ee70710d9c5d7911b","project":"openstack/swift","branch":"master","topic":"proxy-logging-allow-to-add-domain-in-log-messages","attention_set":{},"removed_from_attention_set":{"32333":{"account":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"last_update":"2022-06-22 23:41:58.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-06-22 21:16:18.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Id441688aac1088041e243b8ee70710d9c5d7911b","subject":"proxy-logging: Allow to add domain in log messages","status":"MERGED","created":"2022-06-20 16:53:03.000000000","updated":"2022-06-23 06:43:52.000000000","submitted":"2022-06-22 23:41:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"846792-proxy-logging-allow-to-add-domain-in-log-messages","meta_rev_id":"19dfb3216fc60de096cf67bac5a51b05e9b19784","_number":846792,"virtual_id_number":846792,"owner":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-22 23:41:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-22 17:49:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-22 17:49:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-20 17:53:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-20 18:10:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-21 17:53:28.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-06-22 17:49:05.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"5e9931c70df780f42d468ecd84c233985faa2f4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"date":"2022-06-20 16:53:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdbd7d48c97bb7337494175b8eaf0aaf3d3c5ff1","author":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"date":"2022-06-20 16:55:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b0f9e47f7d347e797266d814ac1d7b2f2254cbe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 17:53:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26b3fce4e4a94d4da06aa655eb7cb936 : SUCCESS in 45m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f2a34b5e96e48588c1148fca31e5fc2 : SUCCESS in 34m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/88add03ab0ed40479d03f15548748fe6 : SUCCESS in 34m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/57fcf32ba4304a04ab97f375de50aa64 : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d86db598fb15216944e74e38f1c59efe11e2a501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 18:10:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccc1307fdf164dd994f1c5ff39468ec0 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/1e80d71c598b491abfeef47c052f823c : SUCCESS in 1h 04m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c5c1abce3dae4a6680dacec8dcdb4776 : SUCCESS in 46m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38107851508940d88017dfa26937cb03 : SUCCESS in 42m 16s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/110c433d90eb417e808fbde131e271e3 : RETRY_LIMIT in 4m 52s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/da633f85141a4b82a405500e2b54453e : RETRY_LIMIT in 4m 54s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1f039e5fac1f4c6bbbb50ce8ed6d7810 : RETRY_LIMIT in 4m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a9896be2b979402da76a1a0daffa7a8a : SUCCESS in 7m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/963c8c9b7b6b49f39ed0026919f9de38 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/65814e85e6914510a12ebe39fbcd9971 : SUCCESS in 16m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c600f953be27465f91fafc41799c6e19 : SUCCESS in 14m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c62409b3fef14b45ad77af381392623a : SUCCESS in 15m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a0bdcfcff37045faa3c76a8d3ced3b2e : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/924755ce735d45c7a188dec358c572d5 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/550a5b4b48714b5387e2afe2db75a12e : SUCCESS in 20m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/99b07c23567c473eb20f068ccc9c1f81 : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1a406ae5d2df40828a07f21d90d32a1b : SUCCESS in 15m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/20d76c804bfb4baeb1adf7a4fb59d3f0 : SUCCESS in 18m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1b3d086007c94a47bfcda76fe3171f21 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f1bb3f7252a24d0c9888176dffcfa26f : SUCCESS in 17m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9d9c7e9fc9a145f79d57094ca2d8be86 : SUCCESS in 1h 08m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8297f572e29f4d2cae9f203a5502a0d2 : SUCCESS in 1h 00m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c616f71ac8a143708962efeefbb70910 : SUCCESS in 44m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/406c5fae37f742909735dd808d6079a3 : SUCCESS in 43m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1746d13527f4f219ca1379bbbfe2ec9 : SUCCESS in 16m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419935982cc94c6b8a13183bea05e5ec : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b7f595b2ac594367944cbec44bf06ef7 : SUCCESS in 21m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/838260d1abfa4aa392ae11d56a16a84a : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":1},{"id":"f8fa1b8efad080c5e67ef8186f17681bdf61e264","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-21 17:53:28.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a8263d839c3579dd45fbc975a470a4f26feac6c6","author":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"date":"2022-06-22 13:58:27.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1a24b3f547bdf663bcefd869fb5cfda843633030","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"date":"2022-06-22 13:59:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0a940b60c7a84786481520a176f6dd72dd585645","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 14:34:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0dc8991d5a324bc1886683d76a2f6ed7 : SUCCESS in 28m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/082f5fca788743848a5da55c9890e2f2 : SUCCESS in 28m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/10bd801b156d486a8424e36f7591b868 : SUCCESS in 28m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f9ec2a8234f4a1f92290f3809594dee : SUCCESS in 28m 01s","accounts_in_message":[],"_revision_number":2},{"id":"c03562295178b6328fedb1ff997f48e707f6a50f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 15:04:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cba4b045311d4f8aba4b2f9d9be6f72a : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/4af1956b0b3e4fcfad6f3ac7469502c9 : SUCCESS in 56m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e6831270ae954e5790d4f4dfd32cc716 : SUCCESS in 47m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19e52bbe96cc4b20baa56e5d9192a6c8 : SUCCESS in 57m 26s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4da74cbbb1594637bbe2fbd8bb5f7cba : RETRY_LIMIT in 4m 29s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c17e2dd7259f4b6bb1feedf3db48bda1 : RETRY_LIMIT in 3m 57s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3019a43b37ec4ce684562f7394a623fd : RETRY_LIMIT in 4m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ccea0adb7bc84479a818223403ace47f : SUCCESS in 7m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2b735d5319184b03a7b287b53f791967 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e032fce2502e46e09f95cccdc1845932 : SUCCESS in 15m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b27edd4e14cb423ba253d93f5f62ba84 : SUCCESS in 15m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/64c3123ddd2040e2839224211c066c94 : SUCCESS in 16m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e6e7ae274de94014b1970e50d3023c49 : SUCCESS in 28m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/50f25ed65419480b89fcf8c6a52706da : SUCCESS in 17m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4037895f95524fbe85b13602eac40bea : SUCCESS in 28m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/14c7e7f9135d4da4b6a5b7766482b203 : SUCCESS in 25m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fda076b1335846b8b1100a07396013c1 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bd3ae5404e03433c834464b201a8c69b : SUCCESS in 28m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/84d3030551934d999bdf5c1d6a022891 : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a44bf5df500a431e98b763ceebbe36fc : SUCCESS in 24m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/43ebfc98058a4fc3a1c42e514d07f2ee : SUCCESS in 7m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4eb11dd896514a778c9ff7044f8cd38d : SUCCESS in 1h 03m 14s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0825b95e6a0b4017803736f04d094f88 : SUCCESS in 54m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/17d0f80477c04f3db7d6b9a6e9f43853 : SUCCESS in 56m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fd446fa8219542b195994432c0a13c93 : SUCCESS in 43m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/332a29604dfc440b8044bae51d320cc3 : SUCCESS in 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccf4fa6983364aa58cee2cc56a5775d0 : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0e3bec76045a48ccb353654ad8c3af3b : SUCCESS in 20m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79b6f79f0ac74ebebc4bcf357ece56da : SUCCESS in 47m 35s","accounts_in_message":[],"_revision_number":2},{"id":"0cec5d1936947a3d391d1cc47cb53e5af72cc156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-22 17:46:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5ae616aba299cff52d45b365279241cf631d01d7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-22 17:49:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b007761924a9c844692c72ca6a4bd3709daea131","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 18:28:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d152fa10365242f4bfa1a28c0ea320f4 : SUCCESS in 28m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9123a8d0e0642e8b2fd4a8816518606 : SUCCESS in 40m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cfd9bd90bb5140268c1fb5ba18d613ce : SUCCESS in 33m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0ab439bd4354ad3be323bfaee2fd2f5 : SUCCESS in 35m 03s","accounts_in_message":[],"_revision_number":3},{"id":"b380382fe9171667c908a4c804741f73fd05b0d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 18:50:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9979105876e34657ba7d908a42387cf9 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/01b7d4a58df24b50ad5e2463c226a729 : SUCCESS in 54m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d1978efc17b4bdc87dcfbc06993b236 : SUCCESS in 40m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/655f1ce65b4741338e41ddba35d09f5a : SUCCESS in 46m 35s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4233033d60e64e98b378199a3d2420ba : RETRY_LIMIT in 3m 56s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d98c70b08a0a4cbc8fbf7de624a465fd : RETRY_LIMIT in 4m 56s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ba18e3fc50ae4d60886b63c51e3196c7 : RETRY_LIMIT in 5m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/11cda62b2fcf4c55942e6daf6145b98d : SUCCESS in 7m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1b7ab6b9295a429ab8b9169b0291ab4d : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/58a958fb382f412f96ced83759e56c1b : SUCCESS in 21m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/de5ebcd3e0484ac68763e113dcd433dc : SUCCESS in 22m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0e44340103004bd084c16fea98474b89 : SUCCESS in 22m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/11698510b8f14a0fa1a83782e11856f0 : SUCCESS in 24m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/be54462a5608423198f678f8b297028e : SUCCESS in 24m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/34c81b8d5c72478dbe1d020780c9de53 : SUCCESS in 26m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/371c282b840048ca9c94a8fac575e464 : SUCCESS in 13m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0bdb38b9fb9b496e96f0995cd8c916a7 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/00d2d5eefe8041269e23d7f00fbca98b : SUCCESS in 25m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7456d97125e5427e9123cc81d2f67b0f : SUCCESS in 6m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/115f83a7aa134e43a31b5236f3e7ff12 : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4199a4f4ada04261846d4a7b8b50fecb : SUCCESS in 6m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e375bd902db94fac883e69d0915634af : FAILURE in 45m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/64a040feadbd48fa8192a4fbd4f14f1f : SUCCESS in 58m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef5fb6d9c766482096b6415975a73b1b : SUCCESS in 32m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b515a7c0c5364686a7433143c92c7a86 : SUCCESS in 39m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d485fe4bfd0548a6b4fb2020e7b101da : SUCCESS in 15m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1f48eba9c5d4816859754f4f35d1df8 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c8097c72ef7a4fa98b800355c30c7109 : SUCCESS in 18m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b0e2cd43d674928b0ec54bb816b06e8 : SUCCESS in 40m 53s","accounts_in_message":[],"_revision_number":3},{"id":"e39be850a4a62af0d84ed1eab203bc049f37ce41","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-22 21:16:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b8c92bc319bcf6cf2b55eb471ae68264eb23c2b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 22:08:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/664da87993314b28b8626fcd7842942f : SUCCESS in 26m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dbc78a993a3e4b7e9fe381eda07060bb : SUCCESS in 50m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1ebf42088794282a2ece47ff632b8fc : SUCCESS in 38m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0e4968b4a52489286953979b2f558e3 : SUCCESS in 40m 34s","accounts_in_message":[],"_revision_number":3},{"id":"1e2c13d74d62b76c2b423831c9f602dfd949a90c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 22:25:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd5fcad8ab6746a3910dd5fd80c1892b : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/b2563e616d5c42d6bdeef98890a525c0 : SUCCESS in 1h 03m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d486d8c6a819410d90d8db4398327643 : SUCCESS in 42m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d42f7ce90ad8453eb2094da470b20cab : SUCCESS in 50m 00s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4a7fcc17011145e8861a0351e92592b8 : RETRY_LIMIT in 5m 02s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cd178430af40437b9cc10f6798efd20c : RETRY_LIMIT in 4m 37s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/12221a077d614faa91b463bd693438db : RETRY_LIMIT in 4m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/21a097134e674c0ca0c59428edcd195f : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6b157cd7a82a45fcba3b7f822b4e9248 : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c6efba423d53444581e1f8f5d2a73a78 : SUCCESS in 15m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bcc3526ab80941ac8623b6b0c9033763 : SUCCESS in 19m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fa3663a30d03466d82bcfc6c4075f33c : SUCCESS in 14m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/61bd36f0a6e04f90a81acaed019b541e : SUCCESS in 24m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/22769f55effb479dbfb06c14ebd43474 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e129716d8dec45b39ffcfe8d7b01fd80 : SUCCESS in 18m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eaafa786b1a24b0691474a21e69b20e9 : SUCCESS in 21m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c98d8285459443658df581065cf2e042 : SUCCESS in 21m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e63f6bcc8db046b6b6f0f6353c78cc31 : SUCCESS in 17m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/920e191023e24f3ab2a92284e71abef4 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/be8d873fc7584a1195f4318b73faca7e : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f0e08818645741d4b13267522a86fd76 : SUCCESS in 6m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cdfefb90aadf4d2189ac387a22bd3aa5 : SUCCESS in 55m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3328cc2da7d54d909ea0b6ad1f55c923 : SUCCESS in 56m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/23507a355895427a9dc9e7356066505f : SUCCESS in 43m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f05432a719bc4947a520629758f0df3b : SUCCESS in 40m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e0c64a473b4480198c619bd32896125 : SUCCESS in 20m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51c875f83aeb4d7ea97208e9d8638437 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/02c687d8774c4d159ec1be0944851b49 : SUCCESS in 29m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/109361260b694bb69736b1bad033aad6 : SUCCESS in 44m 28s","accounts_in_message":[],"_revision_number":3},{"id":"cd933c4b6f630d5c3cd343e09bf0be776c611b02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 22:26:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ba73deec64966defbc1ab3f88ff7ec07aa7ac123","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 23:41:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7184362c9063425cb33913e2db9e1850 : SUCCESS in 10m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/df011b17b41749cb99fb987d01610641 : SUCCESS in 54m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/21ba4b1580a4455f959e3164d61b427b : SUCCESS in 43m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90a19e50f931443680f9c4b369876964 : SUCCESS in 44m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f6c37bd530a24468845370702022b730 : SUCCESS in 7m 07s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/bacefbce13894cf2804168fccf09deb9 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0a4c92c80975455891022acf281b3059 : SUCCESS in 20m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4db7aeede83b42818104574da587098e : SUCCESS in 17m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4e09d995a7684860b1abc14b128b1802 : SUCCESS in 20m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e5c44bc01c814cabaad36495a7aaf1b6 : SUCCESS in 21m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/21154a55da3f45bbbbcfb0ad6a9a4a45 : SUCCESS in 21m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6313eae6c9a1429a90c81a3ada233b06 : SUCCESS in 24m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/244445e6dda6447b92160a3abec67786 : SUCCESS in 20m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9b83c08ea7bf48e59e6c2d26f7517e48 : SUCCESS in 20m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/efa59d433bc74ee0b907fd6d0237a9ae : SUCCESS in 22m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4ea93e45d3784c78afd689022caed3b3 : SUCCESS in 6m 37s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/32853ee71be04050804ad3b48cfbd097 : SUCCESS in 7m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e5cb21d80a654fd4a3c2ed994b64f35b : SUCCESS in 1h 15m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9c833513b6646c8a23b6f8bc5a4a98f : SUCCESS in 1h 09m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/92ed8d4988b0420492445d756c53905b : SUCCESS in 47m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0310456b4ece46b2bad854d11b594423 : SUCCESS in 41m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ed80b377029430491a4347438fb0a70 : SUCCESS in 19m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19dbd24bcc7c45d59c1c4281d3c3ae94 : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ae2a1ef051f435d9157965c33ab8520 : SUCCESS in 18m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ad6bee9185048048980b32759fef49d : SUCCESS in 54m 40s","accounts_in_message":[],"_revision_number":3},{"id":"c24bf7af2c680d5382813547d7c76aa0e67a07af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 23:41:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e4d5de98d45457fa1b4afc31c656e18f7a3aa8da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 23:43:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/594afbd91be34b47b42b72fb4541cd73 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b67bcc6a08f24c5b856257c31bf5824b : SUCCESS in 43s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0c8bcc62c5364a669989f11739913c8e : SUCCESS in 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"19dfb3216fc60de096cf67bac5a51b05e9b19784","author":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"date":"2022-06-23 06:43:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1831658b338d628f18a2e3357e7b7f5d34ec79df","revisions":{"aa7c02dac3a062eafa34d64fef6f651bd59f861d":{"kind":"REWORK","_number":1,"created":"2022-06-20 16:53:03.000000000","uploader":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"ref":"refs/changes/92/846792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/846792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/846792/1"}}},"commit":{"parents":[{"commit":"26b2425ab88ef8d038183d7db58533c08986638d","subject":"Merge \"formpost: Add support for sha256/512 signatures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26b2425ab88ef8d038183d7db58533c08986638d"}]}],"author":{"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@ovhcloud.com","date":"2022-06-20 16:13:33.000000000","tz":120},"committer":{"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@ovhcloud.com","date":"2022-06-20 16:52:56.000000000","tz":120},"subject":"proxy-logging: Allow to add domain in log messages","message":"proxy-logging: Allow to add domain in log messages\n\nChange-Id: Id441688aac1088041e243b8ee70710d9c5d7911b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa7c02dac3a062eafa34d64fef6f651bd59f861d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa7c02dac3a062eafa34d64fef6f651bd59f861d"}]},"branch":"refs/heads/master"},"c8b9f1897cdc6e4d5211953a3041c56a0d102349":{"kind":"REWORK","_number":2,"created":"2022-06-22 13:59:06.000000000","uploader":{"_account_id":32333,"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@gmail.com","username":"AymericDu"},"ref":"refs/changes/92/846792/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/846792/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846792/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846792/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846792/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/846792/2"}}},"commit":{"parents":[{"commit":"26b2425ab88ef8d038183d7db58533c08986638d","subject":"Merge \"formpost: Add support for sha256/512 signatures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26b2425ab88ef8d038183d7db58533c08986638d"}]}],"author":{"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@ovhcloud.com","date":"2022-06-20 16:13:33.000000000","tz":120},"committer":{"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@ovhcloud.com","date":"2022-06-22 12:58:51.000000000","tz":120},"subject":"proxy-logging: Allow to add domain in log messages","message":"proxy-logging: Allow to add domain in log messages\n\nChange-Id: Id441688aac1088041e243b8ee70710d9c5d7911b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8b9f1897cdc6e4d5211953a3041c56a0d102349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8b9f1897cdc6e4d5211953a3041c56a0d102349"}]},"branch":"refs/heads/master"},"1831658b338d628f18a2e3357e7b7f5d34ec79df":{"kind":"REWORK","_number":3,"created":"2022-06-22 17:46:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/846792/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/846792/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846792/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846792/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/846792/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/846792/3"}}},"commit":{"parents":[{"commit":"26b2425ab88ef8d038183d7db58533c08986638d","subject":"Merge \"formpost: Add support for sha256/512 signatures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26b2425ab88ef8d038183d7db58533c08986638d"}]}],"author":{"name":"Aymeric Ducroquetz","email":"aymeric.ducroquetz@ovhcloud.com","date":"2022-06-20 16:13:33.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-22 17:45:01.000000000","tz":-420},"subject":"proxy-logging: Allow to add domain in log messages","message":"proxy-logging: Allow to add domain in log messages\n\nChange-Id: Id441688aac1088041e243b8ee70710d9c5d7911b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1831658b338d628f18a2e3357e7b7f5d34ec79df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1831658b338d628f18a2e3357e7b7f5d34ec79df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
