)]}'
{"id":"openstack%2Fswift~847125","triplet_id":"openstack%2Fswift~master~Ibdafc4a321e41d6fd225c486da831b1f0fead882","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ibdafc4a321e41d6fd225c486da831b1f0fead882","subject":"Ensure clean starting state in test","status":"MERGED","created":"2022-06-22 05:53:22.000000000","updated":"2022-06-22 19:03:31.000000000","submitted":"2022-06-22 19:02:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847125","meta_rev_id":"e3ceefb67add0efb9ca04535dd975fad84e6af06","_number":847125,"virtual_id_number":847125,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-22 19:02:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-22 18:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-22 18:12:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-22 06:57:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b45b855837af2c7b195feba6cbda20f765d61df2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-22 05:53:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d27cb0c85f118d31bf35e69ede7be6349237e573","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 06:57:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15d90b9292c044a08091ad458bdc9547 : SUCCESS in 8m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e53d4c64e7143daa517c2b2ab9b4963 : SUCCESS in 58m 42s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/efd0edacdd78431796924aeb71c6dcad : RETRY_LIMIT in 5m 04s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8216b553d55147a887316c95e148f161 : RETRY_LIMIT in 3m 51s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1c72388489e94489942511a377dfdab4 : RETRY_LIMIT in 4m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/31b25340372745c4a185d983b9ed226b : SUCCESS in 6m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1d07f315437149b0860672447a282473 : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1591b5b27943461eba3b5a5a778763dc : SUCCESS in 15m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29bbe6467e0a4417adfdca5d4f2a790b : SUCCESS in 15m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8d02b9fdb16d4bd186c9839df754cba2 : SUCCESS in 15m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cdc0cadfe2d04ab385f825461c0ec5b8 : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/703d5c8d72374d298def2679a20698f2 : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1274c4767b1d414aa9efe8f25136aba3 : SUCCESS in 16m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c3f9e8a8d3e745fe86b648d4b7eb57c6 : SUCCESS in 15m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e58bedcd11c449c78e25c2d72418bc29 : SUCCESS in 16m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/57685a51e9f54914be1a17452bcce880 : SUCCESS in 18m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ede2fee4d0a14c83bb11b29e5e8472ed : SUCCESS in 39m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4dc23e836a0644e5a25eba74f581d9e0 : SUCCESS in 46m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afcf33fa2b674f1587a615f1b05df1b1 : SUCCESS in 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/478dc37ffb85479e91a23e2827c8f219 : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c6443739a9d945abba042190ced16503 : SUCCESS in 19m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4fb69b7262728a4bb99c359df51fbd6c3305310d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 06:58:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2d72b9f5b67457da7ed802d7b9d419c : SUCCESS in 29m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2b084b68817473cb45a41302eeb4a98 : SUCCESS in 27m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/82f520acc74c4b0d9a2fa51cc80b01c8 : SUCCESS in 35m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/813ef74626884ddd883c4c103752f672 : SUCCESS in 34m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bf861e84f777186ac7d0a91c91641d62a53bfb27","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-22 18:12:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b6aa4f8838f2e64bcfcaac41f53d224c44f8e142","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 18:13:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95c396eb9ade16f6faa0c10bbdf1785f2f96f4a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 19:02:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52c2b31dc4554ac792199a858d6dc62e : SUCCESS in 8m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e12f78983df49cda109b74201409ebb : SUCCESS in 47m 53s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a5dd8b3f53d04a2a880c5add1ab9996b : SUCCESS in 8m 46s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a3572ee8f00147d58288cf79cd78cddd : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f3283435b4b64eb88c2dd9af182c5774 : SUCCESS in 16m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/96694a0d08604553beea6b31c078315e : SUCCESS in 19m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fa3fbc33298649978246333410226db2 : SUCCESS in 19m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/38c636d240f64b29b218c9fef78f4d75 : SUCCESS in 18m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3a0f489b40164e20981ee26e6e72f15b : SUCCESS in 18m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a3c51948e76541308e37e5d55db8bd12 : SUCCESS in 17m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8c31e955f32744858b6bc7d34b0e9ccd : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2350885097a441d896e99819c96abada : SUCCESS in 16m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4ab69973d7644079bdc9e30f5b3c149b : SUCCESS in 20m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5220558c793240098f983dece2de661e : SUCCESS in 7m 31s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/73a1faad448f41a68d91d0eb694bce5d : SUCCESS in 7m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ad62f5464c05421994de4bf3ecdfccb0 : SUCCESS in 37m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b99de4413fe3445fa91b1ebe9e310db8 : SUCCESS in 41m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d00550600a141b9987f1836d646892d : SUCCESS in 16m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/783901865de2474d87aee1373a6f488d : SUCCESS in 6m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e39f159e22254fff99d65d520e01a882 : SUCCESS in 25m 52s","accounts_in_message":[],"_revision_number":1},{"id":"438a14058a746956eed3cd486b699fb2dd956048","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 19:02:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e3ceefb67add0efb9ca04535dd975fad84e6af06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 19:03:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e16adb2f8d41b7b4c5095af5628fd4 : SUCCESS in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a7ad34561f564b088b65def5e660e725 : SUCCESS in 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/032e6d8a103d4e30bde5df52816e29da : SUCCESS in 45s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9f1f4d60fb459990bb96f5b9a42dc92fb55d52d","revisions":{"b9f1f4d60fb459990bb96f5b9a42dc92fb55d52d":{"kind":"REWORK","_number":1,"created":"2022-06-22 05:53:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/847125/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/847125/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/847125/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/847125/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/847125/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/847125/1"}}},"commit":{"parents":[{"commit":"be87b29a72778b52b8d02d8a44f9f342ac26340b","subject":"Merge \"tests: Fix cross-test interaction\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be87b29a72778b52b8d02d8a44f9f342ac26340b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-22 05:52:41.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-22 05:52:44.000000000","tz":-420},"subject":"Ensure clean starting state in test","message":"Ensure clean starting state in test\n\nSince it merged, test_make_request_clears_txn_id_after_calling_app would\nalways fail on the first attempt in the gate, but we wouldn\u0027t notice\nbecause it would succeed on the retry. Trouble is that some tests (like\ntest_catch_errors.py) would pollute the main greenthread\u0027s threadlocals.\n\nChange-Id: Ibdafc4a321e41d6fd225c486da831b1f0fead882\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9f1f4d60fb459990bb96f5b9a42dc92fb55d52d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9f1f4d60fb459990bb96f5b9a42dc92fb55d52d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
