)]}'
{"id":"openstack%2Fswift~848168","triplet_id":"openstack%2Fswift~stable%2Fyoga~I0e0f802fa355060f43f9e63f52897fbcf66816d2","project":"openstack/swift","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2022-06-30 16:37:52.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-06-30 15:09:47.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I0e0f802fa355060f43f9e63f52897fbcf66816d2","subject":"tests: Fix swiftclient/requests log level \u0026 Ignore py36 deprecation warnings","status":"MERGED","created":"2022-06-29 17:45:58.000000000","updated":"2022-06-30 16:39:24.000000000","submitted":"2022-06-30 16:37:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"848168","meta_rev_id":"1da3a237127126e125fad552d8f8462b45aef38e","_number":848168,"virtual_id_number":848168,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-30 16:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-30 15:09:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-30 15:09:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-29 17:45:58.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-06-29 19:48:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-30 15:09:47.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"7dccc39bd4c03ac5dddf20f5e9f97af39e36ef0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2022-06-29 17:45:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ec70d16c52adfd8e5d503b99e0fda708e81560ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 19:48:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/003757bfe4ab4daa9d4205f88c2eb186 : SUCCESS in 7m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d41c9ca6a9d64b369f4b7c287da35f65 : SUCCESS in 44m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8886b097f44d45cca966a4da16ef45e4 : SUCCESS in 1h 08m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/439e38294e124b11931581b356487f1d : TIMED_OUT in 2h 01m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48dc5102f3944138ad3dcd3bbe805e87 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e35877b686683681a9521d190b03a659774bdbd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 21:02:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8df9945f6b8c4efd83d919f0ff9a1606 : SUCCESS in 44m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/81e0c1bbbbde4188b42be6e335e18722 : SUCCESS in 55m 41s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/b399df8784f946b4a042a5c33f138676 : TIMED_OUT in 3h 01m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1542a797102940cb907acdcf36cadcd8 : SUCCESS in 1h 07m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/58d202c7c30949cdac183667262ed5b3 : FAILURE in 1h 15m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2a954aa67b52b5c359f68f4f73f5145be69cf043","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-29 23:38:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14731d5dc4c358bb767926a9e7ac5efe0af97364","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2022-06-30 08:24:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"63aefed1cb675e5cbdcb9e99fb992d6cc388b524","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2022-06-30 08:28:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d1230b2323f4e1313f3746f3609b4f16bdcbeb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 09:39:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89bf9672e6f54fbe8c35606da59ac95b : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/ca2378dc791246c3a5d3402209c8c902 : SUCCESS in 1h 07m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dad8108b84704fa582db11da0ed9223e : SUCCESS in 1h 02m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ece09e14f6b47458a00f897931c9ced : SUCCESS in 53m 32s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/94f2c45e8efb44df98898207bcb0bfce : RETRY_LIMIT in 4m 15s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4e57f5cb6efd4c6b9460182ca777b82a : RETRY_LIMIT in 4m 22s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2ee99545b02b4948904288587cc3f910 : RETRY_LIMIT in 4m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/18e0221ea04f4e9ea672c682aad532b2 : SUCCESS in 7m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9ab2f8f7227c4afa8d9f854386f24780 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5eae7c2f5c604782840139c684ec9b7a : SUCCESS in 17m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3036bc5b613f48bab4ca1c0dbf912d32 : SUCCESS in 16m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d7821600092141ddb2977fd6c9bad806 : SUCCESS in 15m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/99aca16bc42e4c63836115904ffaab83 : SUCCESS in 19m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fa615ce6786d4946878c0f8d9047e0c5 : SUCCESS in 21m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7ae77c911f084fadaa2211755e65b6ff : SUCCESS in 17m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6072882870f04475bcbc28ca99cb4a08 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a07d53343c4b453f827a45a7451d1812 : SUCCESS in 15m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/65c276fb85fa456cae69fc13b6399a5d : SUCCESS in 14m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/07dffc2e12554658904a87aae89477d2 : SUCCESS in 7m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bee835c2a62d45f2b2a24efa9ba857ff : FAILURE in 7m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fd3034a2d1d84b38af24a9c9d11a2319 : SUCCESS in 53m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f5b29e95bbe4f1782850d5e6a1634c6 : SUCCESS in 57m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5a78bcc6d26c4fa6a71caa4d639fe715 : SUCCESS in 32m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8647141d1b7846fab166619ec1cad929 : SUCCESS in 33m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13d253bae1a44ca99830896c94e15507 : SUCCESS in 14m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b31ee45f75147efbe98e0e036165266 : SUCCESS in 6m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92640fb42a584500b21f554456017630 : FAILURE in 8m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e40415fd2a314d2d83c62421b52107d4 : SUCCESS in 45m 51s","accounts_in_message":[],"_revision_number":2},{"id":"9ecd5713c923d8693ac8eb8e64edfdea33a2fce5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 11:30:07.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/57b6a52da0114185953879d5dd220820 : SUCCESS in 58m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7dbbf3c404784341875636fd9c53b411 : SUCCESS in 58m 48s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/64ae2a0fa2464342ab3d87a695233ace : TIMED_OUT in 3h 01m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b48e27c4453044c992a5b876e2dc4a87 : TIMED_OUT in 1h 31m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5e3d590d1e34920a570a47e4de56dd1 : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":2},{"id":"e50be20978b2e31e9e5cc75a9d0cb7729075ddbe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-30 15:09:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"245a2f160b0baa215af87cae3ae7d024b67efc0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 15:10:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d2d46640e0145015c2b9fd315e8c96aea1e1dc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 16:37:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e1a43bd272403f9b5cdc2b699cfd98 : SUCCESS in 9m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/344e0c9e67064cc9b6eee00606ec6db8 : SUCCESS in 41m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9ad13b435cd54447af7f15bfc4310f40 : SUCCESS in 42m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92b91e11fad845b2b3c9f2830ad445a9 : SUCCESS in 44m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/87aeaf8bb17943928971d299a37c6030 : SUCCESS in 8m 30s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/db644c2a54cc45da8a0c6ae633299547 : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d8016317e0dc42d5b1f72860dae68d71 : SUCCESS in 16m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4ec559e2ba9341f78b09f070922a03c0 : SUCCESS in 13m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a51fe56217294c3cbd927042d6c59f64 : SUCCESS in 32m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d0958237236f4eb6801196648b381887 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9afd17f137d1491dad1c1c87a4e9ee72 : SUCCESS in 15m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8cd285452250435ebe8fd7fd6da97045 : SUCCESS in 17m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0ad1e18cd2414bb29db17e2bc298dd46 : SUCCESS in 12m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ceaa781dee30490582854fec01c6cb98 : SUCCESS in 14m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0270e00486f14becb22806703a840ce8 : SUCCESS in 17m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/88d3cdf7f20d4da1a89d09f2abf61f2a : SUCCESS in 9m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ba1c2db26f7c45cc8f692e01f4f5274a : SUCCESS in 55m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/67c1828404604c2fbde080acc4cc5211 : SUCCESS in 49m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6bc95a7a776f4fe89f7358c159d69692 : SUCCESS in 34m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/73e9ad0639454273a3fa18b62b482824 : SUCCESS in 32m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/388e598ea0cb4c4cae16ceb455d7ba1c : SUCCESS in 14m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4677d87d64e4feb8884c467454689ec : SUCCESS in 5m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b70a77b3a9454b69a8739c3edf10f4b7 : SUCCESS in 40m 22s","accounts_in_message":[],"_revision_number":2},{"id":"34504c9a38b8332bf3ade0f9baaf48fd8ec25b52","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 16:37:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1da3a237127126e125fad552d8f8462b45aef38e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 16:39:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38efa22ea650469f9ae66d4ba490b734 : SUCCESS in 1m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/02c7c0be655243ffa969c4303ebb1edd : SUCCESS in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9570a18caca84349898a92d56ad6a9af : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7c888bf6ca81b3fed9a8e6522a078497902bd354","revisions":{"4782ef46b08d7c6dd63d1efa9719d35b7716f126":{"kind":"REWORK","_number":1,"created":"2022-06-29 17:45:58.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/68/848168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/848168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/848168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/848168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/848168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/848168/1"}}},"commit":{"parents":[{"commit":"36ebcc8484db112c9c7998b35aec2e0b004dae33","subject":"s3api: Use constant-time string comparisons in check_signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36ebcc8484db112c9c7998b35aec2e0b004dae33"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-07 21:47:05.000000000","tz":-420},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2022-06-29 17:45:22.000000000","tz":120},"subject":"tests: Fix swiftclient/requests log level adjustment","message":"tests: Fix swiftclient/requests log level adjustment\n\nGotta update now that swiftclient changed how it imports and uses\nrequests.\n\nRelated-Change: https://review.opendev.org/c/openstack/python-swiftclient/+/828821\nChange-Id: I0e0f802fa355060f43f9e63f52897fbcf66816d2\n(cherry picked from commit a5a98d7e3bbcfb6d4e8cca8f4aca08cca4286b17)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4782ef46b08d7c6dd63d1efa9719d35b7716f126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4782ef46b08d7c6dd63d1efa9719d35b7716f126"}]},"branch":"refs/heads/stable/yoga"},"7c888bf6ca81b3fed9a8e6522a078497902bd354":{"kind":"REWORK","_number":2,"created":"2022-06-30 08:24:43.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/68/848168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/848168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/848168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/848168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/848168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/848168/2"}}},"commit":{"parents":[{"commit":"36ebcc8484db112c9c7998b35aec2e0b004dae33","subject":"s3api: Use constant-time string comparisons in check_signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36ebcc8484db112c9c7998b35aec2e0b004dae33"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-28 22:48:13.000000000","tz":-420},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2022-06-30 08:23:19.000000000","tz":120},"subject":"tests: Fix swiftclient/requests log level \u0026 Ignore py36 deprecation warnings","message":"tests: Fix swiftclient/requests log level \u0026 Ignore py36 deprecation warnings\n\nGotta update now that swiftclient changed how it imports and uses\nrequests.\n\nRelated-Change: https://review.opendev.org/c/openstack/python-swiftclient/+/828821\nChange-Id: I0e0f802fa355060f43f9e63f52897fbcf66816d2\n(cherry picked from commit a5a98d7e3bbcfb6d4e8cca8f4aca08cca4286b17)\n(cherry picked from commit 7298038ed90fe5824c250975045a6f0a3eb39d04)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c888bf6ca81b3fed9a8e6522a078497902bd354"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7c888bf6ca81b3fed9a8e6522a078497902bd354"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
