)]}'
{"id":"openstack%2Fswift~848169","triplet_id":"openstack%2Fswift~stable%2Fxena~I0e0f802fa355060f43f9e63f52897fbcf66816d2","project":"openstack/swift","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2022-06-30 21:11:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0e0f802fa355060f43f9e63f52897fbcf66816d2","subject":"tests: Fix swiftclient/requests log level \u0026 Ignore py36 deprecation warnings","status":"MERGED","created":"2022-06-29 17:47:12.000000000","updated":"2022-06-30 21:12:54.000000000","submitted":"2022-06-30 21:11:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848169","meta_rev_id":"640fa2036f95f04b55847164d3fafcb21962a9cd","_number":848169,"virtual_id_number":848169,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-30 21:11:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-30 20:09:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-30 20:09:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-29 17:47:12.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-06-29 20:05:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-30 20:09:57.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"446397b54093a56352a83bd15b61dab0d9ea3b0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2022-06-29 17:47:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a215e9597a0a5f3e7e6ca5d8b04c26f2724241b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 20:05:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c871d72cd6df45949aa432289587e618 : SUCCESS in 7m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75f7c8e95a084216b734112e2e47d036 : SUCCESS in 47m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d143b818f0e74db58104eb1eb95488ba : SUCCESS in 51m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f1321f08bb84b5f9d19f04ba3b04ec2 : TIMED_OUT in 2h 01m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6919e69bf73b4d1e8b6ed900f3ab81bd : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"cc3e2aac546c98a402eedb4ed74cd7acaf980c63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 20:15:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/36b1085e732e41fcb26cb1a2fdfec732 : SUCCESS in 1h 07m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3e99479771042eaa17385b137f52b31 : SUCCESS in 1h 10m 39s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/1c240e426b1548f8b11c12abeab170fd : TIMED_OUT in 2h 01m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/103331f6fba6439382377473a98bc5f3 : SUCCESS in 46m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/55ea48742fb44ee78b51cceb0f2bde16 : SUCCESS in 1h 09m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3569f7588be8a7d777755c1ef4cdc258bd27b70c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2022-06-30 09:58:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"57d02874931ba06cdefab54deec212470743d2cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2022-06-30 10:02:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0cb608f607be6c5be625c7d564b59f3b556ec156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 11:06:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6799376f99e74d9f9c6d59f769f8f8a8 : SUCCESS in 9m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/ec0f381c86c344768e67052519c16639 : SUCCESS in 58m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3855b7d7908b4fcb9cbd365f26c2c529 : SUCCESS in 41m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0ab91efff09482ab06240aff69f21eb : SUCCESS in 41m 30s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/06c5b7a086cc4af796bbe6e7763efc0b : RETRY_LIMIT in 4m 41s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/48fda54f2f5e4921b215b415af9df02a : RETRY_LIMIT in 5m 18s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/889934f8bab1480da08e68c3a958a53e : RETRY_LIMIT in 4m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/076bb56517e044a9b6d2a8d4a6e3ee67 : FAILURE in 5m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/154dcc80052a4112b89b8e7101d65365 : FAILURE in 4m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/17d0cddb213a486da1e38a06f8b02909 : SUCCESS in 27m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/214c050b5ba84531a38c0c2eb05288d6 : SUCCESS in 15m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e4c6883c8a2a4b158864c80ea54cc66e : SUCCESS in 16m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/298d5d5dd5f4482aa69e3252f0714a27 : SUCCESS in 16m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cfdce4cce18e4e7da947b4261bb2f3b6 : SUCCESS in 14m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/35894b6cf9aa41e580fd7f3f288b83fa : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b404d6de1c9340e089b201ee4476eeb8 : SUCCESS in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/36558c76ab3f4ec084f3d9725b2e39f4 : SUCCESS in 20m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0a2b13dbfb0c4801bac074aeed373f8c : SUCCESS in 27m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/81945b7c8b134c04b46bd4353d5895bc : SUCCESS in 17m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/106b72d00968481ba92d0224007f4856 : SUCCESS in 19m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b91a081acfe2418f835938c79885dde5 : SUCCESS in 5m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6f82a153ece342d0896f4062a51b2929 : FAILURE in 7m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c96bb11d30604f4d8352769029527033 : SUCCESS in 53m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc94e3199a1e4696bdef31fa90727417 : SUCCESS in 49m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e7b551f44b1d41a2bab824498cbf382f : SUCCESS in 49m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/776b4a9484234fadb0ac56f0441196ad : SUCCESS in 37m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ace2e8df88146d3b666abf52ee22064 : SUCCESS in 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8959be85dcfc4fa9ae678896ec9708ae : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e5291a0ac8644fe989dcdb53250745b2 : FAILURE in 7m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad7e38d1a9a448e18ac2b4d0bdfe62f3 : SUCCESS in 56m 40s","accounts_in_message":[],"_revision_number":3},{"id":"77d7056a3eeb48e9a2e80b1934f568b840a7777c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 12:15:57.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62cd0227519a4343aa1fc6b1ab2aca10 : TIMED_OUT in 1h 31m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ecc06a207214f92b9468370ef76ea2b : TIMED_OUT in 1h 31m 17s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/4bdc72492182416382c59c4a217f0b36 : TIMED_OUT in 2h 01m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/50196e845b9b45a68c4a9c32148a4c40 : TIMED_OUT in 1h 31m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c72f08898de44ecfa313e2212edabfb6 : TIMED_OUT in 1h 30m 58s","accounts_in_message":[],"_revision_number":3},{"id":"bc74b009d6eac4162868163d26cd41a34e67df17","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-30 20:09:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6a22e0aca23387502b662766c7874e351186cb12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 20:10:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5d38368e5d4b916a91e68d89feb28d29743ed504","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 21:11:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db3c5515ee8045c1ab409a1089c66808 : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/3b72aa7e54a94b9ba736a206abf3bd28 : SUCCESS in 57m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/40b246cc4f9946e4a37be71f5d4a3480 : SUCCESS in 43m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca37f0dba7b5439fb3f0d01992a45428 : SUCCESS in 47m 11s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/af9a1248e4c941ad9eabb5b88afecfa2 : POST_FAILURE in 7m 40s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/cdddbfd825ac4dff830358f905e7d362 : POST_FAILURE in 6m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1c57de53390f4bd4891d74e465aaa3f1 : SUCCESS in 13m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ad53b6d99b664e7c8d10789652821569 : SUCCESS in 15m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/df04245d086040c9b6cb122a251274dc : SUCCESS in 13m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/96559e74f7f94f199cf3873d64763ee2 : SUCCESS in 13m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/631d6a93bcb94129b6164c6e2357952f : SUCCESS in 13m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6e7e621b19b746a0b77eaece4c8a8166 : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7b7893d286844918927d2a3ba6e2b7b7 : SUCCESS in 14m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/32c36701f95d4b9887a5f4b6951aecb6 : SUCCESS in 15m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6f3bcadeb857409d883ec1fbc00036d0 : SUCCESS in 13m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5f58268d1691486783e408fa839be655 : SUCCESS in 14m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/20e9091f94204fd6906aaaced6211e3f : SUCCESS in 14m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a651b274bf994d7cbd51399cce51bc76 : SUCCESS in 7m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/817aeb79d3f94731813628221544a8b9 : SUCCESS in 54m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/05c291f36246402da0d708ee4eda4f96 : SUCCESS in 57m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/344bde6be884467a8b0304689481ca11 : SUCCESS in 38m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/01e3371f8e2f4d21ae7e62eff7abad97 : SUCCESS in 39m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fd0842039fb4677a9e103663a7bf356 : SUCCESS in 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c5d498a49934900a0e36bd0b06a05be : SUCCESS in 6m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2bc5eed4e6a24da7ba7a3d8bae9b14bd : SUCCESS in 40m 55s","accounts_in_message":[],"_revision_number":3},{"id":"8009a9d41743aa588c05f898d18591ba1437443d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 21:11:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"640fa2036f95f04b55847164d3fafcb21962a9cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 21:12:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b31142d98ccc4e51af7c92b0eb34754a : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e56ecdf93b2d498db5fc58e42fa31728 : FAILURE in 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9ae4f7d29cf14cf58b38408928edeed3 : FAILURE in 46s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8804225838f64480cbb9d22b44bc04d2cca5031b","revisions":{"8df189a0b430022bcc55429719bb221028a71c98":{"kind":"REWORK","_number":1,"created":"2022-06-29 17:47:12.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/69/848169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/848169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/848169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/848169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/848169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/848169/1"}}},"commit":{"parents":[{"commit":"0fa745c7f94d60660d28fcbd5b085f8039c3c476","subject":"s3api: Use constant-time string comparisons in check_signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fa745c7f94d60660d28fcbd5b085f8039c3c476"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-07 21:47:05.000000000","tz":-420},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2022-06-29 17:47:03.000000000","tz":120},"subject":"tests: Fix swiftclient/requests log level adjustment","message":"tests: Fix swiftclient/requests log level adjustment\n\nGotta update now that swiftclient changed how it imports and uses\nrequests.\n\nRelated-Change: https://review.opendev.org/c/openstack/python-swiftclient/+/828821\nChange-Id: I0e0f802fa355060f43f9e63f52897fbcf66816d2\n(cherry picked from commit a5a98d7e3bbcfb6d4e8cca8f4aca08cca4286b17)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8df189a0b430022bcc55429719bb221028a71c98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8df189a0b430022bcc55429719bb221028a71c98"}]},"branch":"refs/heads/stable/xena"},"11d953cbde52b05fcbff930c0ffb22e9b9c148fa":{"kind":"REWORK","_number":2,"created":"2022-06-30 09:58:29.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/69/848169/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/848169/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/848169/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/848169/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/848169/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/848169/2"}}},"commit":{"parents":[{"commit":"0fa745c7f94d60660d28fcbd5b085f8039c3c476","subject":"s3api: Use constant-time string comparisons in check_signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fa745c7f94d60660d28fcbd5b085f8039c3c476"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-07 21:47:05.000000000","tz":-420},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2022-06-30 09:57:55.000000000","tz":120},"subject":"tests: Fix swiftclient/requests log level \u0026 Ignore py36 deprecation warnings","message":"tests: Fix swiftclient/requests log level \u0026 Ignore py36 deprecation warnings\n\nGotta update now that swiftclient changed how it imports and uses\nrequests.\n\nRelated-Change: https://review.opendev.org/c/openstack/python-swiftclient/+/828821\nChange-Id: I0e0f802fa355060f43f9e63f52897fbcf66816d2\n(cherry picked from commit a5a98d7e3bbcfb6d4e8cca8f4aca08cca4286b17)\n(cherry picked from commit 7298038ed90fe5824c250975045a6f0a3eb39d04)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11d953cbde52b05fcbff930c0ffb22e9b9c148fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11d953cbde52b05fcbff930c0ffb22e9b9c148fa"}]},"branch":"refs/heads/stable/xena"},"8804225838f64480cbb9d22b44bc04d2cca5031b":{"kind":"REWORK","_number":3,"created":"2022-06-30 10:02:17.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/69/848169/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/69/848169/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/848169/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/848169/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/69/848169/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/69/848169/3"}}},"commit":{"parents":[{"commit":"0fa745c7f94d60660d28fcbd5b085f8039c3c476","subject":"s3api: Use constant-time string comparisons in check_signature","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fa745c7f94d60660d28fcbd5b085f8039c3c476"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-07 21:47:05.000000000","tz":-420},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2022-06-30 10:01:51.000000000","tz":120},"subject":"tests: Fix swiftclient/requests log level \u0026 Ignore py36 deprecation warnings","message":"tests: Fix swiftclient/requests log level \u0026 Ignore py36 deprecation warnings\n\nGotta update now that swiftclient changed how it imports and uses\nrequests.\n\nRelated-Change: https://review.opendev.org/c/openstack/python-swiftclient/+/828821\nChange-Id: I0e0f802fa355060f43f9e63f52897fbcf66816d2\n(cherry picked from commit a5a98d7e3bbcfb6d4e8cca8f4aca08cca4286b17)\n(cherry picked from commit 7298038ed90fe5824c250975045a6f0a3eb39d04)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8804225838f64480cbb9d22b44bc04d2cca5031b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8804225838f64480cbb9d22b44bc04d2cca5031b"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
