)]}'
{"id":"openstack%2Fswift~848195","triplet_id":"openstack%2Fswift~master~I203607a0576582330241172d05bf8fd223bbbb9d","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-07-05 05:31:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I203607a0576582330241172d05bf8fd223bbbb9d","subject":"Emit metrics for tempurl \u0026 formpost digest usage","status":"MERGED","created":"2022-06-29 22:46:39.000000000","updated":"2022-07-05 05:33:04.000000000","submitted":"2022-07-05 05:31:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848195","meta_rev_id":"043277dbb8614d20985bc90e01c916c8f0e1284a","_number":848195,"virtual_id_number":848195,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-05 05:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2022-07-05 04:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2022-07-05 04:24:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-29 23:56:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-30 00:10:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-30 07:08:54.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"06cb77b9dc40f7cf43eb2f71222ff98839daea0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-29 22:46:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b3099167005752353255f7301c6e2c4ec03fe3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 23:56:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fae6502ad8b482c85cb5c859d04f6d5 : SUCCESS in 1h 07m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/452f4c9c878340238487f30d03118f41 : SUCCESS in 29m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5bb3101d195840239d3b0b4a642b64fe : SUCCESS in 1h 02m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1bedd7754d4747c898beb8146b656f23 : SUCCESS in 29m 27s","accounts_in_message":[],"_revision_number":1},{"id":"08a05a1b0856d9fa4af40c5c9b2bcea87dcf0633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 00:10:58.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46b8f0f707044fa5852ce39718bf4cd5 : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/03aa1db936e2425c8be94f5c1f7aaf6e : SUCCESS in 1h 18m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e821bc4dcd74ae4aa0660287fa2b57b : SUCCESS in 49m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4af93e2c47f4aeba4d6066038c724f5 : SUCCESS in 49m 31s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/71222f6988fc42fcbf25a1a30fac0dbd : RETRY_LIMIT in 3m 49s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1e40b4a579cd420897e0717ae6b56194 : RETRY_LIMIT in 3m 52s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/115165cb5e74438ba9d5bc43145faf30 : RETRY_LIMIT in 3m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4133edbd70ef463781a7894cd1b5c2dd : SUCCESS in 6m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/99c93f3c0230499893c5df37b6c2e7f5 : SUCCESS in 4m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4f56466dacf44381a965a10c4aa314f8 : FAILURE in 16m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a4b945b3830547beafff7754f91fac6c : SUCCESS in 17m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3d8775dcf53342a7836be6a69335b217 : SUCCESS in 15m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/575e427d10134a36ab74e58cd670a417 : SUCCESS in 19m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/09a543093e654689906ff24f7b2e682b : SUCCESS in 17m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/38d0f8f9dbda4569bedade317afce991 : SUCCESS in 20m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/81bfa9457c0c40539be4de0bf1b445ae : SUCCESS in 14m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/751edac491a54b339467016a65b1c3e5 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9b7d8298b18c491b9c56186d2173be79 : SUCCESS in 15m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e2eed84737ac4035a41006833c15f26a : SUCCESS in 6m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9dd29d2384174e8d8462ef24a0651b79 : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bc2d83ff77d74a5f8e87b6658ae39963 : SUCCESS in 6m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3a57d004e5304f72a1b4e6c417bbb550 : SUCCESS in 55m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6b88225cbc0e4d108ef930f27cb549db : SUCCESS in 58m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6982b52b8b63494bb836a5ec71d59560 : SUCCESS in 35m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6e9821cba2a64894841fe72deef95c56 : SUCCESS in 35m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/129fe579b7cc4e03bcee5b8968f80d8c : FAILURE in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd06488ba69c4bbf827c8c0ecc14de61 : FAILURE in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7f8144c6d6084b2991a4a00c0fd5f3a3 : SUCCESS in 18m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d866a52d17d4297b22ffe69f536db44 : SUCCESS in 38m 45s","accounts_in_message":[],"_revision_number":1},{"id":"dc58ddd73de1b4d625876ce7de664e49ecea5b01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-30 03:42:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bb104b30ff111072522122169b39ecf34bcf44be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-06-30 03:43:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d64297a8c3331cb76fc73c167e26cb3ab542b9cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 04:44:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/668323910346458ebc615c70a5fcf9a5 : SUCCESS in 43m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9eb816523442448ebb8a6b5811f2b604 : SUCCESS in 39m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b6080d5704a4096b62b813b6f45b291 : SUCCESS in 51m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/866773401ebd44f2abf84e7bc3050b88 : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":3},{"id":"d7f781368b4f81f96d2d4b015a8daf7d4822b566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 05:23:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1480a8e1f68b4c1d9088615ad8b78c47 : SUCCESS in 7m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/6c940d471b334743a6fe975bb00ecfe7 : SUCCESS in 54m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/51793c858d7747de894883a26038792a : SUCCESS in 37m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb62d6cd81334704b6604d3dc7fe83d1 : SUCCESS in 45m 07s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f54aa1b5eabc4a8c921334ac20593a25 : RETRY_LIMIT in 4m 27s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1a97f9acab174ef9928a46bf21214025 : RETRY_LIMIT in 3m 52s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9d36437fa8cf47938a8dcab485c6be51 : RETRY_LIMIT in 4m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/df772c6778c744169b1a60b8570493e5 : SUCCESS in 7m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6aa17ee682c7455e9d77bfcb8949c23e : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0ce6e4f38c8a4073842384e2a6c726ac : TIMED_OUT in 1h 32m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6109545d723f490c8f9caa80dfcbd655 : SUCCESS in 15m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/75d0e806a1bf462a9f799df92bb22119 : SUCCESS in 19m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/28136643e19b4d7bab14038aced87873 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/437d9c8af25e4a21a28fb101db35003f : SUCCESS in 16m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6a1146751fac4bf590a2161af90ab0d8 : SUCCESS in 21m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5dffef5e0d5c4c0ca97f7b769dc84f78 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bd56d1860efb4ebcb78cb8a005308b03 : SUCCESS in 19m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5f7fdb49dbb14e79a569b8b4071ae247 : SUCCESS in 16m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ade5fe0d94994948878a5b769b5fb5b6 : SUCCESS in 7m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ed9a69bf36984eafb4d6d0ce7665cf08 : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7d949c7a732b45e9b020279fbdd9a88d : SUCCESS in 7m 28s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fa713e2cdf4d4c539527fbcfbcf31881 : SUCCESS in 1h 04m 36s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c6534f5b156647eda7cde02c8713b109 : SUCCESS in 1h 03m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b33780a91a504fcaa0825e80dc10d93e : SUCCESS in 40m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6d28b9a5b3e24b97989ca0896327c1cb : SUCCESS in 40m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ef8c3b4c5dd4da6909db6c586457a45 : SUCCESS in 16m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5c66e580c054c8b9705b2e49032b31d : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f26d0cd95ce49e4add1858c3d123721 : SUCCESS in 25m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03f54cf78e564208bc20cec024c89a9e : SUCCESS in 31m 43s","accounts_in_message":[],"_revision_number":3},{"id":"97a26b9eb1ada602be747797f1de28061d1332a5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-06-30 07:08:54.000000000","message":"Patch Set 3: Code-Review+2\n\nYup love it! This is what we need to track digest usages, plus we can do a sum to get the number to track the number of tempurl and formpost requests.\n\nThe test failures dont seem to be be related to us.. at least no that I could see when I followed the links. Will recheck it.","accounts_in_message":[],"_revision_number":3},{"id":"e0a6301cdc47f7b6bef1c3217e0d078ca5a27c30","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-06-30 07:09:13.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b38bb565afced0bca9e38a3f2bed1c34b0963d6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 08:21:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d1f24885c4a4236975026a433f41d30 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/855d37444cf949aca4730c945ac9eeee : SUCCESS in 53m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/348523c0b0c94c169f1c0267bf565a1e : SUCCESS in 36m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d771da8bd804fcc92f51d00c15f6da8 : SUCCESS in 44m 47s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/39d9fe50f4db477d98c9481c7893519c : RETRY_LIMIT in 4m 45s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/efc66a4e9290467d8b4ebc175d524a75 : RETRY_LIMIT in 4m 14s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/181c2a817dec4105a4edf20e20e03d7c : RETRY_LIMIT in 4m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/54e24a340122486e8b6510cf741b7c0e : SUCCESS in 7m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/76f49f9cb50241f2bf7fd276405fd160 : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c83e898bdfa34d219425135af5e76222 : SUCCESS in 18m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f3a93740e1bb446a97f37978aff0b0b5 : SUCCESS in 16m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/41c4a62ea6b143b5a6f1243161742193 : SUCCESS in 20m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1756abdecdd0461eb2ff09a1d8377427 : SUCCESS in 13m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/75d3c39d7c6e4d819132f9426b60040d : SUCCESS in 19m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/06f977f4187149a6bec2fcd6e81f1850 : SUCCESS in 18m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/14878e8d996f405791e6cc97cd70ff6a : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cd72383653bc46b998fca65782a6e7ab : SUCCESS in 13m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/376c0661d52b45f5806450ce5cd65475 : SUCCESS in 16m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/97eef73a6eec4bc9b15a8a37450ddfd8 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2d1ef025298940c2a6fb7310f5e5b98f : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e363baa5616c4afcb5cead2f731f0032 : SUCCESS in 5m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/95b5756fd1534ff79a3c6d786bafb856 : SUCCESS in 1h 02m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0cbc1d28e4645268dcf8ce73543d59a : SUCCESS in 1h 00m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/270d9161358b460aaca3dd1682f44124 : SUCCESS in 29m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/04c9f93d113641759f8e9210285349b8 : SUCCESS in 43m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbabf3bd980b4975b8047fb74db14e42 : SUCCESS in 20m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6443e1e7c87d434284cebbe95ca641c4 : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7fa37d3d6d54402a922613eb11a90af4 : SUCCESS in 24m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83757f568e2b4a28bd3050a9d684d678 : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":3},{"id":"b2ba0e0f33e9821395e07a1914c0cb39aa47acfb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 08:42:36.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b508924e2de44ed19815694e39aa2b0e : SUCCESS in 1h 06m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/325667feb0a24fa5aa43533265c5af64 : TIMED_OUT in 1h 31m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6cc2c5c040f14ea0849b6bfc1fdbd204 : SUCCESS in 1h 14m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0befec6f99d848c8a20b8fe67bead9c8 : SUCCESS in 35m 14s","accounts_in_message":[],"_revision_number":3},{"id":"18357f66f411b20239945df4727b2a436a2defab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-05 00:32:37.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"5244ab8f1ea1367ac306c416d6884bcb45796d98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 01:38:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b85fe30a6bf041598543c5cd700c3fb4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/077f478377e44476a3fbb753afbfb77c : SUCCESS in 53m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9cd4d1de9e241e4b24dc583aa3c7172 : SUCCESS in 27m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ff3310728604dabb5de64c2c2c0ea8a : SUCCESS in 1h 04m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d5e4ffae181433785eff2b8d68a3eab : SUCCESS in 28m 20s","accounts_in_message":[],"_revision_number":4},{"id":"dd7c7c5de4e13bac12bcaed8408e9de739187fb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 01:42:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26262480dc8844c5a40c7c70c0080253\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23bd1103372d4e9eb00a9a1c528161c7 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/d64ee12fc43444b48357fa58b0d5d4df : SUCCESS in 1h 04m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75d634ebca114d65bf5fe3afb4198532 : SUCCESS in 52m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f5582670083467bab1e9e0965bb3e5d : SUCCESS in 50m 30s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c09f2ba50bfa4db89ea2c615a11b8198 : RETRY_LIMIT in 4m 35s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/dafaa6707b934db19a70c4b43a9f7fec : RETRY_LIMIT in 4m 48s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0d955b40ac164e828310a1d930cbe483 : RETRY_LIMIT in 4m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b8f886deb1174ca5bb006d3f4a54b5a7 : SUCCESS in 6m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/46516180664d43ed879a0dd741ac7183 : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fc488d6326af45f1a097086bf0989db6 : SUCCESS in 16m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a1df637eb7c4b4ea6d2d11042b35438 : SUCCESS in 15m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9a687664621c42d493d76a59f0d70d04 : SUCCESS in 16m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b6fefda535e947179e098d99fd852808 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/998966cee2a44675934dbf300129cfb9 : SUCCESS in 17m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/03aeabdc345c4053a508b300243d7232 : SUCCESS in 19m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d6bcd0e0ca984ef49de189eba62116c7 : SUCCESS in 20m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/02dfe6b889c643b08b7830802ae4d9bc : SUCCESS in 17m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/00eb9ca7af6c4ae68c46604390c4fb1c : SUCCESS in 16m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/81de632a1b5f4905a53890cdd2095fba : SUCCESS in 7m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2d3ab7e7a00c4df9813d158112158276 : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/82bdc0a7e0504acab65cb60835697ebb : SUCCESS in 5m 52s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1155226bfee942a7b01d68c684abb516 : SUCCESS in 52m 49s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d8cce37c8bf8457b968e8f199afab0f1 : SUCCESS in 55m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1bf67e54f85b4e2583e2f2c546470ade : SUCCESS in 33m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b2625dd08f54d4aa7561f40a438318b : SUCCESS in 33m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/db68971b0a384a91aa2f97bccd29fb25 : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c00b3f66a93f469eb40552e10c7355da : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/277c67dfc0c845bd8a899c431a2356d4 : SUCCESS in 21m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd06bd88df494565a7e426f90e677103 : SUCCESS in 44m 08s","accounts_in_message":[],"_revision_number":4},{"id":"9ba6455a9797aff4cd8b3cb9738e4392e3fb8aec","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-05 04:24:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"55e26a9970813029063126f0e070e1981afd19c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 04:25:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d5ef5b4cfbc5d7a7359f8f76acf1926138de6911","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 05:31:53.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9c13a777daa4a86bdd59c7eaa8895ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae0afae82d1a4f24b89e1ef3203e3b31 : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/f63a971451724ecc9b616e54bba28513 : SUCCESS in 1h 00m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/16bdd89f52d848c79394d31facaf7a86 : SUCCESS in 53m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d00c46a15a5946bfb63c90a8933574ec : SUCCESS in 58m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/dbaa8fd674e549cdb3025f17b7cfa0b0 : SUCCESS in 8m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ea5c942a1374456d913d91fedfd28d5c : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0f7a3bd4696a4041a3f9ed72e047453b : SUCCESS in 15m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ddac47afde8d4280b49eceeed609873e : SUCCESS in 14m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/226f6acc2c334c01a0216a8349086c07 : SUCCESS in 16m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/973727eda6ee4acd840b11ca2ec934a2 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ed7a402f9c3545debd10e7eb3b44e65d : SUCCESS in 16m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f9c3c31b14e54529b94d8a32d80def89 : SUCCESS in 17m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/66182a3f42974736bae2789fe310b409 : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/94a1ce3b8e224484b18d1211c8c37c92 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/77f26aedccc14f66b08d50e46eafd491 : SUCCESS in 16m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0ba7b398ef6a48eebfae65e1d95ef44b : SUCCESS in 6m 42s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b6f19692e50b4079b0cfce4d41f18544 : SUCCESS in 6m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ee363974365f4e1fbca9b7ce0c7944a4 : SUCCESS in 54m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/383a836842ee4b48a2647b6d47794606 : SUCCESS in 1h 00m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7a87057c58a2441e954a28828dd5d799 : SUCCESS in 50m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/df0f12fdd24748789fb05196f7395dd3 : SUCCESS in 44m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54da69b4d00f4398b43ad9a2f68181da : SUCCESS in 16m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84443c07fe3d42c7a55eecbbfac94e88 : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/97fe8477ccb24f648a49f97710e89560 : SUCCESS in 19m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7230228f3d964337abe5d224803f071e : SUCCESS in 52m 48s","accounts_in_message":[],"_revision_number":4},{"id":"7a50f411d64f0d59bab06ff12020a3e6b7171f2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 05:31:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"043277dbb8614d20985bc90e01c916c8f0e1284a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 05:33:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8310b1f25a04413e8e2e5e284e0c18dd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4393be7913634e49b27dde2c1fdcdb1f : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ec0cecf42ec74eeda7d5de4bfcc426f1 : SUCCESS in 43s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/a883f07678e943ebaf2c5a3d76368946 : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"475cdba65bcd29871b840395576572c0d8399a6b","revisions":{"6138a09833698033cc6cd8b969b60ef03989b7e0":{"kind":"REWORK","_number":1,"created":"2022-06-29 22:46:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/95/848195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/848195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/848195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/848195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/848195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/848195/1"}}},"commit":{"parents":[{"commit":"6af4449268533af57063d5be6e83d7d17c6706b2","subject":"s3api: Better handle 503s in get_container_info, too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6af4449268533af57063d5be6e83d7d17c6706b2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-29 22:44:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-29 22:44:33.000000000","tz":-420},"subject":"Emit metrics for tempurl \u0026 formpost digest usage","message":"Emit metrics for tempurl \u0026 formpost digest usage\n\nChange-Id: I203607a0576582330241172d05bf8fd223bbbb9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6138a09833698033cc6cd8b969b60ef03989b7e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6138a09833698033cc6cd8b969b60ef03989b7e0"}]},"branch":"refs/heads/master"},"22ad73af69498e6f42eca4d8c27bfbe57af16c0a":{"kind":"REWORK","_number":2,"created":"2022-06-30 03:42:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/95/848195/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/848195/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/848195/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/848195/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/848195/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/848195/2"}}},"commit":{"parents":[{"commit":"6af4449268533af57063d5be6e83d7d17c6706b2","subject":"s3api: Better handle 503s in get_container_info, too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6af4449268533af57063d5be6e83d7d17c6706b2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-29 22:44:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-30 03:41:40.000000000","tz":-420},"subject":"Emit metrics for tempurl \u0026 formpost digest usage","message":"Emit metrics for tempurl \u0026 formpost digest usage\n\nWithout this, it\u0027s really hard for operators to feel comfortable\nremoving support for now-deprecated digests.\n\nChange-Id: I203607a0576582330241172d05bf8fd223bbbb9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22ad73af69498e6f42eca4d8c27bfbe57af16c0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22ad73af69498e6f42eca4d8c27bfbe57af16c0a"}]},"branch":"refs/heads/master"},"724d4a56e214637271eab6b862fba7d40573a283":{"kind":"REWORK","_number":3,"created":"2022-06-30 03:43:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/95/848195/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/848195/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/848195/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/848195/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/848195/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/848195/3"}}},"commit":{"parents":[{"commit":"6af4449268533af57063d5be6e83d7d17c6706b2","subject":"s3api: Better handle 503s in get_container_info, too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6af4449268533af57063d5be6e83d7d17c6706b2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-29 22:44:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-30 03:43:00.000000000","tz":-420},"subject":"Emit metrics for tempurl \u0026 formpost digest usage","message":"Emit metrics for tempurl \u0026 formpost digest usage\n\nWithout this, it\u0027s really hard for operators to feel comfortable\nremoving support for now-deprecated digests.\n\nChange-Id: I203607a0576582330241172d05bf8fd223bbbb9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/724d4a56e214637271eab6b862fba7d40573a283"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/724d4a56e214637271eab6b862fba7d40573a283"}]},"branch":"refs/heads/master"},"475cdba65bcd29871b840395576572c0d8399a6b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-07-05 00:32:37.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/95/848195/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/848195/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/848195/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/848195/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/848195/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/848195/4"}}},"commit":{"parents":[{"commit":"6af4449268533af57063d5be6e83d7d17c6706b2","subject":"s3api: Better handle 503s in get_container_info, too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6af4449268533af57063d5be6e83d7d17c6706b2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-06-29 22:44:33.000000000","tz":-420},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-05 00:30:54.000000000","tz":0},"subject":"Emit metrics for tempurl \u0026 formpost digest usage","message":"Emit metrics for tempurl \u0026 formpost digest usage\n\nWithout this, it\u0027s really hard for operators to feel comfortable\nremoving support for now-deprecated digests.\n\nThis patch adds new statsd metrics in the form of:\n\n  formpost.digests.\u003cdigest\u003e\n  tempurl.digest.\u003cdigest\u003e\n\nSomething like:\n\n  formpost.digessts.sha1\n  tempurl.digests.sha512\n\nChange-Id: I203607a0576582330241172d05bf8fd223bbbb9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/475cdba65bcd29871b840395576572c0d8399a6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/475cdba65bcd29871b840395576572c0d8399a6b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[]}
