)]}'
{"id":"openstack%2Fswift~848527","triplet_id":"openstack%2Fswift~master~Ie586f428fd86f1a6acd00271f8e8fb07f7eb7316","project":"openstack/swift","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Ie586f428fd86f1a6acd00271f8e8fb07f7eb7316","subject":"Imported Translations from Zanata","status":"MERGED","created":"2022-07-03 04:22:34.000000000","updated":"2022-07-07 23:51:39.000000000","submitted":"2022-07-07 23:50:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848527-zanata/translations","meta_rev_id":"2a18f662b1902040f5df616e5ad4fe82a35d10ab","_number":848527,"virtual_id_number":848527,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-07 23:50:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-07 23:19:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-07 23:19:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-03 04:32:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-07 23:19:12.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"a251d5f371c54151f416a2c195d8d7479cfce356","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-03 04:22:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"faecec82033226e1a06a4973d7855570b48a3991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 04:32:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f98c574e3afb4fab8f812962aa8ab13f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb63912b545341aea83e8f8f917b4974 : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68af6083bd984ef09e22c2f3671ac947 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4443bfbd7c1d4729b45d3b9e33cb2e1da2a324ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-06 04:25:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"26aa5d9546e81c243b67c4afc249f9c6ca4e6815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 04:36:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58ac947e514d4b53aecbeab3ee1a94fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb43d5e8323c4573a97413cb1d646fa5 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/806224fbe4ef4db7972c9260f48e7923 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"320dfa32664d9e718e1f5210659a853376c74c10","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-07 23:19:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"18001326e70b5b02093820c388c30b9de1520d7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 23:19:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"25aaf8d77d115d2fb0183a2a9ac2b338f48910b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 23:50:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06ac05fc4cc54b2499d08754b2a02a27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3fdb6f168e04c3b8328061459d4239b : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/140f19687702401b80cd6a3c2232bc8a : SUCCESS in 5m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aa118935ff1f4a3ab12a55d31479c9bb : SUCCESS in 5m 07s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a370de916e94e54bbd5f5e60d7178d1 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2},{"id":"17cd28f348044384bbfbb890ccd6beb28f40ad26","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 23:50:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2a18f662b1902040f5df616e5ad4fe82a35d10ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 23:51:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c209be4d87d142a0b90fb552acbfa92d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d11e2d11eb24396a1102a8698fcb99f : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28ed870f2ea3413683afb88c21e5c18f : SUCCESS in 57s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8e1bfbd493904d078a883ebb03ed6e93 : FAILURE in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/6cdeec5cfa814392802293d3b8543c24 : FAILURE in 32s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a55016e57fc68394fc5c8f8df6965895040d6460","revisions":{"baf088c3e005863da42afc5749b6c67d44db53ec":{"kind":"REWORK","_number":1,"created":"2022-07-03 04:22:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/848527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/848527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/848527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/848527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/848527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/848527/1"}}},"commit":{"parents":[{"commit":"6af4449268533af57063d5be6e83d7d17c6706b2","subject":"s3api: Better handle 503s in get_container_info, too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6af4449268533af57063d5be6e83d7d17c6706b2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-03 04:22:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-03 04:22:33.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ie586f428fd86f1a6acd00271f8e8fb07f7eb7316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/baf088c3e005863da42afc5749b6c67d44db53ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/baf088c3e005863da42afc5749b6c67d44db53ec"}]},"branch":"refs/heads/master"},"a55016e57fc68394fc5c8f8df6965895040d6460":{"kind":"REWORK","_number":2,"created":"2022-07-06 04:25:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/848527/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/848527/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/848527/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/848527/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/848527/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/848527/2"}}},"commit":{"parents":[{"commit":"475cdba65bcd29871b840395576572c0d8399a6b","subject":"Emit metrics for tempurl \u0026 formpost digest usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/475cdba65bcd29871b840395576572c0d8399a6b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-06 04:25:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-06 04:25:24.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Ie586f428fd86f1a6acd00271f8e8fb07f7eb7316\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a55016e57fc68394fc5c8f8df6965895040d6460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a55016e57fc68394fc5c8f8df6965895040d6460"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
