)]}'
{"id":"openstack%2Fswift~848691","triplet_id":"openstack%2Fswift~master~I155dd81167c84714fdc82c0caf757ce0aab18700","project":"openstack/swift","branch":"master","topic":"safe-reload","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-07-05 07:47:05.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I155dd81167c84714fdc82c0caf757ce0aab18700","subject":"Swift-reload: maybe use psutils?","status":"NEW","created":"2022-07-05 06:24:56.000000000","updated":"2022-07-05 07:47:05.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2e01ff6c29fb1b8459bb9dbad5c386b169c17ee5","_number":848691,"virtual_id_number":848691,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-05 07:47:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-05 07:18:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-05 07:47:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0ad8f5b9768f69e85668cb7907d748750d726b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-05 06:24:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e1f99230a7cd30da51dd193bc8efa0d27fddbf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 07:18:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64c3ae033aca4e9588dc26f9cd56fe15\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1479e17802d54824a5b1a0492c140cf4 : SUCCESS in 28m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8352926c65974857aaf1573732c1646e : SUCCESS in 51m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e61b808e9fd472b918bf8c49469b379 : SUCCESS in 27m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5615022ad8a45e3b3114bd54c0be95e : SUCCESS in 48m 13s","accounts_in_message":[],"_revision_number":1},{"id":"2e01ff6c29fb1b8459bb9dbad5c386b169c17ee5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 07:47:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4719f7a3b7c7404bb21459cc617801a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a2d9c4fcd79490597797c733eafadac : SUCCESS in 7m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3d5fe378684e44a69757c0ca0f171e8c : FAILURE in 2m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/ae2b13281207437f9a366ca8ecb52db0 : SUCCESS in 1h 02m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eb661344fd894d2f98f73f3f2c4e7f54 : SUCCESS in 51m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/745481743ab640ebab2ff076f2915f01 : SUCCESS in 55m 20s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/dd35971f421b4de08d1691a966a68bfe : RETRY_LIMIT in 4m 39s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8c5f4c0791dc4b18b6281f8830170d90 : RETRY_LIMIT in 3m 45s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0ee1fcde174149b7b2ae48899f5c4dd2 : RETRY_LIMIT in 4m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e3cedc1f4d20473b81f5925bbd0c0cef : SUCCESS in 7m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1e54892950b94d7a951c68ef54f419a5 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/abdadbc8b22f4c3a8fd2744a2dc8af7e : SUCCESS in 16m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7abd8e4d5f2c4667a2736f6c6dc17751 : SUCCESS in 14m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c71f9b0b38284564b9d4bc9243a66e5e : SUCCESS in 15m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8d77e2e8b9a34b98bfe6bd90aa2d2218 : SUCCESS in 20m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/534eac53849e4090b4b19f2c03da077f : SUCCESS in 16m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/514c27c93c814b06916c09bdc6df8dde : SUCCESS in 20m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e859add3aa2c41acaa55162d309922ca : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ddfb786d458644c18374f70e7f3fb3a1 : SUCCESS in 15m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/adcd3a7880064c919aef8d20ea4fd532 : SUCCESS in 16m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a994c5aac2cd49159d5005b6c6fe5bc9 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/468b8bd563b34027b73ecbfbd0b5f69e : SUCCESS in 17m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/41bbf26c492947dc9b86d03eca7df1f4 : SUCCESS in 7m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6a941daa5d334a649ef99732598a807d : SUCCESS in 52m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/97db1be3b08f4588bfd747b1eb68a30d : SUCCESS in 1h 16m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ba7057b100041d6aa22804c5be4ece9 : SUCCESS in 37m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/834432adc29d42fcaea39960b1c1c0ce : SUCCESS in 42m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6611df23d80e42e58c51461527113943 : SUCCESS in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea69f651d3004ac38c738bace3f7b1d4 : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eb39dc09db3943a78a2213f3a7211542 : SUCCESS in 17m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f08b2245c5845adb68e8c65df734c56 : SUCCESS in 47m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6cfcb16a4c15905665d900554eadaa81d05c316","revisions":{"d6cfcb16a4c15905665d900554eadaa81d05c316":{"kind":"REWORK","_number":1,"created":"2022-07-05 06:24:56.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/91/848691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/848691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/848691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/848691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/848691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/848691/1"}}},"commit":{"parents":[{"commit":"3ba4d53afcf5b3bb817f901001e24c33eb2f365a","subject":"Add a swift-reload command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ba4d53afcf5b3bb817f901001e24c33eb2f365a"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-05 06:21:34.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-05 06:21:34.000000000","tz":600},"subject":"Swift-reload: maybe use psutils?","message":"Swift-reload: maybe use psutils?\n\nMaybe rather then calling out to ps with subprocess we could look at\nusing psutil, it seems to be mature and used a bunch elsewhere.\n\nChange-Id: I155dd81167c84714fdc82c0caf757ce0aab18700\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6cfcb16a4c15905665d900554eadaa81d05c316"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6cfcb16a4c15905665d900554eadaa81d05c316"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
