)]}'
{"id":"openstack%2Fswift~849441","triplet_id":"openstack%2Fswift~master~I6052fe4e9d6706bd0390bba0883cf1861ba3b2fe","project":"openstack/swift","branch":"master","topic":"ring-writer","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-07-12 00:45:25.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6052fe4e9d6706bd0390bba0883cf1861ba3b2fe","subject":"ring v2: Add repair function","status":"NEW","created":"2022-07-11 22:09:53.000000000","updated":"2022-07-12 00:45:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ebc1ecded5a6c544b2b34b13d58b88095151b29f","_number":849441,"virtual_id_number":849441,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-11 23:18:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-11 23:18:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 00:45:25.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"a4ba7fe114f818217cb70817acf5e65de8cbcc6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-11 22:09:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd60a824fb087e867ad71016cd99ae49b4bba2fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 23:18:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/961c376e40014ce5ae374f9e31f393be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/926c781f581942d7851362481b68d6fc : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/7087084a3ffc4b6792ac671c363098ff : SUCCESS in 46m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d8e5fd74b1ae413ca38dfe475fc34715 : SUCCESS in 46m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b70e4aba2ec14221bb75b2c909558bae : SUCCESS in 43m 59s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7b54a4e9032247b585085b84982a3760 : RETRY_LIMIT in 4m 31s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ea58fb4aa2d148cb9576bd47f2288503 : RETRY_LIMIT in 4m 25s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/932f9ac9ab274e6080b7727611ffb335 : RETRY_LIMIT in 4m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/23e4c35d0071492ba8208a6aaa5841cb : SUCCESS in 7m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f648c53d812641a78721260013bc1606 : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28f7e4bb86d04ce5a2da97b42c5e46e2 : SUCCESS in 21m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b15c55ff562b426c813c12bffe4ba384 : SUCCESS in 20m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7d773511acce474b84ac572f1e10a4d3 : SUCCESS in 19m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2eb36f4eddf548c8907a2fc51954cbf7 : SUCCESS in 21m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4b1980b109bb474aa2fbc4a255cea901 : SUCCESS in 23m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1baaf4c8241547eb88367ba405d351eb : SUCCESS in 22m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/446b254d115849e782000705fd0b3034 : SUCCESS in 22m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/87a546f4080148f6b4079b838bdc949b : SUCCESS in 17m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/db9e09c7fce845bd9a4cb84162fe1dd4 : SUCCESS in 22m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3c13a2892071476f89e725a1e652733c : SUCCESS in 9m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/db51a176bdce491c9f8a568330283ad6 : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f75382abab8545daae11e6540aadef26 : SUCCESS in 9m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c1422e11fe9a4362910ebd65746b5341 : SUCCESS in 1h 02m 07s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/24d7038509534b168697d9e92f83faa9 : SUCCESS in 1h 07m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f43226b7a6fe49a0bfde61f5804f918d : SUCCESS in 44m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d58f3304bdf4fb9b838260d3b91c505 : SUCCESS in 46m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7afe669c06f6498ba5d920d2aaae0936 : SUCCESS in 20m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1adebec4d764bfc8dd18b2cc2b45744 : SUCCESS in 6m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/512c27f950304b0ea835d2cfe3cd48ea : SUCCESS in 23m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1981777e613c433798cbb93cf69d1b92 : SUCCESS in 51m 19s","accounts_in_message":[],"_revision_number":1},{"id":"6e0f8345c1d1a47d05066eb6ec6bbc02954899e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 23:32:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14d53434f91647dda0696529defd7a70\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2d709231b8d4a4ca2540ba92a3dda52 : SUCCESS in 29m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d45f67b46ee3464a999ceb9992296d11 : SUCCESS in 1h 11m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/361e5c497b214bee9fc5e9589752f26e : SUCCESS in 29m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6835429c173a4eb1953825dc817b4377 : SUCCESS in 1h 20m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ebc1ecded5a6c544b2b34b13d58b88095151b29f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-12 00:45:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"beea93410244bf16c4eca514304c24c669f543ed","revisions":{"beea93410244bf16c4eca514304c24c669f543ed":{"kind":"REWORK","_number":1,"created":"2022-07-11 22:09:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/849441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/849441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/849441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/849441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/849441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/849441/1"}}},"commit":{"parents":[{"commit":"a5dfc19ab37aba4a8c5d6e47d216eaaef90a2a6c","subject":"ring: Add a rebalance history in the ring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5dfc19ab37aba4a8c5d6e47d216eaaef90a2a6c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-11 21:55:51.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-11 22:09:44.000000000","tz":-420},"subject":"ring v2: Add repair function","message":"ring v2: Add repair function\n\nChange-Id: I6052fe4e9d6706bd0390bba0883cf1861ba3b2fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/beea93410244bf16c4eca514304c24c669f543ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/beea93410244bf16c4eca514304c24c669f543ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
