)]}'
{"id":"openstack%2Fswift~849450","triplet_id":"openstack%2Fswift~master~Ic3b3f21ce67c640dfb632b79fd09902f14576eaf","project":"openstack/swift","branch":"master","topic":"p-sharder-deleted-2","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-07-18 18:10:52.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-07-18 18:10:52.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-07-18 17:00:19.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ic3b3f21ce67c640dfb632b79fd09902f14576eaf","subject":"Sharding: Sync container_stat table with fresh db","status":"MERGED","created":"2022-07-12 06:12:56.000000000","updated":"2022-07-18 18:12:10.000000000","submitted":"2022-07-18 18:10:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"849450-p-sharder-deleted-2","meta_rev_id":"ca497409c4628efafb0ec2a8ef94968360c40def","_number":849450,"virtual_id_number":849450,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-18 18:10:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2022-07-14 08:51:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-18 17:00:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-18 17:00:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 07:11:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-12 07:16:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 09:25:38.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2022-07-12 23:41:48.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"630fc532d2f0c4625a9a52168f0f475797ff351f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-12 06:12:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79d51fce8176f0e36965dc24e70590001c27f23d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 07:11:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25507d8f620f478b85c04871f00a246b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4934fe96829b43dab31c199efdebef56 : SUCCESS in 34m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32817188ba834018afcb785006c2c8d8 : SUCCESS in 55m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/883962f17ae84cab962f188c04c8e9a1 : SUCCESS in 34m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2773168a2df548a68527272142e5c45c : SUCCESS in 34m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3a42cf3018a55347acf61c9ade56df7a564cda35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 07:16:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a1e5d71c6f84f598a0b60ef8a66571d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/422f659773674383b7340fd3da73d119 : SUCCESS in 10m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/3ce505e362de462a8d82f89eddfa9fc3 : SUCCESS in 58m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8b5adcc7fb0044718c9aef17a15d22cb : SUCCESS in 51m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d30fc822fea045ffaf874ddd446e8ca0 : SUCCESS in 1h 01m 32s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ddffdd18b6d44535b319a10109667ac8 : RETRY_LIMIT in 5m 19s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/89096e33a0224afeb4f0b47255cd7e06 : RETRY_LIMIT in 4m 26s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/05d6191e12bb4296a96ec28a7e30e999 : RETRY_LIMIT in 4m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b89984cb64ac440fb2965334f75bd81b : SUCCESS in 7m 42s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0a29deeb32824dbdb47bc66ce0b52da4 : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c34c9d6bfe94dddbce7e5d3d965665f : SUCCESS in 19m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/69b5cd76e60f482d9a02f5deb420f35b : SUCCESS in 18m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ae46168f240b49f89c9c144b3814bec6 : SUCCESS in 19m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9a7799266da242abb9043c94a6eddc60 : SUCCESS in 22m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b5472953bf0845e2b798655d9a9eb401 : SUCCESS in 25m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0cae92882dd84df08cf6b392761220f2 : SUCCESS in 22m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4666408b13bd403abce4362c7a34f10c : SUCCESS in 23m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/86fd6b1bf161486380394728799ca042 : SUCCESS in 19m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c6e4c1ba2e6349ddb684ec0d7aaac0ca : SUCCESS in 22m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/572fd87f806e40e793e74a58ceb97502 : SUCCESS in 8m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7a1d08a0461940ada34f4cd3718420e9 : SUCCESS in 21m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4f1538ba4f5f4964ad5c19e42db636e9 : SUCCESS in 9m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/708f8185bce14ff4ac982c931167c24a : SUCCESS in 52m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fd65dc23ae74a9fa50eede3456fbdab : SUCCESS in 1h 01m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5454c122203e4da0aa65d01bf1637c98 : SUCCESS in 45m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c383f64a9727471198ff4fccaf74e586 : SUCCESS in 47m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78dca751624646068473650d82502910 : SUCCESS in 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab2074c072b84d8cbc97a8b5ed620691 : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/38358233c41d4dd09388da40e1f38c44 : SUCCESS in 34m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aee7f463493f42a4921c721ce25d3953 : SUCCESS in 54m 05s","accounts_in_message":[],"_revision_number":1},{"id":"538fd85d9b0e25af9fef5320f12f08df24e0a476","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-12 09:25:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b495252e024a3a6cf1e28f026d354b149bc1c9ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-12 09:28:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"417471e590acc139ba936d7eac41c4003f71d388","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 10:41:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b4679a54d304de391a4e1cc1d598359\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/00c55483a9554619857a9a7178167a69 : SUCCESS in 33m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f8afefcdd7d443d9da1ee38f283dfac : SUCCESS in 1h 04m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/966b62e7c1b24307a09be71f85972002 : SUCCESS in 35m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/637293fc1eb84886b16d704897b38dd5 : SUCCESS in 1h 10m 58s","accounts_in_message":[],"_revision_number":2},{"id":"41f39e5f56bb0ad1f836823b09f7a1f8fc2e4828","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 10:46:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1019be2773d249f29e05e61bdb84f991\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4c568312764282b0ec276a4f0a8ddd : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/96d939da26a54d31a0fbe3aa3e28c7ea : SUCCESS in 1h 10m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/33ff286f60224cc4b854a134d5e32313 : SUCCESS in 47m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/098836fc1a4548ac9e72b36a42279112 : SUCCESS in 1h 00m 10s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/267f2226cea04cb68cf397fee32baa25 : RETRY_LIMIT in 5m 07s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0f143d1cb9cc4c74be5913c5e8446767 : RETRY_LIMIT in 5m 00s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ddb690b12fb54a43b4ab53c3898c181c : RETRY_LIMIT in 6m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32a43718183a4bf08d80b88209c6fd25 : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c5617bfbb3c949f3a70e3736ce33c618 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d2cc9ffcde61463f86e62ca06d7d5c39 : SUCCESS in 16m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b99c2d8eb8ac45eb8f6154e9c5cd420a : SUCCESS in 17m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7c83b1f4563f4d5f871e99cd4a995755 : SUCCESS in 14m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c7af4ebe57d447b69120045b839f8a0c : SUCCESS in 15m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2117b6d112c147699b265ae8bbe5fda5 : SUCCESS in 31m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b10c94a82a4141dfae581dc82a6ad93f : SUCCESS in 31m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a38ed33faa4a4966a4dc8c84308cdecd : SUCCESS in 22m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/25713c68a2a14623873943c12cfafff3 : SUCCESS in 20m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7bc7af9f65054afaad14a5433f8f25af : SUCCESS in 23m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/960b497f11024087bfb3c4e5a179ffcb : SUCCESS in 8m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/475c809902984a97ace69e32b9e217e2 : SUCCESS in 14m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2f260f246e0f4dbeb2cc3b5a8285dcb6 : SUCCESS in 8m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e28b5559e1cc40b3a007de8c9b5bcf8e : FAILURE in 46m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec46dec8dc3c477b86661d96c445260d : SUCCESS in 1h 02m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d509e8a6e3cc4745b6e7640cc4bbc344 : SUCCESS in 46m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/23be7bc33b194ab8889e627a2b627e47 : SUCCESS in 59m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2c8e6ac4f254de2bb26f75dc90a5c24 : SUCCESS in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e36deb596cbc45ccb7e5f26d3b84f8af : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/703de521d688432aaacc1ba927ba9bca : SUCCESS in 19m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36872eed201847718862598587860615 : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":2},{"id":"a0b1677191322b95baa142784cee3664acd6c6fb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-12 14:57:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"40be057a472e34199b2c88c7eac079d9ad598b6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:09:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/573f857c367047518297a8c435d0f476\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b52385ab7cc142a2bc90c3d619a7364b : SUCCESS in 43m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/190299be2ebe4f4f91520a0ae88bcff8 : SUCCESS in 1h 08m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9cb8dd7bc9bc4675bb76a03dd08449d2 : SUCCESS in 45m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/628c777a2eef45e393ac8050aa793bdb : SUCCESS in 43m 54s","accounts_in_message":[],"_revision_number":2},{"id":"44432bec6ab210256faf859de3c90145d2ac86cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:17:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/426767ff6ad14be9946011c6a5975ee3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ee973e61d447f4804753576401dfd1 : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/41864ce91e504bf9b0bf1a93af80956e : SUCCESS in 1h 09m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cb9839cf0a6247d9b9d2dd2e98898a3e : SUCCESS in 57m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86050021c23e404bb5ee9e95204c9aed : SUCCESS in 1h 02m 38s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f48f8de5f7b349e5a29d26f165eb70d4 : RETRY_LIMIT in 5m 20s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/835ee487d7144bdaa3cb4f6145c1559d : RETRY_LIMIT in 5m 35s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5f8b0cb3ce5f49f1850a0c995cb73e33 : RETRY_LIMIT in 6m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c4d4e1dd2bf4365a52ff125ff8ecc92 : SUCCESS in 8m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a3180c03ed7b4815a113763d99a55efd : SUCCESS in 6m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5e7f2828cf4b478ca83ca99298efe023 : SUCCESS in 17m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f791c527c82d4579a5954c4c07c3801e : SUCCESS in 16m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/afa2e7885b30400a9dae90183f5e757b : SUCCESS in 16m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f57216c9f8254e8e9986bae7b56f5ec8 : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ad8d36c39131431eb54bfed22df91c4c : SUCCESS in 18m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/97a485ff2fe54801bf74011ec3df95d2 : SUCCESS in 20m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2e2fae5e4b3146c2a03998e925e61112 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/547308223fae4e7298dfa2d607f89636 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b6e38017c00f4536aeea1f2bf4825668 : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a1f085405ee54e258e842254ce5380a0 : SUCCESS in 8m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2330e9dad94b4bb2892337b6fad143d5 : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/867172c353d84f96ba70484d0d663561 : SUCCESS in 7m 06s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fd8e5c672293456aa864c82e74424d97 : SUCCESS in 55m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/77a2db162a664a799adab15ae513890c : SUCCESS in 1h 01m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d20b40a93c794cb18d1f0d4ce4c2dadd : SUCCESS in 46m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1051afb4b46f4b6aacc87ba8a20577cb : SUCCESS in 49m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b608b93dfede48659e17d8fd1b2589b1 : SUCCESS in 19m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cec1fa825f14890857d6e0569469b9a : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21cdb92d42034a7dbaf06cfca216928a : SUCCESS in 25m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ff5657952af441089c7793e806678bf : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a841256068c1c93507c806518356a0402d5d4e64","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-12 17:52:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8114186e06ef1f13bb26503074b2ba605e9bfc84","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-12 23:41:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dc26b27aab1e0a3253dc4b7c1fe196d1db0bf233","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-13 00:15:17.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e76d4679d664a849052b33c0e0e87ae903be91e2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-13 00:28:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a6887d9b59367ed32443ac1740dc46b1264515d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-13 04:05:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7fa8dcdd1dc935450a8d7c474d65178e6143a664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-13 05:05:21.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0298fb73b804cc95ebac3f22616dc83339a7f18e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 05:42:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0100bac9b2044a3abb3420450e285fb6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ee9d1b6515b483ebdb922ca52208aad : SUCCESS in 29m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b0b447d80d0451bb3e31d1b29f27de8 : SUCCESS in 30m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1e2b29b85c64697a0c53807a57bd2f7 : SUCCESS in 30m 11s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c12058c08d049679911ac8519d69004 : SUCCESS in 29m 39s","accounts_in_message":[],"_revision_number":4},{"id":"5ed133894e60d6e46a66d6ce8a4e618aa969f70c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 06:17:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68f3ca95df614ceba4b0f31b80e15b7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c15cc733fb042e59ad8131e771a60c8 : SUCCESS in 9m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/09c8d70efa19490cb8e31753b622ba31 : SUCCESS in 1h 04m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7f97346c010f44c7b73cc2d266c7d14d : SUCCESS in 55m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0394a88555644e23a949fb36162b6923 : SUCCESS in 59m 28s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c3fc0adb49af44308799da90e5149d24 : RETRY_LIMIT in 5m 34s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7c74570e76cf4cfc9d497d44d685f97b : RETRY_LIMIT in 4m 48s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ec168358687a42ec8243e368c73fe537 : RETRY_LIMIT in 5m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8caa3f34c96444dda12ba48522ebe6c1 : SUCCESS in 11m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/faca3198439b4fbab61c4c2bbe467eb6 : SUCCESS in 9m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/980b7bdb767a41efb5aa1e50b7bfd952 : SUCCESS in 16m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/94022f76219e4d71a0b12ab9674d9556 : SUCCESS in 16m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/41351116ba6e4f0288426c5c942735b6 : SUCCESS in 17m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/be8f6b9baaf74339af3534e048c5ae1c : SUCCESS in 19m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1e78f21603f241008dd00d476ba0daca : SUCCESS in 17m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/efa391380c4149eab4721ddd86f0e113 : SUCCESS in 16m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/416234de5f79461995563c2ba50c858d : SUCCESS in 18m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/749b1baec1164045914aa35c96fad2e3 : SUCCESS in 18m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bae6ce5fa2484885bff251c221f4091c : SUCCESS in 18m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7fee9b3311db412f81c90ed0160804a4 : SUCCESS in 13m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e11ca86484ee49259d40173954d2c9bb : SUCCESS in 22m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d8a6917ad0fd459496f9b9b5fc305ec9 : SUCCESS in 9m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/398c7d82083041f4a6d2f06e6fdec25f : SUCCESS in 1h 07m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e93540750bad411182101c160f973d53 : SUCCESS in 1h 10m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c654a5ae1899486fadb96fa483ef407b : SUCCESS in 46m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e56b320692444d3a9fbdcdd8aa538511 : SUCCESS in 44m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2da1b5bac0a541ad860f4355d433e7d4 : SUCCESS in 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87ecfae799944bbc9a37650a09a49fd8 : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca57a4a7607d4525b336496747992031 : SUCCESS in 28m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e71c94196cbd4d5bbecd1180224f5312 : SUCCESS in 44m 20s","accounts_in_message":[],"_revision_number":4},{"id":"b9fb61d48a0f476c4dd880ae2fdf935a1f47d176","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-13 07:30:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7cba192da0bab42f73dd219b514dd9538c55e498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 08:47:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb088a8daeac482a85776a3398bb54fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fd5e5ad44844dd5ab590370ee73b5ec : SUCCESS in 9m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/261b28ce8b444019bd9c83d36c4f238c : SUCCESS in 1h 09m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fdebed72c9564e85ad622f012e1ce896 : SUCCESS in 58m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4d5df849cec4ffe88fca5bf7605667f : SUCCESS in 59m 22s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/18eef53b76e84a6dab7a0e5ab5dc5957 : RETRY_LIMIT in 5m 26s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/39cb425c3639448fb2745ee894ddaf91 : RETRY_LIMIT in 5m 43s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d53e91151ee047a0ba5855734ce18c3f : RETRY_LIMIT in 5m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d4bda9cf140040f0b54128ad920704da : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f1cf773296aa41fbab7002b8e9bd46d8 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/24b528d519ce4e55b4008b5963206b1e : SUCCESS in 16m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ebaa43d833464eccb21ed11d407d9175 : SUCCESS in 15m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/24cd73aaaf874507bd0d103d50b4ff3d : SUCCESS in 16m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/743eca1b465145608de6b31e36c7c23a : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e55afa4f05a34e0db942c4efc354b559 : SUCCESS in 18m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/86a4290456484427b0c5d3831849b47f : SUCCESS in 16m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/271d6637b84140158ba40578eec7d067 : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6312b3eb45eb4475bbb7a0e2cdb7c34e : SUCCESS in 16m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/75f2f2a330fb472e86b6a85cafc39f05 : SUCCESS in 15m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/197aa431bc0c47fba102b2c6bc472de7 : SUCCESS in 7m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6d59c7d22a1346f585efc26eae7108dd : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/648ff839ef82470ab43cf0d9b187de54 : SUCCESS in 8m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/95f0e7aa58f04534afcf877b5c0ce1cc : SUCCESS in 54m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae4212a728644ca7acd138ad4c3a822b : SUCCESS in 55m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7cfa224eceea4d73b6fb8273a504236a : SUCCESS in 52m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/361646438f9b4965b905d7978c08ae4f : SUCCESS in 38m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b2151e5a60640c89490490306953198 : SUCCESS in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f27a2bcc1d94912a1175d5744fda18f : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f2844e8ce438424a98f3933ce767f9dc : SUCCESS in 34m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49a1798558604c34a929e478e0090c33 : SUCCESS in 43m 27s","accounts_in_message":[],"_revision_number":5},{"id":"37d5796496861b65fec8e310efb437c195d3e3df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 09:01:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/752637b0d4874745a87d632a4990ad19\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b58681daf08949be8a7a9b4252149e16 : SUCCESS in 50m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63c8724dc72446deaa50336e3724963b : SUCCESS in 1h 07m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1b5df5e94e149e9baf019d04ae38228 : SUCCESS in 1h 29m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/31ee95e813a246c2bc2987d33569476c : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":5},{"id":"4d2e1e3d5d733c4d86c728b208ce401997ef26c4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-13 10:31:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c44917984bb8bf13f77302aea7c26354566160db","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-13 16:41:20.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b2e5da1186d5ca0a6409410c0656099f012fffec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-13 16:41:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"66d4ce942b34eb95bd91e8820618d904391adb50","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-13 16:49:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a4194bb7af80bdd9a5f9ae5af80e0b2c21695523","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-13 16:49:55.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"2fb54c28d412adfe18dcf44027b10f9952174037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 17:42:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fad68949c61f4cee9e215d4902ffb11d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b18c6bec47924ba186f17b78ea7520c1 : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/d07d08bb1167492fadf22d4f8587fb2a : SUCCESS in 59m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/62be41f0f472405ebf5fc2191b4c24c1 : SUCCESS in 47m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d71b4ecd19d44c33a480992a6178710a : SUCCESS in 52m 14s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4558d75dfaa842bfbd307fb003ce5fac : RETRY_LIMIT in 5m 03s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/044800d0b5164f2796f0881e58af7039 : RETRY_LIMIT in 5m 28s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6da9ec843f9d4eb98a3e127d1f84ee7c : RETRY_LIMIT in 4m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/649480236c454cd6b1c9f1066b6a1197 : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/87ef4a6d1bdb491eb7e74d9cd52e17a5 : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0ebe1178208e4af49c52165b66e5cab7 : SUCCESS in 19m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b640b624dfc84e3391c32f0eee176f41 : SUCCESS in 19m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1244e0f8ac9845418037095a93a90c67 : SUCCESS in 17m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fc8b023c16e64d9283cdc1623066cf43 : SUCCESS in 21m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a97d9cc6f3424779a0119faa835007c9 : SUCCESS in 21m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6712b2294adf4d90af218b51fccba947 : SUCCESS in 23m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/302a056cb7394d0690e22e8cfc28d00b : SUCCESS in 21m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8bd38b4edb20476181a5a52956b72914 : SUCCESS in 22m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/22ed7fcb7d1441f7b5b1fe5e8e33bc7f : SUCCESS in 23m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b9efe1c611740d99212d0807c54ef0b : SUCCESS in 12m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9a5971aa008a4014b1950c4ab517ac29 : SUCCESS in 22m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c561c9c05ea44054a8c379654b6a9760 : SUCCESS in 9m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0702814f61e948bc8f56f4282cd01153 : SUCCESS in 54m 55s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd08f68ef37246c5bfb393b4a306108b : SUCCESS in 56m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f861876a1c343e4b34e123854ef47ed : SUCCESS in 47m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a139753dfcff4a5abc16a57ca69cbbd7 : SUCCESS in 46m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55ece6ee56164659a35ecb067d449e01 : SUCCESS in 20m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16dd9c3c8369463cb33c7861403045c8 : SUCCESS in 6m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a463fd0d939545fcb02b2cc88e278bc7 : SUCCESS in 31m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bfac56c559e47b686c5a71ec94a75c1 : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":6},{"id":"25053937f08c426c8ca9b826e7d8842dc8e5be06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 17:56:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2832bf433324a36bc786e64e5f36930\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b6df999966b47ec8824dd91d0652123 : SUCCESS in 1h 11m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4bd9c066b6a44ec981a7a2d1d400b684 : SUCCESS in 36m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f320b163c754592aad10f3dec8b19f4 : SUCCESS in 36m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7de3747e464748f3847c39380fc7ab55 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":6},{"id":"b7c9b99d05c3c439fba71417f0caad453a241ceb","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-14 00:56:51.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d51dd46403288eb7e1c67666db621056ecbc55a2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-14 02:27:29.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c203e97e02e6501a6191fe599c61f900bf57b425","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-14 03:46:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"59c0a041d6d7472e3e982d9e10ccc8290df469ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-14 04:05:51.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"990db03cf4d025a163b20ac565fee2706f30a06b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-14 04:46:12.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9fa8548a3f49b01f0d3e631942eaaa44dbb0eac8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-14 04:46:23.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"db027acbe3d202e1f6c7500354e8195aaa679ce2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 05:52:36.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0854d34fbf54908be880b2a4d1cd4b5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5720dc3ba4447d79cb51b4e17a8b600 : SUCCESS in 1h 00m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0cd338a78eda4116ba2d810b99d9f442 : SUCCESS in 30m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/639a4333455e4efcb5031b9263550c23 : SUCCESS in 58m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/47850fe870bc43b5b04348896e9ffc72 : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":9},{"id":"469142e17d9eecfe5daa871896a192b3145167b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 05:58:37.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b44ca068420f467f948af56a8ded9caf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0882a73d733b48c5985bfaea0567bd0a : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/38bf93339c2d4787bbf91d823ba20be4 : SUCCESS in 1h 05m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0ead9c5153c1458d9099db23aca15e92 : SUCCESS in 49m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/588b5822795949e8bb4f2a2d64909adf : SUCCESS in 54m 54s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d7074aa0804f430c8329ae897d2d9347 : RETRY_LIMIT in 4m 45s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8ed4203fbd6d4ed99025e6097b2009f1 : RETRY_LIMIT in 4m 59s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7acd3e24d60e41cdad77afe99ec0b883 : RETRY_LIMIT in 4m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/353ec05d0cbe4aa08147a2c950ba8692 : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4e4bb259167d45ef8fce46a411f6c404 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e01b542ca713445cbbb5124353ad7871 : SUCCESS in 16m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/790e108db0984c9d993f4aedbb75e5cf : SUCCESS in 16m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4680b94ce44b448089132de19aa28220 : SUCCESS in 15m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/186247540d164030a9a1195f49c6c7a2 : SUCCESS in 16m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/754ea4317af24569ae93748e8a1f8f00 : SUCCESS in 16m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1423cc234b5e434eaf672466920460cb : SUCCESS in 17m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ba3bcfc872704439a9d77be71991b3da : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/57aacf481e7145118485bdcb50fc0621 : SUCCESS in 14m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/84d7bd0ae7a84135afe4f32d982921bc : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d047d29ae393421db01cc19b93199ef5 : SUCCESS in 8m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fad2dd40ab4c4d82ac165de931bf64de : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/670ede0aed944eaa95130e2e36311c7b : SUCCESS in 8m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9e00ba8ea29a40278fbc5740f5137a6c : SUCCESS in 53m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d03b1f333824ad38734e4bee6d522b7 : SUCCESS in 54m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ad3d8dd42c54ae783356c2682aa760c : SUCCESS in 44m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1cdc8c22da984d369b0d984131281486 : SUCCESS in 49m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e5af500c54547c9b0d02398403b9508 : SUCCESS in 23m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49fe696ca55d4644988c46d66460e877 : SUCCESS in 6m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e78cd265e2ec4be0bbb2fa55fe583269 : SUCCESS in 33m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9c6969cdf1e429caf7b67fb2d002ce9 : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":9},{"id":"d01b2d3435cce6a04416a7fd4f919aaeff561e44","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-14 08:51:26.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5445afec179b15bc002a5f900864eafbafb29286","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-18 17:00:19.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"06479858280d1e3e88eee6770450d79e41463b5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 17:01:02.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"6dcd1545d637fc911bfb9140c3923c1c7ed30ce2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 18:10:52.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9d1b8617daf4af7839b86fdf7db807e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0079d309832414d8ef7ae077135618c : SUCCESS in 12m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/79c60e1d716442d896131e57346e78fd : SUCCESS in 1h 08m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75b91d4c6fa94749ac10e108615e38c8 : SUCCESS in 58m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5e11ec06180431ead61d784848cd1a4 : SUCCESS in 1h 01m 19s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/53a9cb56eb404494984c522b3f3b9892 : POST_FAILURE in 9m 29s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/885c4b2ee8f6477c8b7e8a5e6f6586cb : POST_FAILURE in 9m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/095c3b6c72db4abcac719ab52099cd1d : SUCCESS in 17m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/170e439174bc46c4b035c2075787bd7b : SUCCESS in 16m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7472f0499e094048bb522df6901ecc37 : SUCCESS in 17m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/75ba243fc1934b51963ea94612dd1dfc : SUCCESS in 21m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7cb48b7bb5b140b59d57aeb807f16bc7 : SUCCESS in 21m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd561c88bb3949d884a699ab732e9602 : SUCCESS in 21m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6e3a0afa036446c7aa02a48ddbc0823f : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2020509aee4442f08d483aeac35e18b4 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9a29b7d11b154a92956916580dc632c9 : SUCCESS in 14m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eaf954cc3b1b4acea65768812105cb89 : SUCCESS in 13m 13s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/55dc940fbc4b4facbff7381cd4d426ab : SUCCESS in 9m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5d1eebe69b6a420389d54f719bda469d : SUCCESS in 44m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb3b91a6ceab4fbfb64e4c18e6efe62f : SUCCESS in 58m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/16ad4f447b504946a511274a731d6b7f : SUCCESS in 51m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9b4aa4d58b944ffa93ccb4def1e825b7 : SUCCESS in 45m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e11407a81a84883a9cca2265d813747 : SUCCESS in 13m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f2bb50eb4304f1c97aa85813a1dd8bd : SUCCESS in 6m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fdf86f11bf6143d3ba753bc95d702917 : SUCCESS in 35m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68de3e54f1f5438288597ab4ffda9032 : SUCCESS in 53m 46s","accounts_in_message":[],"_revision_number":9},{"id":"795bba41ce12a3806686f2eb26c559615f2d9f41","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 18:10:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"ca497409c4628efafb0ec2a8ef94968360c40def","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 18:12:10.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/364e62de85ae40c8a607ff94ac177ed5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4d2928afa5640bc97b3e6135b8ef216 : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6b0d2d30d750409989b0e422b08cf599 : FAILURE in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/7bad54b94a484128988e9ddf83d9d151 : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"45e13ff4c56748345c9d31bbee3bc6e7abdb5b43","revisions":{"bf1a8b669be17f92c8f794c27e4e7cdbab734a25":{"kind":"REWORK","_number":1,"created":"2022-07-12 06:12:56.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/50/849450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/849450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/849450/1"}}},"commit":{"parents":[{"commit":"2e2251638ce747781e836e64f8b9aa41f9eddd38","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e2251638ce747781e836e64f8b9aa41f9eddd38"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-12 06:00:03.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-12 06:00:03.000000000","tz":600},"subject":"Sharding: persist delete_timestamp and status_changed_at","message":"Sharding: persist delete_timestamp and status_changed_at\n\nWhen moving an UNSHARDED broker into the SHARDING db state the important\ninformation from the retiring DB needs to be moved into the fresh DB.\nHowever, currently we don\u0027t update the delete_timestamp or the\nstatus_changed_at. This means we loose some historic metadata that could\nbe important.\nDelete_timestamp for instance, might indicate the UNSHARDED container is\nactually deleted, but there are objects inside that need to be dealt\nwith. Currently, when transitioning to the SHARDING state it effectively\nundeletes itself.\n\nThis patch, when moving into the SHARDING DB state via\nbroker.set_sharding_state() also persists these 2 extra bits of\nmetadata.\n\nChange-Id: Ic3b3f21ce67c640dfb632b79fd09902f14576eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf1a8b669be17f92c8f794c27e4e7cdbab734a25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf1a8b669be17f92c8f794c27e4e7cdbab734a25"}]},"branch":"refs/heads/master"},"f9158071343ac889a450aca311bfbfcf28d3609a":{"kind":"REWORK","_number":2,"created":"2022-07-12 09:28:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/849450/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/849450/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/849450/2"}}},"commit":{"parents":[{"commit":"2e2251638ce747781e836e64f8b9aa41f9eddd38","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e2251638ce747781e836e64f8b9aa41f9eddd38"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-12 06:00:03.000000000","tz":600},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-12 09:21:05.000000000","tz":60},"subject":"Sharding: persist delete_timestamp and status_changed_at","message":"Sharding: persist delete_timestamp and status_changed_at\n\nWhen moving an UNSHARDED broker into the SHARDING db state the important\ninformation from the retiring DB needs to be moved into the fresh DB.\nHowever, currently we don\u0027t update the delete_timestamp or the\nstatus_changed_at. This means we lose some historic metadata that could\nbe important.\nDelete_timestamp for instance, might indicate the UNSHARDED container is\nactually deleted, but there are objects inside that need to be dealt\nwith. Currently, when transitioning to the SHARDING state it effectively\nundeletes itself.\n\nThis patch, when moving into the SHARDING DB state via\nbroker.set_sharding_state() also persists these 2 extra bits of\nmetadata.\n\nChange-Id: Ic3b3f21ce67c640dfb632b79fd09902f14576eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9158071343ac889a450aca311bfbfcf28d3609a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9158071343ac889a450aca311bfbfcf28d3609a"}]},"branch":"refs/heads/master"},"245f5ce9f4b86d7c3cad31dacb1dc2edf9c3a210":{"kind":"REWORK","_number":3,"created":"2022-07-13 04:05:01.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/50/849450/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/849450/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/849450/3"}}},"commit":{"parents":[{"commit":"2e2251638ce747781e836e64f8b9aa41f9eddd38","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e2251638ce747781e836e64f8b9aa41f9eddd38"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-12 06:00:03.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-13 04:03:45.000000000","tz":600},"subject":"Sharding: Sync container_stat table to fresh db","message":"Sharding: Sync container_stat table to fresh db\n\nWhen moving an UNSHARDED broker into the SHARDING db state the important\ninformation from the retiring DB needs to be moved into the fresh DB.\nHowever, currently we are missing a whole bunch of container information\nfrom the retiring db.\nWe are keeping the row_id so for 1 we should keep container_sync and\nreconciler sync points. Secondly, we dont sync across the\ndelete_timestamp, status or status_changed_at metadata either. These are\nimportant bits of information from the retiring DB.\n\nDelete_timestamp for instance, might indicate the UNSHARDED container is\nactually deleted, but there are objects inside that need to be dealt\nwith. Currently, when transitioning to the SHARDING state it effectively\nundeletes itself.\n\nThis patch, when moving into the SHARDING DB state via\nbroker.set_sharding_state() now also uses the broker API to persist the\ncurrent reconciler and container_sync points and also does an SQL update to\nupdate other bits of metadata. Namely delete_timestamp, status and\nstatus_changed_at.\n\nTo get the container status, I had to add status to the get_info SQL.\nThe fact that it had status_changed_at and not status feels like a bug,\nso adding seems to be in scope for this change.\n\nChange-Id: Ic3b3f21ce67c640dfb632b79fd09902f14576eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/245f5ce9f4b86d7c3cad31dacb1dc2edf9c3a210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/245f5ce9f4b86d7c3cad31dacb1dc2edf9c3a210"}]},"branch":"refs/heads/master"},"819b13713ff7e7cb74ae14dde314c76e6c2b62cd":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-07-13 05:05:21.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/50/849450/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/849450/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/849450/4"}}},"commit":{"parents":[{"commit":"2e2251638ce747781e836e64f8b9aa41f9eddd38","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e2251638ce747781e836e64f8b9aa41f9eddd38"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-12 06:00:03.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-13 05:04:51.000000000","tz":600},"subject":"Sharding: Sync container_stat table with fresh db","message":"Sharding: Sync container_stat table with fresh db\n\nWhen moving an UNSHARDED broker into the SHARDING db state the important\ninformation from the retiring DB needs to be moved into the fresh DB.\nHowever, currently we are missing a whole bunch of container information\nfrom the retiring db.\nWe are keeping the row_id so for 1 we should keep container_sync and\nreconciler sync points. Secondly, we dont sync across the\ndelete_timestamp, status or status_changed_at metadata either. These are\nimportant bits of information from the retiring DB.\n\nDelete_timestamp for instance, might indicate the UNSHARDED container is\nactually deleted, but there are objects inside that need to be dealt\nwith. Currently, when transitioning to the SHARDING state it effectively\nundeletes itself.\n\nThis patch, when moving into the SHARDING DB state via\nbroker.set_sharding_state() now also uses the broker API to persist the\ncurrent reconciler and container_sync points and also does an SQL update to\nupdate other bits of metadata. Namely delete_timestamp, status and\nstatus_changed_at.\n\nTo get the container status, I had to add status to the get_info SQL.\nThe fact that it had status_changed_at and not status feels like a bug,\nso adding seems to be in scope for this change.\n\nChange-Id: Ic3b3f21ce67c640dfb632b79fd09902f14576eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/819b13713ff7e7cb74ae14dde314c76e6c2b62cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/819b13713ff7e7cb74ae14dde314c76e6c2b62cd"}]},"branch":"refs/heads/master"},"714e96cb716ec5376afc23d0a2300eb2b77c1cb7":{"kind":"REWORK","_number":5,"created":"2022-07-13 07:30:10.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/50/849450/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/849450/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/849450/5"}}},"commit":{"parents":[{"commit":"2e2251638ce747781e836e64f8b9aa41f9eddd38","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e2251638ce747781e836e64f8b9aa41f9eddd38"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-12 06:00:03.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-13 07:29:58.000000000","tz":600},"subject":"Sharding: Sync container_stat table with fresh db","message":"Sharding: Sync container_stat table with fresh db\n\nWhen moving an UNSHARDED broker into the SHARDING db state the important\ninformation from the retiring DB needs to be moved into the fresh DB.\nHowever, currently we are missing a whole bunch of container information\nfrom the retiring db.\nWe are keeping the row_id so for 1 we should keep container_sync and\nreconciler sync points. Secondly, we dont sync across the\ndelete_timestamp, status or status_changed_at metadata either. These are\nimportant bits of information from the retiring DB.\n\nDelete_timestamp for instance, might indicate the UNSHARDED container is\nactually deleted, but there are objects inside that need to be dealt\nwith. Currently, when transitioning to the SHARDING state it effectively\nundeletes itself.\n\nThis patch, when moving into the SHARDING DB state via\nbroker.set_sharding_state() now also uses the broker API to persist the\ncurrent reconciler and container_sync points and also does an SQL update to\nupdate other bits of metadata. Namely delete_timestamp, status and\nstatus_changed_at.\n\nTo get the container status, I had to add status to the get_info SQL.\nThe fact that it had status_changed_at and not status feels like a bug,\nso adding seems to be in scope for this change.\n\nChange-Id: Ic3b3f21ce67c640dfb632b79fd09902f14576eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/714e96cb716ec5376afc23d0a2300eb2b77c1cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/714e96cb716ec5376afc23d0a2300eb2b77c1cb7"}]},"branch":"refs/heads/master"},"eb64f3cf56fdfbd42f5b16385e6903b24eb7208b":{"kind":"REWORK","_number":6,"created":"2022-07-13 16:41:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/50/849450/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/849450/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/849450/6"}}},"commit":{"parents":[{"commit":"2e2251638ce747781e836e64f8b9aa41f9eddd38","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e2251638ce747781e836e64f8b9aa41f9eddd38"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-12 06:00:03.000000000","tz":600},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-13 15:40:46.000000000","tz":60},"subject":"Sharding: Sync container_stat table with fresh db","message":"Sharding: Sync container_stat table with fresh db\n\nWhen moving an UNSHARDED broker into the SHARDING db state the important\ninformation from the retiring DB needs to be moved into the fresh DB.\nHowever, currently we are missing a whole bunch of container information\nfrom the retiring db.\nWe are keeping the row_id so for 1 we should keep container_sync and\nreconciler sync points. Secondly, we dont sync across the\ndelete_timestamp, status or status_changed_at metadata either. These are\nimportant bits of information from the retiring DB.\n\nDelete_timestamp for instance, might indicate the UNSHARDED container is\nactually deleted, but there are objects inside that need to be dealt\nwith. Currently, when transitioning to the SHARDING state it effectively\nundeletes itself.\n\nThis patch, when moving into the SHARDING DB state via\nbroker.set_sharding_state() now also uses the broker API to persist the\ncurrent reconciler and container_sync points and also does an SQL update to\nupdate other bits of metadata. Namely delete_timestamp, status and\nstatus_changed_at.\n\nTo get the container status, I had to add status to the get_info SQL.\nThe fact that it had status_changed_at and not status feels like a bug,\nso adding seems to be in scope for this change.\n\nChange-Id: Ic3b3f21ce67c640dfb632b79fd09902f14576eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb64f3cf56fdfbd42f5b16385e6903b24eb7208b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb64f3cf56fdfbd42f5b16385e6903b24eb7208b"}]},"branch":"refs/heads/master"},"932bae928ae8a140a777335669250a9ec3412fe3":{"kind":"REWORK","_number":7,"created":"2022-07-14 03:46:52.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/50/849450/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/849450/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/849450/7"}}},"commit":{"parents":[{"commit":"2e2251638ce747781e836e64f8b9aa41f9eddd38","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e2251638ce747781e836e64f8b9aa41f9eddd38"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-14 03:30:45.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-14 03:45:31.000000000","tz":600},"subject":"Sharding: Sync container_stat table with fresh db","message":"Sharding: Sync container_stat table with fresh db\n\nWhen moving an UNSHARDED broker into the SHARDING db state the important\ninformation from the retiring DB needs to be moved into the fresh DB.\nHowever, currently we are missing a whole bunch of container information\nfrom the retiring db.\nCurrently we dont sync across the delete_timestamp, status or\nstatus_changed_at metadata either. These are important bits of information\nfrom the retiring DB.\n\nDelete_timestamp for instance, might indicate the UNSHARDED container is\nactually deleted, but there are objects inside that need to be dealt\nwith. Currently, when transitioning to the SHARDING state it effectively\nundeletes itself.\n\nThis patch, when moving into the SHARDING DB state via\nbroker.set_sharding_state() now does an SQL update to\nupdate bits of the container_stat metadata that doesn\u0027t have direct API access\nfrom the broker. Namely delete_timestamp, status and status_changed_at.\n\nTo get the container status, I had to add status to the get_info SQL.\nThe fact that it had status_changed_at and not status feels like a bug,\nso adding seems to be in scope for this change.\n\nChange-Id: Ic3b3f21ce67c640dfb632b79fd09902f14576eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/932bae928ae8a140a777335669250a9ec3412fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/932bae928ae8a140a777335669250a9ec3412fe3"}]},"branch":"refs/heads/master"},"f347d0b2fec6701fe9bea57f32f43462ec72f300":{"kind":"REWORK","_number":8,"created":"2022-07-14 04:05:51.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/50/849450/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/849450/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/849450/8"}}},"commit":{"parents":[{"commit":"2e2251638ce747781e836e64f8b9aa41f9eddd38","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e2251638ce747781e836e64f8b9aa41f9eddd38"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-14 03:30:45.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-14 04:03:47.000000000","tz":600},"subject":"Sharding: Sync container_stat table with fresh db","message":"Sharding: Sync container_stat table with fresh db\n\nWhen moving an UNSHARDED broker into the SHARDING db state the important\ninformation from the retiring DB needs to be moved into the fresh DB.\nHowever, currently we are missing a whole bunch of container information\nfrom the retiring db.\nCurrently we dont sync across the delete_timestamp, status or\nstatus_changed_at metadata either. These are important bits of information\nfrom the retiring DB.\n\nDelete_timestamp for instance, might indicate the UNSHARDED container is\nactually deleted, but there are objects inside that need to be dealt\nwith. Currently, when transitioning to the SHARDING state it effectively\nundeletes itself.\n\nThis patch, when moving into the SHARDING DB state via\nbroker.set_sharding_state() now does an SQL update to\nupdate bits of the container_stat metadata that doesn\u0027t have direct API access\nfrom the broker. Namely delete_timestamp, status and status_changed_at.\n\nTo get the container status, I had to add status to the get_info SQL.\nThe fact that it had status_changed_at and not status feels like a bug,\nso adding seems to be in scope for this change.\n\nChange-Id: Ic3b3f21ce67c640dfb632b79fd09902f14576eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f347d0b2fec6701fe9bea57f32f43462ec72f300"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f347d0b2fec6701fe9bea57f32f43462ec72f300"}]},"branch":"refs/heads/master"},"45e13ff4c56748345c9d31bbee3bc6e7abdb5b43":{"kind":"REWORK","_number":9,"created":"2022-07-14 04:46:23.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/50/849450/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/849450/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/849450/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/849450/9"}}},"commit":{"parents":[{"commit":"2e2251638ce747781e836e64f8b9aa41f9eddd38","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2e2251638ce747781e836e64f8b9aa41f9eddd38"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-14 03:30:45.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-14 04:25:56.000000000","tz":600},"subject":"Sharding: Sync container_stat table with fresh db","message":"Sharding: Sync container_stat table with fresh db\n\nWhen moving an UNSHARDED broker into the SHARDING db state the important\ninformation from the retiring DB needs to be moved into the fresh DB.\nHowever, currently we are missing a whole bunch of container information\nfrom the retiring db.\nCurrently we dont sync across the delete_timestamp, status or\nstatus_changed_at metadata either. These are important bits of information\nfrom the retiring DB.\n\nDelete_timestamp for instance, might indicate the UNSHARDED container is\nactually deleted, but there are objects inside that need to be dealt\nwith. Currently, when transitioning to the SHARDING state it effectively\nundeletes itself.\n\nThis patch, when moving into the SHARDING DB state via\nbroker.set_sharding_state() now does an SQL update to\nupdate bits of the container_stat metadata that doesn\u0027t have direct API access\nfrom the broker. Namely delete_timestamp, status and status_changed_at.\n\nTo get the container status, I had to add status to the get_info SQL.\nThe fact that it had status_changed_at and not status feels like a bug,\nso adding seems to be in scope for this change.\n\nChange-Id: Ic3b3f21ce67c640dfb632b79fd09902f14576eaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45e13ff4c56748345c9d31bbee3bc6e7abdb5b43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45e13ff4c56748345c9d31bbee3bc6e7abdb5b43"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
