)]}'
{"id":"openstack%2Fswift~849548","triplet_id":"openstack%2Fswift~master~Ia6ad92042aa9a51e3ddefec0b9b8acae44e9e9d7","project":"openstack/swift","branch":"master","topic":"sharder_fallback_to_same_device","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-07-28 22:25:20.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-07-28 22:25:20.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-07-27 20:45:07.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ia6ad92042aa9a51e3ddefec0b9b8acae44e9e9d7","subject":"sharder: process deleted DBs","status":"MERGED","created":"2022-07-12 15:07:14.000000000","updated":"2022-07-28 22:26:35.000000000","submitted":"2022-07-28 22:25:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849548-sharder_fallback_to_same_device","meta_rev_id":"96c10d2cfaf4d43f5c8e305f851a1751fe0b1bef","_number":849548,"virtual_id_number":849548,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-28 22:25:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-27 20:45:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-28 21:27:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 16:40:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-12 16:50:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-13 01:14:45.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2022-07-23 05:18:54.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"5e6f53172c5dd0d1420f5736ffccbe972da0e945","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-12 15:07:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85a5adae0b0d38c9f327ef246f170dd84040c004","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:40:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13f0f613d447454baaaa013e59c5a9c2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7ca76d75a374431a2a1af27e9280b9d : SUCCESS in 56m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa1a6a26f9e94aa69732f698a57bca9a : SUCCESS in 37m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/87566cecf619453180099b82fd5beb77 : TIMED_OUT in 1h 31m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2dece18eb78a41febe97f0a8f2d8ca09 : SUCCESS in 37m 26s","accounts_in_message":[],"_revision_number":1},{"id":"32f25464fcbea419af9dc215138fb0c707214bbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:50:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b199b96f4c34c20abfa0fbe15e05e9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd826ba32b749a18ce9f136aa57a4fe : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/323da9995c464c71b04b1d765cbeb53c : SUCCESS in 52m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2300da6a9d944f0ca13922e400cbdc7b : SUCCESS in 45m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8774cbef2ef544a0aecbf9563a584cbb : SUCCESS in 54m 00s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fafe28bcbb374e2cbea466edc6585214 : RETRY_LIMIT in 5m 27s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/410cb02815d54a98a197ef4eadce9425 : RETRY_LIMIT in 5m 26s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fbbcd39c59eb49f6bb22a576a40c3e3b : RETRY_LIMIT in 7m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5139dd3d857c45258f3e2c780ca43e7c : SUCCESS in 9m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b92efadcf8b24bcb9304ec3f90bcbe54 : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/75c5f7dd02aa497dbebe284b810ba170 : SUCCESS in 15m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9c04e9e26dc743b4b66f61edd00c3813 : SUCCESS in 15m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c6755f985baf4027bdbd2558cda665f3 : TIMED_OUT in 1h 31m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e7ac326ba4784e0485d7fc2552e463d8 : SUCCESS in 18m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7614f1231363433bae174ebbaeed654b : SUCCESS in 14m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9cb9ad60ebbc43b69ad202dbdced8ba5 : SUCCESS in 17m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0e4be5a449b2436d9bc7db5782db9b89 : SUCCESS in 13m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/196b3bd079654ca6ba1c7a26b8d8934c : SUCCESS in 13m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a07f835078f348e788f0a47819d7e865 : SUCCESS in 19m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/44b07e9484f14909a4827352dc84bd14 : SUCCESS in 8m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2caab3647a6049578447fcd5c959892f : SUCCESS in 17m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cd47de34dab34313807245bd0b7bc0cf : SUCCESS in 9m 07s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/988549495207415191e02e74b8718aa3 : SUCCESS in 1h 00m 08s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/176d03b412a7479eb2e1be34d064f54e : SUCCESS in 1h 05m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c7a5c2a850864ba29f72aaec5c276b9e : SUCCESS in 49m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/50a5604c82094aaab97f91c7cf7e09e6 : SUCCESS in 53m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37657059596b4a9c9c2de9012e888245 : SUCCESS in 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da79737af7cf4690af559e29875abdb2 : SUCCESS in 6m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0cd1c4b163544636976360b7f276d4db : SUCCESS in 32m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a63baf9670b74f0ca3b72742defed109 : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":1},{"id":"191d237ff7d2f4dc27959eb5d48132cbeb58404b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-12 17:29:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7230fd7a9015d0da31042050e399fb0e4eb142ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 18:37:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab4d598c124e4dadaf03fc35ecf14d32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0188e3fc25704a56ab0d7f0009b1352f : SUCCESS in 9m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/6643f0892c714133b05f44a40dca6098 : SUCCESS in 1h 03m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4d4e0969e3d74a5c81efb8acb2d9d37b : SUCCESS in 46m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5994394ed59448bb8fc6a6ddca8a45e2 : SUCCESS in 48m 40s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d104bf3d411643ad83a6f32b9b855426 : RETRY_LIMIT in 7m 01s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/821465327b3a4268bba4be0f26f9b6bc : RETRY_LIMIT in 5m 58s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/beb9d5225a024210b4ced0555e274c8b : RETRY_LIMIT in 6m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cbc13bf736474d7ca97b0cec24d830ba : SUCCESS in 6m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4c4ca24c454a45739fb5cf8ac57ac0dd : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/391b0488b835412e9446c1c899a540f4 : SUCCESS in 16m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/64142641209f4118bedd9098a4cf5b38 : SUCCESS in 16m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e328deda4963468e86e77a4d01bcc4c7 : SUCCESS in 14m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d3e02c1174904a099d3b7053987cb710 : SUCCESS in 21m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/840a1ac442d249fc9f6d26caed5c9b79 : SUCCESS in 18m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bafa2e5052bf4228bc0332e2ce891384 : SUCCESS in 18m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1a59c8ded64f441191e83ca81605bcb6 : SUCCESS in 13m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ec67ea1af97841d6b9c5f33162411e3f : SUCCESS in 14m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9c191158189344bda48bc5009d8be891 : SUCCESS in 14m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b25ef99b59654fb5b0f0591198598b22 : SUCCESS in 8m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3e7ada7c83c544dea7f027ae7830ac0c : SUCCESS in 14m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a7536ac5f5224ef9a9a067915a827795 : SUCCESS in 9m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ca6bcbe19c294b979e56787f39092148 : SUCCESS in 49m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/86d43770a0db4828b5c93b10173c25e1 : SUCCESS in 52m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a2eba631c8e24b53b358ccbc31a9df05 : SUCCESS in 46m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2af3446d794244f3beff143b7a47dcab : SUCCESS in 41m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81cf4f6e3b534927bca0216183cf44bd : SUCCESS in 17m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/816355a8b3764c348b6c83991ec2584d : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/596c1aeef3a64df9a2279e735319db40 : SUCCESS in 27m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20fdaff8c56e4600a3a772811623d080 : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a07ab09e09dea6c8116e6e487e3e720b4cbe429a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 18:38:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a1d820246c244ffb29b72f7590641e2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb872e6ffb4f4a12852ab3f6de71a729 : SUCCESS in 1h 00m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/445d434b13b54dd5a363421811505aea : SUCCESS in 1h 00m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b03df3e43254fb7a9671cbae6f2db02 : SUCCESS in 1h 00m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/45f9787ae27249929b4aa56ed2a07994 : SUCCESS in 59m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d5066e6a20ba50220248f14108b4a5ceeac95853","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-13 01:14:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b4375259fea72e451de019b2111998ed65e91092","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-13 07:45:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"3a9057bd7908bfbd5ea606e4119cdf68bbdd6e9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 08:31:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06dfa4b20a0f416ab04d2acec8aff793\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5bebe9fc266f4bbda49350fdc4e95e80 : SUCCESS in 29m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25433d216c504928a166478b93d6414f : SUCCESS in 44m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/28c013f5beb5417bba03695784a69847 : SUCCESS in 29m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7eb687c5537146c88a7052e406df5dd9 : SUCCESS in 28m 57s","accounts_in_message":[],"_revision_number":2},{"id":"46a31f2202fe4f4100a3ab90d832d2c61897a3ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 08:57:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af4d6798a1e74cf1a762fb53e379bb3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed42b686a06145bcb5f7d10362152e13 : SUCCESS in 9m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/7bc17def51854bbab14ccb7434b391cf : SUCCESS in 1h 05m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd6f61eea7e14a1aab7ebfe25ec4ff5e : SUCCESS in 56m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fb2695bc7494002894ec1a7325a8edf : SUCCESS in 1h 01m 08s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2a4c9c35cb00446ab61e0eeeffa013bb : RETRY_LIMIT in 4m 23s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6f2d645d6fc240acbdecee3c4ae6ac69 : RETRY_LIMIT in 5m 03s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3c8d264f1979403481ac65bbb26f580a : RETRY_LIMIT in 4m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c8ed34048be042da9b1f956c2b02725c : SUCCESS in 8m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e663ba98ca8c4a49aba6eb0ed3b9f4d7 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/249a6be62d9e47418889c038f912fdd3 : SUCCESS in 16m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/801743b3547741df91eeff8059ffa936 : SUCCESS in 16m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/96f96955bcff422e863574d4b0cd4a0a : SUCCESS in 17m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2f38462a71cf4a5aa4e4b31efc321a49 : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/79d73159d841479e8744d42d818e0950 : SUCCESS in 19m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c81f99c1eba74db7a3501ea338aaecd3 : SUCCESS in 19m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f1ad2924262843d285bdb03c1ed22b8e : SUCCESS in 13m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/67cd0c3c96db4ce7814739bee9b8908b : SUCCESS in 13m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/be0c0e693f0c4aa7b6b099cb5f85e75a : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cc17344588fb4a8291026c74abf90263 : SUCCESS in 8m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1070903bd8fb4c2ea03ccabab6379765 : SUCCESS in 16m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ccb8ffd22e4d417c9a497ef73e395042 : SUCCESS in 8m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/36115497a8db40278ed5aa479e9bfefb : SUCCESS in 44m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc623e1040f642df92bc8e0498597f9f : SUCCESS in 1h 04m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0ac5aef986aa45bfb34015c97c485497 : SUCCESS in 47m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aaa52275c5d0421da0ed142def27535d : SUCCESS in 38m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/303322b2e56d4a5993c2f16e2cdafeb1 : SUCCESS in 13m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdb1d205b57a4c96ad14f416c9a87726 : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7405ce970fbb4c308c6a41258829b2d7 : SUCCESS in 22m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8454a83b5691408cbad059796454518a : SUCCESS in 1h 04m 55s","accounts_in_message":[],"_revision_number":2},{"id":"cf221bd56af3d10c77d211cdda3d227c5b96a064","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-13 10:20:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d5bb8a2d876597d797d028c2c5d1a4802c579d3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-13 16:41:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"297ce84477ac969a634ddf0a9dd5fa4573ce90ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-13 16:47:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"05a7746c32495c9453d3e883b9144839e979bef3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-13 16:48:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"df44087b8691feab308c9977850f896463790725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 17:53:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6182e4469de4955a083ff20e8598485\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5550c1adf30448afb525ced42700b3d4 : SUCCESS in 10m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/328bbda878d349c8ac3ea357bc32c261 : SUCCESS in 49m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6efd19cf06214656889fb291db0b96b0 : SUCCESS in 43m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c663aa609b04782bdf0cca64c84757f : SUCCESS in 42m 50s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/aa9a586d4f67430c8e84992dce0e3a79 : RETRY_LIMIT in 4m 21s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0b9324dbc9624822a55e43542e8a356b : RETRY_LIMIT in 5m 22s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/37ba5a028688419db8ae42fc6d3c5ae5 : RETRY_LIMIT in 4m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2f5870fd0efa4cb18366abed15a81ebb : SUCCESS in 9m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0fb7e0a2a6a2432d8872dd56afcfa6ea : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a28b93b39bbe4e91a6c1d665dee6f30a : SUCCESS in 14m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5d30609a87b943869bbf0de7fa74ac01 : SUCCESS in 20m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ab3290b9205d484e82f1a9e68657550c : SUCCESS in 14m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a0144541cbe04268baa1c8d960bb59bd : SUCCESS in 20m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a83d4912f32b447fb04ba67dc2b0674d : SUCCESS in 16m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9f3c6d2dd40144a8bfdfbc640905dc1c : SUCCESS in 22m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0d34e6f9db7d4892a0bbb5a000200144 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b541665df0414444955a9f036621d858 : SUCCESS in 19m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/087dad2e47204e1e98b7709ab75577a8 : SUCCESS in 20m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a9fa81ee82cc48d0b0bb03e1a80227d5 : SUCCESS in 11m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7c6baf284086419fae1feb39f6169945 : SUCCESS in 17m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f9a95606bbeb47a5bafbcd7a654e12c0 : SUCCESS in 10m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fa32d1a061144948a820925862906139 : SUCCESS in 1h 03m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5e9aa2a7a38145ed99bb21613317ea93 : SUCCESS in 1h 03m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0fc973c0046d44e497afee7db9615bc4 : SUCCESS in 46m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d49f02cf44c2468593968fa59539334f : SUCCESS in 44m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3b68e697dca4fe3bb234f590303ff8a : SUCCESS in 16m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15559009ac0b4f8aad1a07de0a896391 : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6a3d4ae895bc42fdac71a41ddea198cc : SUCCESS in 26m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e70d93def89f47af9f98d5c3809e23a9 : SUCCESS in 53m 28s","accounts_in_message":[],"_revision_number":4},{"id":"a60a785795ed5c6c0030e5b423a6342bbf493580","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 17:56:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f964efcf38c487688e157513cd369fa\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6336cb3c3aa14fe8aa41632f1d68020c : SUCCESS in 35m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/400c306f6278429fa062ae26647b2a19 : SUCCESS in 1h 07m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/21346789f8214d7a9112538d032d9652 : SUCCESS in 34m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a2ae786eb3e461c8db2494fcba3855b : SUCCESS in 34m 16s","accounts_in_message":[],"_revision_number":4},{"id":"d2eb703d79ab12a9e39c917b5a248883b9fdcb29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-14 03:46:52.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"384bec69eb51dfa8557282271838af7701b9fc7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-14 03:53:27.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"dab77d06839301d46dd015a7a861caadbca40174","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-14 04:05:51.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"7d1a3fd60326da6abfc78b063bb8399364d63c85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-14 04:46:23.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"f3ad7438c1c7a5c4df435373dbf6df864d93cf42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 05:51:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79cd02888c0c443c9d9763b836a396cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93ff4f17b08f4d61a5d1c5bde28a007c : SUCCESS in 10m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/94ec8c9541e14aad898bbf978ee58f1c : SUCCESS in 59m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/79e2d310ba624679a3f8c94fe7df1645 : SUCCESS in 50m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93b8bb6b943048a59f34ec7399540ce4 : SUCCESS in 53m 47s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/72500965fb8c4034adfa76d20b9c24fe : RETRY_LIMIT in 4m 39s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7181bc82b55a45319074790e1a4b2ecb : RETRY_LIMIT in 4m 44s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3a09efc6d2fe4f0c9b6940dfdeb0d6b3 : RETRY_LIMIT in 5m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7e89e34eb6f546a79ff05822cd8985e2 : SUCCESS in 7m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/21565ae6d3e4426ebc1411f007a455ae : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/434f797b6a6f4c6cb74e73a87f95b6e7 : SUCCESS in 15m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/74180bdbe08c4dc1a6844836432abb5d : SUCCESS in 13m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/80b589bcbe204cd0a663d13ed3ac245f : SUCCESS in 17m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aa636e2374814300b44e900d5f644717 : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9ad23267b7794aa6874e95ba30073e03 : SUCCESS in 17m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e2ba9e04444b47038ad6e7a5dbb471c4 : SUCCESS in 21m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a76d589983254da881640eee4bc651a2 : SUCCESS in 18m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0bf8e68024e24c1996ec45a5099f3c45 : SUCCESS in 14m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3bd19a38963646af957e570e9b8f48bf : SUCCESS in 15m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2559b14b634f4e339b0c4316998b14d2 : SUCCESS in 7m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6af3459358a74842a0ad0f26edc83be5 : SUCCESS in 17m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6e2b5ebb276c4abe9635178d1be892f0 : SUCCESS in 8m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8fa062b4d2f34fa38585347ed4b9b9e2 : SUCCESS in 56m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/430cd9d3129a47a398108f107c11d5cf : SUCCESS in 52m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6a7a6c4deb5e40aea59a7c9f3f1ba349 : SUCCESS in 37m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e340b962af4f4e459d9f0e4228a58519 : SUCCESS in 47m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/705cdf786ca84a23bc46611b4584a237 : SUCCESS in 17m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a647ee1adc4f4cbba470982dd647ab34 : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/131ee943a3c347f2aa91422249429087 : SUCCESS in 22m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b126a3f2fbbd4ceebbe309f47106531d : SUCCESS in 44m 53s","accounts_in_message":[],"_revision_number":8},{"id":"cb1442a052e79f2afd77171d804f6006a615b182","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 05:59:40.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6537207309634c858d748b6ee508a2ef\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/952321f75c2b4a089718731a7aee56e8 : SUCCESS in 54m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e0fdf914c1084a1da0cb76e6459cd731 : SUCCESS in 50m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/737473b4e2ee48118c7fbe22395e15c5 : SUCCESS in 1h 07m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/022861a52638456b92fc39e76a43cbe7 : SUCCESS in 1h 10m 49s","accounts_in_message":[],"_revision_number":8},{"id":"4d5e553aec8dfc34bd466d1c0ce3cf905c746382","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-15 17:15:59.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"5a2b1ca52db91198b395bc3f6abf6e879315abb2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-18 03:45:41.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"aabf415f5a83f21aa6c3d31907d56db8fbcc8d3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-21 15:09:46.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"74fe74706d6fbb1ba9a3dedcadb4c88a190fbebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 16:15:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82a31c53e26f45839648b3edce69ba58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b1c670da318402eb0b25df5703f64d3 : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/ec445e1ba14343729c53b95def354698 : SUCCESS in 51m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c8a99a787ed247a28cd7ea44817bb458 : SUCCESS in 51m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ef1b161a3d9428a8fe5d9b661062a64 : SUCCESS in 1h 02m 22s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fc38e8bfc65a4d9f834669bdc9e71471 : RETRY_LIMIT in 4m 55s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0b1044d4a4cf40aba018b91ebc6fabee : RETRY_LIMIT in 5m 19s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5e1f321647584dee816e62483d023a6e : RETRY_LIMIT in 5m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/00aa2743e3e24f21b7312c9d6e05698c : FAILURE in 4m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dd61c0e3248149e69ebc67971d245220 : FAILURE in 4m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2a02f0c2bc114f6d968f23d6466d3a42 : SUCCESS in 28m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7e5cc4079c2d4066ac2521fccae5d0b0 : SUCCESS in 32m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ae3b052d6cdf49e2ac9a142ae9a23f4e : SUCCESS in 31m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/088c8779d06e4433ab53b987bcfe0f1c : SUCCESS in 34m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9535c63f860e43b192a84d611867f3bf : SUCCESS in 16m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8f457276dedf4e6daaaabb6c7590376a : SUCCESS in 44m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a34c74487297472f8d9f35158843f039 : SUCCESS in 13m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9c1e040167c94c27a5d6099992ca5944 : SUCCESS in 13m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e3b2c7fa7f3d47bba34ef18e55c7181d : SUCCESS in 13m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/37394c8a92d84da7b13b62e2fff373d5 : SUCCESS in 7m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/58659588be2a4e6e8fd1d264f277ac19 : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8bfa8830053c4f48881ae045c6af67c0 : SUCCESS in 8m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c902a837fe4746acb951ce16bf56d478 : SUCCESS in 46m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c90ed38c99004a8d9f21512aadc5d6ed : SUCCESS in 1h 02m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4da5afe8b5fd46939df8a32ed3bc5947 : SUCCESS in 39m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d99016ba4354965b9bb5edaa2d7d46e : SUCCESS in 38m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a39fcc09440844fd983ac160af0428e9 : SUCCESS in 14m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/490e925ece6748709dcf565d44f88350 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/74d86b2448ee4b1d8d596ae064266f6e : SUCCESS in 25m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9e99f09a1dc4b2dbf7385b17a4844c3 : SUCCESS in 1h 03m 44s","accounts_in_message":[],"_revision_number":9},{"id":"eeaad12345c08d4582dcfe087ece76664e93313e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 16:19:31.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/590d124563d84404934170f14c221054\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49c2aed9e92f4c1eb0dcb75d047b5740 : SUCCESS in 1h 00m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37553288d659419384a0f3eb9c4a10ca : SUCCESS in 49m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/13acd249584441c0b7e4c3660907a87b : FAILURE in 1h 07m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c63264c365b4a6ba15b4e5a4bfb03ba : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":9},{"id":"2c7178b34e48de0a167b6b4c824ca3c54e47e28a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-23 05:18:54.000000000","message":"Patch Set 9: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"994bef044be498e0527f87e6e0f778a710bcf943","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-27 14:50:44.000000000","message":"Patch Set 9: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"20b20af41100551582e4811542ae17e324bf00f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-27 14:51:34.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3367964391635975ee5d3bd8664dfd35b6044b6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 16:04:29.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b942dfe56c3f49c1a0a739e12a1c73a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1a56e7c8a894da28f46e2c317535f74 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/b3d325beb77d4e8bb17f923c40ce2269 : SUCCESS in 49m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/63070b5711c64f6cb3c7bd93a76002a1 : SUCCESS in 38m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c55b26142be44e4aa810d7c2806e7976 : SUCCESS in 41m 10s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8d18af1f30864ed1bc0b8d47f8339e68 : RETRY_LIMIT in 7m 24s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/94f8b6f4e5a34c73a452fad29f525b7e : RETRY_LIMIT in 4m 51s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e091600a6d294b39a738237e949647da : RETRY_LIMIT in 7m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5ef66c8863764dbf8615d6f1f49b064b : FAILURE in 4m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/25543558b6de4a3780b60f2b3e5be72c : FAILURE in 4m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/02f974424a184677bc59d3860ab835b8 : SUCCESS in 13m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2dffc5516d1b4160b1fa58ed8ebf2bd5 : SUCCESS in 14m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fbcbc2c5b1d14b39a8ad497ce8cda616 : SUCCESS in 13m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d382489763b54596bc01bd6f4c40c66c : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/298dfa0b262f4817af387e4076b2a9eb : SUCCESS in 15m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a9a53b8d8d4a43feb73aeec5c89e373e : SUCCESS in 16m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/313af582187a455e8337bd068ebe15ee : SUCCESS in 14m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/485a2b44afeb43d3b484764f6e68adac : SUCCESS in 13m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4d56e79d711e43b28177a4f2cfde0974 : SUCCESS in 14m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/689987f6fbc84f05984c86ba933ac66a : SUCCESS in 8m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dff20f105a37432380b1c2fdc41b41c0 : SUCCESS in 15m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3a10d50983d74095be969b668e113245 : SUCCESS in 8m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/558cd6de12dd420190d3e64580b2752d : SUCCESS in 52m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8dd2a296817f4cd1a39f6fede85a52ad : SUCCESS in 55m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a6440b3ff7d546668c4d3803295fd75f : SUCCESS in 40m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a7be4d2a3f204271a126dbd2495e038e : SUCCESS in 37m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5dcec09bf4a0447faedc696722e3b8d7 : SUCCESS in 16m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68daf2765f834429bf88e37fcd9d9fd0 : SUCCESS in 6m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/259729785a934e8eba46a9e9a0e185ed : SUCCESS in 23m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc8e57298b714b8c87f3fdcd1cb8434c : SUCCESS in 51m 21s","accounts_in_message":[],"_revision_number":10},{"id":"e4b7e4d587992fa6754360789593a378db3adc6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 16:16:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cd2af1348fa435db89c6da74c445b5c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/609469c1d0924d13915f06c4241fa658 : SUCCESS in 38m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c12c76e46ca64486946a8eccff380888 : SUCCESS in 42m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/62a2b11acec348bda61907bbfeef702f : SUCCESS in 1h 13m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ffa123b11354feb98169395a4ccc12a : SUCCESS in 1h 21m 51s","accounts_in_message":[],"_revision_number":10},{"id":"4c571bd13b3b1a9a407f84c62eeaf6ce9b855a71","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-27 20:45:07.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"4f1d0ac40ea8b04835d386e46149abb2213f81da","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-28 21:27:47.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"70d0fbcec6ba7c9e4a3909ec3bca553b4d1746d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 21:28:22.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"7f214b675c90893a5ec3a7642c896562d1ba83be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 22:25:20.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/908afbcbe85d4ca3be431b107ddab6bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8df7ade34bbd4af6a64d7c80307e995a : SUCCESS in 12m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/28677e9f0dbf4cceb9b89809d97e2b36 : SUCCESS in 50m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fd152ba4393d495a84bcf8188106c9d0 : SUCCESS in 41m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85039808a01941fd9b45be7adffcabfc : SUCCESS in 38m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f9d5d2ec11aa4ff792cfacc978d1fca5 : POST_FAILURE in 6m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b57a7623800e4d99bf7c92a18c8cc373 : POST_FAILURE in 6m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/637b8724ef7e4b4aaf52b02e13bb2568 : SUCCESS in 17m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4898167e7bab42d19fce4599d67781f5 : SUCCESS in 13m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/85c54daa705c4e6eba713a849fbd5ff7 : SUCCESS in 16m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d6a7aa2c093a4b1c8974fac864fd7abe : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f6cc0910a0544af0a5dde85400216ab3 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6b16c2efbf0a4ae6bf65323ef1fa93fb : SUCCESS in 19m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/970e425e03ca4c8cbafe95ac217aae7e : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7fd664da58de4e9996196cc79a63528b : SUCCESS in 18m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/922db1636e3446ceb80d7ac1e3bbd908 : SUCCESS in 14m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cba1fca094f54b2d8955241b68010a14 : SUCCESS in 7m 46s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c40421d7b8be448c8b2478a86b49792e : SUCCESS in 9m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a6cc50811674494e8b591f13e035fd87 : SUCCESS in 52m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/68a7e55c3669499189e1a679d3ddb6ff : SUCCESS in 55m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/35549d5595f74534a5ce0db63cbc5743 : SUCCESS in 50m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb81281a9cd64edb831409a3874096c8 : SUCCESS in 46m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49e4610418da40c982f82292260d309e : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5df3ff00d30b4d38aab78fbf648bbff0 : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ea8d83d777b44d66a2241f57067b75c5 : SUCCESS in 25m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acd7c8984d1a48ef86cf9dfdcb8663b9 : SUCCESS in 47m 02s","accounts_in_message":[],"_revision_number":10},{"id":"fb484b23a96a66638f5644b85805f5109850ac95","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 22:25:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"96c10d2cfaf4d43f5c8e305f851a1751fe0b1bef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 22:26:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a184d985a036436b9bd0327d1b378378\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8660fa51821a440986526b61adb4fff2 : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e29f6aeadc944c0fb6c05ad40eb63e66 : FAILURE in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/cf42a4ed52564d3fb479f48316774f37 : FAILURE in 29s (non-voting)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"38271142eb403a1cdf587616b6d26bfc6b7fb382","revisions":{"f365974e2255c40a6fb1e90750d0b8a7e984a0fe":{"kind":"REWORK","_number":1,"created":"2022-07-12 15:07:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/849548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/849548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/849548/1"}}},"commit":{"parents":[{"commit":"f9158071343ac889a450aca311bfbfcf28d3609a","subject":"Sharding: persist delete_timestamp and status_changed_at","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9158071343ac889a450aca311bfbfcf28d3609a"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-07 17:29:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-12 15:02:35.000000000","tz":60},"subject":"sharder: process deleted DBs","message":"sharder: process deleted DBs\n\nIt is possible for some replicas of a container DB to be sharded, and\nfor the container to then be emptied and deleted, before another\nreplica of the DB has started sharding. Previously, the unsharded\nreplica would remain in the unsharded state because the sharder would\nnot process deleted DBs.\n\nThis patch modifies the sharder to always process deleted DBs; this\nwill result in the sharder making some additional DB queries for shard\nranges in order to determine whether any processing is required for\ndeleted DBs.\n\nChange-Id: Ia6ad92042aa9a51e3ddefec0b9b8acae44e9e9d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f365974e2255c40a6fb1e90750d0b8a7e984a0fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f365974e2255c40a6fb1e90750d0b8a7e984a0fe"}]},"branch":"refs/heads/master"},"9c0e7e7f51d447235b7e97ec496492b6e91ac7cb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-13 07:45:17.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/849548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/849548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/849548/2"}}},"commit":{"parents":[{"commit":"714e96cb716ec5376afc23d0a2300eb2b77c1cb7","subject":"Sharding: Sync container_stat table with fresh db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/714e96cb716ec5376afc23d0a2300eb2b77c1cb7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-07 17:29:06.000000000","tz":60},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-13 07:34:36.000000000","tz":600},"subject":"sharder: process deleted DBs","message":"sharder: process deleted DBs\n\nIt is possible for some replicas of a container DB to be sharded, and\nfor the container to then be emptied and deleted, before another\nreplica of the DB has started sharding. Previously, the unsharded\nreplica would remain in the unsharded state because the sharder would\nnot process deleted DBs.\n\nThis patch modifies the sharder to always process deleted DBs; this\nwill result in the sharder making some additional DB queries for shard\nranges in order to determine whether any processing is required for\ndeleted DBs.\n\nChange-Id: Ia6ad92042aa9a51e3ddefec0b9b8acae44e9e9d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c0e7e7f51d447235b7e97ec496492b6e91ac7cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c0e7e7f51d447235b7e97ec496492b6e91ac7cb"}]},"branch":"refs/heads/master"},"c19f0831375519758f5e93330bc6e96d85cf1ead":{"kind":"REWORK","_number":3,"created":"2022-07-13 16:41:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/849548/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/849548/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/849548/3"}}},"commit":{"parents":[{"commit":"eb64f3cf56fdfbd42f5b16385e6903b24eb7208b","subject":"Sharding: Sync container_stat table with fresh db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb64f3cf56fdfbd42f5b16385e6903b24eb7208b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-07 17:29:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-13 16:41:43.000000000","tz":60},"subject":"sharder: process deleted DBs","message":"sharder: process deleted DBs\n\nIt is possible for some replicas of a container DB to be sharded, and\nfor the container to then be emptied and deleted, before another\nreplica of the DB has started sharding. Previously, the unsharded\nreplica would remain in the unsharded state because the sharder would\nnot process deleted DBs.\n\nThis patch modifies the sharder to always process deleted DBs; this\nwill result in the sharder making some additional DB queries for shard\nranges in order to determine whether any processing is required for\ndeleted DBs.\n\nChange-Id: Ia6ad92042aa9a51e3ddefec0b9b8acae44e9e9d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c19f0831375519758f5e93330bc6e96d85cf1ead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c19f0831375519758f5e93330bc6e96d85cf1ead"}]},"branch":"refs/heads/master"},"51a076df7ae6ddd49a02ae8e022f2c6ceb1bc446":{"kind":"REWORK","_number":4,"created":"2022-07-13 16:47:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/849548/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/849548/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/849548/4"}}},"commit":{"parents":[{"commit":"eb64f3cf56fdfbd42f5b16385e6903b24eb7208b","subject":"Sharding: Sync container_stat table with fresh db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb64f3cf56fdfbd42f5b16385e6903b24eb7208b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-07 17:29:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-13 16:47:04.000000000","tz":60},"subject":"sharder: process deleted DBs","message":"sharder: process deleted DBs\n\nIt is possible for some replicas of a container DB to be sharded, and\nfor the container to then be emptied and deleted, before another\nreplica of the DB has started sharding. Previously, the unsharded\nreplica would remain in the unsharded state because the sharder would\nnot process deleted DBs.\n\nThis patch modifies the sharder to always process deleted DBs; this\nwill result in the sharder making some additional DB queries for shard\nranges in order to determine whether any processing is required for\ndeleted DBs.\n\nChange-Id: Ia6ad92042aa9a51e3ddefec0b9b8acae44e9e9d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51a076df7ae6ddd49a02ae8e022f2c6ceb1bc446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51a076df7ae6ddd49a02ae8e022f2c6ceb1bc446"}]},"branch":"refs/heads/master"},"9e1233a9212238f118f7203a90f33d2b3c851f8c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-07-14 03:46:52.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/849548/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/849548/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/849548/5"}}},"commit":{"parents":[{"commit":"f05ec8e8a7e98dc6629002dcadee2b80c7d80662","subject":"Sharding: Sync reconciler and container_sync points","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f05ec8e8a7e98dc6629002dcadee2b80c7d80662"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-07 17:29:06.000000000","tz":60},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-14 03:46:21.000000000","tz":600},"subject":"sharder: process deleted DBs","message":"sharder: process deleted DBs\n\nIt is possible for some replicas of a container DB to be sharded, and\nfor the container to then be emptied and deleted, before another\nreplica of the DB has started sharding. Previously, the unsharded\nreplica would remain in the unsharded state because the sharder would\nnot process deleted DBs.\n\nThis patch modifies the sharder to always process deleted DBs; this\nwill result in the sharder making some additional DB queries for shard\nranges in order to determine whether any processing is required for\ndeleted DBs.\n\nChange-Id: Ia6ad92042aa9a51e3ddefec0b9b8acae44e9e9d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e1233a9212238f118f7203a90f33d2b3c851f8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9e1233a9212238f118f7203a90f33d2b3c851f8c"}]},"branch":"refs/heads/master"},"f8f75191ad9f3988d6d8cc5a565adf215234639e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-07-14 03:53:27.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/849548/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/849548/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/849548/6"}}},"commit":{"parents":[{"commit":"932bae928ae8a140a777335669250a9ec3412fe3","subject":"Sharding: Sync container_stat table with fresh db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/932bae928ae8a140a777335669250a9ec3412fe3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-07 17:29:06.000000000","tz":60},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-14 03:51:50.000000000","tz":600},"subject":"sharder: process deleted DBs","message":"sharder: process deleted DBs\n\nIt is possible for some replicas of a container DB to be sharded, and\nfor the container to then be emptied and deleted, before another\nreplica of the DB has started sharding. Previously, the unsharded\nreplica would remain in the unsharded state because the sharder would\nnot process deleted DBs.\n\nThis patch modifies the sharder to always process deleted DBs; this\nwill result in the sharder making some additional DB queries for shard\nranges in order to determine whether any processing is required for\ndeleted DBs.\n\nChange-Id: Ia6ad92042aa9a51e3ddefec0b9b8acae44e9e9d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8f75191ad9f3988d6d8cc5a565adf215234639e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8f75191ad9f3988d6d8cc5a565adf215234639e"}]},"branch":"refs/heads/master"},"d284c57d9b4ee3f9a9712003377d3f22bb469295":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-07-14 04:05:51.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/849548/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/849548/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/849548/7"}}},"commit":{"parents":[{"commit":"f347d0b2fec6701fe9bea57f32f43462ec72f300","subject":"Sharding: Sync container_stat table with fresh db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f347d0b2fec6701fe9bea57f32f43462ec72f300"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-07 17:29:06.000000000","tz":60},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-14 04:03:56.000000000","tz":600},"subject":"sharder: process deleted DBs","message":"sharder: process deleted DBs\n\nIt is possible for some replicas of a container DB to be sharded, and\nfor the container to then be emptied and deleted, before another\nreplica of the DB has started sharding. Previously, the unsharded\nreplica would remain in the unsharded state because the sharder would\nnot process deleted DBs.\n\nThis patch modifies the sharder to always process deleted DBs; this\nwill result in the sharder making some additional DB queries for shard\nranges in order to determine whether any processing is required for\ndeleted DBs.\n\nChange-Id: Ia6ad92042aa9a51e3ddefec0b9b8acae44e9e9d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d284c57d9b4ee3f9a9712003377d3f22bb469295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d284c57d9b4ee3f9a9712003377d3f22bb469295"}]},"branch":"refs/heads/master"},"34991abf80dad04bde314723841e31062ad00d60":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-07-14 04:46:23.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/48/849548/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/849548/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/849548/8"}}},"commit":{"parents":[{"commit":"45e13ff4c56748345c9d31bbee3bc6e7abdb5b43","subject":"Sharding: Sync container_stat table with fresh db","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45e13ff4c56748345c9d31bbee3bc6e7abdb5b43"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-07 17:29:06.000000000","tz":60},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-14 04:26:02.000000000","tz":600},"subject":"sharder: process deleted DBs","message":"sharder: process deleted DBs\n\nIt is possible for some replicas of a container DB to be sharded, and\nfor the container to then be emptied and deleted, before another\nreplica of the DB has started sharding. Previously, the unsharded\nreplica would remain in the unsharded state because the sharder would\nnot process deleted DBs.\n\nThis patch modifies the sharder to always process deleted DBs; this\nwill result in the sharder making some additional DB queries for shard\nranges in order to determine whether any processing is required for\ndeleted DBs.\n\nChange-Id: Ia6ad92042aa9a51e3ddefec0b9b8acae44e9e9d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34991abf80dad04bde314723841e31062ad00d60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/34991abf80dad04bde314723841e31062ad00d60"}]},"branch":"refs/heads/master"},"a303319f0c21cff5eef74a229d5131d3a6b0ef34":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-07-21 15:09:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/849548/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/849548/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/849548/9"}}},"commit":{"parents":[{"commit":"c3aa1ce66c20c0c9148a4d3e22d3ff5bcd12bfdb","subject":"Merge \"sharder: always set state to CLEAVED after cleaving\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3aa1ce66c20c0c9148a4d3e22d3ff5bcd12bfdb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-07 17:29:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-21 11:25:55.000000000","tz":60},"subject":"sharder: process deleted DBs","message":"sharder: process deleted DBs\n\nIt is possible for some replicas of a container DB to be sharded, and\nfor the container to then be emptied and deleted, before another\nreplica of the DB has started sharding. Previously, the unsharded\nreplica would remain in the unsharded state because the sharder would\nnot process deleted DBs.\n\nThis patch modifies the sharder to always process deleted DBs; this\nwill result in the sharder making some additional DB queries for shard\nranges in order to determine whether any processing is required for\ndeleted DBs.\n\nChange-Id: Ia6ad92042aa9a51e3ddefec0b9b8acae44e9e9d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a303319f0c21cff5eef74a229d5131d3a6b0ef34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a303319f0c21cff5eef74a229d5131d3a6b0ef34"}]},"branch":"refs/heads/master"},"38271142eb403a1cdf587616b6d26bfc6b7fb382":{"kind":"REWORK","_number":10,"created":"2022-07-27 14:51:34.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/48/849548/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/849548/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/849548/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/849548/10"}}},"commit":{"parents":[{"commit":"c3aa1ce66c20c0c9148a4d3e22d3ff5bcd12bfdb","subject":"Merge \"sharder: always set state to CLEAVED after cleaving\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3aa1ce66c20c0c9148a4d3e22d3ff5bcd12bfdb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-07 17:29:06.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-27 14:48:40.000000000","tz":60},"subject":"sharder: process deleted DBs","message":"sharder: process deleted DBs\n\nIt is possible for some replicas of a container DB to be sharded, and\nfor the container to then be emptied and deleted, before another\nreplica of the DB has started sharding. Previously, the unsharded\nreplica would remain in the unsharded state because the sharder would\nnot process deleted DBs.\n\nThis patch modifies the sharder to always process deleted DBs; this\nwill result in the sharder making some additional DB queries for shard\nranges in order to determine whether any processing is required for\ndeleted DBs.\n\nAuto-sharding will not act on deleted DBs.\n\nChange-Id: Ia6ad92042aa9a51e3ddefec0b9b8acae44e9e9d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38271142eb403a1cdf587616b6d26bfc6b7fb382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38271142eb403a1cdf587616b6d26bfc6b7fb382"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
