)]}'
{"id":"openstack%2Fswift~850357","triplet_id":"openstack%2Fswift~master~I7862c8ef153416c00c8ca7d6bf2f3556a1776d8c","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"28499":{"account":{"_account_id":28499,"name":"Romain de Joux","email":"romain.de-joux@ovhcloud.com","username":"rdejoux"},"last_update":"2022-07-26 01:05:32.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-07-22 22:28:32.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I7862c8ef153416c00c8ca7d6bf2f3556a1776d8c","subject":"Drain and close response in StaticLargeObject.get_slo_segments","status":"MERGED","created":"2022-07-19 14:47:30.000000000","updated":"2022-07-26 01:06:39.000000000","submitted":"2022-07-26 01:05:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"850357","meta_rev_id":"f9fcbb6986dbe8c9c4c0c33f961a8340cc6d198f","_number":850357,"virtual_id_number":850357,"owner":{"_account_id":28499,"name":"Romain de Joux","email":"romain.de-joux@ovhcloud.com","username":"rdejoux"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-26 01:05:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2022-07-25 23:59:05.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-22 22:28:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2022-07-25 23:59:05.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-19 15:25:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-07-19 15:39:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-19 16:39:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-25 23:59:05.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"d62c13d30882f25436bf933ea70abea83e575d6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28499,"name":"Romain de Joux","email":"romain.de-joux@ovhcloud.com","username":"rdejoux"},"date":"2022-07-19 14:47:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"026ac57c484195439f66998d0cbc05d5e5e18e99","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-19 15:25:52.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"646b513b7e7d69e4050c7c6e097c5941fe8a395f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 15:39:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c4c0eee33cf43398e65461abcb114b7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ed6725f5ee547ccaab8f13c7618ff47 : SUCCESS in 36m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5d62d2530874c28afa7a1a3f8c5fa78 : SUCCESS in 31m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c656f93550e44ccabc99b6f53ee86bc : SUCCESS in 49m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa17e1f8bd8c4a67b2feb232e82d7e3e : SUCCESS in 30m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5b8d1049fc4a92406586fbebb51ad33e786db748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 16:39:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e27c73d6dc0f408db648a47520882da8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0463f5a124d49cea2b20ca082c00f4f : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/550af60f5779443aabc86797c33cae28 : SUCCESS in 1h 09m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/94a648976f5d468dbdfd556d4b5a5b61 : SUCCESS in 44m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73dbb7bce3b14dd9bceb6657747fa73c : SUCCESS in 54m 21s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/06357090c1824fcca7757c5f7c15080d : RETRY_LIMIT in 4m 18s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/eb7e4558a6924404bcb5d78f0045a07e : RETRY_LIMIT in 8m 39s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0cc0e244115c40b695ef24e51f7fef57 : RETRY_LIMIT in 4m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6b23bd14fd124bb4882079d0d88de9b7 : FAILURE in 4m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0d590d2054a44926bdbe3c4718e6980f : FAILURE in 4m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28007357efd540b5a8b2b006041a289c : SUCCESS in 18m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5f2d51ee0fe443088972f3ed42d1beac : SUCCESS in 17m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a52c3ee5b73646c884c8ec8155ba86e2 : SUCCESS in 18m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b39ec76bb1e243c68ea323f342d8ff31 : SUCCESS in 20m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cc36324154ae4ee18a645ac0c2a273e2 : SUCCESS in 21m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/98d2f9c914cb412db222e638e8704fb6 : SUCCESS in 20m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e1ed3d0f74af4b4fbd102f18ff35dc5a : SUCCESS in 18m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/03c9a7b27e364a97a0cad456e4c5c5df : SUCCESS in 19m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/65901e36f262447eb8ea51edb260c906 : SUCCESS in 20m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c51c51d150d14bfea09b19f566fcc847 : SUCCESS in 6m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a806a6b34daa4621aa672bc2e7c6dc2c : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2e31dcebe83a4b95958a1eb08e498c50 : SUCCESS in 7m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4a6f8291c29346d8bd50e2a109e11030 : FAILURE in 1h 50m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c79149b7674e4d12a188c63d4ef785e0 : SUCCESS in 1h 49m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/11a0d170f0984e3ab12566dc6ad2bcf5 : SUCCESS in 44m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4a2520d7a2e94cbaab42737657bd1a07 : SUCCESS in 50m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e082213754804bc6b64a5ed4ba64117e : SUCCESS in 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3135dab506f64b6f856c7df1bad17ee9 : SUCCESS in 6m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/18d603239e3a4f3f8e980d73559f3afb : SUCCESS in 20m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ef4e2b9c4d44c2c927f6b016211b68e : SUCCESS in 46m 34s","accounts_in_message":[],"_revision_number":1},{"id":"dc16996c5378c44e15eb4bf82bc55d1e14eb883c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-19 17:54:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d85cf4159fb67d91559994aa025a56d8f4e5ae5","author":{"_account_id":28499,"name":"Romain de Joux","email":"romain.de-joux@ovhcloud.com","username":"rdejoux"},"date":"2022-07-22 07:56:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"79037aced312e560b32c23329d617abaa692b063","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-22 22:28:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3829645adec6d99da66f807adcdadfb5364e3a30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 23:32:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f12a9498ad5943ffac8c0355f30e27eb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a7388915ba54f708f5c4a050084d352 : SUCCESS in 58m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b9c2abdd28e74bc3a9f06a7cdd44dbe0 : SUCCESS in 1h 01m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c99c755a4d94235b1f6c206b10c43df : SUCCESS in 1h 02m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f73624f9a554b81bd7985ddb89269f9 : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c49a1cbd6bdb672d4dda2a6fcd86a6629d1af275","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 23:48:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ab4e30b43c9490cb1c6c777d1035687\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f33a47c6ed444c58b6a63a343b4a482a : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/0a459116416f4c27aa6baf2d48efc8eb : SUCCESS in 1h 12m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b54edd774554701863f314d7d12920a : SUCCESS in 38m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/658be49b724140ee8948400e067b27d4 : SUCCESS in 43m 32s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6183879cba9f43a7b2db04844fa0532b : RETRY_LIMIT in 4m 08s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0cea881609b64a5bafd143ca544c3c37 : RETRY_LIMIT in 4m 47s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d838c10efeb94a7181a73e1b6797326a : RETRY_LIMIT in 4m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f72f725162e047f8b610a14f45ee9b04 : FAILURE in 4m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aff4a59e32fa457293a2d17f7c2de525 : FAILURE in 7m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/106ccaf31ae74267afc25fbb88d518d9 : SUCCESS in 17m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/529004bf55324cd1af4d887db080c81c : SUCCESS in 14m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/151cf80cf6654aeba8e4f9b966e61eb8 : SUCCESS in 15m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2537d1911aba448a98d14aee0806b390 : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4193d17142c54881ba8ed3031299da98 : SUCCESS in 16m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e7459dadbd2a4779a96704cd80624bb1 : SUCCESS in 17m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2cf2958cbd214d5e8f0a809475d75f3b : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/af246e4719a04fcd9c63b46abd2741ea : SUCCESS in 17m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/39fa420eda9f41d4b30410647a58f0a6 : SUCCESS in 17m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d55f10638d5f41839c95d60983557a6b : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/988141af87b3476291ef8b9fd13c97a5 : SUCCESS in 17m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a1ad8dd2ea6492099f355cf2c08a00c : SUCCESS in 13m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/797c069484414bad9b138a6826a513fe : SUCCESS in 51m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/69d3ae5cfb574cd6bcd43f9ddbe8e6c5 : SUCCESS in 1h 01m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c272204a40e240a4b449436d92cb865d : SUCCESS in 49m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/190469a8cc3a42ac9f9ce206452e4779 : SUCCESS in 50m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48bc4efbdd0046068c99ed28261a2fb3 : SUCCESS in 16m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25cd8551a1ba4a34b884fc1eaebb157a : SUCCESS in 6m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cac20faaa91d49f8a7aae90b012a222f : SUCCESS in 33m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a26b54bb12e4637984c27832e688375 : SUCCESS in 54m 01s","accounts_in_message":[],"_revision_number":1},{"id":"4f075f9b21a71d5c9568d6d58abd9b0bd3fc2ea3","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-07-25 23:59:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8316909d54d73b1e8a3886dbfd09066f129ca6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 23:59:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9b040511d436e6d597775cf348b978e8e59d6e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 01:05:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c65d461af5b4846a38560a91c93117a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9132c41cdeb6498ea07182f3a22fce54 : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/b3f343c7f7e446be861c997fb6f664dd : SUCCESS in 49m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b6e5596c4aa34aeabadaee712684b5f1 : SUCCESS in 41m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0587c28d7e74c64b188fe5ef1fcd099 : SUCCESS in 50m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f0d504f5992b473e9f11f876d3c2fcb8 : POST_FAILURE in 5m 56s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3797213d069041ceba1a7d331c097897 : POST_FAILURE in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f7f727534d9440be9046102663bf78b6 : SUCCESS in 23m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fb00442a9ebf4bbc8a9084d1e8656f2a : SUCCESS in 20m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/53737e42d6a9494493b5f98679e21cdf : SUCCESS in 20m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2179ac4a4ed2491dbdb00923c8866a0a : SUCCESS in 24m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c2ed9a1dbd0d4c9698b606c64ee6fe58 : SUCCESS in 23m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/73bb82b2fddb4b5fb273d50e345e4f1c : SUCCESS in 23m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/049cf0e5f7be4c3eb165cde95b065ca7 : SUCCESS in 24m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9bf8bb91d34a4422834ecdc8896448f9 : SUCCESS in 22m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b6a27aed77c44a3897e596a5b919c992 : SUCCESS in 25m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7b02eb76d1954410a67f655edda0f0b2 : SUCCESS in 7m 40s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/171f3981de1144429499ea0e1c3681f7 : SUCCESS in 8m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/faff585ca677424b901b75271e5d8d72 : SUCCESS in 59m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4f4c4ef561f4279ac06887b5087c066 : SUCCESS in 1h 04m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/07c088082677498e94ce82d814dd8661 : SUCCESS in 42m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4747f1126c38411780a992434e1b8221 : SUCCESS in 41m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14c0d9d0a5da4647aefa5cc2b40da165 : SUCCESS in 22m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/094a4c9700164472a02843b6303a2572 : SUCCESS in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a86d2310423e46d6a71245a1f3108705 : SUCCESS in 27m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/636283400cf244efb0c235b651e195b2 : SUCCESS in 44m 51s","accounts_in_message":[],"_revision_number":1},{"id":"4e0619d99efa6bfa006eafb3ceb2e6f44f9a35b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 01:05:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f9fcbb6986dbe8c9c4c0c33f961a8340cc6d198f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 01:06:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/423e6d975aed4100a1f57c070b0139ce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6a36d6234074eba8e6436944ea61189 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e8ae6314293d4dc799540516fa2623c7 : FAILURE in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b8a769ee18754a1dbcb74c49d21b0cb6 : FAILURE in 26s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5c1444faa827dc2e3d85468d42622acea1d435e","revisions":{"a5c1444faa827dc2e3d85468d42622acea1d435e":{"kind":"REWORK","_number":1,"created":"2022-07-19 14:47:30.000000000","uploader":{"_account_id":28499,"name":"Romain de Joux","email":"romain.de-joux@ovhcloud.com","username":"rdejoux"},"ref":"refs/changes/57/850357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/850357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/850357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/850357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/850357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/850357/1"}}},"commit":{"parents":[{"commit":"c3aa1ce66c20c0c9148a4d3e22d3ff5bcd12bfdb","subject":"Merge \"sharder: always set state to CLEAVED after cleaving\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3aa1ce66c20c0c9148a4d3e22d3ff5bcd12bfdb"}]}],"author":{"name":"Romain de Joux","email":"romain.de-joux@ovhcloud.com","date":"2022-07-19 13:22:31.000000000","tz":120},"committer":{"name":"Romain de Joux","email":"romain.de-joux@ovhcloud.com","date":"2022-07-19 13:22:31.000000000","tz":120},"subject":"Drain and close response in StaticLargeObject.get_slo_segments","message":"Drain and close response in StaticLargeObject.get_slo_segments\n\nIn get_slo_segments a GET subrequest is processed to get SLO manifest,\nbut if the object is not a SLO the response was not drain/closed.\n\nCloses-Bug: 1980954\nChange-Id: I7862c8ef153416c00c8ca7d6bf2f3556a1776d8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5c1444faa827dc2e3d85468d42622acea1d435e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5c1444faa827dc2e3d85468d42622acea1d435e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}}]}],"submit_requirements":[]}
