)]}'
{"id":"openstack%2Fswift~850597","triplet_id":"openstack%2Fswift~master~Ic38698e9ca0397770c7362229baef1101a72788f","project":"openstack/swift","branch":"master","topic":"sharder_fallback_to_same_device","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-07-29 18:59:40.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-07-29 18:59:40.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-07-29 17:36:50.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ic38698e9ca0397770c7362229baef1101a72788f","subject":"Sharder: Fall back to local device in get_shard_broker","status":"MERGED","created":"2022-07-21 02:38:32.000000000","updated":"2022-07-29 19:00:43.000000000","submitted":"2022-07-29 18:59:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":61,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"850597-sharder_fallback_to_same_device","meta_rev_id":"7dd29460eb1c69006adb43823b9d9380eaf776a8","_number":850597,"virtual_id_number":850597,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-29 18:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-29 17:36:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-29 17:36:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-21 03:46:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-21 03:59:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-21 15:08:51.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2022-07-25 20:16:28.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"16328ce378541fd9b17f050b4bc028db3e8aa5b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-21 02:38:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ebad78b7fa4129a1a4beb8d8947897de6995767","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 03:46:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef954bba825d41dab0742c7231f08617\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3ddf65ea1754b1599ad0f701f6a2b37 : SUCCESS in 59m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac23940b20884f50b04cb4ac692e2f1b : SUCCESS in 31m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4963978402e49738f8d1193040d8031 : SUCCESS in 32m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/90ab26cfc317457ba1011b9d6f7bc31d : SUCCESS in 52m 04s","accounts_in_message":[],"_revision_number":1},{"id":"72821ffce6905c7c7c68d1bdf571eb4faed7961e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 03:59:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43d488ad689d45f592de6e7c02c2cf88\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6086f03d3bb43f28550ea161fe88bbd : SUCCESS in 19m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/37a6573a185d4bbba7681fde1b2d4423 : SUCCESS in 1h 04m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e397a1217381458f80c129a4607c88f3 : SUCCESS in 59m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7ac67bbe185405f9866cf685ad7cb11 : SUCCESS in 53m 31s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d8c72b5b4e954af7a19b582618e165cc : RETRY_LIMIT in 7m 02s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9b37b87b06d74860be39d9321b15693d : RETRY_LIMIT in 5m 22s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ba033836860a4033ae8fa7b8b15ae358 : RETRY_LIMIT in 5m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/549842b61c7c4c07804e0ce17c118e30 : FAILURE in 3m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/74df960c6b3a4ad08fcacba00886582b : FAILURE in 4m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/95e900c3d5de44af9da9ad49a7c62b47 : SUCCESS in 16m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/46c05c57c17649ba824519d06c7ed7f4 : SUCCESS in 14m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/896870c33b0b45189e5604ab22113f03 : SUCCESS in 14m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/250fd53df89641fe8f15124d60dbd621 : SUCCESS in 20m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6d07233d875f43ecb5bbd1af144c364b : SUCCESS in 18m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c7d77d35c5dc4ffb855d4015091e0aa1 : SUCCESS in 17m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3fc6d68b236544cb9dfaba55ce95f3ab : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fc892e9410f442bbb56e87c82563ce40 : SUCCESS in 19m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0afb5c0c44e948bba054659bb1abfe80 : SUCCESS in 24m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4bfdeb0de9e74a2baaa70088985072aa : SUCCESS in 9m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ac59af50729c4b638a9b8c8d104c02f5 : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/accfcd664d734076a8c04642ecc80fb1 : SUCCESS in 10m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ed2bc726062548e49a3cae432e641ae7 : SUCCESS in 45m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/dcba4199d5824d7c910d6968f685962f : SUCCESS in 52m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4f5fcca551994238bb50c7e3e3d4ad1a : SUCCESS in 32m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb6654fa089645f29d5685b6d78a2dcf : SUCCESS in 47m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee1e283923c44c7cbc0a86758372f090 : SUCCESS in 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/014a2946b9df4db697a66d14088236df : SUCCESS in 6m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b3e6b8f0a1724ae4b88ab511283c7312 : SUCCESS in 26m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68d8d53383eb494eb72585a602fa7328 : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0c218fce8a179d6be4f6949242ea8c19085a4c48","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-21 15:08:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"856676652cf787c572c4ccd78864662d0a7c4bef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-21 15:09:46.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"45851ba9b52edda46277606a60aa7b9959f93a85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 16:10:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4152c038e79e464d917e6e21dfb0fbf9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e141c9c040c4c7ca3437cc90e3c0ee5 : SUCCESS in 51m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54ba114de5ac466d9e07898501c71c74 : SUCCESS in 51m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/feea957fe8c64f2ba1963202879a59ce : SUCCESS in 53m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fdea3762b86c423cbb09a30176188378 : SUCCESS in 53m 14s","accounts_in_message":[],"_revision_number":2},{"id":"ca2c992427e919ac4e3951e3ec610272f05268f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 16:23:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da414cf9c6ea481291b51296ce61ac6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5243b1d9f711477a9ff6b90701cd56e4 : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/f491f6e4d7c947c2868bf8597074f647 : SUCCESS in 1h 06m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4d20e48411824fbe9eff5d209147d6ab : SUCCESS in 51m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f938e3978da942f09b1b1859f914f5dd : SUCCESS in 51m 39s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cd77a5e3a22a4c2087abd7a6eda6c728 : RETRY_LIMIT in 5m 24s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5fe9d5a25a444dbb8127b04c2f89431f : RETRY_LIMIT in 7m 55s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/06e2baefd2424bc6a35e43be089fd02f : RETRY_LIMIT in 5m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/768ccd849151408cbb8c47394b4a7f29 : FAILURE in 4m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b84af3b7be4e49c781b9cc08f9f9fe8f : FAILURE in 4m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d1696ec7cddc41eaa279c4055674935c : SUCCESS in 18m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f341b969497e43c99ffaf6f29e6332a5 : SUCCESS in 20m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/66eeec478dde430e930fbbbbaf1a3ab6 : SUCCESS in 16m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/93df67d3dc634847805c690a93357452 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/414495c78c634ca588226bb40d160f1d : SUCCESS in 20m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e34958dbfbb741d2a938e265a64f6c03 : SUCCESS in 20m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/25fb4c3b32cd4938ba4deab5e7d7ab09 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c7c6eba47b2b4f2aa0ae875fce5e0b22 : SUCCESS in 17m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4bded2526fe74b89be7eb8f1d8cd8bdd : SUCCESS in 18m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/844f453660c1463b8bfe9b407f42c4e0 : SUCCESS in 8m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d162a30231894cc191d8332b84c40908 : SUCCESS in 23m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7d4471441f7b4dcd9b6826d45e30afe9 : SUCCESS in 9m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7491b4009ea34d5a889cc91924975cd5 : SUCCESS in 55m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba37e3922916428bb50e96cba2613ee2 : SUCCESS in 58m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c649b92ca7f54e1fbb34f3aa73b78097 : SUCCESS in 37m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/34aa45b5777c4efcb47a91942fb119cd : SUCCESS in 52m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/393a4ad1c3dd4acf941b2af4a22e2f37 : SUCCESS in 21m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59166beb21c24811854b16c1fa243183 : SUCCESS in 6m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d49809fa76f04db0984f4272dd49c4a7 : SUCCESS in 29m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f137a085b644a6085641c53dfd164a3 : SUCCESS in 1h 12m 35s","accounts_in_message":[],"_revision_number":2},{"id":"1ef8c046405282116ff55c685f8b66f0e880ab5b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-22 00:42:09.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"814288f1c1aa1d5652d06e2a6970b4b3c87c1a36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-22 02:22:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5e889f62b8fbd0dd1431b3c53fb3c53a444f089d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 03:34:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e3e0ec414f9462cb294d46bc094a4f0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b722c8db6674944bdf5da64b07d9743 : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/5aabb438f53a47759748096be8cb66f0 : SUCCESS in 1h 01m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/02f1fbb621754c0b886d23b0a3c67082 : SUCCESS in 52m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2458a0905edc4ae581fb3d371410e53a : SUCCESS in 53m 07s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5973dae27f2c41e58d731858b22dcfb8 : RETRY_LIMIT in 4m 24s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e592609a871441fabf8ad5db0b5033a1 : RETRY_LIMIT in 5m 06s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/53ee09cd9ad64e4ebaacc8f74fb1e6f5 : RETRY_LIMIT in 5m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/db9d68db1158402f8029476c63fd8267 : FAILURE in 5m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ce4b52bb907f4fbb839e7918c3e62314 : FAILURE in 4m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/730785fba8604f238948ea553f6c58f2 : FAILURE in 18m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/19140345acdc43ad9a8b92c3475cf4db : SUCCESS in 25m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6ca5bcfd9dfb4287a89d3c3479653896 : SUCCESS in 27m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bc3a890089124055b51241a6a04a3354 : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f1f4ac03936949acb239344af373521a : SUCCESS in 15m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6dcde921cb094e4cabe76acffa4d9963 : SUCCESS in 16m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a4c46087e1b64ff6a6436d14bc25dcc8 : SUCCESS in 13m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c83a34f5bc38495aa9c2025704d70dc6 : SUCCESS in 16m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/633df0e2b0a642969dfb72f1b84983f0 : SUCCESS in 19m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1c193567a52e443c9dfb2b75a17b5a0e : SUCCESS in 9m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/702c30d6a19745b6a761a1c6f3cce487 : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/94a7d7bd682947dc9f5eef0e4b0af771 : SUCCESS in 8m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a266ee69358e4499b2a3b938a86a4f89 : SUCCESS in 56m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0f7c5e29a5784e8084e3881cf55a2adf : SUCCESS in 57m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a9e9071cf9064e36ad4444782f0f8e16 : SUCCESS in 43m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/20575e32e8644c1eb3c6ad7739a164b0 : SUCCESS in 43m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3b924d6925e445db28663dc0ac69a03 : FAILURE in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62b0ea13950c4e98aa99a2b11250cdcd : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7f42053d58f94d0298e62f125e5949ad : SUCCESS in 30m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e7a3c53df834a959840e4657040d623 : SUCCESS in 1h 04m 20s","accounts_in_message":[],"_revision_number":3},{"id":"d3f1e89fc0df39f37da70fbb461f202ea2e1666b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 03:42:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0e105d6bc8a4ced99b088a810535d41\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e0922515c1942ec908d2f22c39aa2f2 : SUCCESS in 33m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce788aba18754a5c8836a982c8881786 : SUCCESS in 1h 05m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1d91f3dde25462d9632659166d530ec : SUCCESS in 33m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0fe0e476333d4d26a0dc9edc5b0f8392 : SUCCESS in 1h 15m 19s","accounts_in_message":[],"_revision_number":3},{"id":"27b0e0342ad87e7d140546fc6ec933a8b495b22f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-22 09:26:02.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"48fa11dc3ce359a2441267a003f353347579d714","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-25 06:58:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6de72558c942d21a100b36d1d42256fee7c1549c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-25 06:59:36.000000000","message":"Patch Set 4:\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"947bbebe2a957b2328453e925d1e15d2176172b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 08:01:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f6dc16d07c442268135ba22194f42e5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76894d6dfda84ed7a8d5e791af3f29eb : SUCCESS in 31m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c54e8897f2104127a16b01be06317498 : SUCCESS in 30m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/41fe3a97966249f89d24d622afd8554b : SUCCESS in 30m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/08809a65f5d84e4ab6432839165e28ed : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":4},{"id":"fd4e2a3897dbdd63ccd7ee3f911aa04e64112598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 08:09:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b064399141a4a6d98593bc55b07bad2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9de01fed28924319926804d6dba71eaf : SUCCESS in 11m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/ee53d7281e2341efb824062ec4ff3842 : SUCCESS in 57m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/24b0faa243c34abbb4b9e19289b68294 : SUCCESS in 47m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7e287849f5b4aa7b71fa45a4f99b1c5 : SUCCESS in 50m 10s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4706061191ed4d8195893f1ce45f63b1 : RETRY_LIMIT in 4m 56s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/76cda0a8809f4a09ac2eda954c2c2ac4 : RETRY_LIMIT in 5m 29s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ad40a55db8fb488e97af42c9c58880c3 : RETRY_LIMIT in 5m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c27b816a4103469989f7d23405967d93 : FAILURE in 4m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a8b12c8fe31248a999f1c4789b2bc59a : FAILURE in 7m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/888ba2e484f642698a2c2cb51bc26db7 : SUCCESS in 17m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9fa306efab334397abe6c7aaff297447 : SUCCESS in 18m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7918b7f3c3e147958b841baa363fcf5a : SUCCESS in 17m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aaea51c43a2346b198a70fc432586c5c : SUCCESS in 30m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7cd47752d6a14ab786893822c24cbbcd : SUCCESS in 35m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e06addcc8b134c80902687cc380aea59 : SUCCESS in 18m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cc8f385e88a24aae87355b6beb6feb0c : SUCCESS in 28m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/54a365b5d0614378bafdb3665cd1d4b3 : SUCCESS in 19m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7d63efe30e1b4938b45e6f5afcdddc3e : SUCCESS in 25m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4a351ac929a946388bc923cda9eb8127 : SUCCESS in 8m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e0219e7bb9d746b1a911c67faa2d2556 : SUCCESS in 20m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/25ba3d71dad74e9db8c5b8c70bbbfc28 : SUCCESS in 6m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5e8d27f9e84e419088399efe413b5f14 : SUCCESS in 1h 08m 57s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/99db784954964a4f9edcf1cf8b7af3f6 : SUCCESS in 1h 00m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3a09ecb423a0410daf0814f28a54b610 : SUCCESS in 47m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7507e2410b9a412a803375d198f5a731 : SUCCESS in 47m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6c76bfa32a04482bacd702872189bcc : SUCCESS in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52f8c365ce014da38d62f51a6df03bd0 : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39c562e1da5f410da43a6bf7624f558d : SUCCESS in 30m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b18f6f9622814db1a45c809a9e4bb7bf : SUCCESS in 51m 23s","accounts_in_message":[],"_revision_number":4},{"id":"8e55d9875559a61cbdcc41c7fcd9c21dc3f79154","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-25 20:16:28.000000000","message":"Patch Set 4: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"41fccfd722e851f8d95aabcbb1b2588548bb69fd","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-27 05:51:27.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"08d4e55d150cbd4a8073b9fdf6b11d6f8482e367","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-27 11:28:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"cf5b4efda44d529c65646ce98597c79014ac0bd9","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-27 11:32:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0d867c06d5620811bfa7e7b5f32629e7c93597e2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-27 11:34:21.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0ac35a47e43c7d4782a1c7494b11b13596f83047","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 12:37:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/810116f2840141d4a9480eaec42af49c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3bc22369e4746479553696737274a17 : SUCCESS in 12m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/4f27b4c456bb4733bb001911680c8358 : SUCCESS in 1h 05m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/82c5642ec6544baabc00d7be080b5e32 : SUCCESS in 52m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e800293a97e402cb2b1041db0c67020 : SUCCESS in 53m 28s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/368ef01c067042e487b4605ce612a8af : RETRY_LIMIT in 4m 35s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1895c01759fc4067af41821cc2513d7c : RETRY_LIMIT in 7m 05s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d06645ff6d044e84b6593f9072c5fe57 : RETRY_LIMIT in 4m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a83771280db84c68b1aeac80b36786d4 : FAILURE in 3m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6b52042ac1f44b5ea2b802b0d50e65d6 : FAILURE in 7m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d5648b70bec343989a5fca94c7026a57 : SUCCESS in 20m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/391d83e4af9647b6acffee12a26014a9 : SUCCESS in 20m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb1b1689322d4cc3a8621375c46d64c7 : SUCCESS in 16m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/14ce082d5e054d11af2ed91426ea62af : SUCCESS in 18m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/84beee61723b4c4e83ec25e40b1ff79d : SUCCESS in 18m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/57dc71a23a064553b2c7028434826a82 : SUCCESS in 19m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/266a2d7b681b4e6192d23384e8ef4d71 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/792d5ea80f3248e9a80531748ba92c7e : SUCCESS in 16m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4ea8abac03954d4cb8034767dba1adf9 : SUCCESS in 21m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/658c0723437b4d6fb4cd324706b04529 : SUCCESS in 9m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/acecbb6baa014f5ea68695404573ce5c : SUCCESS in 22m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8c3ea7254f5d411a811b07b90e2c56fa : SUCCESS in 13m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/461070c331244c838709f2a5fa422d73 : SUCCESS in 54m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/72aa5809493e478e9e840de20ca25d7b : SUCCESS in 1h 01m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1aa96d9a7c1e4d7b80de0a492f65008b : SUCCESS in 46m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/02f6e5b3bf784d44a41829331adf30f9 : SUCCESS in 51m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1251861cbc294621a56184192d997bf9 : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0a269896fa64e87a04f58c1f3943650 : SUCCESS in 6m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/daddbcf3cf384bc088509455e75e4481 : SUCCESS in 34m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f3fa837b2f4477d949b0f430f6c8ca7 : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":5},{"id":"cf26f02a51eb2fa13dcc509495287147302aa332","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 12:48:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2463187bcbfd45e7836c2b09d77e522f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/77469ab1924d4aa497ac59593b2d80f7 : SUCCESS in 30m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a609e42c9b984603860befd38d1f64c5 : SUCCESS in 1h 18m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f018ab221d64336a9c0ac789e64fabb : SUCCESS in 30m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c89ad0e55b8471593f5436f3821f1e5 : SUCCESS in 1h 11m 10s","accounts_in_message":[],"_revision_number":5},{"id":"920d0d388965b04f09e69b7f1de53c8130fe5d52","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-27 15:47:46.000000000","message":"Patch Set 5: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4553b41967851d48c1c129b4a8bcc8e25a98d270","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-27 16:04:22.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"31c56721b09bff45641376a12e1b70b3f19bb6cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 17:20:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dc477d6c97247aab5a136cc182a73b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66dc78480f974aae8787ec352ad05bf2 : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/498f6c3215db41c3b059a1a7546e3c1a : SUCCESS in 1h 15m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e7dce42ea1a14250bdc1608468fb07ff : SUCCESS in 55m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21d7a4875d2044eb82c455fafc69ca5b : SUCCESS in 1h 09m 54s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/87f8c33081c0403383934a5e806b2e8b : RETRY_LIMIT in 5m 02s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6a264d60bae34b7e9f5f5091ac1c6009 : RETRY_LIMIT in 5m 06s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8da99a057a99408d9215bae3532a87b4 : RETRY_LIMIT in 4m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8fdcd95b51cf490c9a31cc0e1b880826 : FAILURE in 4m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9a64b0fb777d48d987ef61382073ecd8 : FAILURE in 4m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1167665ada694ef2848a7a25f6a73e5e : SUCCESS in 16m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9812e176a1f3403b8cf51a2d31f4cca0 : SUCCESS in 15m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0ecdfc443f22469999a19a4986cd21fb : SUCCESS in 15m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2e8fa8742cc444e78211b3430374d824 : SUCCESS in 16m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ad25b92458374d8e944b2f8bddda31f9 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ecd59f8b15d64b99b4c9350d68cf9e65 : SUCCESS in 17m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f96cc94c36d24894a1333eb4184b9a3a : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c0c66e95bf4f48579b9916faef94604a : SUCCESS in 18m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/757693dcc6984174952d4205229ccca0 : SUCCESS in 15m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e4aa110937ad4776834452f918dbf8f9 : SUCCESS in 8m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/88850bd408504a7c9cbc171b7be2d2f2 : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f534c1e9b18b4dc5ae9e4575cd0f2bbf : SUCCESS in 8m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a93182f584274c949817cb182f842252 : SUCCESS in 54m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef2575ec7cf148fe94da5201053f2ea1 : SUCCESS in 1h 01m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eedd945bf9d242bcbe6fa28c9146bba2 : SUCCESS in 46m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/83f248c450f14176bce29386db154bc6 : SUCCESS in 49m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be32fd4ccd704a0c9b8aa53079674953 : SUCCESS in 16m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/810394e7bc41404eb16fde537ddacc72 : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/28b87f7da0ce4d3195e299aa2045a5d4 : SUCCESS in 25m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7ae1d12bf1140a2a48e2f2e7ee8bf45 : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":6},{"id":"984e2eeb9976c1ed76b7c6fb0cae40891a13cc6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 17:42:07.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d549b02041b341a5a4571df1361db2d9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/58228af9a63b42f9b50ff4da8e05ffa1 : SUCCESS in 34m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1abdac48467349abaee2557682a02827 : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c49078158ed74f3d9251afa095cf49f4 : SUCCESS in 33m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/90957992f5084cde836c1146f12ee370 : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":6},{"id":"559054308a7fbb8df9614431551b55a0ca2160a7","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-28 07:24:21.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"16c27a238693678db73833a1d7b680ad71b51ca0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-28 07:24:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5b0fbc5b415ec531873925ba65a8f6e08f8bc3db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 07:25:12.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 850597,7","accounts_in_message":[],"_revision_number":7},{"id":"2b060913c406c2fb4104566daa411e5f0046ebe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 07:25:13.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 850597,7","accounts_in_message":[],"_revision_number":7},{"id":"1d9c85a0dbd55cc79cbe90bfc85f7f58f40d97a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-28 08:53:39.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"dce09c53df5f54f39107a6dee6c87b023888913c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-28 09:44:57.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"263bb7f2b24af717b9a9c6a18a8fe6313c53efaf","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-28 09:45:42.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"670ca84fd1cff4bda43f8e7268e952f7a7834da5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 09:55:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44e7b3031f56419aaaf120b498f6519b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c86e4681bc243ba807c828da9766579 : SUCCESS in 47m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b5807f46eb19440d88d2490fc2316fd0 : SUCCESS in 48m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/23cab62a47f2481b901740bf3d42c82d : SUCCESS in 59m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/85ecf710533b48afa1d09992e8157945 : SUCCESS in 47m 46s","accounts_in_message":[],"_revision_number":8},{"id":"fcf94be3811ff109b4caa89f1cbb3e0725b1a859","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 09:58:06.000000000","message":"Patch Set 8: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f4fb50d68364dbbaf215c349ceb723b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7766d5b844314bc0ad39a6734770a91f : SUCCESS in 10m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/51edf1314f2343e79f490fe12799c486 : SUCCESS in 1h 02m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd9627074a4349babc21dc87690ac700 : SUCCESS in 48m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e287009108448d2af46fb6418fdb121 : SUCCESS in 50m 09s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/059a0a0cae444c1d89c2eb8760ada1b8 : RETRY_LIMIT in 4m 07s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/fee1df3e21c14aa8bb8abbd94d1ea824 : RETRY_LIMIT in 5m 16s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e8f065445be743e0ac810c657c1f4e74 : RETRY_LIMIT in 5m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/33c029b1251749679a42308915ab03e3 : FAILURE in 5m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ae18bd613637493e923f901011f89951 : FAILURE in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/418c0fd8b3bb4622836c7a088a65402f : SUCCESS in 17m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/af3b4e9b875942809fa7c05e89667dd3 : SUCCESS in 20m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1b9b8142d24845c8a5ac13bc9d90ea22 : SUCCESS in 17m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0e5094aba65943e58c3b570eb54f84ce : SUCCESS in 22m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/43622f5544af4b8dacb73410e63de216 : SUCCESS in 19m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6891a6c4c55941679fdf7f5a19ad428d : SUCCESS in 25m 40s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/19366613ffe54ec98c0e0d71b00804d5 : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0ede5869b94942ebba057f5f713b3f7d : SUCCESS in 22m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d715bf1e576d4f3995fd378c111a24e1 : SUCCESS in 13m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5b29455f5a6a440caabf36de06239881 : SUCCESS in 7m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/01e6b82279a34c1b9fd4e846ae2eda64 : SUCCESS in 17m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fab67c08a08c42bf87ac2f99abc8b420 : SUCCESS in 8m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e710953acd8746c59f9f271699c03a5f : SUCCESS in 50m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6cab5daa6d1c40a7b6e4a40087724487 : SUCCESS in 56m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4b6f6b398c94a3e999e5223d48863a9 : SUCCESS in 36m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a358d14cfbff4a6fa2c68fd95196bbc8 : SUCCESS in 34m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55bc7d8de9ae425aaacef116681ba736 : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d166dd18b93d4a8894a4474677f95ea9 : FAILURE in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d10784182ce42048a4ac8424c620794 : SUCCESS in 21m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad0fcf1425634f3db98e01f1d9871c80 : SUCCESS in 52m 51s","accounts_in_message":[],"_revision_number":8},{"id":"422dcbcf893b100ddb68e84381cebab9c917b5e2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-28 21:18:28.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"95aa5b99e6064882b565af63760c4c89025ce6e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-29 14:02:57.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"12eec467af691ec3a2d7878c3779ca7f401e2415","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-29 14:04:44.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d8b4e31efa64c7ef7b0484cd1bfeb9adcea6f10e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-29 14:22:22.000000000","message":"Patch Set 9:\n\n(9 comments)","accounts_in_message":[],"_revision_number":9},{"id":"f1aa96d7771b9827306a97a52bdb2d988ae83842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 15:25:50.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a98ab358e3894fb6a3731e6035e5b378\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfe861dff8f14b93a60502ff8f7eb8df : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/07b47dd5644b4de993f3abda1ac3d681 : SUCCESS in 1h 01m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1f9c20f9d5ae4787a60e4df153b1b460 : SUCCESS in 57m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a328805af514e6cb9cfac3beef84497 : SUCCESS in 55m 51s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/074fd8ae44954b52a9310add755e2742 : SUCCESS in 24m 56s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/963b3a50679546bbbf61bff495c5878c : SUCCESS in 33m 06s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5d998fedf25540b28b7ffe7befc35cb5 : SUCCESS in 25m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4aac9800b1f540bf9c937432cb01f3cf : FAILURE in 6m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1b4ec3ff38b246d8afce22c12c03982a : FAILURE in 4m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/298b77b61f5040dc8503859dc7e46486 : SUCCESS in 27m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6ace8884c44245aaa395737bc594ea5d : SUCCESS in 16m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/83b6b42e5ba74c3f9bbcceadd6f8398c : SUCCESS in 20m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cad6597d4fad4c898fee48b974b24e1d : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1fa6ae4d85e940be995dd02fc288822b : SUCCESS in 22m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b921027d2dd640db95c57cd01f1c4dd7 : SUCCESS in 18m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ee780f56f77e4f2c8dac5d27b6beb167 : SUCCESS in 20m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/71f6a083702e4f738ca1af054c5c4efd : SUCCESS in 17m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e3b982ee496347579c57f0a0f3679c56 : SUCCESS in 21m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/664a78ed0a94447db01ee56aec014332 : SUCCESS in 9m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1084a751a6242a2b4605d8d12714c46 : SUCCESS in 23m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/39a7495cab174ccfa79560af6353ea6f : SUCCESS in 9m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/352206de34664d97950cbf236a96ac3e : SUCCESS in 55m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/61f340d3e1744540974bb257de5d25ab : SUCCESS in 1h 08m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eb15378bc7df4835b0fac4dc28924101 : SUCCESS in 51m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e253a01faed44a8b233fb7a80bc4818 : SUCCESS in 45m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95ae32ffce314cdca514cfdd0a4d319f : SUCCESS in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e46903558184948a20c419bb595ed9a : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4be7c57e3e00444caffc0b51be10371f : SUCCESS in 25m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cfb6930839545a998852c6f3bbbee3b : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":9},{"id":"07640a5ff12b5cfa224ab4226cf73e1330d8c6dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 15:35:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25abefea738a4a48b497d96b6a920f95\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c6a7b28bfef4b32ad810fbfd344cfcf : SUCCESS in 1h 31m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12f903c957324629ba745e916b3e216f : SUCCESS in 1h 20m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/def2985cba2c416f89556a1a33c9a7d8 : SUCCESS in 1h 21m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ba05a8a9a8e439681336e66917fdcf9 : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":9},{"id":"3a7f90da7a46d2ae8356e07ed9a4cd760cf364e4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-29 16:02:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7516d53288fc368154e3d84a11a63da016d5957e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-29 17:36:50.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"416786e413d8c9e0da1380e7cacac017160b64ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 17:37:41.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"0a0ca9a049f0b2e4ff37e5cde56a5c0eeca7e9d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 18:59:39.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de8bed6677484cfb800986f15f91c40d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/308bf9e225ee4dafa77e6d495e299eb3 : SUCCESS in 9m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/28e40d113a0348218bb0ec7505e3245a : SUCCESS in 1h 06m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f4907d77521c45f1b5a959b8844e660a : SUCCESS in 55m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94d26877c66c49c4a4aaba823eed61c5 : SUCCESS in 57m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6fbde0c3bd82444bad7cc439ce836e83 : POST_FAILURE in 7m 03s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/26f380be5eca458f86204148a07f9666 : POST_FAILURE in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4b13fdd393e241c1a2135140f8afef1d : SUCCESS in 16m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2f9cf77a3a5a46a1b60ce3ceb67448db : SUCCESS in 14m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/641ac23e16474e7c8b6008d8473c650a : SUCCESS in 15m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9c31a4c536fa4dbeae3d2a2e6d8e7067 : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9661175e28c6447385271039ebd058be : SUCCESS in 15m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0ae1cf59844241869f78360b1a1efcac : SUCCESS in 18m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2b0b8af6b32c4754aaa6ebdd846b7ce0 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c48789968d1c4ba7be5df252fd7c8399 : SUCCESS in 17m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4d50ee63b5644c6a9e21982d8d4511a3 : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/728d8c6893ed4554889949dcde11e21d : SUCCESS in 8m 22s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/245d1dd888cc406fbb5176b7475a2e11 : SUCCESS in 10m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b6059b829f424feab64c210f1979ff11 : SUCCESS in 57m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a82007e5ff3a4f58a5ff9a6d074568dc : SUCCESS in 1h 15m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d9c9136fddb14effaea5617dffef2826 : SUCCESS in 43m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bf7f18ee2a104505b70b2f7f5e9ca608 : SUCCESS in 46m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed0961661ab34fc88c0058040b53fa25 : SUCCESS in 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa6433baa65d46e298baba4c59941cfc : SUCCESS in 7m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/47722cfdaf1648d5bbb23c77ad8475b7 : SUCCESS in 32m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc1ddcc03e2b46d0bba585c4b688885b : SUCCESS in 51m 19s","accounts_in_message":[],"_revision_number":9},{"id":"1cc67620088568d81dd3411bbca906e783fe95fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 18:59:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"7dd29460eb1c69006adb43823b9d9380eaf776a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 19:00:43.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/954f2e7e1035427b9b0247357ad7f671\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/657a6687423a443997a0fb072cc25654 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5501cef960204429bc9db112cdd658f1 : FAILURE in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/335806cd8533474292bb1ca83e7d9c14 : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"c4e00eb89f34de79d5fb123dd044621ef4df679c","revisions":{"cf4e688347cea13c98c65bf469b03260c57ee980":{"kind":"REWORK","_number":1,"created":"2022-07-21 02:38:32.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/850597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/850597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/850597/1"}}},"commit":{"parents":[{"commit":"c3aa1ce66c20c0c9148a4d3e22d3ff5bcd12bfdb","subject":"Merge \"sharder: always set state to CLEAVED after cleaving\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3aa1ce66c20c0c9148a4d3e22d3ff5bcd12bfdb"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-21 02:32:27.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-21 02:32:27.000000000","tz":600},"subject":"Sharder: Fall back to local device in get_shard_broker","message":"Sharder: Fall back to local device in get_shard_broker\n\nIf the sharder is processing a node that has 0 weight, especially for\nall the devices on the node, the `find_local_handoff_for_part` can fail\nbecause there will be no local hand off devices available as it uses the\nreplica2part2dev_id to find a device.\nHowever, a 0 weighted device won\u0027t appear in the replica2part2dev table.\n\nIf the `find_local_handoff_for_part` fails then we fall back to a local\ndevice identified by the `_local_device_ids` set that was identified in\nthe sharder. This uses the ring.devs, so does include 0 weighted\ndevices.\nThis allows the sharder to find a location to write the shard_broker in\na handoff location while sharding.\n\nChange-Id: Ic38698e9ca0397770c7362229baef1101a72788f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf4e688347cea13c98c65bf469b03260c57ee980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf4e688347cea13c98c65bf469b03260c57ee980"}]},"branch":"refs/heads/master"},"ba0104ff9bee79dda9d3ca8b8c0036d9cf7f8941":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-21 15:09:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/97/850597/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/850597/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/850597/2"}}},"commit":{"parents":[{"commit":"a303319f0c21cff5eef74a229d5131d3a6b0ef34","subject":"sharder: process deleted DBs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a303319f0c21cff5eef74a229d5131d3a6b0ef34"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-21 02:32:27.000000000","tz":600},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-21 11:27:12.000000000","tz":60},"subject":"Sharder: Fall back to local device in get_shard_broker","message":"Sharder: Fall back to local device in get_shard_broker\n\nIf the sharder is processing a node that has 0 weight, especially for\nall the devices on the node, the `find_local_handoff_for_part` can fail\nbecause there will be no local hand off devices available as it uses the\nreplica2part2dev_id to find a device.\nHowever, a 0 weighted device won\u0027t appear in the replica2part2dev table.\n\nIf the `find_local_handoff_for_part` fails then we fall back to a local\ndevice identified by the `_local_device_ids` set that was identified in\nthe sharder. This uses the ring.devs, so does include 0 weighted\ndevices.\nThis allows the sharder to find a location to write the shard_broker in\na handoff location while sharding.\n\nChange-Id: Ic38698e9ca0397770c7362229baef1101a72788f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba0104ff9bee79dda9d3ca8b8c0036d9cf7f8941"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba0104ff9bee79dda9d3ca8b8c0036d9cf7f8941"}]},"branch":"refs/heads/master"},"0576b4619e31845993a01c49cca83e586d541bfe":{"kind":"REWORK","_number":3,"created":"2022-07-22 02:22:37.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/850597/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/850597/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/850597/3"}}},"commit":{"parents":[{"commit":"a303319f0c21cff5eef74a229d5131d3a6b0ef34","subject":"sharder: process deleted DBs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a303319f0c21cff5eef74a229d5131d3a6b0ef34"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-21 02:32:27.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-22 02:22:15.000000000","tz":600},"subject":"Sharder: Fall back to local device in get_shard_broker","message":"Sharder: Fall back to local device in get_shard_broker\n\nIf the sharder is processing a node that has 0 weight, especially for\nall the devices on the node, the `find_local_handoff_for_part` can fail\nbecause there will be no local hand off devices available as it uses the\nreplica2part2dev_id to find a device.\nHowever, a 0 weighted device won\u0027t appear in the replica2part2dev table.\n\nIf the `find_local_handoff_for_part` fails then we fall back to a local\ndevice identified by the `_local_device_ids` set that was identified in\nthe sharder. This uses the ring.devs, so does include 0 weighted\ndevices.\nThis allows the sharder to find a location to write the shard_broker in\na handoff location while sharding.\n\nChange-Id: Ic38698e9ca0397770c7362229baef1101a72788f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0576b4619e31845993a01c49cca83e586d541bfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0576b4619e31845993a01c49cca83e586d541bfe"}]},"branch":"refs/heads/master"},"eb0cecf682005d99d16a154472faf878b6bfadb2":{"kind":"REWORK","_number":4,"created":"2022-07-25 06:58:01.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/850597/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/850597/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/850597/4"}}},"commit":{"parents":[{"commit":"a303319f0c21cff5eef74a229d5131d3a6b0ef34","subject":"sharder: process deleted DBs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a303319f0c21cff5eef74a229d5131d3a6b0ef34"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-21 02:32:27.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-25 06:56:21.000000000","tz":600},"subject":"Sharder: Fall back to local device in get_shard_broker","message":"Sharder: Fall back to local device in get_shard_broker\n\nIf the sharder is processing a node that has 0 weight, especially for\nall the devices on the node, the `find_local_handoff_for_part` can fail\nbecause there will be no local hand off devices available as it uses the\nreplica2part2dev_id to find a device.\nHowever, a 0 weighted device won\u0027t appear in the replica2part2dev table.\n\nIf the `find_local_handoff_for_part` fails then we fall back to a local\ndevice identified by the `_local_device_ids` set that was identified in\nthe sharder. This uses the ring.devs, so does include 0 weighted\ndevices.\nThis allows the sharder to find a location to write the shard_broker in\na handoff location while sharding.\n\nChange-Id: Ic38698e9ca0397770c7362229baef1101a72788f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb0cecf682005d99d16a154472faf878b6bfadb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb0cecf682005d99d16a154472faf878b6bfadb2"}]},"branch":"refs/heads/master"},"a210079b8feac07cc9893e4d6f17858b57b349e0":{"kind":"REWORK","_number":5,"created":"2022-07-27 11:28:19.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/850597/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/850597/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/850597/5"}}},"commit":{"parents":[{"commit":"a303319f0c21cff5eef74a229d5131d3a6b0ef34","subject":"sharder: process deleted DBs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a303319f0c21cff5eef74a229d5131d3a6b0ef34"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-21 02:32:27.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-27 11:17:43.000000000","tz":600},"subject":"Sharder: Fall back to local device in get_shard_broker","message":"Sharder: Fall back to local device in get_shard_broker\n\nIf the sharder is processing a node that has 0 weight, especially for\nall the devices on the node, the `find_local_handoff_for_part` can fail\nbecause there will be no local hand off devices available as it uses the\nreplica2part2dev_id to find a device.\nHowever, a 0 weighted device won\u0027t appear in the replica2part2dev table.\n\nThis patch extends `find_local_handoff_for_part`, if it fails to find a node\nfrom the ring it\u0027ll fall back to a local device identified by the\n`_local_device_ids` that is built up when the replicator or sharder was\nidentifing local devices. This uses the ring.devs, so does include 0 weighted\ndevices.\nThis allows the sharder to find a location to write the shard_broker in\na handoff location while sharding.\n\nChange-Id: Ic38698e9ca0397770c7362229baef1101a72788f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a210079b8feac07cc9893e4d6f17858b57b349e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a210079b8feac07cc9893e4d6f17858b57b349e0"}]},"branch":"refs/heads/master"},"592523654f4531d1c3fb69c884d9e7792066500d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-07-27 16:04:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/97/850597/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/850597/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/850597/6"}}},"commit":{"parents":[{"commit":"38271142eb403a1cdf587616b6d26bfc6b7fb382","subject":"sharder: process deleted DBs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38271142eb403a1cdf587616b6d26bfc6b7fb382"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-21 02:32:27.000000000","tz":600},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-27 16:02:39.000000000","tz":60},"subject":"Sharder: Fall back to local device in get_shard_broker","message":"Sharder: Fall back to local device in get_shard_broker\n\nIf the sharder is processing a node that has 0 weight, especially for\nall the devices on the node, the `find_local_handoff_for_part` can fail\nbecause there will be no local hand off devices available as it uses the\nreplica2part2dev_id to find a device.\nHowever, a 0 weighted device won\u0027t appear in the replica2part2dev table.\n\nThis patch extends `find_local_handoff_for_part`, if it fails to find a node\nfrom the ring it\u0027ll fall back to a local device identified by the\n`_local_device_ids` that is built up when the replicator or sharder was\nidentifing local devices. This uses the ring.devs, so does include 0 weighted\ndevices.\nThis allows the sharder to find a location to write the shard_broker in\na handoff location while sharding.\n\nChange-Id: Ic38698e9ca0397770c7362229baef1101a72788f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/592523654f4531d1c3fb69c884d9e7792066500d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/592523654f4531d1c3fb69c884d9e7792066500d"}]},"branch":"refs/heads/master"},"24398845e7e08867117bc93b79de3b9d10319b1e":{"kind":"REWORK","_number":7,"created":"2022-07-28 07:24:50.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/850597/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/850597/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/850597/7"}}},"commit":{"parents":[{"commit":"a303319f0c21cff5eef74a229d5131d3a6b0ef34","subject":"sharder: process deleted DBs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a303319f0c21cff5eef74a229d5131d3a6b0ef34"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-21 02:32:27.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-28 07:23:22.000000000","tz":600},"subject":"Sharder: Fall back to local device in get_shard_broker","message":"Sharder: Fall back to local device in get_shard_broker\n\nIf the sharder is processing a node that has 0 weight, especially for\nall the devices on the node, the `find_local_handoff_for_part` can fail\nbecause there will be no local hand off devices available as it uses the\nreplica2part2dev_id to find a device.\nHowever, a 0 weighted device won\u0027t appear in the replica2part2dev table.\n\nThis patch extends `find_local_handoff_for_part`, if it fails to find a node\nfrom the ring it\u0027ll fall back to a local device identified by the\n`_local_device_ids` that is built up when the replicator or sharder was\nidentifing local devices. This uses the ring.devs, so does include 0 weighted\ndevices.\nThis allows the sharder to find a location to write the shard_broker in\na handoff location while sharding.\n\nChange-Id: Ic38698e9ca0397770c7362229baef1101a72788f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24398845e7e08867117bc93b79de3b9d10319b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24398845e7e08867117bc93b79de3b9d10319b1e"}]},"branch":"refs/heads/master"},"ac013541bed82ddabbb3e97dfb74fc120523094c":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-07-28 08:53:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/97/850597/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/850597/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/850597/8"}}},"commit":{"parents":[{"commit":"38271142eb403a1cdf587616b6d26bfc6b7fb382","subject":"sharder: process deleted DBs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38271142eb403a1cdf587616b6d26bfc6b7fb382"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-21 02:32:27.000000000","tz":600},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-28 08:51:14.000000000","tz":60},"subject":"Sharder: Fall back to local device in get_shard_broker","message":"Sharder: Fall back to local device in get_shard_broker\n\nIf the sharder is processing a node that has 0 weight, especially for\nall the devices on the node, the `find_local_handoff_for_part` can fail\nbecause there will be no local hand off devices available as it uses the\nreplica2part2dev_id to find a device.\nHowever, a 0 weighted device won\u0027t appear in the replica2part2dev table.\n\nThis patch extends `find_local_handoff_for_part`, if it fails to find a node\nfrom the ring it\u0027ll fall back to a local device identified by the\n`_local_device_ids` that is built up when the replicator or sharder was\nidentifing local devices. This uses the ring.devs, so does include 0 weighted\ndevices.\nThis allows the sharder to find a location to write the shard_broker in\na handoff location while sharding.\n\nChange-Id: Ic38698e9ca0397770c7362229baef1101a72788f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac013541bed82ddabbb3e97dfb74fc120523094c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac013541bed82ddabbb3e97dfb74fc120523094c"}]},"branch":"refs/heads/master"},"c4e00eb89f34de79d5fb123dd044621ef4df679c":{"kind":"REWORK","_number":9,"created":"2022-07-29 14:02:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/97/850597/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/850597/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/850597/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/850597/9"}}},"commit":{"parents":[{"commit":"d7a931191b805f9f14f57d91532d88bba88cf362","subject":"Merge \"CI: Add nslookup_target to FIPS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7a931191b805f9f14f57d91532d88bba88cf362"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2022-07-21 02:32:27.000000000","tz":600},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-29 14:02:26.000000000","tz":60},"subject":"Sharder: Fall back to local device in get_shard_broker","message":"Sharder: Fall back to local device in get_shard_broker\n\nIf the sharder is processing a node that has 0 weight, especially for\nall the devices on the node, the `find_local_handoff_for_part` can\nfail because there will be no local hand off devices available as it\nuses the replica2part2dev_id to find a device.  However, a 0 weighted\ndevice won\u0027t appear in the replica2part2dev table.\n\nThis patch extends `find_local_handoff_for_part`, if it fails to find\na node from the ring it\u0027ll fall back to a local device identified by\nthe `_local_device_ids` that is built up when the replicator or\nsharder was identifing local devices. This uses the ring.devs, so does\ninclude 0 weighted devices.  This allows the sharder to find a\nlocation to write the shard_broker in a handoff location while\nsharding.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ic38698e9ca0397770c7362229baef1101a72788f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4e00eb89f34de79d5fb123dd044621ef4df679c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4e00eb89f34de79d5fb123dd044621ef4df679c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
