)]}'
{"id":"openstack%2Fswift~850787","triplet_id":"openstack%2Fswift~master~Ic0f573aefd920087c11fcb047bffc3c9c6ee0a11","project":"openstack/swift","branch":"master","topic":"bug/1733634","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-08-05 18:50:39.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2023-04-18 21:44:33.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-04-18 21:44:33.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ic0f573aefd920087c11fcb047bffc3c9c6ee0a11","subject":"info: Add support for sha256 and sha512; deprecate sha1","status":"NEW","created":"2022-07-23 04:41:37.000000000","updated":"2023-04-18 21:44:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":14,"unresolved_comment_count":8,"has_review_started":true,"meta_rev_id":"64d5dd20efde01f0d6de3e572b71751634c35b1c","_number":850787,"virtual_id_number":850787,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-01-07 01:17:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-23 05:18:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-23 05:42:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-04 07:38:04.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2023-04-17 23:06:14.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"d6104bf5e8d0de4b3b187e1d5c14d0af7effd68b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-23 04:41:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b97aaa618ff3b12e1ba5577573dc5827f160c760","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-23 05:00:03.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f3eafdbe9465ed1603badd7e10a1d32b0e72b0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 05:18:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24a0cdd6850741709882b3d81c28fbcb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/881b71288a5d4ef4b1216fda687604b1 : FAILURE in 29m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c50cb74d8c32400483699123ed03150e : FAILURE in 29m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/be926f8af7e843a486b37d3fd40e2eb9 : SUCCESS in 29m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/609f2cd56d76413fa593f5bd3c2f782b : SUCCESS in 28m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d9fa27e2401f349da1d45ebffdbaca2da3156940","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 05:42:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/607d26e7dad74d57999133d3d421f3ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f3b33b8057a4a3b8696e05835652b4c : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/c7a14a8d87eb4d22a6d3ba72c23de186 : SUCCESS in 49m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e259b0e746f140c092e72befa1911379 : SUCCESS in 49m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4e6bc6a0f3b4c839ca49093984a97b7 : SUCCESS in 55m 45s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/3e088fb056a54d2e82095e22150e523f : RETRY_LIMIT in 7m 25s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ef54ed640a9c4c45adc0b4394cbf358b : RETRY_LIMIT in 5m 29s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/068e5cfe10fc4ebcb122a691a150d7d4 : RETRY_LIMIT in 6m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cde1b55480da41a687042fa3511a347c : FAILURE in 3m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ddefe2b1e5d9411bb2c3da4fd53a1028 : FAILURE in 3m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b3dbb233e79d41f9b2bac649ada6dabe : FAILURE in 15m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/617a3d62e4da4ae6b3fbe6e6ed69c607 : FAILURE in 15m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/089967e147894136a3cb75e8d52aface : FAILURE in 14m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3933ece74dd34362a976e8556b80e478 : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/72e75f3ce0d948cdacf0737076abd216 : SUCCESS in 15m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3838554fb78f4dbba5f81ad970348051 : SUCCESS in 15m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/73c181c1f0ae4223b0a165dcbdf62029 : SUCCESS in 13m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/60ba4eab134e43d89712f9bcd42c4154 : SUCCESS in 13m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/90ed2680eb4645baaea663513553df50 : SUCCESS in 14m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/227fcff805b64628b047ec6051f0702e : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3b5be0f11b0443d7bb8b7a387162929e : SUCCESS in 14m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/19426382038f4e298c1cb0d1929616e4 : SUCCESS in 6m 25s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a84d1efc265d4494a6db4d222cc5fad4 : SUCCESS in 50m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/da157ae0635e4152bbf86343a49e1e11 : SUCCESS in 53m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2062ec765e7a4130a642b21905808852 : SUCCESS in 42m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1124a94384734615a7f8157993ba74f1 : SUCCESS in 42m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc590d7629714658bfceb06eede526ed : FAILURE in 15m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bff0ecb51cb14d4c8e81e0195c3f61ff : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f8ae1e6c7e3e4b78bc7ba5d5e812346b : SUCCESS in 19m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a8016f128514568852660fd2ac00dcc : SUCCESS in 44m 24s","accounts_in_message":[],"_revision_number":1},{"id":"6ddaa3af2827b8812614b6abaf562c29c051d975","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-30 00:33:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"114396d9fbafb954517f13bc1a4d4834f5babd59","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-30 00:34:09.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1952130499719c6c5cb7e20f0029d2e1ad9be8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 01:29:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/307e2b96329e4f28b18d5014b545b367\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9681576378204cabb35b5b99fa3117d2 : SUCCESS in 26m 37s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc7fc75252c24bc693f80ae3a59f25aa : SUCCESS in 54m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/830726366559403d8fa240bde0c29739 : SUCCESS in 26m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/899d4ed4684447468d042298c38680ea : SUCCESS in 44m 58s","accounts_in_message":[],"_revision_number":2},{"id":"c129ff72ad9f63d131f3fb84fea8dbcd275a69db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 01:47:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f46631151f943fa93995074f3352a84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f34314e9ebcf4e4b8ca50c21a419bfdf : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/235efefb08e444468e237ff7c22f5782 : SUCCESS in 1h 03m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f7300d50a8cc4fbbb9b374a8e5ad7cb2 : SUCCESS in 38m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7afafa90f7824ec68ecbd7f2055e0181 : SUCCESS in 54m 26s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c0ee5bc844414096bde1eab6013dab2b : SUCCESS in 24m 17s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2838549dcaa84af3848e86738716d3cf : SUCCESS in 16m 38s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/89684a76c65c417fb648110252770999 : SUCCESS in 24m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47be08e4ae474f06ba4f8b88c27b95b6 : FAILURE in 4m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a083fb3f10254d418ca036a3fd22e234 : FAILURE in 4m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ebc573c7ac3491eb346d95f2eb5c49e : SUCCESS in 12m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/82c1ea1a56654b93b5fe7e168a17139b : SUCCESS in 20m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/10f56b7df17f4f5cb1d212db5c1b7620 : SUCCESS in 11m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/be9dd84289cf4a63ba57563840be4124 : SUCCESS in 13m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bdf84ac80f684f0ab8f974b763058e7c : SUCCESS in 21m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/420ed7502e094603a165402d3abc7d78 : SUCCESS in 13m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/516c96c1949046fda38d6e367cc927e9 : SUCCESS in 12m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/79591930aaf54630a5961187ddb57bb7 : SUCCESS in 11m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ca7a9be8369d498db3302ded94109477 : SUCCESS in 18m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/efafd0b8306a403db9e9d3214ac63333 : SUCCESS in 8m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b3538741de9a437b9c7165f79527ece8 : SUCCESS in 14m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c8433aa0a8b9445284e79620e356440d : SUCCESS in 13m 44s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/95f4a91c7fb144588510667f4afdc424 : SUCCESS in 42m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/eda6661a71804738b0bac8f77afdd61c : SUCCESS in 1h 07m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c2754eb4f424e34bf0112709d3d75ec : SUCCESS in 50m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f70646ef2f6f4cc0970f4cc4b34b0d6d : SUCCESS in 46m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b2c314ba1084f2094c37f047e1ebf39 : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f97c8d41d1434d21988752dfb9a73690 : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/714ea09364aa4270ac581665bc207eaf : SUCCESS in 33m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5368ca9585d846c9b4e3ba12d71369e3 : SUCCESS in 43m 58s","accounts_in_message":[],"_revision_number":2},{"id":"4ecf85d7b260294f7542865550e115ed82efe66d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-08-04 07:38:04.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3b31b666f8ea696a0c0d09f2cb62dc1fdaeb49f6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-05 18:50:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e1712dfba8fdfece7ff5c293ee8911129353529","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-05 18:52:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"333c00b4a93d049040e897feb1acf65557ab0cfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 19:33:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f56c242f160843198e56490d36b9dff8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c701979c2fbb43fb9e3791ccf343e083 : SUCCESS in 32m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf63b3d55c2844a1af77f571e2797fe0 : SUCCESS in 39m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/39b4e477a4e5432c9bb81da4c9fe02e2 : SUCCESS in 32m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b700387aebc0417098939bcc9b397820 : SUCCESS in 32m 00s","accounts_in_message":[],"_revision_number":3},{"id":"00f4369033bc206922b91895cc8f6caeb4d58643","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 19:54:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/557611c0f939472ba0c8d011b121511a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c3fdc17daa84dd893390f445412ab0b : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/aa05f8aef5394eef88f349b5884b75f4 : SUCCESS in 49m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f8841191c91d45bc806daad5c9a92ddf : SUCCESS in 51m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9560217926ad4c129aa85b27c862d720 : SUCCESS in 42m 03s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f4e621fe5fbe44e9a2099dffa6232e4a : SUCCESS in 21m 48s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/dd1af540e2f947e7bd40d29b8d147fba : SUCCESS in 21m 35s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d26a4a5571ad424b85e369a02151ad35 : SUCCESS in 22m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/680e63757fa9480783915fc799276989 : FAILURE in 4m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bdee8c730b81467faaf543ff1b62215f : FAILURE in 4m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5ea93013658f4d2f8f891ead76795abf : SUCCESS in 17m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d95125f35f8248f1bb3eb7db496aa6c6 : SUCCESS in 16m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/51e1a1ea73294e92a5eb58d6dbbb4423 : SUCCESS in 18m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ba65a038c4b248978811f7a5585b16a4 : SUCCESS in 20m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b94f3393f3e843b7b064bc20e8e894c7 : SUCCESS in 19m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/965a643b53db401db9e4e57f23798adf : SUCCESS in 21m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1e96b3ffbeea4d6e993b0588198a6f61 : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0d0f5c07b70345368763e4c562f922a1 : SUCCESS in 19m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d9e22401ceac46abb77e5e5102983115 : SUCCESS in 19m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/890a9912fffe4d1f820c2633407b4f4f : SUCCESS in 12m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a50112462663423fb852d2d5721a947a : SUCCESS in 20m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/611ce6fa2bf0496e859d66d58c703318 : SUCCESS in 12m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c38398679c9b4a77bcd368e955c6cc24 : SUCCESS in 53m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8cb476ef591f43e2b961ca8f230b1d94 : SUCCESS in 59m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5a2b4c6b326146ae8a275dcff7d7a14a : SUCCESS in 41m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1c85a3eebc8d45389e9a3fd96bc3784e : SUCCESS in 46m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6717bd09d7c94075a69f08b360271105 : SUCCESS in 17m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21a36ad47a2445219ff6da6e0799c488 : SUCCESS in 6m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2b48c3f4a0e247ae8cad223d03969e5b : SUCCESS in 24m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebf3549a6e184e10a100a653bf9b9ef2 : SUCCESS in 53m 53s","accounts_in_message":[],"_revision_number":3},{"id":"ba3cc0a08bb3e51a3a1783630baaffb4daea4db2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-10 20:53:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fa0965237b46300e1c9b0f025954ac4069a9b591","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 21:55:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/066055eb40964ec29d7ddbda154234e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe340ad505114e09bafc163ed8ed203d : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/991fbfa6cc5f43f5922d17f8db1cb78a : SUCCESS in 56m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5f96b04149ad4eb88cc2d3e315f33f38 : SUCCESS in 52m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a73320601d404ce2b9e58d1e3154e9f1 : SUCCESS in 51m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a0a83bed24b4f7ea2ed2a3de61cb45e : FAILURE in 18m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/891584b2c82340afbc2e9bb40f31be13 : FAILURE in 17m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d80a922e4fb497391aeac8ec6ba6ef2 : FAILURE in 17m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/78ea69b2e9084532800a93b1d16403f9 : SUCCESS in 9m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/36a4d9dcdc624ab2869042685d74774e : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/276084a9b3ab4085903683f16353a0c8 : SUCCESS in 16m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/efec4a4dea5441e8b8a93dbc56db79aa : SUCCESS in 13m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/83b15ab0a21749cbbff59a7b9f7229c3 : SUCCESS in 13m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/43f85fe78c384a948633c07733842b68 : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7d3cdf578d834d2d9ba4ae496e6d873b : SUCCESS in 16m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f861431efad3471a9894248aeb7f7fdc : SUCCESS in 15m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ff436aef87d944f68b0989f7bd3c189c : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/551c916a71ab4cc28618ac490919536c : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e7f2ca9c767e4cdda2d1216fed84c7ae : SUCCESS in 14m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cf062ef9885a4a57bb090df017c84873 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b4a58454707944a4bc62ce9e4fab4a51 : SUCCESS in 15m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3b84c2b9539e42a795c3971f79cf3e65 : SUCCESS in 6m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/45ed0ed4065b40ecb3eeedfae30c09a6 : SUCCESS in 54m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/fad85309e3e443809dc368dda6ced2b5 : SUCCESS in 57m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c2ef33c62887408f85d478f14a9d9234 : SUCCESS in 41m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/890281f884c54f918e554e4c96bc6964 : SUCCESS in 44m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d390b19aec4b45019bec72a5960c381e : SUCCESS in 15m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18b42fd1b3f740ec95d426a182a9a324 : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/16a3213a2e84439b91fe537aa872630e : FAILURE in 25m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80ae8b297b3f4aceb8b53c02efbe3052 : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":4},{"id":"ee874b8d3c65497d10a967c880890b66d17dfb70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 21:55:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6f720b7990c4032910fbd9fe9e70416\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/acf48be8a99548aba7b4828fad8a0ce0 : SUCCESS in 29m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0a02fe704384572bb7305eeb2185466 : SUCCESS in 29m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/336372fcda95458eaff94a88eeed1f68 : SUCCESS in 29m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0c8454e369643efa153faaf3506ea3a : SUCCESS in 1h 00m 46s","accounts_in_message":[],"_revision_number":4},{"id":"6ed836ff56b78f27a833f39c81db97e464bd8d00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 23:49:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f4c78f3c61c6b25b9ed8bf26131a00e62452b985","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 01:13:15.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8e1faa62f88496fa7aada8777cf4f15\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a0c7fcc425042398a968e2942f30c53 : SUCCESS in 49m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4da93a42ba994849945de7ba632d5bfd : SUCCESS in 47m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1bb7054013a4a939529e129ccfff0b9 : SUCCESS in 1h 00m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b7055fd70bd42a1b336630fd08c9eb9 : FAILURE in 1h 22m 15s","accounts_in_message":[],"_revision_number":5},{"id":"f6468f1ecf35ad3b4495f0fb77ddd6b3bcae2195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 01:17:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c246eca2a86a41598a792761c94d0d21\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9f77b8c133048398372c4237bc13811 : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/cb894784d4214e96bc8d50c14447cb2e : SUCCESS in 58m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a3d5354206fa4807b663c64670a3f4de : SUCCESS in 57m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6b872c9018d4a1e828b7d13fc941d8e : SUCCESS in 53m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63c3e20dfcc8484e97c775fc8d974c85 : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45733fa51e73494fa419e05c8acab20a : SUCCESS in 17m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c30110e76c64d348994eb2732c3b870 : SUCCESS in 16m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b28052a5e61d4159b76ddcfc259862c0 : SUCCESS in 8m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3d73f590efe64aa8bf8651effb7265b7 : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fa9c07072c3b47ed812013f17cec757f : SUCCESS in 13m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/671f2845203f45cb8652375f3a6bb957 : SUCCESS in 16m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cef5013f902d45b5940a1bfad906de05 : SUCCESS in 14m 44s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b0248adc2663438b890949b30998a08a : SUCCESS in 12m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ea3238df5edc4779b24ea829671a76c7 : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/643a573db426462ebaddec3c4703a2f6 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d5b06162656d4236860a289d45ef7915 : SUCCESS in 15m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5e1cf799b137464caafb1fd1412dbc52 : SUCCESS in 11m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fee1f01a56a94d10b055f1437d9928ca : SUCCESS in 15m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b92175664e3a4646bd22e7bb51ca7288 : SUCCESS in 16m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e7eeab6bc0b41aba6973bd6b0011274 : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9182aef0831e4a3dae0f7730bdf026a3 : SUCCESS in 13m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/be7f85460db44ad7a402973e5f5e0b8b : SUCCESS in 7m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/44a05c2ec1e0415993c6e21b87ea2b43 : SUCCESS in 1h 23m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/22e621d1e504410e885b84d56f019a14 : SUCCESS in 1h 09m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/516821e22fd24d2b914acdb4aba15a33 : SUCCESS in 36m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3290d77c07984b7b80d3bb2e63a6dace : SUCCESS in 40m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/641aa09b5c64483db01ed3f4a71c0a24 : SUCCESS in 23m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cb4393ecca9486d8a4d2ca3e6a5b033 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f7c172473a5745c090d9212cb9799110 : FAILURE in 27m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c1b30a7507f44b0b3794981061572dc : SUCCESS in 50m 46s","accounts_in_message":[],"_revision_number":5},{"id":"5c7a5642509c305603f49347780f9846590611bc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-17 23:06:14.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"64d5dd20efde01f0d6de3e572b71751634c35b1c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-18 21:44:33.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9d78dbad9cf746f8a90ee1f7222668d52a19d826","revisions":{"b2b4d35b59889b148e02e076bd48989919a435fe":{"kind":"REWORK","_number":1,"created":"2022-07-23 04:41:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/87/850787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/850787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/850787/1"}}},"commit":{"parents":[{"commit":"2c8fe050b5b17ed57b46e578d2989db0429e9980","subject":"formpost: deprecate sha1 signatures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c8fe050b5b17ed57b46e578d2989db0429e9980"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-23 04:40:21.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-23 04:40:21.000000000","tz":-420},"subject":"info: Add support for sha256 and sha512; deprecate sha1","message":"info: Add support for sha256 and sha512; deprecate sha1\n\nThis brings it in line with what tempurl and formpost are doing.\n\nChange-Id: Ic0f573aefd920087c11fcb047bffc3c9c6ee0a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2b4d35b59889b148e02e076bd48989919a435fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2b4d35b59889b148e02e076bd48989919a435fe"}]},"branch":"refs/heads/master"},"663389a60be8af152127a69ef2279ad23063b775":{"kind":"REWORK","_number":2,"created":"2022-07-30 00:33:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/87/850787/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/850787/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/850787/2"}}},"commit":{"parents":[{"commit":"c4e00eb89f34de79d5fb123dd044621ef4df679c","subject":"Sharder: Fall back to local device in get_shard_broker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4e00eb89f34de79d5fb123dd044621ef4df679c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-23 04:40:21.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-30 00:33:36.000000000","tz":-420},"subject":"info: Add support for sha256 and sha512; deprecate sha1","message":"info: Add support for sha256 and sha512; deprecate sha1\n\nThis brings it in line with what tempurl and formpost are doing.\n\nNote that we only consider which digests are allowed if an admin_key is\nset. If it is not, no deprecation warnings will be emitted and no\ninformation about allowed or deprecated digests will be published in\n/info.\n\nChange-Id: Ic0f573aefd920087c11fcb047bffc3c9c6ee0a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/663389a60be8af152127a69ef2279ad23063b775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/663389a60be8af152127a69ef2279ad23063b775"}]},"branch":"refs/heads/master"},"46dd3bc62760fc5a307c80e04581ebacc7e960b9":{"kind":"REWORK","_number":3,"created":"2022-08-05 18:52:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/87/850787/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/850787/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/850787/3"}}},"commit":{"parents":[{"commit":"c4e00eb89f34de79d5fb123dd044621ef4df679c","subject":"Sharder: Fall back to local device in get_shard_broker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4e00eb89f34de79d5fb123dd044621ef4df679c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-23 04:40:21.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-05 18:51:54.000000000","tz":-420},"subject":"info: Add support for sha256 and sha512; deprecate sha1","message":"info: Add support for sha256 and sha512; deprecate sha1\n\nThis brings it in line with what tempurl and formpost are doing.\n\nNote that we only consider which digests are allowed if an admin_key is\nset. If it is not, no deprecation warnings will be emitted and no\ninformation about allowed or deprecated digests will be published in\n/info.\n\nChange-Id: Ic0f573aefd920087c11fcb047bffc3c9c6ee0a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46dd3bc62760fc5a307c80e04581ebacc7e960b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46dd3bc62760fc5a307c80e04581ebacc7e960b9"}]},"branch":"refs/heads/master"},"ff417e2c25ea48926f7c2a6ba8204e50fcd697f2":{"kind":"REWORK","_number":4,"created":"2022-11-10 20:53:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/87/850787/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/850787/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/850787/4"}}},"commit":{"parents":[{"commit":"03d8f823ef0b24d40376e6f8db995eece9215e47","subject":"Merge \"Mark rolling-upgrade job non-voting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03d8f823ef0b24d40376e6f8db995eece9215e47"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-23 04:40:21.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-10 20:52:21.000000000","tz":-480},"subject":"info: Add support for sha256 and sha512; deprecate sha1","message":"info: Add support for sha256 and sha512; deprecate sha1\n\nThis brings it in line with what tempurl and formpost are doing.\n\nNote that we only consider which digests are allowed if an admin_key is\nset. If it is not, no deprecation warnings will be emitted and no\ninformation about allowed or deprecated digests will be published in\n/info.\n\nChange-Id: Ic0f573aefd920087c11fcb047bffc3c9c6ee0a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff417e2c25ea48926f7c2a6ba8204e50fcd697f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff417e2c25ea48926f7c2a6ba8204e50fcd697f2"}]},"branch":"refs/heads/master"},"9d78dbad9cf746f8a90ee1f7222668d52a19d826":{"kind":"REWORK","_number":5,"created":"2023-01-06 23:49:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/87/850787/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/850787/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/850787/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/850787/5"}}},"commit":{"parents":[{"commit":"b4124e0cd25b7f50b7d6be284fe2b936d7191acc","subject":"Memcached: add timing stats to set/get and other public functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4124e0cd25b7f50b7d6be284fe2b936d7191acc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-23 04:40:21.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-06 23:48:41.000000000","tz":-480},"subject":"info: Add support for sha256 and sha512; deprecate sha1","message":"info: Add support for sha256 and sha512; deprecate sha1\n\nThis brings it in line with what tempurl and formpost are doing.\n\nNote that we only consider which digests are allowed if an admin_key is\nset. If it is not, no deprecation warnings will be emitted and no\ninformation about allowed or deprecated digests will be published in\n/info.\n\nChange-Id: Ic0f573aefd920087c11fcb047bffc3c9c6ee0a11\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d78dbad9cf746f8a90ee1f7222668d52a19d826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d78dbad9cf746f8a90ee1f7222668d52a19d826"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
