)]}'
{"id":"openstack%2Fswift~850905","triplet_id":"openstack%2Fswift~master~Ia43e91a545a822cace41a0e814ab6c3bd89f8402","project":"openstack/swift","branch":"master","topic":"p-sharder-db-stats","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-08-01 08:46:49.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-08-01 08:46:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia43e91a545a822cace41a0e814ab6c3bd89f8402","subject":"sharder/replicator: emit stats for DBs created or existing","status":"MERGED","created":"2022-07-25 15:56:46.000000000","updated":"2022-08-01 08:48:07.000000000","submitted":"2022-08-01 08:46:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"850905-p-sharder-db-stats","meta_rev_id":"a63ea391391ce6fb3c4fe89c22d019cb235923cb","_number":850905,"virtual_id_number":850905,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-01 08:46:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2022-08-01 07:36:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2022-08-01 07:36:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-25 16:56:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-25 20:49:43.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2022-08-01 07:36:06.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"8fcfd31e89243d8bd3ca7a6579ae1f17402e2d74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-25 15:56:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c31e5057c86ad431c2b0dea51d45a0cea9186c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 16:56:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/452b514dc3e14fd583cc4e183527e5d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f7db292884044baa490a1c96940201e : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/a506689fd0bc42f2905cf308175ce07d : SUCCESS in 53m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/67a1e927d0e3444a8ad750fcecc27702 : SUCCESS in 50m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9fa9cf6d2314ba4ac7f40fd52721a9a : SUCCESS in 45m 28s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4aec6db22806494a87d41e0e789a8ba8 : RETRY_LIMIT in 6m 01s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1b5bb34a87df4c55bc3a831b53ea4e46 : RETRY_LIMIT in 5m 36s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4a1194f6f90643f7907869f14bdb9c8c : RETRY_LIMIT in 5m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4af20305de25480ab602f10c66cb6380 : FAILURE in 4m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/62beded2155740758ae6bb223073caea : FAILURE in 5m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bfbfdad7f84948859686f245577ebe62 : SUCCESS in 13m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f8b4a9d9f1194e9884c49fd68dffb2e9 : SUCCESS in 11m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dbf099181396437f90ce89ad30f92840 : SUCCESS in 17m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fbacfdfdfc824957bdd9318152fbaf2b : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5b5c0bda52e24a3b87298556eda03128 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b064cddf173840ba9c9c9781dbb36ff0 : SUCCESS in 21m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/43bb6183808247228c41e01335402b0d : SUCCESS in 13m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/98e97a76f25f4920b771f4e2d87aaac5 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/927af16a60f94e0388c9444d65f12cdf : SUCCESS in 14m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8e15d15712144c54a4961c18c6df87e6 : SUCCESS in 9m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e8cf919d3614c66a06b20a90cd3ebc9 : SUCCESS in 17m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5c92135fd0a74edb8a0d31dbc7a43cdd : SUCCESS in 8m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/587b40710ef54b93a2212afec730d0a7 : SUCCESS in 56m 03s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2d340f3e770e48e6ae4ebe975b2fb5dc : SUCCESS in 55m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a75c4c5c03b14b64a708a591f84f5422 : SUCCESS in 54m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/54f60b84d1c34128a5a3f933f0f05642 : SUCCESS in 47m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfd824877a3c438090e6e125a5511c34 : SUCCESS in 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c20d204ea524428f944789d838a2d5de : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/07a16e73150c4aab82da11865feb0b08 : SUCCESS in 31m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddc7250023b341dc87aebd0471138f47 : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":1},{"id":"93e0dbeb51d641e040843967fb96a9d7c3d2837b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 17:38:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e818895d03684b57b1465eb6c669c959\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f706910a8a9347858688ac39b7e6024a : TIMED_OUT in 1h 30m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea12fa03b8934d4eba0d33fd33a3b052 : TIMED_OUT in 1h 30m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d46824979dfd4b52a67c5efb9182a6de : SUCCESS in 59m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f65a5e4e52c045cfaddf348871c5ed8f : TIMED_OUT in 1h 30m 50s","accounts_in_message":[],"_revision_number":1},{"id":"6fe07808f27a8f4001ee8fa5a68d2a1314057d6c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-25 20:49:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7fe62ae8855acd1baa4fa9e6e70dff2b25e614e3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-27 10:03:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab3c39a34259ce6c2ba372737dc8cb88ae1416d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-27 10:04:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ce45a15c493250dad5eb8e94dbbb4c2e1d161494","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 11:11:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ff0ce6a129a4e1e940a0778e52b16f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c865ef8f8b664b2691263a4ec95f7e94 : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/2b4ba923505d49958a9cdd0cb129bbc5 : SUCCESS in 1h 00m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/25f7411821e24006a7be01f84428c90b : SUCCESS in 51m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d467a0e8639a426295f9f9d694ee4a39 : SUCCESS in 49m 50s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4c5ff99a81864bc798e32d6cd4d3831c : RETRY_LIMIT in 4m 58s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b677e7f5f61a4bb7aff1f8fbb3bcc75d : RETRY_LIMIT in 4m 05s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0913e694b60c4492bfaaed221453e122 : RETRY_LIMIT in 4m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f95cd9f80ac74fe694fa0262dd6f1110 : FAILURE in 5m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bfb6c79ad8ad48278d4bf40aac3a3528 : FAILURE in 4m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38ed7120807c44b38df3117552a32616 : SUCCESS in 12m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/90662ae855e44763927ab21ccc48ced6 : SUCCESS in 18m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/24a112e220984a1fbf27861e69ceff61 : SUCCESS in 12m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7afa29bc3bb3413697c84104874bf938 : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9cab959055504000bb46bcabf067d46b : SUCCESS in 22m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8045bae55b0b4c818826d923d3b6193d : SUCCESS in 22m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f9e7243497d94bf69558f702a467c943 : SUCCESS in 19m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c3e491211f67481a8a8475cde88205dd : SUCCESS in 19m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7d84e575063345378a64d08746acfb28 : SUCCESS in 18m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e61297102a264739bb837d6641a76735 : SUCCESS in 9m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a64a6b0ea370439dad65bf1f4643ddb4 : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b8c3c07c520e4e078b7233eab3e8657c : SUCCESS in 9m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/59f69e468f874354b1a30e102cec6625 : SUCCESS in 57m 29s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5485fdb503c74fc1aeac29a2b1c93efc : SUCCESS in 1h 01m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5752fcc6be3341319dbb5b73637e0f06 : SUCCESS in 44m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3f9d9b81e06148198051917e37167f1f : SUCCESS in 48m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/399b02283c6c4367bbfb8573ad9a268d : SUCCESS in 18m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/500d8ffc19664dfe95e3b3cf3de91090 : SUCCESS in 6m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e42ac730fb8d45ac8b88ccbed63d8ae3 : SUCCESS in 30m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed400334f3714cd09b77dca62380b412 : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":2},{"id":"55d4e16fd9f3f573fd3766dd13b7984e06e5f675","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 11:23:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b0d53485d52475b9a3da657f9ee8357\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74bda56ac1cf40a78ee20160ff0e53b6 : SUCCESS in 1h 04m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5262e226c3174e68b05f00c4c11e5932 : SUCCESS in 53m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/954d4f64bbd84384837147b2ce30ab3a : SUCCESS in 1h 06m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8413d0484f504cd7b0947be7084295d0 : SUCCESS in 1h 05m 01s","accounts_in_message":[],"_revision_number":2},{"id":"4fb0035a0050db4d7406f9ff0c8e3b6c0e1ec1af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-27 13:03:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"080bd10151fbb9a39bd42fcf3cdf3a8137ca6e2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 14:16:22.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf7ca76886f2419aacabaf0a05bbb96c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f17c48a9ade7448a803d74b340bc431a : FAILURE in 1h 10m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b6d1245c31cc4bb1a1b3a2dab3a09fe2 : FAILURE in 58m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3764fca50de4ced9d12d0b92cd81c06 : SUCCESS in 1h 02m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e130b7e87cd4e54aaf2db9297419406 : SUCCESS in 1h 00m 44s","accounts_in_message":[],"_revision_number":3},{"id":"dcbbeca4fd58994219621f2043ee780fd1ed47c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 14:22:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f8e7bdf27a84ef19545bbe482c4c3fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef2fb7743f2d4cd999e5ea7f322b516c : SUCCESS in 12m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/0526dfc878d646a196cef1a88a4785c8 : SUCCESS in 1h 07m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0367a2ccbee74129a635a4be7542bc20 : SUCCESS in 57m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f13bf3a534bb44b08f191eea16e27bbf : SUCCESS in 1h 01m 36s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6ff7d26d52f34dafb6e327a4a72990a6 : RETRY_LIMIT in 4m 48s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c4cbdf5fc8184efb8138a93c1e74708d : RETRY_LIMIT in 5m 01s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1f11e206cc3d4adab3a765c96879f1a7 : RETRY_LIMIT in 5m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aa31ee6555b34b658911b97d887f2dd1 : FAILURE in 7m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bbc14924cf0d4603b912e0cb717ff714 : FAILURE in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/79a6ce8aa7eb4a1aaa25ca590d207265 : FAILURE in 16m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e03ff7d8533e46739887a2166eeb370d : FAILURE in 18m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c00c284f1ff84c92940a532256a63ebf : FAILURE in 16m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/954d7e02744645299446e73f19f1e565 : SUCCESS in 19m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c1c7c09898354daabcd050d3ccc33796 : SUCCESS in 19m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8e69c630164f46eabc5606de2e134ea7 : SUCCESS in 16m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/16bf2799137a4d7eaaede9013f4b9958 : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d5ecf4d4b8864c20a5d3ed3984d7d144 : SUCCESS in 17m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c2c50c380a154829ae58e7818907b44e : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5f09d514faab4fb9b38c96c681fce810 : SUCCESS in 12m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/466b603f4b654e76b36bb3f8b95d4212 : SUCCESS in 16m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6bcd134e77fe41e09f27abb003c1a7f0 : SUCCESS in 14m 07s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e1a58d52d7a046e48f6706addfcb3152 : SUCCESS in 56m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/af8ff551fe27462c8afec898aacf6d84 : SUCCESS in 54m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d71ca49bb25c4af0a3360817157fa7b3 : SUCCESS in 51m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c2b75ecb597d4e86a1185090b05705df : SUCCESS in 48m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f39a673415764c3bb79b22c8292ed10a : FAILURE in 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e5e7461b84b4012ac8b2514a8aee6e8 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2302ff3fa0d645498b024cae52faaa8a : SUCCESS in 26m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81ce9407c3a7468a8b63382f75775b4f : SUCCESS in 56m 11s","accounts_in_message":[],"_revision_number":3},{"id":"cd80cc1983c7c1ff3ffb80dfad5f9b180cc87871","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-27 16:00:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"52773448b23ccfe29461f965a48bebfcab63210d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 17:01:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0167d51c8c04b909068362f689ea611\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7925040b34444b66998a47eaeda6d26f : SUCCESS in 43m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e786888143a14cd582c9d021f01d0193 : SUCCESS in 33m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1344155e067143d8b2a0a5fd20cced2e : SUCCESS in 35m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/132f0a47ccdc4ee2b8615b799e025ede : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":4},{"id":"27c15cb02ab10036c8c8293d7cdd1223417459da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 17:23:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a48253db30a54778ac1d9921f3039e15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbbe7c8969b84a3799c5a2933363b964 : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/4a7e13854d7441aca47508882179185d : SUCCESS in 1h 03m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4c4dc761a83943d0af25e84a75370c1f : SUCCESS in 56m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73384feaf7d8448a9cb8d4e3bff96de2 : SUCCESS in 1h 06m 38s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/897b5ce3fe154a70ab3a8263fe2cb236 : RETRY_LIMIT in 5m 33s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f2ec468c55b044aeabc23a62e1706d3a : RETRY_LIMIT in 5m 20s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8ba154be821f4d01840fa72801febf63 : RETRY_LIMIT in 4m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/611bae643fef4c7485a15f58f89b0e0c : FAILURE in 5m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2c4c1d41c31f4e998c63a1904b2369c7 : FAILURE in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/665cc76abf1b4b62a587bd4bd15327de : SUCCESS in 43m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9eef2260daa04e06b8c1352c075ea144 : SUCCESS in 42m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/64a8493c5c5b4549bf95c418ea9b76ce : SUCCESS in 42m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c40398253bfd46e08e774fe1ec9651e6 : SUCCESS in 44m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/17bdc396b0534c5faae8e89399c7e9d1 : SUCCESS in 44m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/65a6cf08592b4d738eb662777eafcfb3 : SUCCESS in 17m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8fdbc991daca4d81814d0764685e9b69 : SUCCESS in 18m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/51efed547b534e12ac43c3343897d73d : SUCCESS in 45m 07s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7776a6f3e40c4754bdf71674f6748638 : SUCCESS in 50m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e4d307cc98d946cda03fd2ad34997636 : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8a18687e24ce45d3b49945e76bb1df65 : SUCCESS in 44m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ac39efff54574cb9a2c14260a0569d8f : SUCCESS in 8m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/59574328720148c6b1284b6adddaf4c7 : SUCCESS in 1h 20m 04s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/29dc9305561c4b15a6ea141b89c0648c : SUCCESS in 55m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d942f89d9104c08985d64eeaf0d1943 : SUCCESS in 1h 06m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5657026441214213b392412abac33f9b : SUCCESS in 44m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1378cc8e0a714419b36864d10a51c34d : SUCCESS in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c96af9b88904217baaef5f14bbc3d04 : SUCCESS in 7m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8f184419571a452db2425331320ef240 : SUCCESS in 21m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3385993634e4060938778911c27777a : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":4},{"id":"aa30d306e1b2b9246ba365c31753cbf1c2d46dcb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-29 22:33:50.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f9c2739d22f843397f2d8931f0eb2c22f3bee8f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-30 16:30:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0c3148c4e2859c8fc97dc240c66d0bc8fbf57a0d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-30 16:35:03.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2beac8d3bffb7fdc93dd4ba506e9a7b756a4f6c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 17:28:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/061b30a3f40c4f1cac73fc1f6f1b2187\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c31ad55e98e44bb2a52cf58bde051e4c : SUCCESS in 55m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f0fed2e43b742d9b7fb9ab20d1f6781 : SUCCESS in 29m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ed7e42b9d9e4556b9a6dca433bea760 : SUCCESS in 49m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/14f750901fb14f6b8b4be1349feb633c : SUCCESS in 28m 54s","accounts_in_message":[],"_revision_number":5},{"id":"2309bd01ebad340f0879663a86f91a5530451b60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 17:51:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1751a05a8b20443e936a17b28e74e758\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9a4240f1b724240a09092b8d39addd4 : SUCCESS in 7m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/e7905736ba754caca0b03ffec4c1f679 : SUCCESS in 1h 14m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/929d61d0b2f3458d83829e437c434585 : SUCCESS in 52m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54b42a4f73d24f4e9322b1de986afd28 : SUCCESS in 50m 33s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/43b03e6f730040e884e457e70954f2c8 : SUCCESS in 21m 08s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c54ac93a05a3405a908ac61f36698219 : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2a022d1dbdba41c98c75c15146929618 : SUCCESS in 20m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f264405fdf6d43998f99e4eb1924b404 : FAILURE in 3m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8d7958db222d4d4eb19412994a620bd8 : FAILURE in 3m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f6c640f306c2406f82da6c64c6dd5229 : SUCCESS in 16m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e5615897a566435c9b2cea83a62ffda4 : SUCCESS in 14m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/abfc8dbc31cd4310919f190a9821dd47 : SUCCESS in 14m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ee52ce5363494dc4a395425cd3a475d0 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d382782e61da4e88bfffaaa6274273a6 : SUCCESS in 16m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5ccb3ae73f30472f8135ee8a84073e59 : SUCCESS in 16m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a8d5af19f65c440dabf010004fdd6383 : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/39f85f2048b4465386b8191114e3a811 : SUCCESS in 19m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/24ad6787d47f4feb8db3ff93e3345dea : SUCCESS in 14m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d8cb71d3bdb5484cbbfc9f033578d2ce : SUCCESS in 8m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/581abcd76f9a4100a0e6a0db3235fde8 : SUCCESS in 16m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5a086ccb7d34460fb6d7660dd0ccbbe7 : SUCCESS in 6m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1c32981338ef4f4a9cef41732b8bd13a : SUCCESS in 52m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/bca6a927c4aa4cf6872f364f10e6ae0f : SUCCESS in 56m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b6fb37166eae4a9ba60603df70b52150 : SUCCESS in 40m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4b6bc6e3fc4f48c999337f27f6e8b4d0 : SUCCESS in 42m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54273b3b4244447f907f2f3bd1142b96 : SUCCESS in 18m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1264447bffec4c5eb90c21f8b60b190b : SUCCESS in 5m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ed68ec75b997477ea6e50ca94b9776f0 : SUCCESS in 25m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42482ca270664c3ca0bced23ad76fb1d : SUCCESS in 52m 55s","accounts_in_message":[],"_revision_number":5},{"id":"8815c2c798310fc926730846af201935efb4e958","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-08-01 07:36:06.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"af837519adc4de779a5c7129ceb27ec322043fba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 07:36:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c5dcc992e5b1788718a467712d935ca956d8c80b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 08:46:49.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67ea469a5be5449ba7c8a45f0e49220d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ad573230a649fc8ef2641d361c5c5d : SUCCESS in 10m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/f1019eb711fd4e7181e0fe421998150e : SUCCESS in 1h 07m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/78f68fc2b8a14d10a9660e9d3051cc5a : SUCCESS in 56m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d32ec1577074009999cb0a311e68819 : SUCCESS in 1h 00m 27s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2402ff79304b4d6fa870ba991092acb8 : POST_FAILURE in 6m 20s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/9028363a6e6b4c58ba6901ecfbe253b4 : POST_FAILURE in 7m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eafe38a79efc469f8c5b59f7e586202f : SUCCESS in 16m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/41b778c779cd47248dbd8e036c5481d9 : SUCCESS in 14m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f18d079814c6481aa86f7847e8e7d0f0 : SUCCESS in 15m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8b59f4aa4b274b8188c2ba11bdc517ff : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f30dcc95d5274f7f87a1201643029198 : SUCCESS in 15m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/28fd752e7da342b2862d888888a4f934 : SUCCESS in 20m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/33cbc34bb96b4a7fb528b69967225aff : SUCCESS in 18m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/031bc4ca4ac54bda9fbdfa795412855c : SUCCESS in 13m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c96822af052644ce8bfddf789f34a3b8 : SUCCESS in 20m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/78e56831aa9943a399d8d5f5b143a308 : SUCCESS in 10m 30s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/23ad04031e50439c9a90fb01b2c43dfd : SUCCESS in 10m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0ff6b78f61114b998ec36f95a15938e7 : SUCCESS in 45m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a85c1cc11f3248f1a2437466b05a2200 : SUCCESS in 1h 03m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/61094c963efc47a4be4267549ad69943 : SUCCESS in 48m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a830c6823a57489b8d87d7538c133fe0 : SUCCESS in 44m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fffeb1ae24a8451ba4eabf93d500fa60 : SUCCESS in 15m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d94786f7be214757b432d03b56dbf374 : SUCCESS in 6m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fbd7208a11ba4e0ea2dcbac44f94d9d3 : SUCCESS in 29m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca5b884975844283986644d388a218ee : SUCCESS in 53m 31s","accounts_in_message":[],"_revision_number":5},{"id":"43c7a73cc74f7242432a40dd900787d0425bf5cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 08:46:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"a63ea391391ce6fb3c4fe89c22d019cb235923cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 08:48:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caa8eab9f7754b65a43ab17f4facc363\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c93927b37bb42a0b131065a11894f60 : SUCCESS in 52s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/289a8759795d47a3827b26cb01079ba9 : FAILURE in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/973c143593ce49989b332d8e8d230442 : FAILURE in 40s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"517738ac9a6efe97d227d979c47d5222fed47aae","revisions":{"8c32875989a69eeae124e260182873869ca1a599":{"kind":"REWORK","_number":1,"created":"2022-07-25 15:56:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/850905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/850905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/850905/1"}}},"commit":{"parents":[{"commit":"a4a8103c8c8ad33ed8131f000149d1f09d31c4c6","subject":"sharder: emit misplaced stats to statsd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4a8103c8c8ad33ed8131f000149d1f09d31c4c6"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-25 14:09:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-25 15:21:11.000000000","tz":60},"subject":"sharder: emit stats for shard DBs created or existing","message":"sharder: emit stats for shard DBs created or existing\n\nThe number of container partitions that are in handoff locations can\nbe used to track progress of a container ring rebalance. However, the\nsharder also creates shard container DBs that are often in handoff\nlocations, which distorts insights into rebalance progress. Emitting\nstats for sharder-created DBs will help quantify any such distortion.\n\nNew stats are:\n    cleaved_db_created\n    cleaved_db_exists\n    misplaced_db_created\n    misplaced_db_exists\n\nChange-Id: Ia43e91a545a822cace41a0e814ab6c3bd89f8402\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c32875989a69eeae124e260182873869ca1a599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8c32875989a69eeae124e260182873869ca1a599"}]},"branch":"refs/heads/master"},"b836635ccb8bd1f4a8cbdba9430a2df6b7df5dc4":{"kind":"REWORK","_number":2,"created":"2022-07-27 10:04:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/850905/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/850905/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/850905/2"}}},"commit":{"parents":[{"commit":"21fab529c9f0d1681b88aa6b847696717dd34065","subject":"sharder: emit misplaced stats to statsd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21fab529c9f0d1681b88aa6b847696717dd34065"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-25 14:09:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-27 10:00:31.000000000","tz":60},"subject":"sharder: emit stats for shard DBs created or existing","message":"sharder: emit stats for shard DBs created or existing\n\nThe number of container partitions that are in handoff locations can\nbe used to track progress of a container ring rebalance. However, the\nsharder also creates shard container DBs that are often in handoff\nlocations, which distorts insights into rebalance progress. Emitting\nstats for sharder-created DBs will help quantify any such distortion.\n\nNew stats are:\n    cleaved_db_created\n    cleaved_db_exists\n    misplaced_db_created\n    misplaced_db_exists\n\nChange-Id: Ia43e91a545a822cace41a0e814ab6c3bd89f8402\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b836635ccb8bd1f4a8cbdba9430a2df6b7df5dc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b836635ccb8bd1f4a8cbdba9430a2df6b7df5dc4"}]},"branch":"refs/heads/master"},"448427388778a9622a46cf8288ab652f31be3257":{"kind":"REWORK","_number":3,"created":"2022-07-27 13:03:38.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/850905/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/850905/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/850905/3"}}},"commit":{"parents":[{"commit":"21fab529c9f0d1681b88aa6b847696717dd34065","subject":"sharder: emit misplaced stats to statsd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21fab529c9f0d1681b88aa6b847696717dd34065"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-25 14:09:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-27 13:03:25.000000000","tz":60},"subject":"sharder/replicator: emit stats for DBs created or existing","message":"sharder/replicator: emit stats for DBs created or existing\n\nThe number of container partitions that are in handoff locations can\nbe used to track progress of a container ring rebalance. However,\ncontainer DBs are sometimes deliberately created in handoff locations:\nthe sharder creates shard container DBs while cleaving and also while\nmoving misplaced objects; the container-replicator creates DBs to feed\nmisplaced objects to the reconciler. These DBs can distorts insights\ninto rebalance progress. Emitting stats will help quantify any such\ndistortion.\n\nNew sharder stats are:\n    cleaved_db_created\n    cleaved_db_exists\n    misplaced_db_created\n    misplaced_db_exists\n\nNew container-replicator stats are:\n    reconciler_db_created\n    reconciler_db_exists\n\nChange-Id: Ia43e91a545a822cace41a0e814ab6c3bd89f8402\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/448427388778a9622a46cf8288ab652f31be3257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/448427388778a9622a46cf8288ab652f31be3257"}]},"branch":"refs/heads/master"},"289f59416f964cf705e10c24dbd566c6767c8726":{"kind":"REWORK","_number":4,"created":"2022-07-27 16:00:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/850905/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/850905/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/850905/4"}}},"commit":{"parents":[{"commit":"21fab529c9f0d1681b88aa6b847696717dd34065","subject":"sharder: emit misplaced stats to statsd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21fab529c9f0d1681b88aa6b847696717dd34065"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-25 14:09:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-27 16:00:46.000000000","tz":60},"subject":"sharder/replicator: emit stats for DBs created or existing","message":"sharder/replicator: emit stats for DBs created or existing\n\nThe number of container partitions that are in handoff locations can\nbe used to track progress of a container ring rebalance. However,\ncontainer DBs are sometimes deliberately created in handoff locations:\nthe sharder creates shard container DBs while cleaving and also while\nmoving misplaced objects; the container-replicator creates DBs to feed\nmisplaced objects to the reconciler. These DBs can distorts insights\ninto rebalance progress. Emitting stats will help quantify any such\ndistortion.\n\nNew sharder stats are:\n    cleaved_db_created\n    cleaved_db_exists\n    misplaced_db_created\n    misplaced_db_exists\n\nNew container-replicator stats are:\n    reconciler_db_created\n    reconciler_db_exists\n\nChange-Id: Ia43e91a545a822cace41a0e814ab6c3bd89f8402\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/289f59416f964cf705e10c24dbd566c6767c8726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/289f59416f964cf705e10c24dbd566c6767c8726"}]},"branch":"refs/heads/master"},"517738ac9a6efe97d227d979c47d5222fed47aae":{"kind":"REWORK","_number":5,"created":"2022-07-30 16:30:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/850905/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/850905/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/850905/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/850905/5"}}},"commit":{"parents":[{"commit":"ec2bbc0e14ee1e25e04dded9dda62cb980a534db","subject":"Merge \"sharder: emit misplaced stats to statsd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec2bbc0e14ee1e25e04dded9dda62cb980a534db"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-25 14:09:41.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-07-30 16:30:19.000000000","tz":60},"subject":"sharder/replicator: emit stats for DBs created or existing","message":"sharder/replicator: emit stats for DBs created or existing\n\nThe number of container partitions that are in handoff locations can\nbe used to track progress of a container ring rebalance. However,\ncontainer DBs are sometimes deliberately created in handoff locations:\nthe sharder creates shard container DBs while cleaving and also while\nmoving misplaced objects; the container-replicator creates DBs to feed\nmisplaced objects to the reconciler. These DBs can distorts insights\ninto rebalance progress. Emitting stats will help quantify any such\ndistortion.\n\nNew sharder stats are:\n    cleaved_db_created\n    cleaved_db_exists\n    misplaced_db_created\n    misplaced_db_exists\n\nNew container-replicator stats are:\n    reconciler_db_created\n    reconciler_db_exists\n\nChange-Id: Ia43e91a545a822cace41a0e814ab6c3bd89f8402\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/517738ac9a6efe97d227d979c47d5222fed47aae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/517738ac9a6efe97d227d979c47d5222fed47aae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[]}
