)]}'
{"id":"openstack%2Fswift~850951","triplet_id":"openstack%2Fswift~master~I5a043598e1c5b6a77d571a000e0cf2da77e465d9","project":"openstack/swift","branch":"master","topic":"bug/1733634","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-07-26 00:42:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5a043598e1c5b6a77d571a000e0cf2da77e465d9","subject":"squash: rework get_allowed_digests API","status":"ABANDONED","created":"2022-07-25 21:27:45.000000000","updated":"2022-07-26 00:42:58.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bc8ff694bc4b15e0fd60440a3aaaf5f128a336ac","_number":850951,"virtual_id_number":850951,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-25 22:37:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-25 22:21:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 22:37:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-26 00:32:07.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"9d98541247606714a4d0e899784f8e63a865c8a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-25 21:27:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91806f8311da46bb750ddcaed8fea3145855cf62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 22:21:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef4152e0c1384c86b1de6085942c0525\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/972ab4c651164db6a6472e9ecdae705f : SUCCESS in 28m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33f3bf7f273c4374b8811aa7d212a571 : SUCCESS in 51m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/651bebb2e49246f5819881ef6317ace9 : SUCCESS in 29m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba8ab47edbbb42288bd1d9a2e4a96325 : SUCCESS in 28m 06s","accounts_in_message":[],"_revision_number":1},{"id":"77f027287a533db0303a34953549831922c22b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 22:37:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d87477eb7d546e78911f45aebd8e710\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07d48245cf534752ae03f3d95c4a8e93 : SUCCESS in 9m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/8b6507af7cc04d71a1bd36a3880794ad : SUCCESS in 1h 08m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1b9ba66b7636476ca7f69c74330e712e : SUCCESS in 59m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b227549f92947dab8f2b2ae72899245 : SUCCESS in 1h 00m 56s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0e886515e11c4527a32849104f45058d : RETRY_LIMIT in 7m 03s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/e6c4c9be2c2847c4bb869d32e4e28a97 : RETRY_LIMIT in 5m 24s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/238a1b9fef6a49fea00c825580e4b369 : RETRY_LIMIT in 5m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c122aec3150434188d4d57653145c0c : FAILURE in 10m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9e2dfb396b914b98af33f0df9ba1edaa : FAILURE in 4m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/409ca17ec2224901a93db8d634bb25b2 : SUCCESS in 16m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8691fa1b5fdc45d3a95fe77228e90012 : SUCCESS in 12m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f990779e756048888014e89dfdeb96ed : SUCCESS in 16m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fb66b67cdcc344248d430a846219b0e7 : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3e25f39bfc8648edbc64132fd188dc84 : SUCCESS in 18m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/95711df3c4e94cd4a8c568e99907bbe6 : SUCCESS in 16m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2c805ed13e2e44e5becdf2d8795cb4b8 : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b0ac8c5128a642508ba091cb2811b884 : SUCCESS in 13m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/99026cf7f95041948f01b25e05570294 : SUCCESS in 18m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0073f736c4064bfab38631d8a7dbdec9 : SUCCESS in 8m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6cae2a0184ab40f7af6f73a01f4b1f2d : FAILURE in 13m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/17bb280a64014c159404d08d9e31ff30 : SUCCESS in 8m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fbe4da79d99846fb845e13f43f3d221b : SUCCESS in 57m 11s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c4ccf37279645f4bbc384172b64b5e4 : SUCCESS in 47m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d44850a2926b4a099bfd2c17380258a8 : SUCCESS in 50m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/df408470d1c64a44bea7d38c2a0e2815 : SUCCESS in 36m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da2a53539b4a4d63a7a9e8d05440d030 : SUCCESS in 16m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb882bab90584dff92b633c6ac08e1d3 : SUCCESS in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6af7df25a9bb4042aa2c758d95884bcd : SUCCESS in 34m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6a0e9af37f349bfaaea28e5b8403cb5 : SUCCESS in 43m 41s","accounts_in_message":[],"_revision_number":1},{"id":"311a0e1bb487ef18b64861987a143412f458aa46","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-26 00:32:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bc8ff694bc4b15e0fd60440a3aaaf5f128a336ac","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-26 00:42:58.000000000","message":"Abandoned\n\nSquashed into parent.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a4e0424010ef3665191058d5ee4a84f0e124650","revisions":{"4a4e0424010ef3665191058d5ee4a84f0e124650":{"kind":"REWORK","_number":1,"created":"2022-07-25 21:27:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/850951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/850951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/850951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/850951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/850951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/850951/1"}}},"commit":{"parents":[{"commit":"2c8fe050b5b17ed57b46e578d2989db0429e9980","subject":"formpost: deprecate sha1 signatures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c8fe050b5b17ed57b46e578d2989db0429e9980"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-25 21:26:54.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-25 21:26:57.000000000","tz":-420},"subject":"squash: rework get_allowed_digests API","message":"squash: rework get_allowed_digests API\n\nHard-coding the option name like that doesn\u0027t feel right; callers should\nbe responsible for their options.\n\nMay as well make the logger optional, too.\n\nChange-Id: I5a043598e1c5b6a77d571a000e0cf2da77e465d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a4e0424010ef3665191058d5ee4a84f0e124650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a4e0424010ef3665191058d5ee4a84f0e124650"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
