)]}'
{"id":"openstack%2Fswift~851100","triplet_id":"openstack%2Fswift~master~If48cda4ae206266bb41a4065cd90c17cbac84b7f","project":"openstack/swift","branch":"master","topic":"py310","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2023-01-13 02:52:36.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2023-01-16 09:33:52.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-01-18 16:25:26.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"If48cda4ae206266bb41a4065cd90c17cbac84b7f","subject":"Clean up a bunch of deprecation warnings","status":"MERGED","created":"2022-07-26 22:13:05.000000000","updated":"2023-01-18 21:02:28.000000000","submitted":"2023-01-18 21:01:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"851100-py310","meta_rev_id":"fb412d7fc69425186a5634bb7010ff76297cb8ce","_number":851100,"virtual_id_number":851100,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 21:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2023-01-13 02:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-07 02:54:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2023-01-13 02:52:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-26 23:10:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-26 23:25:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-27 08:03:12.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2022-07-27 10:41:57.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2022-12-12 10:42:56.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2023-01-12 01:03:20.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"e8b811c0f59413d151dbbfb8450af1bf06e92cde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-26 22:13:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"859dd0d437b6c3a56876dab88fcaaafede5ec2c5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-26 22:22:53.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"295758350e84155de80f6f240b83f4c9a4681590","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 23:10:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9d57acd3b05474e8b219daa0eff80b1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d9fb800d2fd428f84d8da5dd445fb3a : RETRY_LIMIT in 2m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05e1bb5514584322aef45494ecc4b500 : RETRY_LIMIT in 2m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3c7ea8a5d1347e0bf4c3a8015f4613e : SUCCESS in 40m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/893f09a435594527b9385913b8fd7a54 : RETRY_LIMIT in 2m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e2560b12d1a0fe6295c76a977069259de14993d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 23:25:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06ac2d9a6b3c488085d77961cc60bbc0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d98f1dec1daa483990e7cc021b0f9ba1 : SUCCESS in 9m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e584d77209494b00bae67f046f03d9d9 : SUCCESS in 1h 00m 27s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9bb20011782840b8ae9d79c1092b594e : RETRY_LIMIT in 4m 08s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d0cb5687bf13461797906bb98e8fff73 : RETRY_LIMIT in 7m 04s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/64e0e2f01cda4675a96957325106f679 : RETRY_LIMIT in 4m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6d0289b1dcc94e6e8605b32cf452e536 : FAILURE in 4m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a2eb540c5e9a4ce19d277eeb4ce2027d : FAILURE in 7m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4218787d53e1418cbe2a77369f4c677a : POST_FAILURE in 10m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/13f698a15bca4888840d8dac903b86f6 : POST_FAILURE in 13m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cafaecc0cc214230ba12ccbf1c2580c7 : POST_FAILURE in 10m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4dc25c68def34233b050fd12acc94781 : POST_FAILURE in 15m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e0ffd550098c47128e4cb840846f073c : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c88f5f1d4c144901976d76f804c9e3bf : SUCCESS in 18m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5397efe977b146dba2864640d5bce4b9 : SUCCESS in 23m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4560a6d3f97f4d7e9482ede75f2a46a5 : SUCCESS in 12m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/27421ab499ac443696d68b11f4c028c2 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/10ae3c5ea95b4d9fa19fad9a6dbde3e4 : SUCCESS in 13m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5185463e6a644150ba87d21008f9ff45 : SUCCESS in 48m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7a077d64e88e4c42843e8de4ed41abb7 : SUCCESS in 32m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/706b36a563e749b9a9718084844a8227 : SUCCESS in 14m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c6141666b5c4dc49de29a5495629666 : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c5777d4c859f4bb4b4ca0de654fa7e07 : SUCCESS in 30m 21s","accounts_in_message":[],"_revision_number":1},{"id":"58df62da0c1e6b3883168d1547586ac04ad119e3","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-07-27 08:03:12.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nSo I assume things starting with Test are getting hit. Do classes starting with a \u0027_\u0027 break any other python style rules? I\u0027m in a kids sports class so can\u0027t really do a quick check. Just worry _TestReplicator might come back to haunt us.\nI guess if it is an issue we could rename them to FakeReplicator or something.","accounts_in_message":[],"_revision_number":1},{"id":"c1cfaffbffd27b31f7a6e57963e149bff2efee27","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-07-27 10:41:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40361142253689b0a5499369c5a260e230c4e2d5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-21 20:17:06.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e7d0bb15c079d692ec14a1ca30ad13f967800fae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-09 19:25:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"dc1be7c040df6cb250aa927d4722800d7e67827e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-09 19:38:32.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"3422688dc9de48f3068f4f0f21f0ae2ffac3812b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 20:22:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/817eb9811fd6409c9d0418ee6c9cb973\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11931bbc9ee94379975b159b6b9209ce : SUCCESS in 9m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab64230558734a7b9ae713624ba904de : FAILURE in 24m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f69341d494b4eca804a1ddb1e46da54 : FAILURE in 22m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd4b9810fae7457a80bbe7f71e95a648 : FAILURE in 26m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0e5e95b8deb34a958df3ba0303e08720 : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a09af78d07d340eeb6b036a1055dfc1c : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8d7793661bfa4bb4a1da7163bb9b6a8e : SUCCESS in 21m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d190da6b501e4ec1933b5b128091a867 : SUCCESS in 20m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/549a82d1b6bb41109772144cb74284f2 : SUCCESS in 20m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/507e154bdf5f4a60b6c33bbd8539be21 : SUCCESS in 20m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6790435ed87344e592a908da00955021 : SUCCESS in 22m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a64034e7783a4e41bac028d99e88d606 : SUCCESS in 17m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1f5daed8415f4b4c9a8852cd43f46ce6 : SUCCESS in 23m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/400444e5ee9c41b8a4f762e382b315aa : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f00825dc497f4fdc918c58a6cceb8c68 : SUCCESS in 22m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3f564cc8b8cc4eb7a00f10b8cb320e86 : SUCCESS in 23m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d91f624d62c4931be23978b009abe26 : SUCCESS in 41m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/493ccca9820e45d4ba1ddd3787fcb17c : SUCCESS in 31m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de1ddcb987c4419f9006984e3ccc6679 : SUCCESS in 14m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eb8054897094966a5592622e4c5bf17 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4761079084db4ef388a27cd94401f60b : FAILURE in 24m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"93316701c28b291a6c9011e21a0c92dcdfbd1973","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 20:41:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e65cfbddf194d0d9f95ed884a729a82\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b218b2903c04ee097855ab8abf62b4a : SUCCESS in 44m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85fd428398db4927a2d856ac2d5b32c6 : SUCCESS in 43m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/12fbb17cad85451b9b07997baa27f704 : SUCCESS in 54m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/61cb9ddf3ca645dcb44ca2640695f040 : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":3},{"id":"7e3e99bb30228b559adbe2101099f8fb218924e8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-12-12 10:42:56.000000000","message":"Patch Set 3: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"784f584cc0fe0d10ac66f8e423476f5b343151a5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-12-13 06:02:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"01c8325422acbb2e89220247ecf1adcc14caf8aa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-13 19:57:41.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"65e736707c16abe0d8bc36a4d3fc2b1b6b1a707c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-13 19:57:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fe50fc9ade5889972021220e81532db25ff1e6cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 21:07:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8238f1449eee48f792a1ee60a3ebeacf\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4aebf5dd0e3460d88e52c70172a8495 : SUCCESS in 59m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/140a6754cb8242b5ab37d3251b1fbf9e : SUCCESS in 1h 02m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0ec3fbe7f2ac4a0db024bcde15f1b87d : SUCCESS in 1h 01m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/693fc38947324eb08589ecf7dcc4af21 : SUCCESS in 1h 07m 22s","accounts_in_message":[],"_revision_number":4},{"id":"1a9585c7d20167f8f854a42385d45f911c759bf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 22:04:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02c7898ef8f94475aead39949daf61c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acbbc5856dd54c2aaee7b48e47ac501f : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/7b3135a8f7454b2b9ecc4ec96b6259b1 : SUCCESS in 52m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/babcf2e48f3e4ac78eb51b10e7961c6c : SUCCESS in 45m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57a07618f42d492a892ee33af2d57bd0 : SUCCESS in 44m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/871aab00827b47eab8128a340bee1772 : FAILURE in 19m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39aadb76bed5465caa2a50347012477a : FAILURE in 19m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e63cdf2098643c0bd7420f7cfa36c9a : FAILURE in 34m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0946f9177e8042cc823c9ed781301544 : SUCCESS in 8m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/056758f0acfa494588b39259061d7c91 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/30e58c68bbf445f68022ceee9a8eb3a7 : SUCCESS in 15m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/13f6a2822ddf4bf2b82f9e40774c2fe5 : SUCCESS in 25m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/45934d60ac5344e9b26759a15abdb3d8 : SUCCESS in 14m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9ad5d220310a4c90882980798f1a1ba7 : SUCCESS in 14m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2cbd5a0fd62f48e38ad8bdee84b87342 : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a995908633fe48a9aa51afc2123ba56b : SUCCESS in 25m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3c9bb0980497485bb73ac857cf98444e : SUCCESS in 18m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/751ea09ee3794e259080f261d87fe75c : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/460200a3269d4eb38f13be16abaf0299 : SUCCESS in 25m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d9ac45c8c53c418590ee12f6f362093e : SUCCESS in 15m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d1d61196cc6c4312b51bb5905098ea37 : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/122d3fdf346f4e8ba48733b1129bb4a9 : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/75c4733012844c50ab25b0a0debb929a : SUCCESS in 7m 06s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/140ec2100a4c489498093f6fa029f923 : SUCCESS in 1h 19m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1ab672e4c924fcabe9e4d4d5353a72f : TIMED_OUT in 2h 01m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b680a6fc0e604d24bfb9a908cad931c7 : SUCCESS in 45m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3ee6bc03480f41509ec7908aa846524a : SUCCESS in 39m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/284d42f6b18d4e43a050083320e2cf68 : SUCCESS in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c80504a016f34e339d87774a77f8a435 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11f9a001c9d4422284c4640cb0765f4c : FAILURE in 25m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e11fd440ac1e45dfadcf0f3cced0a319 : SUCCESS in 51m 30s","accounts_in_message":[],"_revision_number":4},{"id":"aa5b836e46ce2af43f3726cee4720abc73e44143","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-13 22:15:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd5001dfb93574ff51f9fcd3e3e6caafa7d8d480","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-19 20:15:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4b3aab615e4fb22ce0d5a57b4e44aefa815ce32a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-19 20:18:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8009f8d870a35e7f5e7a9b1b8fdcb4948d9c5c69","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 21:26:30.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/200c8761dbe74d3286ce1d7a2cdf7439\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9987ea3f2a184ce7b85abbbe48b104a6 : POST_FAILURE in 51m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a6d27842a25448d1a137b0d697e16af2 : POST_FAILURE in 49m 21s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb998ecd2e024bc08ea80071b5e6fc88 : SUCCESS in 55m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c409d8a6bed42a49f026b90b8dc0eb3 : SUCCESS in 1h 06m 25s","accounts_in_message":[],"_revision_number":5},{"id":"082bfd12637c2661089c64f7ba4592ef54731f02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 21:27:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4df9436af6b34bb5bfa82f9b18427fd8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8802b4e216684bd9bdc541be33e3b009 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/270573434fcf43cdbc672dbb4a9a2784 : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97d90a2e71544fad8ceb6f0d4d40d36e : SUCCESS in 5m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/ddb9940f89e44160baee935c9038dd70 : SUCCESS in 1h 02m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/be3550dab75e41c8b881fbef69e170a2 : SUCCESS in 49m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc94c8c69fa3403a87f97897983adfbd : SUCCESS in 57m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c13cd4e3abfa42739f4dd20ec610ed1f : FAILURE in 17m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bebfe783ed24464c810935249f9df562 : FAILURE in 18m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3eb3df47b1124b679526809dd81ef9bc : FAILURE in 19m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/43653f1473fa49c689ac5c7cd840c871 : SUCCESS in 9m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/37db13e2c35c4d04913ab443b2646cd3 : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7ebe25f8edde46df8c35165a699d961c : POST_FAILURE in 9m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8c0b347ac1104d4ba982005c55b4b54f : POST_FAILURE in 12m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d31a766967584442aa4675a49122e04d : POST_FAILURE in 12m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d21402787d014d3c871637d98508edf6 : POST_FAILURE in 11m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/183e2987f03448a98d051e848080936f : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4b28d2abe4704b988934c8bd184d45ce : SUCCESS in 15m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5ff5c51750d240719504d23a8c165e15 : SUCCESS in 18m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c1229b599baf4ea5a1ba03c88aff6d10 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/95240b419f174b85ad6358983926132d : SUCCESS in 13m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1ad02e094c3749b19fc62e3cb7cfe34e : SUCCESS in 12m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/37bba171131a4492a54cbbf1a39f20d7 : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f7c78f468924a8f8505f44b0e914f48 : SUCCESS in 16m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ccab4f4ea24c428298b67544cb0606b8 : SUCCESS in 7m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/49e1af2e3508498c924311a97a7683fb : SUCCESS in 57m 30s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e90d4cffd9014605b6e2b00314d9956e : SUCCESS in 51m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/52cd4b22a22b43bc969349e48242d28b : SUCCESS in 37m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/84f51721275f4f8b9c244ed35b199de9 : SUCCESS in 42m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a97d28244d424745bb59eeb5263a3aa7 : SUCCESS in 15m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/506a14f4f8ad4094ba84fbc1065f10f2 : SUCCESS in 4m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d7605b7a3e3549a0afb70a4bb8009175 : FAILURE in 19m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9a8c43eb62043fdb19d94fe953380a6 : SUCCESS in 38m 17s","accounts_in_message":[],"_revision_number":5},{"id":"646f857a45666ffdb2eaa043e8a120eeb436e2ee","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-22 00:04:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"37a6c05c82a72edb4837d548b650975b6b6a34c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-27 21:34:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b0ec91b77c63b56bda9019f32fec34285bc1b337","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-12-27 21:34:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a32f6527d9cd05e512cbdf54862f345d83cf01e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 21:42:55.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10cb0dd44b5245b2a554cf34388068b6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dff107eb0756489788aa68a6abe1ea86 : FAILURE in 6m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9ab70d1bb4042338a173f347de17ea5 : FAILURE in 6m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a673864969b14c098fe7b5b13d09390a : FAILURE in 6m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5fdbac783dee4ee681b95875d79a83e1 : FAILURE in 6m 13s","accounts_in_message":[],"_revision_number":6},{"id":"2d65bc54477e20820e9ac78fce37911223b27806","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 22:11:43.000000000","message":"Patch Set 6:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e8fd7d0fd664f26ba6339ff7c04475a\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f7faa4c85a0145428d35f2f3f5d1bc74 : SUCCESS in 16m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6f454db31a074c98a873bae0c9a09e96 : FAILURE in 3m 47s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/85be51b1369a4e5bb625e1182e938f5d : FAILURE in 3m 26s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/f6c6c6df466d404ca5da372dd3203d78 : FAILURE in 3m 17s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/a2e4340bf1a0422e838cf6d3ade5b5b9 : FAILURE in 3m 22s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/282b1ecc7a2d4482a052dbe34b8197c0 : FAILURE in 3m 15s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/b33bc889979e4e2ba4028670eb0535b0 : SUCCESS in 16m 20s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/55728f97c8f64b3a8a873ede37572f73 : FAILURE in 3m 30s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea1d98e04c0d47c293fc9cbde5018764 : FAILURE in 4m 12s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce6e40dd704a4f75b7c1f3a8edd61988 : FAILURE in 3m 22s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/16529ae5c22e4b9f99958199ae272bd8 : FAILURE in 19m 38s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/d0c416269fc24d1c991d591b94fa91c9 : FAILURE in 19m 06s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/61544dbc05404d73bef4ca7602e613ec : FAILURE in 22m 35s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/5a29df72f2f84cf7b96e247668f9aeb7 : SUCCESS in 23m 25s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/22bfa5714a38403a8a8451db48bae704 : SUCCESS in 21m 55s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/7f6c80434a1c4bceb9e7ce65620cc626 : SUCCESS in 22m 35s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/4d9d2c8075774bb29cfddd9f639c5ae6 : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":6},{"id":"a1f21737713a24ef905490cff4115b1e226a3d80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-27 22:57:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/59869dcb51794fb29020e31470bd0f20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7645091ecf074126819e76c1d71f1440 : FAILURE in 5m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/5f5f962e46384780a2370ea19d401ed3 : SUCCESS in 45m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e4f1dca8bc54457a9a6ca216bc893268 : SUCCESS in 51m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fc50e7010304c21b0fd6c0e22204c9a : SUCCESS in 46m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c3a9d37c1cf4819a4d9e778ba36da98 : FAILURE in 5m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aaf35db375264162a19ee8f3c01a8fbe : FAILURE in 5m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/952c7ab99717497a8f35de55c96a007c : FAILURE in 5m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/16abb2aae9004f449a50cf7acb18d37a : SUCCESS in 7m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/09c6ec305a3147468227e255beb373e2 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0b7b7649cdeb4b77a72ad3ee20db847c : SUCCESS in 17m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/da7c0e783b2f463f8a041ec742c9c79b : SUCCESS in 18m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5b2639978d4d489ba5e9ce31a3ceabda : FAILURE in 2m 52s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c41075c6f979495aa61952b193cfa689 : FAILURE in 2m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e587d9cb899447009bca1ae43d696bff : SUCCESS in 14m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8a53350fbf89487688692e564e29b95e : SUCCESS in 18m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bab5acebbea44956abfada732b4deb97 : SUCCESS in 16m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5ac50279460c4b14b965afd3e89e9c4d : FAILURE in 2m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e15e51e6accd46c5913b4f96f571b269 : FAILURE in 2m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ffeb5d1523234e899540629eabba741a : POST_FAILURE in 2m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/425047a4e09d49ceade0b5a632697646 : SUCCESS in 7m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f0c392477ea049348370a794c976420b : SUCCESS in 12m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b9bb661e33a845b0960231d44b4d4a9d : SUCCESS in 7m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4ab988f8234544edb8ef09d58f265224 : RETRY_LIMIT in 2m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7feacadfc2dc4594ae0d210c4ecd8ff0 : SUCCESS in 1h 18m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/77e3f58bd639453db9b1c886e4b0aed9 : FAILURE in 11m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/844e568abd914eb3b153cf86e56aec87 : FAILURE in 12m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30cd0bd22fb84c1c98bca8c8dd2f6a42 : FAILURE in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9289e986ae594c8c8048e2f3779ab451 : FAILURE in 2m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/06e9624f48c640b1af5bbabdfb617242 : FAILURE in 22m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da16df70b0124875879ae5f79859e747 : SUCCESS in 39m 51s","accounts_in_message":[],"_revision_number":6},{"id":"088056eaa79585e10f922e9d0f626de3fe5ca9e5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 23:34:35.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"52eae15935669a1ae68ade2d154e310aef316383","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 01:01:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee0a9c487ac1423d8a5b5b06687fbebb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31178eaa5c7c4819972f960783cc528f : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/63a38d06ae3b415ba7f667a56199bb27 : SUCCESS in 1h 01m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aa30288ede9c4dfb8fea5d710b0c801d : SUCCESS in 52m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45d599e1e94e436a8813f73e09e011e8 : SUCCESS in 1h 03m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb1438c840154075933118121f4c713b : SUCCESS in 17m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b78cce1702aa4cec9ca17c3bae131d61 : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26628d498c9748738010984c37fedd38 : SUCCESS in 23m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/124613b75a4740a492bf404cf06dc8c2 : SUCCESS in 9m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d2fd980142154978be29c8ea1250532d : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/320fabebc9874abfaace98b1ba314c52 : SUCCESS in 18m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/824b94cd810a4ebeb351c62184b5d320 : SUCCESS in 20m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1971e69ef0274725a3f7ad463dd564b9 : SUCCESS in 18m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/ce55573327194f9c8e4ad7858748f029 : SUCCESS in 15m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a140dc1551b442bfbc81ee244bbe7c2f : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c4f3733a7ab847cb8a8fac5dcd45f65c : SUCCESS in 23m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3c77cc95f48e445f87a32e7650da69eb : SUCCESS in 17m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/08d85082f99b4784bb954f211e37da37 : SUCCESS in 12m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/02923d31e93b42e3883bbc91fe51118b : SUCCESS in 21m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9519338ef28c48fdaf3eee2468625902 : SUCCESS in 15m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f6195ccf8de94dadad17a6163892f56e : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/02a4064e65234d24b7890005bec02a87 : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b3fdd290752a4417a42beb820e8d80dd : SUCCESS in 8m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e32bc99c5fba4b839315c6fe9657725a : SUCCESS in 1h 12m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/019f83f66bc542b7a5ea74e8318d1bf3 : SUCCESS in 1h 12m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1e7c487bcfc84e1085ec3c5e4420a79d : SUCCESS in 43m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e5f7ef351cd345fcb49e8b38e76f8c94 : SUCCESS in 38m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90a04a70f99748268cf7dab55117f307 : SUCCESS in 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1360cd6dc1554bf38db5cff85617b59a : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cf34b7a0fad744b995550d87da478ebc : FAILURE in 24m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/784cc054c9774dbba0deba78024d5b4e : SUCCESS in 43m 04s","accounts_in_message":[],"_revision_number":6},{"id":"d2363c902cde135af6cc3fff6502d89b8153e2ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 01:04:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daa47d5cac7a46128c51aea702490fc4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd846e987fdb4ceb9600b326edb0fe04 : SUCCESS in 1h 14m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/330a98b2fa1b433aab8294f6b0b92817 : SUCCESS in 43m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/89c1b422986246f0ba891228cd47bdb1 : SUCCESS in 58m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/168d555ba1164740962d4bb38a65194c : SUCCESS in 1h 28m 36s","accounts_in_message":[],"_revision_number":6},{"id":"268f094965e377f6ef989ea753300fd6adc8c4f2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-07 02:54:01.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e1866edc3e5eed442c2183d06759a23ee9e59058","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-07 23:14:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"60b29b9014339d923d05554e00061552c1c611ff","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 23:58:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48a7593d369443d89391c3d0863e7dd5\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ba4d1db7a3074a80869dd0959666a402 : SUCCESS in 17m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2056bf2e67a144e38bafbbaf08319e51 : SUCCESS in 12m 25s\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/84fd5162393c4038b3cea1953385bf9b : SUCCESS in 17m 08s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/f1b285678ac7423e808fedbdf2596162 : SUCCESS in 16m 03s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/f1b66735711a463ea373012e1f68051c : SUCCESS in 23m 02s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/4949caca008d4978966d0662da372f1f : SUCCESS in 19m 21s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a1f11993b8742a4bae5a1874dee7568 : SUCCESS in 19m 12s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f1a20fa14c4417a99bd29f87985c401 : SUCCESS in 11m 46s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/60d60463130748d89d6542873be43cb5 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/656e84c11fa2421b94d78990d29176f3 : SUCCESS in 16m 23s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/830a400e8ad445159b7d7f0b6603f81b : SUCCESS in 17m 32s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/0e9bb24f75684d4e93252b5e87bf397d : SUCCESS in 15m 41s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/c54975f6c8904ee68f7fc695974e6cb0 : SUCCESS in 21m 10s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/2ff5bc86d79e4b95b843cba694f3e1e8 : SUCCESS in 15m 23s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/d5025b634fb7433899ed4143d69f55ba : SUCCESS in 18m 25s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/7c07f4b3d71c4925a030d5d94972f0cd : SUCCESS in 17m 31s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/a0549a24afeb4fd8851dc8ec4ac2986b : SUCCESS in 15m 47s","accounts_in_message":[],"_revision_number":6},{"id":"e05356c612681161bcff04d8416a71af57cf9404","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-01-12 01:03:20.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c4a552222ed13dd5700d748ce1d9bfa7b38ce80e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-12 17:52:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ce96bc211dd7634e4762d5b1182cfe2561e80608","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2023-01-13 02:52:36.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7af05e264d6838b3c5ff42af153302588cdbbed6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 02:53:53.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2f7e91290e7e73dd67a10139b66b3f5e5c4ce234","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-13 04:29:55.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37bd63d1138145d8889f8c55d2a971d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1290323462fa40c7aa2b12f2c9cbae02 : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/2f30c623d2c6428b90667941ccf65d3a : SUCCESS in 1h 00m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d0bb78694c104966a6ec559cbdd80c05 : FAILURE in 32m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e27a3d7adcd1440d9dc8703db717e4f7 : SUCCESS in 1h 02m 38s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3caef84727b04846bdfbbd66ac55d763 : SUCCESS in 8m 41s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f9133d6fa86f466aa9b22d2d2f048e91 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/27dd50a19fdd4dde948d0667e4d42551 : SUCCESS in 16m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a930c49e5b7a482293d48a6d99f9bfc4 : SUCCESS in 18m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ba879f9d4f7f47a8a0a1a78ac3b10657 : SUCCESS in 14m 22s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7b71f7b6a1d64ab191917b1327067449 : SUCCESS in 16m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1f40f0d126f04cdb89d07c0c0b204724 : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/906972cdc84c4bfca71ee3a210d6eaf6 : SUCCESS in 15m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6f6ad6b00abb4c3ca15313723f5338ea : SUCCESS in 50m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/67a13922ba574b4ea0e728cd93018bc5 : SUCCESS in 31m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9fdb2a007cc84e4ea51420bae27a32e6 : SUCCESS in 17m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/31e400e6d84447a59d32f99e5beda636 : SUCCESS in 13m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9d644ccf14954457b25a88ec427830a1 : SUCCESS in 10m 14s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d699777730d147be9d76f73cb4f4f1b4 : SUCCESS in 11m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bf5941fedbb44ef6bab0a59f6eacd72a : SUCCESS in 1h 26m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/3907bf31bb154fa6b019c2ac4fec04fa : SUCCESS in 1h 22m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d72d447bd9e4eae8f89fcd1512e3b11 : SUCCESS in 50m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/30ea850189ef4721bc0e43aef4f211a5 : SUCCESS in 38m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/943f8fd4dc714cf49aa12ff8e850aa85 : SUCCESS in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ff5e3d79a864db981d9712c054703bd : SUCCESS in 6m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0b17134d246431f9eb08ae702903c27 : SUCCESS in 1h 01m 27s","accounts_in_message":[],"_revision_number":6},{"id":"a78c234bbe76134388d286e8bec2878c0503b0e9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-01-16 09:33:52.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"923c6fe0b82fab10ce2394989cd1bf612e87fac8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 11:08:49.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6ce7007f90b449a82ade23afd77b1b9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db4654d8a662430fb09ba0b84a65a519 : TIMED_OUT in 1h 31m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/437eb63056474e23a0e83c2d1a70d161 : SUCCESS in 1h 14m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0835c17a4533417cb375f35bc15363df : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd615edc81534df7887b204660e5bc84 : FAILURE in 1h 11m 46s","accounts_in_message":[],"_revision_number":6},{"id":"9226a5e7c9afef78b9b9bbd44da7b8c25a934332","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 11:20:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7251cc3c55f4d029a0e80052802f61b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/300f91aaed8e4eb58778acb3f2c8e9ec : FAILURE in 5m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/9c4dbd65b9e84b59bf1b0e57b4bc58d8 : SUCCESS in 1h 00m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4bd99c511f4942dd82759536dada2f23 : SUCCESS in 53m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af92b721ad544d1d880fb5b97cf37759 : SUCCESS in 1h 04m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9062ba95090c43c4b42e155ff8c9edeb : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a5e792cbe034657ae5a2988ba678544 : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afa6269c30ca4f7ca0f3dbc4ca64c2ee : SUCCESS in 17m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2900e48733f142d596ec0fe5fc8cdba9 : SUCCESS in 9m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e34eab9ad6f04b9fa8dc414c9b846dfd : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f56909af6f964cb99cb1011cdc1b11fb : SUCCESS in 16m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1544cd49b68a4825965356c817df5547 : SUCCESS in 16m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/602bb094d96f47bbb66f17189243b4af : SUCCESS in 17m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/581b3d5420c34774b7487864856b2a7f : SUCCESS in 14m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e059b4fb6e9742a5898b0e5ddd637d5b : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a76fe02c322749cb9d11cce3ef70f08c : SUCCESS in 17m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/956612f466874b71907660516a86d4d2 : SUCCESS in 15m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/88664975f80c4c8d82703c55fa437539 : SUCCESS in 11m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5160ae3034bb4de8abf430b5c9864688 : SUCCESS in 11m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6cc319ddd6724c96ae5d110056a258bb : SUCCESS in 15m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7f7b318fc34146938f3baeeae975fb90 : SUCCESS in 5m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3492fa75e9834491a850f3f8940c97ae : SUCCESS in 12m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0ed47086e3cf4e778b32e7cb0e395eb4 : SUCCESS in 6m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/df100cc9ddea4b6fb01f9b30b19bd719 : SUCCESS in 1h 14m 56s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/75ee54837ef44ac884944def0a66349c : SUCCESS in 1h 26m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/13bf0bc4b94042099f3a110164f3835c : SUCCESS in 44m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/37335a1d0a264d33aee6ba50ff75b9e3 : SUCCESS in 34m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ab92f5936514237abf6ca0593835f93 : SUCCESS in 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c99baba3f08f43fcbacab368153f5821 : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/599249f928b84dcd88b558fcceec9425 : FAILURE in 21m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cff80c8612c4452bb0682ffbc51a7c2a : SUCCESS in 46m 10s","accounts_in_message":[],"_revision_number":6},{"id":"edf579382dc568ee21247b96fa1b603ad54f51dd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-17 22:14:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"19c4ad881940b63649362e5257fd33744af7a378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 00:13:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc7ade32adec4366bd243be3e091336e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b5197057fe4e3b84727edafb36c304 : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/da34c3ea38f04b4582d6376530fd8cf8 : SUCCESS in 56m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f4aa311611414bfbb7776292e6f0466b : SUCCESS in 42m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48373f5e8bf64628bbbb3ba590ac328c : SUCCESS in 53m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb330235cf2d4e72b055f2799864de55 : SUCCESS in 26m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c7e57a2421042a0b37392e7d442fab7 : SUCCESS in 26m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/873da0cca3f24b20879778dca17a76e3 : SUCCESS in 28m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0e4a48aec3da4ccbb3f87949b223d830 : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/181f48f6bf7b4b3c9e97abf14f7dcfdd : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8e1fd302e89d4a43b03ff6862887fd21 : SUCCESS in 25m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7fe8fce9eaad42d7bd02444ac3dbab86 : SUCCESS in 14m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/28d2d9d398e842019bfaf03a0ce8e284 : SUCCESS in 24m 07s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6690f9d36ef54361a4e2c51df0c2a1cf : SUCCESS in 15m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4d2779358ea24cb9b1c573adff3fec9e : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/72883ba6e0724a279332bbaa13545a1c : SUCCESS in 25m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ead1a57d7f7c47bb9781523e273f2bbd : SUCCESS in 25m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b13686108a0645f0b96474ab5a797b72 : SUCCESS in 13m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bb00b2dcb75d4796a6b6bfd72df0b3fd : SUCCESS in 14m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0dfbb3462a414aeb94614995c44e1a59 : SUCCESS in 23m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/66a8d8eba0b84087a9e564bef15b0c4d : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0c48d08bb73d41d885e6453ddf33b679 : SUCCESS in 15m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6fc57eef846b4ad4810196054fee717e : SUCCESS in 8m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7ceba321c25941f2aaea060651300b72 : SUCCESS in 1h 05m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0078d5d12544ebca1ff8659eca712ed : SUCCESS in 1h 29m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/54049efdebbb4203a64a133875248430 : SUCCESS in 56m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fe5c4d91f82d4d4abc0299b998a20bec : SUCCESS in 52m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4afd7ff214b54f139a498e6931a7faf6 : SUCCESS in 15m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f19c3226fc554961b297e4c79a8983a2 : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a097d6bfa00494587254787bc430585 : FAILURE in 18m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd6e6e64af3c4937ba70d6db85268a4a : SUCCESS in 45m 14s","accounts_in_message":[],"_revision_number":6},{"id":"fcd393bfdecba19611771b9c77d6a13a8c5c5c0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 00:13:46.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9a5768144b03957a68100bf9394baf7927e78ca2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 01:57:32.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40c3c1f242de421e9642b673c780b80e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22e2f4ca8fba41f2870fee357cdef4f5 : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/7a0aced166514384a81cfdf6fe1d3aae : SUCCESS in 48m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9d48291fc963407daf11074ba66f3f52 : SUCCESS in 39m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fe9ddf5a2b949c29514d9c00a8b2709 : FAILURE in 31m 04s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/16d90e6399ff476589608337d501b802 : SUCCESS in 10m 17s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c257cd5b5a724cd597357dfdf660a73d : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/abf338c7855b4a189176f115072982a3 : SUCCESS in 13m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/206775d6c25741409975b3578dc8712b : SUCCESS in 10m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d24e5c346eb84027a6965c9ffa88c18b : SUCCESS in 13m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a59520cfcbf6494184ff233542853fc7 : SUCCESS in 18m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e831bffc5c6a4e7d96c466fdc45a39cd : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/17dd471779e2432999df419abfd64a8f : SUCCESS in 12m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3439320950e74435abeab43f5fafc27a : SUCCESS in 18m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a63d414d8a2a4070ac77af7c988ba58c : SUCCESS in 12m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/624cc72386db44148aa7cca7eb6d430b : SUCCESS in 16m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/79e4409dc578465290667696fc08f9f8 : SUCCESS in 13m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7a397cfb69f14a20adf935b7bff73f62 : SUCCESS in 6m 21s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3d45940d7913455b9ea5942bc229937b : SUCCESS in 7m 54s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6ea2379b7e794ba7b8b4ce1f1566bb3f : SUCCESS in 1h 20m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4d2ee8c1c82e45eaa529f9e6a4a6fe73 : SUCCESS in 1h 25m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b744446e2f59465d998940c324c4fdcc : SUCCESS in 48m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d802971ffdb24447af0f7b0020aaed3f : SUCCESS in 39m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d29bd920979b45e48c50a9f65e07a6ad : SUCCESS in 21m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4e38eaa9a9146c4aa6f8ba5a32f8e03 : SUCCESS in 4m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd95ea2e30164456895458efa781bb67 : SUCCESS in 39m 20s","accounts_in_message":[],"_revision_number":6},{"id":"52ca029b597c4fdc007e12d89633c16e20913027","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-18 02:34:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"752c2786841657c6fe3bfa9d7c30e0afa0b221f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 03:39:59.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb0cc044736a4d24a0be6dff473869db\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26dfa85961454b748a3b571475956c3d : TIMED_OUT in 1h 30m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9e315d254644d299d08c093719e3340 : SUCCESS in 50m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c51c5984e0a74c708b5379f4af8ad4b7 : SUCCESS in 1h 19m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f105fe0f5d1640e192e754f22fe7e07e : TIMED_OUT in 1h 31m 15s","accounts_in_message":[],"_revision_number":6},{"id":"f967c182aa1203cc13acf90fb0d5bf2bafed0a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 04:10:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef1e2dfbe4eb4c4d93457d06a7d94097\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50cb58badfed422a8b9ef9f2bfe549ea : SUCCESS in 7m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/5d832af4aa1643dabb6c5a49aca22efe : SUCCESS in 55m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/47b3995ec8f346c585b3a34dc6d421ae : SUCCESS in 52m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67142c8326e0400d9e3f7d40c00537df : SUCCESS in 1h 04m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0098bc03db144d88a3e0db7e4db245df : SUCCESS in 17m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6cc409d4895b420b9015a62825d1bb06 : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbeb87471edc4fb6b418d5237d620926 : SUCCESS in 20m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0635e2c963914c9abd165b6cf9db1c61 : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dfa53d31c01a49d38ec3cc43fb4b6713 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c51239d9b2b840409605e8a68934cc65 : SUCCESS in 17m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b1e66e789fe743cabf7c0872df4065a6 : SUCCESS in 20m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b15a228b33e04418b7f295e2cabd3f98 : SUCCESS in 16m 08s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b45a012527c643e6928fd4262554d4f5 : SUCCESS in 17m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0113f1c3e97b447f867398ea8d71bf95 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/05723ad16a8e47cf9437c8b37ef24281 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f6d15c7d587d4f349bcc71dee8b2c7f5 : SUCCESS in 18m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ef9c185781bd41b38361feacb9144b32 : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c538cb3653b644368b64d46f5e84a4e9 : SUCCESS in 15m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5282e30a817348e5b23624c511737e30 : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bc4d2ff297f1449c96372e5fe11d0403 : SUCCESS in 6m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/04a0a22d09a24d78a07dd5b7bab0a6d7 : SUCCESS in 32m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d961b52007404187851c27f57532b9ad : SUCCESS in 7m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/92c31e731e39435880057c4b523bd4a8 : SUCCESS in 1h 00m 09s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3d19d7f040b41b8a41d92cb80cd2c11 : SUCCESS in 1h 10m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3463bf241ce640bdad150d1cf2657e0b : SUCCESS in 48m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/448e1d995a9e41c288b052e02f179d5f : SUCCESS in 52m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c982d431251141c58ce3362f645f0973 : SUCCESS in 18m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec2db3e824d54ab2bf9ae98df7aab53c : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/876a688ec53c4739ade322698b43be81 : FAILURE in 20m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f66f19ea890402392ea07618ddc03f1 : SUCCESS in 51m 26s","accounts_in_message":[],"_revision_number":6},{"id":"2fd1aebb2328301468dfab8bf9208db941b945b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 04:10:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"48dcefcd9f769b01624c2efd74335b6febd5527a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 05:34:44.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c737b864f254284aef76e1f47b209e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e81f6f27a86469ba932396788c143b1 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/5c6c93c4d7314f37bb51d3593e8d1372 : FAILURE in 11m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4d0a193f6643404a9cea69cacd87eaad : SUCCESS in 37m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/634b84dd2b8a463689c7447c785f05cc : SUCCESS in 54m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9498db44b2f04f088b0694b72e07472d : SUCCESS in 8m 52s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/13b4bbce0ad941fbb483e0f52f80a8cd : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a45538b65c3845f29a8d37c2a5c85911 : SUCCESS in 25m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d5786eebb01402fb98c648e362eb6b1 : SUCCESS in 24m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/08eae6d1e8c84528a7e9ff10ae0f8b0b : SUCCESS in 23m 56s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/52cc6f41f98441d0828404a5a7e0cd35 : SUCCESS in 25m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6fe47c9dbcbc409fb019865620fb1e98 : SUCCESS in 25m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ed30ae9d219f4868ac677a2ee1742c0f : SUCCESS in 25m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5f7575324a5344899434a92bfdf7d098 : SUCCESS in 25m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e0858143e89841a6b5af0694addfde85 : SUCCESS in 23m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/10dbb78e6fd44b59a87c699c5ef48d2d : SUCCESS in 24m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/195417bd626c4604a6911c2ae520bd12 : SUCCESS in 26m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ccb2db10a6574f31b463f7998cee0568 : SUCCESS in 7m 20s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/209f643d36c146a8ab67e9f859638080 : SUCCESS in 7m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8795246d20d043f2bc1a0e69e2dc3aed : SUCCESS in 1h 08m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb2dd73a8d1d4f9ab6ca96c20e2d129b : SUCCESS in 1h 11m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a5e7a317ecaa4b4995702ec2ced43a59 : SUCCESS in 53m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/65f704cd95b54661bdf82d1d4b870a04 : SUCCESS in 50m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/191b2ba2a3294f3ebafb067c859b7d6e : SUCCESS in 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/655f9d873cf946eeb2474c2d8e72bdeb : SUCCESS in 5m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad10195bc1444bd385aed892cad15fc9 : SUCCESS in 37m 26s","accounts_in_message":[],"_revision_number":6},{"id":"5d4da54575b09801b9c9feabf46a48da852bd895","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-18 16:25:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0b65c6645138f2b67ebf1d1d8d6aa99f3f163a98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 17:41:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a0b431d1b0e4df781d49c99f479fc48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a72e18bf81744a48a84116ddbf0fe303 : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/ad675ec95f5c4538843b3e8e58288757 : SUCCESS in 57m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/525210b3e6e04e3fbbe1d58e5a0b387d : SUCCESS in 44m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d6212b6016249edbb65699aa450fa7c : SUCCESS in 52m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc8bb90bebf94050aad16260c0e937c2 : SUCCESS in 17m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a742a48d8804f7ba40b2cf5a30b5361 : SUCCESS in 16m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9983f948665d40959f6f820fdfdf9a4b : SUCCESS in 19m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7790a67efd44de883cb7b8b9200d6c5 : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b91c523a5fe64c28a30b9f21db6ae2c3 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3a7c8a7979424ecfa0ce848d4707365b : SUCCESS in 16m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4f0bb1ff6ced449ab9fae8cb8ae29123 : SUCCESS in 16m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1b8aa23eef8947898f818b5bc5961229 : SUCCESS in 14m 20s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e87db0532f7e450c8e3190d4fbea4b55 : SUCCESS in 15m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/06fb895f347542cfa51e70f08cfbeba1 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0108f9c47f264a7f8a5300087ad10eae : SUCCESS in 16m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fe33ba2e49384df2803993e476c8d753 : SUCCESS in 19m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e7c41d9dc9e74d4c9830a5c2223dbce6 : SUCCESS in 13m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/024cdfab3e6b4beb9676a09c930f0e59 : SUCCESS in 36m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eee45405e6f64114ad02ac4774b2671a : SUCCESS in 17m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/88dd061b971b474da2e763dba5766a17 : SUCCESS in 7m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/407cea584ef14b1c8b9658da6642cf1d : SUCCESS in 16m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fdea155ff63f470094d3401694d6598c : SUCCESS in 6m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3f4d3d7c540d498da5956d0d299a3343 : SUCCESS in 1h 02m 18s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd1976c6a3e34170a6315b61811dbe2e : SUCCESS in 56m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/86d6ad7da008436694aa8722e1cb2f09 : SUCCESS in 47m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b5ed244091b647f88539bd56742178e9 : SUCCESS in 1h 00m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/932cefec04f14426af1948e5028febfd : SUCCESS in 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cae84c19c9d491396e14d3e97f31703 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e1b49a4a01d46d5b35186abda3807e1 : FAILURE in 18m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c95d8f91b5b94ab395e6e7e5a1b72b73 : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":6},{"id":"5f38f8b209c88da58bec4807795940421abd2003","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 17:41:53.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c57c9cf262e1e449e8305b50ce311810e1a2a289","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 19:46:33.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd2f4dc0b02d45cd96885e335b0270a7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9bb6398d2c6f43eaacb04092287a556a : TIMED_OUT in 1h 31m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3c0ce0303714717830d0309376a6c46 : TIMED_OUT in 1h 31m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0fea1e551612447e99e53d6337eab815 : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c6bfe281a934553afab769c775c7417 : TIMED_OUT in 1h 30m 57s","accounts_in_message":[],"_revision_number":6},{"id":"86c36d53972a02a4233579ddd0a22c88d32bea3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 21:01:11.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9c60bc084b3402eab30771bda50be32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f852ba8b25f44f30bf4d591b7902b2c3 : SUCCESS in 9m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/dbfbbc3b6cf142719e3e28543c81143f : SUCCESS in 51m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8774f23876f94991865a8413b4e7c187 : SUCCESS in 47m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8482062ec4f47b1a21831aca3172587 : SUCCESS in 1h 02m 29s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e6aa0ee8e9ae4663b2e4298c91072d2a : SUCCESS in 10m 06s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2f294475a99b4d1d8a08822b4577e243 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83eedd7698e84cbb8885f358de8ed4e9 : SUCCESS in 30m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0765d0347d664670b6786abcddd621c4 : SUCCESS in 15m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/435c842b18fa4953a00acdea51fa48e2 : SUCCESS in 28m 03s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e4331aa796514d85bd26a0a20c94fc80 : SUCCESS in 14m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4b887a6e89534a45a97a289627e85b00 : SUCCESS in 30m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bbf2e8ee32f34a4faee71c18c5f43a79 : SUCCESS in 16m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/782f06be9c5c4d47ab036fcbb54e0c7b : SUCCESS in 15m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/104a9cc54b344c818d7fd4c4ae4d10eb : SUCCESS in 13m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/75021f1b276b438fa67187f29d0fb6aa : SUCCESS in 16m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3c43850e65df4ee48308b315b043b28d : SUCCESS in 14m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7eb1ce0a84ab41798928f2610757d5d7 : SUCCESS in 7m 24s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/16396523c9994864bc513630a0a6b59d : SUCCESS in 7m 37s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/91cb247b6856484b95d2cacaf79088a9 : SUCCESS in 50m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d587e74c82e142c79fc8b02dcdcd118f : SUCCESS in 1h 03m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/300d1ce9a8fc4dff8bdb8952917de817 : SUCCESS in 47m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/71f97545b0524ef38fd4f4fc0d570099 : SUCCESS in 48m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f0a17e1a116408fa3413fd94a84b3eb : SUCCESS in 18m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cab66c20afa4fada750730343701605 : SUCCESS in 5m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/650fd9e0f6b5415a95a3c5d5329b9e34 : SUCCESS in 48m 41s","accounts_in_message":[],"_revision_number":6},{"id":"3602f78bb6d754507d30a286c32f50b256684c99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 21:01:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"fb412d7fc69425186a5634bb7010ff76297cb8ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 21:02:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27c2b3392c5340a0aca728b6b7568ce2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5acf72ac7453407fa6ee15dcfcaae1f5 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6a7c460772444084b282d94c34ea25a4 : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f2fdd74002f344219c1b6f8534eca55d : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"20b48a6900c8c53f009416256279ce8c89cecf27","revisions":{"3addd81d1d73df18638db9e6d60ec6792aaa3a7d":{"kind":"REWORK","_number":1,"created":"2022-07-26 22:13:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/00/851100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/851100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/851100/1"}}},"commit":{"parents":[{"commit":"5467a305a0b789e7283beb82888c1bfa4fe37893","subject":"CI: Add py310 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5467a305a0b789e7283beb82888c1bfa4fe37893"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-26 22:11:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-26 22:12:48.000000000","tz":-420},"subject":"Clean up a bunch of deprecation warnings","message":"Clean up a bunch of deprecation warnings\n\npytest still complains ome 20k warnings, but the vast majority are\nactually because of eventlet.\n\nChange-Id: If48cda4ae206266bb41a4065cd90c17cbac84b7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3addd81d1d73df18638db9e6d60ec6792aaa3a7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3addd81d1d73df18638db9e6d60ec6792aaa3a7d"}]},"branch":"refs/heads/master"},"0c26d430948f4213716c19f5ab9b511f54453fec":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-12-09 19:25:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/00/851100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/851100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/851100/2"}}},"commit":{"parents":[{"commit":"30102868eaadbe462e042c11e733ee3744319abc","subject":"CI: Add py310 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30102868eaadbe462e042c11e733ee3744319abc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-26 22:11:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-09 19:25:15.000000000","tz":-480},"subject":"Clean up a bunch of deprecation warnings","message":"Clean up a bunch of deprecation warnings\n\npytest still complains ome 20k warnings, but the vast majority are\nactually because of eventlet.\n\nChange-Id: If48cda4ae206266bb41a4065cd90c17cbac84b7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c26d430948f4213716c19f5ab9b511f54453fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c26d430948f4213716c19f5ab9b511f54453fec"}]},"branch":"refs/heads/master"},"401a6d6b0a051c9ec7f37e3592354ca1b9a4b202":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-12-09 19:38:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/00/851100/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/851100/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/851100/3"}}},"commit":{"parents":[{"commit":"223cb2e384f93bce058ebc68ccc921f9b03116e5","subject":"CI: Add py310 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/223cb2e384f93bce058ebc68ccc921f9b03116e5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-26 22:11:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-09 19:38:02.000000000","tz":-480},"subject":"Clean up a bunch of deprecation warnings","message":"Clean up a bunch of deprecation warnings\n\npytest still complains ome 20k warnings, but the vast majority are\nactually because of eventlet.\n\nChange-Id: If48cda4ae206266bb41a4065cd90c17cbac84b7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/401a6d6b0a051c9ec7f37e3592354ca1b9a4b202"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/401a6d6b0a051c9ec7f37e3592354ca1b9a4b202"}]},"branch":"refs/heads/master"},"cfd02fbdf945379f2e12e21061fd5ea9e1fa48aa":{"kind":"REWORK","_number":4,"created":"2022-12-13 19:57:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/00/851100/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/851100/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/851100/4"}}},"commit":{"parents":[{"commit":"2d0fea6a570cc4bf22525a3c680b052345003a46","subject":"CI: Add py310 unit test job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d0fea6a570cc4bf22525a3c680b052345003a46"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-26 22:11:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-13 19:57:05.000000000","tz":-480},"subject":"Clean up a bunch of deprecation warnings","message":"Clean up a bunch of deprecation warnings\n\npytest still complains ome 20k warnings, but the vast majority are\nactually because of eventlet.\n\nChange-Id: If48cda4ae206266bb41a4065cd90c17cbac84b7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfd02fbdf945379f2e12e21061fd5ea9e1fa48aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfd02fbdf945379f2e12e21061fd5ea9e1fa48aa"}]},"branch":"refs/heads/master"},"3fab0c925a98811aeb03838588b4ace90fcc982c":{"kind":"REWORK","_number":5,"created":"2022-12-19 20:18:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/00/851100/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/851100/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/851100/5"}}},"commit":{"parents":[{"commit":"9dc12a89029526f0ac8cd4aa30e121e4fdac1641","subject":"Merge \"CI: Add py310 unit test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9dc12a89029526f0ac8cd4aa30e121e4fdac1641"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-26 22:11:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-19 20:17:48.000000000","tz":-480},"subject":"Clean up a bunch of deprecation warnings","message":"Clean up a bunch of deprecation warnings\n\npytest still complains about some 20k warnings, but the vast majority\nare actually because of eventlet, and a lot of those will get cleaned up\nwhen upper-constraints picks up v0.33.2.\n\nChange-Id: If48cda4ae206266bb41a4065cd90c17cbac84b7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fab0c925a98811aeb03838588b4ace90fcc982c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fab0c925a98811aeb03838588b4ace90fcc982c"}]},"branch":"refs/heads/master"},"20b48a6900c8c53f009416256279ce8c89cecf27":{"kind":"REWORK","_number":6,"created":"2022-12-27 21:34:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/00/851100/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/851100/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/851100/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/851100/6"}}},"commit":{"parents":[{"commit":"9dc12a89029526f0ac8cd4aa30e121e4fdac1641","subject":"Merge \"CI: Add py310 unit test job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9dc12a89029526f0ac8cd4aa30e121e4fdac1641"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-26 22:11:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-12-27 21:34:00.000000000","tz":-480},"subject":"Clean up a bunch of deprecation warnings","message":"Clean up a bunch of deprecation warnings\n\npytest still complains about some 20k warnings, but the vast majority\nare actually because of eventlet, and a lot of those will get cleaned up\nwhen upper-constraints picks up v0.33.2.\n\nChange-Id: If48cda4ae206266bb41a4065cd90c17cbac84b7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20b48a6900c8c53f009416256279ce8c89cecf27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20b48a6900c8c53f009416256279ce8c89cecf27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[]}
