)]}'
{"id":"openstack%2Fswift~851474","triplet_id":"openstack%2Fswift~master~I7976e6bb151802dcf0c969e2468e686f2dfa78a7","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-05-20 21:41:28.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-07-29 18:44:24.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I7976e6bb151802dcf0c969e2468e686f2dfa78a7","subject":"replicator: Ensure handoffs can clear with large handoff_delete","status":"MERGED","created":"2022-07-28 22:45:05.000000000","updated":"2024-05-20 21:44:05.000000000","submitted":"2024-05-20 21:41:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851474","meta_rev_id":"71e9c5ad13e63364d9140ad4fcaa52e444871c88","_number":851474,"virtual_id_number":851474,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-20 21:41:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-20 19:44:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-20 19:44:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 23:18:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-28 23:52:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-29 16:07:54.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"eae873bbcd2351d385def2ec39f92c55930fb6f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-28 22:45:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e41218395504f585cee6217354379f6c4c46f64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 23:18:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f786bf2a39a446f3a7bbd8928cbad7a2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbe215084e9f4ff8a345bc22892270f3 : SUCCESS in 28m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96b42d9c5bc94e4dbff19db633f5c695 : SUCCESS in 28m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/49a15dc9b28b46b98a083146682dc575 : SUCCESS in 28m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d47122347c754843a5c1878aee658d6b : SUCCESS in 28m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b126d4406c0870d79a2ad66cfd057a8b898d6f03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 23:52:43.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d66c74b2b9ef4308acba6263c33b296b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cacba6e5f21d4ec589cd01cb1fe69686 : SUCCESS in 12m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/43c6f0046d9c4a73a8d204d9d684ce26 : SUCCESS in 51m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6f16e3ab7b4a42089fb52f2f6083fc70 : SUCCESS in 47m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2eae3a113d4d4a919dbca8cade9bfaca : SUCCESS in 50m 01s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b8b3d05b21214960b1fcc09e9cb86a26 : RETRY_LIMIT in 5m 12s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ae8bfd24dc7b4befa2da3a032dbf9e37 : RETRY_LIMIT in 5m 06s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/59802985692944fba5a8fe994d19d3ef : RETRY_LIMIT in 5m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f0239a0c8f91444084f8f3d6e5361a02 : FAILURE in 4m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/76eb4925311745308c95cadcf9c59ed9 : FAILURE in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/be35e3be5c0846048b773de658fc606d : SUCCESS in 27m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/658fa356523f4babbc81971b1e7c61f8 : SUCCESS in 27m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3799e72ad7f041af8541df032bc3284c : SUCCESS in 26m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/70f0757c33c04cf0ba32acad0a6db346 : SUCCESS in 29m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c99eb2d2aaae42f8bfa4ff8bb888e0e7 : SUCCESS in 28m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/13f192f0abe14e66a2eea572f7a299b3 : SUCCESS in 29m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/892db36d7bd4465cb4a0676c2def8bc5 : SUCCESS in 27m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cb4c7bbd86b44d32b0dcc090e7f3b9ea : SUCCESS in 17m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/36087ed271c746d984e98c4d80a9e35e : SUCCESS in 29m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dd80367c047f4812972186f81c0a660c : SUCCESS in 10m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/01325a9b1c8747d6885abe6ae209ee55 : SUCCESS in 27m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7c2721259fb947488f15f280ca9eff68 : SUCCESS in 11m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/85f865279cdf499b99457c183d4bf5bc : SUCCESS in 1h 00m 46s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca6abd1465f94805994768450c12516b : SUCCESS in 1h 06m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6f12fc5926e346cc8f64c0cc2d9b78d6 : SUCCESS in 48m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c33378f1fd35449fae01cb62e5b97ed0 : SUCCESS in 45m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebeb3ed651d6467197e85ce3cfdd2bc7 : SUCCESS in 27m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eccf3b80646544eeb10bf0e30ff29aa0 : FAILURE in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b5347671e3834cebbd15ffa588855872 : SUCCESS in 24m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d85b7aa7a0b4d909f40e908fa90f07b : SUCCESS in 52m 58s","accounts_in_message":[],"_revision_number":1},{"id":"366f73799e8fd4e070dad11b5598c9fd00fc1fa2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-28 23:59:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"78449065f83347fb9ea2d74168d252ed3eb2b4cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 01:06:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/217909265d854b60abdd86cece00e740\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/daf98c7d1b6a41baaa5b158a55d4a399 : SUCCESS in 42m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9bb1bfb0b2f48b9b2f22e2b53b75d3d : SUCCESS in 57m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/492fea02ed5c478aa592a5301e997de3 : SUCCESS in 43m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4a3377ff0624ecd8dbba5fc1567bcfb : SUCCESS in 1h 04m 55s","accounts_in_message":[],"_revision_number":2},{"id":"d811ccb1a3d1216fbd62171c75760f0683e6d7f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 01:09:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/489468667a5349d78d0e1b97fa2db4a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5053a1464d4e4cf68a017355886e3177 : SUCCESS in 7m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/a4676abf8e364d79b9657a9b338acc7d : SUCCESS in 50m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a79e7e0c5fc042599e81908f7af608bf : SUCCESS in 42m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/173b9101a44a4c3e9b194fc414d5a4ae : SUCCESS in 44m 35s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ca76b281f2ee4d86a6b71b41b3e3a914 : RETRY_LIMIT in 6m 59s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7a3f6a22f2254d1c8f0bc4c2ef563f03 : RETRY_LIMIT in 5m 12s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b1c2cba265a6476d9ac5795fba88bfa1 : RETRY_LIMIT in 5m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9009115694e441d09dae6cbb88870c39 : FAILURE in 10m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c935c2bc656a47b58d9e0b8f21bf0eb8 : FAILURE in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/42240bedc02243cbb823c4e2d2bfc944 : SUCCESS in 14m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/09bf4f4435824882b3291cf355e35fed : SUCCESS in 15m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/08080154c26e43eba2be0c5713809327 : SUCCESS in 16m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eabfbf573e4548f99d299c8f9a00b640 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1a3b95fc354540f5ae016076f251f450 : SUCCESS in 21m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b8f5bf395e47405aa13b80054df57bdf : SUCCESS in 15m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/639f443e2e8543bdaa8d3a94ed1035c3 : SUCCESS in 19m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/403c2bd51fc445deb6232b7236ab5505 : SUCCESS in 18m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e662d443205545e5ba5576b9c7f0b454 : SUCCESS in 13m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/359fd75be7ec49888d610e90431e0329 : SUCCESS in 13m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a93e0bc0f95848879e7bef6430ce9a20 : SUCCESS in 14m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ac2c1bdc97184fa6813c3dfd143386b7 : SUCCESS in 8m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c63ad92ed7414986b52e3705d9760c95 : SUCCESS in 53m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d22d00dbbbd04e68b1f51f8df667889b : SUCCESS in 1h 01m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5f690c4983034910a1592e51e6d3f319 : SUCCESS in 52m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c11ed45654b84859819b0677ad2d63bd : SUCCESS in 44m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b5471829e4c48248105334a9b3b383a : SUCCESS in 15m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/110482a984bf44d0824a629583cb52fc : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2830e9e7783342ba88ff887bf8bb82ae : SUCCESS in 30m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c896987018e64917ac9aeb0b1dd34bdb : SUCCESS in 55m 21s","accounts_in_message":[],"_revision_number":2},{"id":"80d473a96bf25501ee3f865b036b7fbbea1d0aeb","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-07-29 16:07:54.000000000","message":"Patch Set 2: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7c56253dff7f3b5bdab7d24261609f64a7123972","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-29 18:44:24.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cf0746f580e30e1468d5a868676855f7682a11a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-29 19:01:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a1028d8154d085822f105890ceac2096ad02eae7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 20:10:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27e4cfe6882a4005917f3102c0243cfe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6239ccc1e4c04c9ba20830e9527d24e6 : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/b6ecf06ca7c14d5287c03d4152ea7542 : SUCCESS in 1h 05m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7719f94160e847888a88137bf2dcbfa8 : SUCCESS in 44m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2fd2db4f63b4ed7a7122f7cce14a2f2 : SUCCESS in 50m 36s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/641bb12928c1418889d6a489c75cffbb : SUCCESS in 23m 38s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/074b0813b5f1444d93fbf2614969bcfc : SUCCESS in 22m 28s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b722c043db7a4a1c978553a0d353563f : SUCCESS in 22m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4fd12ea1dc184ce2a99ffdd430469e70 : FAILURE in 5m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/77dd65ac84a74958a2c9c33017c921fe : FAILURE in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/51dd80dd44544885a791cdd6851dad4a : SUCCESS in 19m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1957e86e5ad4451c98d3e52a36d57234 : SUCCESS in 16m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9bc2e99db9774e24a0d7377231a6190e : SUCCESS in 16m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a83fa984db7b460c9e94e097bb5e5a79 : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/28f811509353412db640ddd6437e50c5 : SUCCESS in 18m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/44378fe2a95e47828314f1b06a53a2d5 : SUCCESS in 22m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bb08527e08174c0fab4cfec4ca80e814 : SUCCESS in 19m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d9040e6f41204254bd357b56318f0c1f : SUCCESS in 17m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b08eb0a9f3224b18a58f94de8827eb9d : SUCCESS in 17m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b4aaa7570d1d4c90b74191bf6357ba58 : SUCCESS in 9m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/38801ff5083d4624a0d284a7beeb2147 : SUCCESS in 21m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c6ee12c4f2d0442cb9fe9fa5b92ffc95 : SUCCESS in 10m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c95a7ae721ce4ce9a0760f6508a65ccf : SUCCESS in 1h 06m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/73c86a4c8a42457284352bbad7e1d82d : SUCCESS in 1h 00m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2d8ab59960fd421885e7723e1bdfebb0 : SUCCESS in 45m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0dc6b4c02802478d8bcb0a91a247c2db : SUCCESS in 46m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5161c258dc6e4f24af514cb520aa897f : SUCCESS in 17m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96a6e8efdec244958a1d393cf9103b61 : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f6494f06954471592b8b40185033e25 : SUCCESS in 23m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08ed1a342fe74cf69fc1e4978b5feecb : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":3},{"id":"ba138590a61dc6e55eeda75d16b126661a7d4084","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 20:10:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/296188eead6f470985981d80b44ff5f1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7233d95a0c86468d89dfc9fbd9ff5501 : SUCCESS in 28m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22b5d297008d4efe9b4720be2710dfe0 : SUCCESS in 1h 07m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/22d9598ba0d7454c884af76f07eda113 : SUCCESS in 28m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d88ad9f07b541a4bb89477df8b5bf91 : SUCCESS in 28m 40s","accounts_in_message":[],"_revision_number":3},{"id":"0b83e6b38a772217021c4e04d6852400f5807e3b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-22 20:34:43.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b8c235ee0422fff04b3508f5738152d8c10f163d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-20 19:44:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4989b75b47b3f3a800ee0c654fdc6222bc223ee4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 19:44:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fa1f46ad2266e9251cfe6b26c1d2f62f1c533454","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 21:41:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07eb538d51ba4f7983ace37756291d5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ff2fdd863cb43289b5739cfdb223eeb : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/554ef43e87b54a7eb9be069a8c3b7ccc : SUCCESS in 1h 12m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da5843c889d242688c955dd37cfa62a6 : SUCCESS in 41m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4025c7771eb4c3b813a0f1cf00e5eb0 : SUCCESS in 57m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a4a917826f0b41c0bb68f876b3a716b9 : SUCCESS in 10m 36s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b93d089cef764db0a1c5ef51f6d77b75 : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/df47174bfae949b484ee508bbed96300 : SUCCESS in 17m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/674746f497b74916a4b2b90c8766113e : SUCCESS in 25m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c30a4dd0c5f04482bfa0a3747e565d77 : SUCCESS in 28m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cffaf9ebdf704c6983d1dafbeed3e7f6 : SUCCESS in 13m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9a6452313a0746ab8a8ab94d345b6508 : SUCCESS in 25m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e65871d14f274ae4a0151f2caa1d5d7f : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b41daff1639b40008e232eacba1952a9 : SUCCESS in 20m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b75c966a2d104bfc8820f294b62d2253 : SUCCESS in 17m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e38e18d0672044feb22caeb187ed4a67 : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2b79fc1662584966b3c22904693bee5a : SUCCESS in 15m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cabb6c8af9044ef9905d570ac9daffca : SUCCESS in 18m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5596f59c2cdd46ddbe0884cd0b26c0ec : SUCCESS in 7m 38s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/736791014e704daea5d4acaa96e3a9ff : SUCCESS in 7m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/17c8e992cb9143dbb33a366f5d070341 : SUCCESS in 56m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d9367d6330c4129bb0c34ab092d6807 : SUCCESS in 50m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/68a70d9fd363402d8f23013cc96c49c3 : SUCCESS in 35m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b217db2ce1745daa81961e356905d59 : SUCCESS in 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41b55c645e564d10a756ca575e470f11 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f8c288e11624cc98714088b914e0db2 : SUCCESS in 24m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e51e4a1b852143b1822529d1d289f709 : SUCCESS in 58m 39s","accounts_in_message":[],"_revision_number":3},{"id":"3ddf6a265c0c1f4b3c35926b605ea0eb134bef51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 21:41:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"71e9c5ad13e63364d9140ad4fcaa52e444871c88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 21:44:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d669f4ff27e942bca319c057e4267950\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f741c7846b4591923f102ea6537e63 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/84836ac4f15241fdb87b8efc1b5a83f4 : SUCCESS in 52s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0dc696c73ae74129971ef2ca5daee76c : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"336c97bbf8a577a5f74dacec4dd245663d626289","revisions":{"ce23bcb9877ed39dc3c71199e7d2f1af9b28ea06":{"kind":"REWORK","_number":1,"created":"2022-07-28 22:45:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/74/851474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/851474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/851474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/851474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/851474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/851474/1"}}},"commit":{"parents":[{"commit":"9b0e5ea9757a1a96c7207334cd80e3309fc6a7f6","subject":"Merge \"formpost: deprecate sha1 signatures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b0e5ea9757a1a96c7207334cd80e3309fc6a7f6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-28 22:33:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-28 22:44:26.000000000","tz":-420},"subject":"replicator: Ensure handoffs can clear with large handoff_delete","message":"replicator: Ensure handoffs can clear with large handoff_delete\n\nChange-Id: I7976e6bb151802dcf0c969e2468e686f2dfa78a7\nCloses-Bug: #1982201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce23bcb9877ed39dc3c71199e7d2f1af9b28ea06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce23bcb9877ed39dc3c71199e7d2f1af9b28ea06"}]},"branch":"refs/heads/master"},"8235feff4c74b2838ef47244f265911a42718e77":{"kind":"REWORK","_number":2,"created":"2022-07-28 23:59:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/74/851474/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/851474/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/851474/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/851474/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/851474/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/851474/2"}}},"commit":{"parents":[{"commit":"9b0e5ea9757a1a96c7207334cd80e3309fc6a7f6","subject":"Merge \"formpost: deprecate sha1 signatures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b0e5ea9757a1a96c7207334cd80e3309fc6a7f6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-28 22:33:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-28 23:58:58.000000000","tz":-420},"subject":"replicator: Ensure handoffs can clear with large handoff_delete","message":"replicator: Ensure handoffs can clear with large handoff_delete\n\nChange-Id: I7976e6bb151802dcf0c969e2468e686f2dfa78a7\nCloses-Bug: #1982201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8235feff4c74b2838ef47244f265911a42718e77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8235feff4c74b2838ef47244f265911a42718e77"}]},"branch":"refs/heads/master"},"336c97bbf8a577a5f74dacec4dd245663d626289":{"kind":"REWORK","_number":3,"created":"2022-07-29 19:01:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/74/851474/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/851474/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/851474/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/851474/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/851474/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/851474/3"}}},"commit":{"parents":[{"commit":"9b0e5ea9757a1a96c7207334cd80e3309fc6a7f6","subject":"Merge \"formpost: deprecate sha1 signatures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b0e5ea9757a1a96c7207334cd80e3309fc6a7f6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-28 22:33:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-29 19:00:46.000000000","tz":-420},"subject":"replicator: Ensure handoffs can clear with large handoff_delete","message":"replicator: Ensure handoffs can clear with large handoff_delete\n\nChange-Id: I7976e6bb151802dcf0c969e2468e686f2dfa78a7\nCloses-Bug: #1982201\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/336c97bbf8a577a5f74dacec4dd245663d626289"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/336c97bbf8a577a5f74dacec4dd245663d626289"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
