)]}'
{"id":"openstack%2Fswift~851598","triplet_id":"openstack%2Fswift~master~I7837a2ec7dee9a657e36147c208c524b5a01671d","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-08-10 15:18:27.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-08-10 16:35:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7837a2ec7dee9a657e36147c208c524b5a01671d","subject":"s3api: Be more consistent about CompleteMultipartUpload logging","status":"MERGED","created":"2022-07-29 17:31:11.000000000","updated":"2022-08-10 16:37:11.000000000","submitted":"2022-08-10 16:35:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"851598","meta_rev_id":"1fe7deb5373253ee0b3c527739af93b5a4284c53","_number":851598,"virtual_id_number":851598,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-10 16:35:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":2,"date":"2022-08-10 15:18:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":1,"date":"2022-08-10 15:18:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-29 18:17:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-29 18:36:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-01 21:37:40.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2022-08-02 10:42:32.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"ef41e4be75aa6b4878206e66258510a42f479479","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-07-29 17:31:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7a73735f5401abb3b645bd9d4e37700cc5ac3f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 18:17:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6baa4f442e20437ba829f5c930e1a426\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba30bb3f8774445fbc5667051a7724a6 : SUCCESS in 28m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc04d13090ad4e05a78eedb7a0f8eaee : SUCCESS in 44m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/838cd655160e4ff880c69cf9d0ca9431 : SUCCESS in 28m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/08cbbc517e544c04ac5e906142a6615e : SUCCESS in 27m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8d6b75d0d2226cd367a5e94a32493ed426798ea4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 18:36:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4cfc6b8cf234c5d899a88b92fec625b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c9f71d5b9fb404ca48e5b814bd1906a : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/836aa40bae0c4194823ec58260c674f8 : SUCCESS in 49m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e15522fa76db4c09a0c399d383dad67e : SUCCESS in 39m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3404c272f0f647f5b1d4aee795a9cf6f : SUCCESS in 44m 21s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/1fe824d122cf4285a4228ca070ade3c4 : SUCCESS in 17m 40s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c1e96c1f820041e7912c29500d457d22 : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9997d170f1eb4e92be543fd2bfdb26a1 : SUCCESS in 19m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/437620e1e8ba4ef2a6f260db8da86444 : FAILURE in 4m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/102e8c0e67374b2b8fc26651123a137f : FAILURE in 3m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2ab7a2327b824123b061e708a99edf01 : SUCCESS in 15m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/db23cf236e2e4ee1b7ceeabf113692e2 : SUCCESS in 14m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7e3da872220c454f8d83aa7a98e49844 : SUCCESS in 15m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d09befb1afc243ab98cbc47f3534d7c2 : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/25d94312392741028ac17f9fedf4b427 : SUCCESS in 15m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/74a05deecb0a4f3185b92aded8945e9d : SUCCESS in 17m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4e3bbd8d55c54398a13934f9e2bc41b9 : SUCCESS in 12m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/eaf7ae88cb6b476a93be40c62320451b : SUCCESS in 19m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/58524f52e9f549fb8d2966466ca69a2a : SUCCESS in 13m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1bfd4bccdfb448919787a3163379bcc2 : SUCCESS in 8m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fff2fd3f920a402da4bf31e8a9240869 : SUCCESS in 19m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f4b43f781eb9490fad136286a75030be : SUCCESS in 13m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/faa4adb858fc47a68e8f04f5209e6236 : SUCCESS in 52m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b32c2ac730345e584738dd67da4e58e : SUCCESS in 48m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bbf02c48742f4e678e51e06ca702ecdc : SUCCESS in 51m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97c08e4d9bc846af853d6bb6c4275252 : SUCCESS in 52m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f53400cc526d49bf9c3c225d3faff72a : SUCCESS in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60881a77a81e41b5b89179721c68c196 : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b9e4bfc46af74262a61b636a7146f590 : SUCCESS in 25m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cfe65ead99f4af0b150f0fd4142129b : SUCCESS in 54m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b36133ae9262323c8196e1d8a834d28fdf4fc713","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2022-08-01 21:37:40.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b40fc0a6008033f76d82ef1d30d22f2031cfb923","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-02 10:42:32.000000000","message":"Patch Set 1: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c024aaf3acbc5a398a0bff47da4b7c1ec5ebc318","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-02 10:46:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7cab30a4145f1f481e280897202b8940bbcb1788","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-03 00:33:33.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cb3ffc8e9f956563c0cbdad7c5f921079f76cd57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-03 00:34:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b6834767130e3191fac7ad411ae4d626d23924aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 01:11:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b175ec6a983e4c8c887c24863f7ea7a8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5a53491f5c84fa59d2188782dd6732b : SUCCESS in 28m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe48a7cae37d486083052993c6a41a35 : SUCCESS in 35m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0fd2f9dfb654fedb821ac8f02914e36 : SUCCESS in 29m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d93aa2e0e314e1abcef5a69eee0b9fb : SUCCESS in 27m 46s","accounts_in_message":[],"_revision_number":2},{"id":"905b898771fc15593703f508485bf5d02cc2171e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 01:48:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31a5ffa51b79423e84b919d25ca9923f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbeeb8d03ba94c45a45e3e217c9f4748 : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/bbca54497c7a4a28ae57c03dbd4b8c3b : SUCCESS in 1h 07m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/73dbb133326d43eab67901e626febb32 : SUCCESS in 50m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c49ac117889f45209afc5ae2301a3078 : SUCCESS in 58m 54s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/923e8f86e25b4e4088758a86ace3136e : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/81fa19a03c3348dc829bc8fff498eed3 : SUCCESS in 21m 19s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7e86a8c6698c427cb23f66046630ca20 : SUCCESS in 22m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0b19ae202bd947cea8229840728e80b8 : FAILURE in 7m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/563d3fe654ce407397d6de9351483146 : FAILURE in 4m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bda470fec96a490797897d000dc604e5 : SUCCESS in 11m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2321dfebbf51431b817c72ee974eb6b7 : SUCCESS in 16m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9c369caaa5e3465d89b6ecf4d1fd48af : SUCCESS in 15m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b88ae6a1930f491180ec1153f4655bc6 : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dfd34b345fd6402f89034a4e0c5cf9cf : SUCCESS in 18m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1409e13edb1f47fdbfd25319db690cb3 : SUCCESS in 18m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/12c431866d404b819c3f5748e3ffc066 : SUCCESS in 14m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2c4409ec9d284315824ba8a7226febc4 : SUCCESS in 14m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c5de425fb6db48ac9ee903bf21efe592 : SUCCESS in 18m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/95ec33687e794513baa307ef6a548f59 : SUCCESS in 7m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6644a38f2c04404092747d1ea9db2194 : SUCCESS in 17m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9e53ab255bbf4cc3a41e59e9543b89fd : SUCCESS in 9m 34s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3be58a4d3373436d97c7fcb3b289a67c : SUCCESS in 56m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef51fbcb30b44356a6f503a28c5dc1bc : SUCCESS in 1h 01m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4c2167b32fbd4e14988443658568241c : SUCCESS in 55m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f5732827af464ba18dc2319e169e208e : SUCCESS in 46m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aac25ce31c2349d6ad36dd822bcd9c1b : SUCCESS in 16m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/987bc7888d10443388d3c3bf45223ce6 : SUCCESS in 9m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/db2188bbfb0946b68de0dbbd29cdb2db : SUCCESS in 22m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3bed81dce3d4f70b5d5137476771af7 : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":2},{"id":"17fdb747a8d484c0e990ce4b17e8ff4139d10f39","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-03 08:42:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d3e36e4d8b1f81aee74fffd5d485627d9a68616a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-05 19:17:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"361556774d6541434a4a37d896d4b715f4b9d172","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-05 19:18:12.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"39cef10fd8015ff430197105684fe92963163383","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 19:59:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72f55fcb0d1740cb98ea301b2fa0b2a8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/42d36fdb0bfb4b78bd2f7d1785a59752 : SUCCESS in 29m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae255acf017a4130a71e09c6a7a7bf66 : SUCCESS in 28m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5710666aa8bf4d40bca63629d476a0e5 : SUCCESS in 33m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2e577f68f1024b868cfb14be614f01a5 : SUCCESS in 27m 47s","accounts_in_message":[],"_revision_number":3},{"id":"fb62526a89d1eee062873a9731638306fe88209e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 20:38:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85e2c9c8bbf640d2b565022fed423d0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/706e41487dd941f0959858167bdec7eb : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/9052667dce6648bcaa103f1063eb4df0 : SUCCESS in 55m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/67d0b89fd9664575816fda9a0e07bfdb : SUCCESS in 45m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01549f440e5a409192e1ed9a1282e577 : SUCCESS in 48m 22s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ca2c085f0c7c45789b6eb3b0481add59 : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/657958aa87a4430f9f50e53759a7ba17 : SUCCESS in 17m 33s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/2c4aa6b5faff4eb4b898c3e1881a3b1f : SUCCESS in 18m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2babada11a2d4828ad6dab79fe03e846 : FAILURE in 5m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1578e06caaed4fe8aa41a08aeb752785 : FAILURE in 4m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9299ed0373b14b8592d3fe99dc1fe45f : SUCCESS in 15m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dc834750e6e542cea1b03d7dbcc0ebfe : SUCCESS in 11m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/55a848c1e61548b786c75d83c1f6194f : SUCCESS in 13m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/41761bc6fbe84b0a9375c42254513dba : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/62122d667dbe43869d5e5dc81efbb813 : SUCCESS in 16m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/124a13b2eb19471a9c8d60d642b26279 : SUCCESS in 15m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/454bd309d89349eca3132f96a70db049 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5d506c1d68294bcc8c38d527b61c84fc : SUCCESS in 13m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2b72806120044b6ca3e150376e8eba69 : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2fd9a5144154436fbb8b1f155735a3a9 : SUCCESS in 9m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6d10d44199cc431ea002dad26cb2a23a : SUCCESS in 16m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/487dc4d00a37401db2765a9ce676fb45 : SUCCESS in 8m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2c1ea45e4bf64e69a25283671f167b68 : SUCCESS in 45m 16s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b00f0d0593504a9b9ff9055961fb4696 : SUCCESS in 48m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8f9633ca5a444edcbd3431dfe4a9a428 : SUCCESS in 35m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff0c46b69abc4ebc9d67596bdc2ea4df : SUCCESS in 32m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/440070313d3049fb835a653d34afb7da : SUCCESS in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43322429ffaa45f786bded4174f4ece7 : SUCCESS in 5m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5431a09a991e47638b017de098e740af : SUCCESS in 21m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/784e467107e4493da0c4b7135e843103 : SUCCESS in 45m 13s","accounts_in_message":[],"_revision_number":3},{"id":"943ddf3e1c612eab770ce05d4d70fb1c9355defe","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-10 15:18:27.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"14f0c099c490ec4c6e0038da9e074713e0f7812f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 15:19:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"da609c711d6e054ef1a1131e1ea57792b6e04101","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 16:35:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/344610d2b22c4cebb464c8a7bfb52851\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/561da91d7eb64a468946b3dd58582b81 : SUCCESS in 11m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/d983827f7cab4ef39b823219040ae343 : SUCCESS in 1h 07m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7391a0401d7747a0a4bfa3611151020a : SUCCESS in 47m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6bae003f32b45d69c610aad7ca451f3 : SUCCESS in 54m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/acf7e8e3982a4ecbba957b51c1504611 : POST_FAILURE in 7m 32s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/640f035a2e594827a37f725ed3010424 : POST_FAILURE in 7m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/967fb5b9f3a248bc95662d8178b66cdb : SUCCESS in 20m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/792d90603d7c491e8e51f024028ca74a : SUCCESS in 19m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/26cdf95bcc0b4b039ee7d4daa43933da : SUCCESS in 18m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/000fdc3d8b1b46a5bed2afd4b48feae4 : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cb5cea00480e421f98507f72971cb643 : SUCCESS in 18m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/26ea2d70d82f483f8e26163bb3c97983 : SUCCESS in 21m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8ff41ae42cb7479b84d1121ff96fc2dc : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/82616456967e48e2a9f7853e18e0424c : SUCCESS in 15m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ae72c1fd312c48028e4eb5bdb05d507e : SUCCESS in 23m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a1c3ddb603ea445097367e112f9c4f7f : SUCCESS in 11m 10s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/11ca0e82935c45f8a69282b2c5ac1572 : SUCCESS in 11m 44s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4912ab6d9c8b42d090f68f75e76429c4 : SUCCESS in 52m 05s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d36b071fce274502a66313c443329e83 : SUCCESS in 1h 05m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bbbaf040828f4c32b06909333a482e24 : SUCCESS in 50m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d816fb846cae4f5e91d32396d87b563a : SUCCESS in 44m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e770ceef8d5b4090be43e79baab83a91 : SUCCESS in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e862c521d10b4a4e9af589be6bf30092 : SUCCESS in 8m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/04797eb2b55b455589cc36139d7aa12d : SUCCESS in 32m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86e9088308c243aea66f2461cdab0705 : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":3},{"id":"47439fa73c756d826b9550c0e98dbbc352dfa788","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 16:35:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1fe7deb5373253ee0b3c527739af93b5a4284c53","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 16:37:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdf4a391e4324edf8449931b03f57ff3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c991c5b836db4f989daedecd83a1122b : SUCCESS in 1m 12s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9532c4dd68a14fc7b9abbb56f2448ee4 : FAILURE in 46s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1b55cc6897b042ea885b0a174f69508c : FAILURE in 41s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d24678dd58e23e90dc053797caa66941e43eadc3","revisions":{"b14a875deab3a8ae30ecb6f070620f1a66e8a8c4":{"kind":"REWORK","_number":1,"created":"2022-07-29 17:31:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/851598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/851598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/851598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/851598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/851598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/851598/1"}}},"commit":{"parents":[{"commit":"9b0e5ea9757a1a96c7207334cd80e3309fc6a7f6","subject":"Merge \"formpost: deprecate sha1 signatures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b0e5ea9757a1a96c7207334cd80e3309fc6a7f6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-29 17:28:40.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-29 17:30:48.000000000","tz":-420},"subject":"s3api: Be more consistent about CompleteMultipartUpload logging","message":"s3api: Be more consistent about CompleteMultipartUpload logging\n\nChange-Id: I7837a2ec7dee9a657e36147c208c524b5a01671d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b14a875deab3a8ae30ecb6f070620f1a66e8a8c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b14a875deab3a8ae30ecb6f070620f1a66e8a8c4"}]},"branch":"refs/heads/master"},"f488bb352c38ca95427745106400d03510081dae":{"kind":"REWORK","_number":2,"created":"2022-08-03 00:34:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/851598/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/851598/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/851598/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/851598/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/851598/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/851598/2"}}},"commit":{"parents":[{"commit":"9b0e5ea9757a1a96c7207334cd80e3309fc6a7f6","subject":"Merge \"formpost: deprecate sha1 signatures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b0e5ea9757a1a96c7207334cd80e3309fc6a7f6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-29 17:28:40.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-03 00:33:53.000000000","tz":-420},"subject":"s3api: Be more consistent about CompleteMultipartUpload logging","message":"s3api: Be more consistent about CompleteMultipartUpload logging\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I7837a2ec7dee9a657e36147c208c524b5a01671d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f488bb352c38ca95427745106400d03510081dae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f488bb352c38ca95427745106400d03510081dae"}]},"branch":"refs/heads/master"},"d24678dd58e23e90dc053797caa66941e43eadc3":{"kind":"REWORK","_number":3,"created":"2022-08-05 19:17:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/851598/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/851598/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/851598/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/851598/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/851598/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/851598/3"}}},"commit":{"parents":[{"commit":"9b0e5ea9757a1a96c7207334cd80e3309fc6a7f6","subject":"Merge \"formpost: deprecate sha1 signatures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b0e5ea9757a1a96c7207334cd80e3309fc6a7f6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-07-29 17:28:40.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-05 19:17:32.000000000","tz":-420},"subject":"s3api: Be more consistent about CompleteMultipartUpload logging","message":"s3api: Be more consistent about CompleteMultipartUpload logging\n\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I7837a2ec7dee9a657e36147c208c524b5a01671d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d24678dd58e23e90dc053797caa66941e43eadc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d24678dd58e23e90dc053797caa66941e43eadc3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[]}
