)]}'
{"id":"openstack%2Fswift~851788","triplet_id":"openstack%2Fswift~master~I130b44f0f17eddc5e3df4e4be2b886ba009b95b3","project":"openstack/swift","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I130b44f0f17eddc5e3df4e4be2b886ba009b95b3","subject":"Imported Translations from Zanata","status":"MERGED","created":"2022-08-02 04:22:22.000000000","updated":"2022-08-02 20:26:02.000000000","submitted":"2022-08-02 20:24:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851788-zanata/translations","meta_rev_id":"f7dcaed4a5a433d986b311bf806b9d10eff18084","_number":851788,"virtual_id_number":851788,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-02 20:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-02 15:48:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-02 15:48:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-02 04:34:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-02 15:48:49.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"47d6704ebca10513dfe5d827941ba4ffee34f093","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-02 04:22:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a3a73ac75291c45e43f46e053a3bb94cb37c4f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 04:34:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5d323b16d5e4f3286a621c98d313f34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a9f17e2aed40a1be787122c1aa99d4 : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b38101e34870416eb17d8477e68205c8 : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"cdf56c27184d1baafdb1eeab1afe9ce7cb8cdca6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-02 15:48:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca33974b1134208ee89b268990fd34bd766ed522","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 15:49:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d43c45a5a391ca23ffe9dd729818404de1cdcd08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 20:24:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8480d2fcbbd64f60a2027ce9df694965\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c809b5429de4804b171c926369c6dad : SUCCESS in 9m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3715fb9436e24153b668848a7d670d72 : SUCCESS in 4m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/90c2a9d949e945d7a2ff8efabc16ca88 : SUCCESS in 13m 41s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0a2c044f46084d218c74033a44a83895 : SUCCESS in 9m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4935cd6c119c4a92c5418074d0d5fcfb4428031e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 20:24:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f7dcaed4a5a433d986b311bf806b9d10eff18084","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 20:26:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1636ca26dbca451ea4fa9a96d59b4438\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/033756f598e7400b85f3442ecf23bcf3 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a58e23ffc54e4e6a994ffe109e9ba10b : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2eb4e51d40d44e4fb9b00d97bf078c0c : FAILURE in 26s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b7739ddc70fe4581829a4797be89ec4e : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"91cfb0d6d99ca6148495f5966c1b762a1579029b","revisions":{"91cfb0d6d99ca6148495f5966c1b762a1579029b":{"kind":"REWORK","_number":1,"created":"2022-08-02 04:22:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/88/851788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/851788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/851788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/851788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/851788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/851788/1"}}},"commit":{"parents":[{"commit":"f2841081989a8557eed1d800f316bde04ebbdf93","subject":"Merge \"py310: Fix formatdate() call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2841081989a8557eed1d800f316bde04ebbdf93"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-02 04:22:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-02 04:22:20.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I130b44f0f17eddc5e3df4e4be2b886ba009b95b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91cfb0d6d99ca6148495f5966c1b762a1579029b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91cfb0d6d99ca6148495f5966c1b762a1579029b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
