)]}'
{"id":"openstack%2Fswift~851934","triplet_id":"openstack%2Fswift~master~Ib1f8ef5e36825b9c241d2a4d838ea01b3df70da0","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-08-10 10:52:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib1f8ef5e36825b9c241d2a4d838ea01b3df70da0","subject":"Update \"Getting Started\" requirements","status":"MERGED","created":"2022-08-02 21:29:01.000000000","updated":"2022-08-10 10:54:31.000000000","submitted":"2022-08-10 10:52:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851934","meta_rev_id":"c883f3dbba82dad3eb02ac0ccddb72b7b319cc51","_number":851934,"virtual_id_number":851934,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-10 10:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2022-08-10 07:43:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-05 19:08:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-05 19:08:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-02 21:39:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-10 07:43:49.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"f2c2115b4030d0fac0198de519471726f692a0e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-02 21:29:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d1c649d2da1ad997b5af84ed47fcca28a22b703","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 21:39:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edd1dd753545424fa39ad56fd7ccfb22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff7ddfcce6db476d92783ee1e7684fe8 : SUCCESS in 9m 17s","accounts_in_message":[],"_revision_number":1},{"id":"81765893d2ac9e1e6073a9d4e78352f56a1fa2b7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-05 19:08:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cebe7c6ce17ba69148f7b76ad6ba3c8247b21c54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 19:09:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23012eeab43e2e2d4f7f3bc1c70c76a040364ec1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 20:12:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ac207ad88bf4200b400dab552cac699\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1f4f18739d94ea1965c4cfe82c7213c : SUCCESS in 11m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/80d521fdaa88484294fb1fce86fed626 : TIMED_OUT in 21m 29s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d0a5b4d919c24021912b3fb547097703 : SUCCESS in 22m 14s","accounts_in_message":[],"_revision_number":1},{"id":"995c69c24d3d9716a815d8e72939e9d5c0c1d5e6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-08-10 07:43:49.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b5e9296661d187da8f575a608fa65aff01f70bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 08:03:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58e88ca2d83048508c47ebeebbc634e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946b95ba93e1452cb8e598954cf2d91c : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":1},{"id":"12cd7a06fae99e2d2f7a6968ce164605ef8053f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 08:03:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad6e1515c9d1080aef55403364264ce259584efb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 10:52:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e8a62ab16ce46e9af77fe572f8305b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff5335a1aa984b03a6f1ce4952f12c7a : SUCCESS in 8m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/78df24136b1741ab8289d2cb13b924e8 : SUCCESS in 7m 15s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3c3511fc6a3541b689865acbc9c07f26 : SUCCESS in 9m 10s","accounts_in_message":[],"_revision_number":1},{"id":"71c06accea54c1b644e65a974a70245029ecd5ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 10:52:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c883f3dbba82dad3eb02ac0ccddb72b7b319cc51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 10:54:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/636097ac0e6042dda8657719b29d1aea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dfaf57fcf1e4f009ec83a4aa55654ff : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/738ac1d5de8841b08285d0d5f1555565 : FAILURE in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/807dc60218fb417f942c887aff3f397b : FAILURE in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e5c78423f5c5482bd36596b23c78e20ab9c40bf","revisions":{"7e5c78423f5c5482bd36596b23c78e20ab9c40bf":{"kind":"REWORK","_number":1,"created":"2022-08-02 21:29:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/34/851934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/851934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/851934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/851934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/851934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/851934/1"}}},"commit":{"parents":[{"commit":"91cfb0d6d99ca6148495f5966c1b762a1579029b","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91cfb0d6d99ca6148495f5966c1b762a1579029b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-02 00:13:25.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-02 21:24:30.000000000","tz":-420},"subject":"Update \"Getting Started\" requirements","message":"Update \"Getting Started\" requirements\n\n* Call out liberasurecode as a requirement\n* Include more py3 versions\n* Use anonymous links\n\nChange-Id: Ib1f8ef5e36825b9c241d2a4d838ea01b3df70da0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e5c78423f5c5482bd36596b23c78e20ab9c40bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e5c78423f5c5482bd36596b23c78e20ab9c40bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[]}
