)]}'
{"id":"openstack%2Fswift~852203","triplet_id":"openstack%2Fswift~master~I448350ec59a697650b6f99ddfa8ff87f6f2e6a81","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-08-24 21:19:10.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-08-24 21:19:10.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I448350ec59a697650b6f99ddfa8ff87f6f2e6a81","subject":"proxy: Minimize cached shard-ranges for updating","status":"NEW","created":"2022-08-04 18:11:22.000000000","updated":"2022-11-02 20:50:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":12,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"396cd99ba0dc62d93f747f3aadbc815e5fee6b8d","_number":852203,"virtual_id_number":852203,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-11-02 20:50:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 18:57:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-04 19:30:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-22 15:40:44.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"005a3fc5adae3de953a3efcb29ed39260f173eec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-04 18:11:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"545270ae2f315834e3a3fc3187719ae4486044ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 18:57:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc5dbf64144b4934bc9801765685c906\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/82a860b739b24c059fc2c969a82c138a : SUCCESS in 38m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82990f5451a3416c9c714cd26f77bda9 : SUCCESS in 37m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb2143e826a24bad861da14a6f1a1c71 : SUCCESS in 38m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e3ac8f10f774a519963495bb59b4d7c : SUCCESS in 37m 43s","accounts_in_message":[],"_revision_number":1},{"id":"604a53453cde3df8acc702b860a433d1cecab3a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 19:30:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2d6156f0ebe461381f2156cd3ab010a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1d7c25af94a4a5f94cbd4dbab6af7fd : SUCCESS in 7m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/4c27abf25d3d44198e8ebb9c3fa200a3 : SUCCESS in 47m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a255a1e240fc48b49b5a8c2547d0f7a1 : SUCCESS in 53m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfc14c997f5a4b29bffd3aa69651c502 : SUCCESS in 39m 46s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a5fae244033c4da7bc8f777ec01594e5 : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b4f19297cc07494c94403495db3e0a04 : SUCCESS in 20m 56s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9051dbd9b5914319a92d942f57ba5164 : SUCCESS in 20m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/faca5d0881034f489a4d6fda91d8dd8f : FAILURE in 3m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/51f73e1005804146b94da39b5bb87a06 : FAILURE in 4m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b41dc094130f4e9abe093c7dfbc5cae4 : SUCCESS in 14m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f4b7f6223ae848868e7f03db07ccc052 : SUCCESS in 17m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7186ca09661044dd82d3b40873af1200 : SUCCESS in 17m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/193756c0ea8b44c5bc9b2383d225e545 : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/83cac0f832104b8e8c3b9b69f3bed445 : SUCCESS in 17m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8b7c8859fadc43378e797e9595802e53 : SUCCESS in 19m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3d1b4d788dd14dbd9f9379b1f448e37b : SUCCESS in 15m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/590c2301d6264dea88d43bb6cc80a247 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2cf53f581a7141c7b962066f703e435d : SUCCESS in 15m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/37d8d79b13454516a6073e9b69d5cbdf : SUCCESS in 7m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7666dad2aaec4ed2b9b638c3e3a87a60 : SUCCESS in 17m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9744b4d7c0c74fcd851468906f6905fe : SUCCESS in 8m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ff9454ac0d364dcea2fbcfce92cbc93f : FAILURE in 1h 09m 21s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2dc3716886a84d1db3aa7c64ffa449ea : FAILURE in 49m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dfb626d76bcf4889b8c998848da827e1 : SUCCESS in 43m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e2f3ac9b02884f7a95309a666827ae87 : SUCCESS in 44m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/215d844828694da49b204eab14f467c1 : SUCCESS in 15m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d803417e275c4662b7fbcdfd8a4e3fe4 : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9639e0c6559049c98c64598c912d39f1 : SUCCESS in 24m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be05c3bf5c084f4cbde2cd334495c45f : SUCCESS in 43m 26s","accounts_in_message":[],"_revision_number":1},{"id":"0a83744198e1a3aa1092b93dcaf498db9f2c6e06","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-04 19:42:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aa60baeb46f7dc2ff5728739bcd4529752998099","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-04 23:37:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"42811cdb71626457b07587b9b3d84ea81e186cc4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-22 15:40:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0c5edc3ccb681525eb06f241e7e589a7e2315508","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-24 20:52:18.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ec4f8c5e262ed080d7de212e9b1696132e0d6666","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-24 21:19:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"537110f372d5cf3cda12bcd9046e8e6409f1a31c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-02 19:13:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3809ad3c9cf72fdaa4ef8d09e8f5ff553441d081","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 20:15:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feda36a550ed469e9780ee340d3a042a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ddf37d7c103041dfa1ce6c775a83ad2f : SUCCESS in 58m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa706deb4c6d4f3da2038b59d1caec53 : SUCCESS in 28m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fca3ffc8975c4e319630a85ae64b0fb2 : SUCCESS in 28m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/537f107148714835b38131f62c725944 : SUCCESS in 27m 22s","accounts_in_message":[],"_revision_number":2},{"id":"396cd99ba0dc62d93f747f3aadbc815e5fee6b8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 20:50:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77c0432568584d6bb7551efb5b0400b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d673dbd415594d91b7ea6772d1c7895a : SUCCESS in 7m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/55f25eaa9d1e448c8779e946883b65d1 : SUCCESS in 1h 01m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3c6317efa80468bab9a8fac0cd68e83 : SUCCESS in 31m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d72cdc82c01499883dc39b76bc70717 : SUCCESS in 59m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b6a723d2b794ef1a1cae2ea76c87cfd : FAILURE in 18m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/493690b4671949148c4b584f99bc75db : FAILURE in 19m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ff96ff7f02f14fcba9196933a4e42ce6 : FAILURE in 28m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dce4629ef68c4f8bb47f11dd38cd61f7 : FAILURE in 3m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7893a89ead0148faa7be086e0db452a7 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5fd764beb493460696b747e5aaf63d59 : SUCCESS in 11m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/66d52236f4e148d38c964290c04aa1dc : SUCCESS in 10m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5279b80e96d14e0f90cdd6718d4a3e79 : SUCCESS in 12m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4d51d2cc60864fad8dc1cfd987823139 : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8d193900f61447689fa4a5e2b0bcd424 : SUCCESS in 16m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f198caca99bb4447aa5c1df74ecc936e : SUCCESS in 17m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/14e9270d98f64af49c6d7185e73c1e60 : SUCCESS in 15m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/507e519acd944d1c92cbfef1f9a065a3 : SUCCESS in 17m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dea613765fbb43de8347f069a63a376b : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e77c5ccdc546431fb0b5358d811be7a8 : SUCCESS in 5m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f2fc9a9b98d24771b80d3ed5be93369d : SUCCESS in 16m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d29eca8d8432486eb7a9a4c54877a7e6 : SUCCESS in 6m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9d8ec3564cc44e7dbb1d847279489cbe : SUCCESS in 1h 03m 59s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1f9479017eb241889d82d3986c8236c3 : SUCCESS in 1h 06m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/48ba99632f464a318b0c39ae02111a88 : SUCCESS in 41m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/91b6ae38544d4e66b371694b3a3174d3 : SUCCESS in 28m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/204296bf1c02461d9ca0d601f60210f3 : SUCCESS in 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15d879a7cbcf4cabbc80bcb85ab6c12a : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf1cf25b34c5401e9301a46ca7c45901 : SUCCESS in 18m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d3ad3f209764611beb324cc70cbd908 : SUCCESS in 38m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1ddd4742cf9eed991bee0f5fca942f42d746ef48","revisions":{"824f132933d32903acd5c2733865c23dccdf01b6":{"kind":"REWORK","_number":1,"created":"2022-08-04 18:11:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/852203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/852203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/852203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/852203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/852203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/852203/1"}}},"commit":{"parents":[{"commit":"3a7b89506d4b1d5ec28c28c4edf0cfeaccf19dc6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a7b89506d4b1d5ec28c28c4edf0cfeaccf19dc6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-04 18:07:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-04 18:07:45.000000000","tz":-420},"subject":"proxy: Minimize cached shard-ranges for updating","message":"proxy: Minimize cached shard-ranges for updating\n\nWe don\u0027t need to be dumping/loading whole shard ranges, just the name\nand bounds. Hold off on instantiating the actual ShardRange object until\nthe last possible moment.\n\nChange-Id: I448350ec59a697650b6f99ddfa8ff87f6f2e6a81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/824f132933d32903acd5c2733865c23dccdf01b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/824f132933d32903acd5c2733865c23dccdf01b6"}]},"branch":"refs/heads/master"},"1ddd4742cf9eed991bee0f5fca942f42d746ef48":{"kind":"REWORK","_number":2,"created":"2022-11-02 19:13:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/03/852203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/03/852203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/852203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/852203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/03/852203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/03/852203/2"}}},"commit":{"parents":[{"commit":"62a15d885cbeb19a1b36ad781d9b7121fcd58e21","subject":"container: Stop returning sharding ranges when already covered","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62a15d885cbeb19a1b36ad781d9b7121fcd58e21"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-08-04 18:07:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-11-02 19:11:37.000000000","tz":-420},"subject":"proxy: Minimize cached shard-ranges for updating","message":"proxy: Minimize cached shard-ranges for updating\n\nWe don\u0027t need to be dumping/loading whole shard ranges, just the name\nand bounds. Hold off on instantiating the actual ShardRange object until\nthe last possible moment.\n\nChange-Id: I448350ec59a697650b6f99ddfa8ff87f6f2e6a81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ddd4742cf9eed991bee0f5fca942f42d746ef48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ddd4742cf9eed991bee0f5fca942f42d746ef48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
