)]}'
{"id":"openstack%2Fswift~852430","triplet_id":"openstack%2Fswift~master~Ic9ffcc042399f3834027a7935b64292d1fffe8d5","project":"openstack/swift","branch":"master","topic":"p-smsr-merge","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2022-08-31 06:39:52.000000000","reason":"removed on reply"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2022-09-09 05:45:58.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2022-09-09 05:45:58.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-09-09 05:45:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic9ffcc042399f3834027a7935b64292d1fffe8d5","subject":"swift-manage-shard-ranges: add \u0027merge\u0027 subcommand","status":"MERGED","created":"2022-08-08 13:56:14.000000000","updated":"2022-09-09 05:46:58.000000000","submitted":"2022-09-09 05:45:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"852430-p-smsr-merge","meta_rev_id":"6a86540ba5116edb377f45ea3434f9291c8c937f","_number":852430,"virtual_id_number":852430,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-09 05:45:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2022-08-31 06:39:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2022-09-09 04:37:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-08 14:45:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-08 15:13:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-08 15:13:48.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2022-08-15 05:12:40.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2022-08-29 04:16:19.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"157dde422ed143cc6ba685b51056ec3b867e36c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-08 13:56:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10e19562a8b839c54e8d846341446ff238e0a16d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-08 14:36:32.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"1dcfc8a8e109b14286b909010eed3c863e28a17c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 14:45:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ac52d3905eb48a091bfad41529cb78d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5da2d4802424e79b07af66d06dfe589 : SUCCESS in 30m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ca4c332387247789a9e499d6244898c : SUCCESS in 36m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c893882d9864a36ba5608d0d1b1af2d : SUCCESS in 30m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d44489201cc4e21a921a685aed170dd : SUCCESS in 47m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d5e8d0d42ed3f5ff589f5ffc1933f2640a4929c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 15:13:35.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ada73cb7354486497ec9d368c1b70a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e03b71a61704749a1b30acb66abe61c : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/b990e0ebc7d843a4838a72d0534a78c7 : SUCCESS in 1h 07m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8f5e01418d4849d4ab1d695f420accf3 : SUCCESS in 53m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/68c220984aeb4d19b5d34a0280372712 : SUCCESS in 57m 11s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5d45c97483564d9891e4c6f68ff8a64f : SUCCESS in 21m 27s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c4112b4c21554753b757732a5b75417e : SUCCESS in 20m 35s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/4248bafdab674f6188bd7d6d8b78414c : SUCCESS in 22m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ca021aa978d4519a48d694ca1a3761b : FAILURE in 4m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/de7b5f58e88c4ba9885b6f6e6718dd31 : FAILURE in 4m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cff73e47cc544936898a6e6ecccf8d89 : SUCCESS in 16m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/21a31a0158844d6c97a8877606731dfa : SUCCESS in 16m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/528af97cddc947caa9d11dda7e0f418e : SUCCESS in 15m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/251481fedd764b3ba7e60dabda6d5abb : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cc015e535058467aa052fad0c5208e1f : SUCCESS in 19m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4aa0408ab291440e8a13401c291c5118 : SUCCESS in 18m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a2503489df9d4dd69a8fe94066267ce5 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/224df96301364bbf89d5254d15dd8757 : SUCCESS in 16m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6d31bff40e4543d8a10fb2ab6bd433a4 : SUCCESS in 17m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ac3096b9621348a1b217f3bc44a7eefb : SUCCESS in 9m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ae1c7672be34d2da3c983005f61ba8e : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bdfceb4c8f2643c0a03e24bf6d3cbc91 : SUCCESS in 8m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7681e0b0e9e4477099cea901c6deb41a : SUCCESS in 52m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c7941b16f594871bf618533a8fab106 : SUCCESS in 1h 02m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/39d1b4b7653e4d11bc9be0a579903cfa : SUCCESS in 51m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/105163d899ae41f893c90630d5bb6409 : SUCCESS in 50m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e9c92bf73e548ea8de6afe80c7b2d07 : SUCCESS in 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c34af98957b4148beb617df52332a2a : FAILURE in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ba14bf4c60da4ef796894295a8159b89 : SUCCESS in 22m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/887f9b994e6244328f0f97749cbcc358 : SUCCESS in 56m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8921f65d8d3fe4d1f0dc6dad2bbf154887121b21","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-08-08 15:13:48.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d327a6879a36ac6630d322b0e00a80ba932b21ac","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-08-15 05:12:40.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"457c006fc4c98798905c52e4bb1cd03c1ce42a7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-19 14:43:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"50f60fccf1488b27f1853f7cbe99830fadc5c49e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-19 14:45:30.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d203ae098789b2fa5227cdaa440ba7b947d2ce2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-19 14:47:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d7a10440d36c2df7ec4e4c0ce9975f8fc13d2755","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-19 14:53:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"33ea23dcb973c0974c930f6a9c3c7be188deb08e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 15:39:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c62c0e5a618a4d41874da6b19050a12f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/10d3fda0549845268c9499517a0a4e97 : SUCCESS in 34m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91557e9d32af42118c178b0631532f7f : SUCCESS in 27m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac8377d7c480437faf87863c16194f4d : SUCCESS in 39m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a99ce0f547454a56b8e70b43b7a35fb1 : SUCCESS in 43m 41s","accounts_in_message":[],"_revision_number":3},{"id":"030610e1c86e066b9b5005f4df02b9811edf9593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 16:06:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abcc92cb6b064c41b1b34c3778a41177\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eba3bdffc564420584c68cd2193ae448 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/7973ca2f5c274af2ac9cd86944ec15b4 : SUCCESS in 1h 06m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/32d788eac30245f9be81213bd68cd72c : SUCCESS in 59m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a101d6d15b784433aa2c2e80ff38daa9 : SUCCESS in 1h 04m 53s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/a0be33b138094def809bbe89b52a7135 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9d331fa1569549828f2a0794728db0ae : SUCCESS in 21m 47s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/f4988999acdc460abb1796e9ee3b3597 : SUCCESS in 23m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9827a405a9164d3fb8c29d6220512e07 : SUCCESS in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1e5ef0435f184ddb8aa0865f53aed141 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38eccc460fa04ad4b4fc00a31c94e81f : SUCCESS in 16m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/67172c10acb340079f65ae70a0e77f03 : SUCCESS in 15m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/499426dc3d444ff7ab70f0d0a5dd3fab : SUCCESS in 15m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/76edbe14081b4fe0947b350394ecd904 : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c38ec73c6a534d0ea81b8534f22dcbcc : SUCCESS in 17m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/82f538a32124473395b6b1d27c4a986d : SUCCESS in 17m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3b831f611704497f8ae68466626b2521 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d6b8f944cc5f4eb6b6a132ae84c84ecb : SUCCESS in 16m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/229cd0e320124d7a9ae0bd3fa626a071 : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/87a4d0dc1a4f4b2c8c5b8efac1901b5b : SUCCESS in 8m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/298e5c08ea42422fbb6ccd4868081809 : SUCCESS in 22m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/50b49f0e5cb747e8807e2c3d7b94ee17 : SUCCESS in 8m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1faddd1eaef2453886f537b0fb2a6bb4 : SUCCESS in 56m 26s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d296c331432b41dba9a430dca3f3f33f : SUCCESS in 1h 04m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d08dc041f471471c8dad21fead526466 : SUCCESS in 47m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/76a156a689714fa396963b82d28efef4 : SUCCESS in 49m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f21a15e70c324b3b9b0db65d435b961b : SUCCESS in 21m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e430c1d63e146b38540be4f6f8e697c : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c2b42b3f4bef4e58a5db3b5f5de45780 : SUCCESS in 24m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65883d311113403390ce3fbdf7cc2f07 : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":3},{"id":"21d489ef56b394d09b6ef5a1620e2a25cbd5f548","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-24 15:08:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d97d9689d120f691abda7f4a3fd4d8d7865d32d7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-24 15:08:48.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"bb74e3c355a1a5ff20c88718b60fd007c1d9a723","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 16:10:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff083b19c1474a7f9f8c568c80912c8a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e875e783c9aa479380289cde0ca64745 : SUCCESS in 45m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2400672ea6e04a22985e77e1285bc55e : SUCCESS in 1h 00m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3221e580b79543468f2dfb027cfd70f5 : SUCCESS in 55m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a50888ed76824adbb25f42d624922ce7 : SUCCESS in 45m 14s","accounts_in_message":[],"_revision_number":4},{"id":"e70d6a5cc78f4195e4eb9dc960790da65cfd5d41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 16:20:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d0a1543bc684117a246c2b6661f4c83\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e58186bfe2b4a269abcd02031c21ac0 : SUCCESS in 9m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/e2df1e7b9fc149739b8b31252175828b : SUCCESS in 1h 00m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4d4eb796ad2a4bf3a4d016ffb5e1559b : SUCCESS in 55m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88582f39c6ac43f79f0203ced81dc939 : SUCCESS in 53m 20s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ca4a530af4924876b2c71d482637c64c : SUCCESS in 24m 04s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5570ae7b6a4f4711ba16d77f788c3863 : SUCCESS in 25m 16s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/abb1bdcac634465baf845174dabfa229 : SUCCESS in 25m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/30afd62d227549a2bcc9b8c24ded4bfd : SUCCESS in 9m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/36b223a745e145f4ae5e8884e5fd1cf1 : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e5f5b1d04a044a3998629f97cadda5df : SUCCESS in 18m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/afa34e7ffe3646f0a0c26e57d1220c72 : SUCCESS in 14m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cec122ea500a4b65ae9250e075d65808 : SUCCESS in 16m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ad0bbd40e914896a142a278af7adb10 : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8e05e6ac3fbb41c7984dbff885961ed5 : SUCCESS in 19m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eb9c60dc44cd4620a115c07480ca09dd : SUCCESS in 23m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fc6f36bf768c4a54a6ccb52b35150dcf : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9b7e4d70433c49afb96efc560c2e433a : SUCCESS in 17m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eaa0126dc04644f7b67a23145c668c83 : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bfccab831fa643a594d18a4013e651e2 : SUCCESS in 8m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e800ab5bccd64b06a87cad93937ccc94 : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ab48e779f51740089dca8b0ae8db6dce : SUCCESS in 14m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c176501561834c9fb0d171e67d6d90fb : SUCCESS in 50m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/59e632a9cda64db9b13c9511f128cd62 : SUCCESS in 1h 00m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd3c73733bbe4ae69fe3fbb5d9ffad65 : SUCCESS in 40m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0286b59f703b468d83c200669ed7036f : SUCCESS in 40m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c621fd2ae8634c23a87dc15d37cd9ea3 : SUCCESS in 21m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7046d04aae3492694ed6c5714b4030a : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9cd8a1032d214082befdcaee8b9e6eee : SUCCESS in 25m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/384b5f6032e34c35911debc5f6b1eb3b : SUCCESS in 1h 07m 52s","accounts_in_message":[],"_revision_number":4},{"id":"788259294b232fc722a912fc1adc1b8202615546","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-08-29 04:16:19.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ff26c20920df8291b637c3258c4fe81c990be06d","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-08-29 09:30:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"de2f9d70a6200cdeb82569f5735ef8b411946ab5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-30 12:00:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f2614fe0b6ba97da3fb0dc127188f35ed0c1141f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2022-08-30 12:01:52.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"05a10a4ea1ecc0bd6b2ce006cbe11cd2a9789707","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 12:56:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de8785e86217417ca2f328dd02d8406a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3381f74ee30417890f50be79a697d58 : SUCCESS in 27m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d8b28b980f4049279ca31cde27c90b83 : SUCCESS in 53m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/133b33dc0abd4df49306df9c61a2b64c : SUCCESS in 27m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ddc8b363b69040a89e517a5e26deefa8 : SUCCESS in 26m 37s","accounts_in_message":[],"_revision_number":5},{"id":"16fee51501b0d77f41e1e3f8413e8bf8c58d658d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 13:08:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39c814b2dddd49a396776586c83bbac4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b0dea8f6f241b98d3716a8e65d7b44 : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/4e47c238844f4a15b6bc0a2360e3037e : SUCCESS in 1h 00m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ab8f29df8f164401b52499bc836f114d : SUCCESS in 54m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1c3039ca6894e6db350065414c1e438 : SUCCESS in 1h 01m 07s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/53871928508c4717bf86c53f5e7e28d4 : SUCCESS in 23m 54s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/5ecd77bd4e29403c8333a8d8f000647a : SUCCESS in 23m 28s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/12371cd4b0e74d6597c7255c5a68590a : SUCCESS in 24m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f2ac521001664d3d8fbf4c9cbd169378 : SUCCESS in 9m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/26f19af478824f7a90e74765f36ad785 : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/762044d2c7104256aaf27ac717ca3f05 : SUCCESS in 16m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f3107802a0244fe9ba8ccd14e24edcf : SUCCESS in 16m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3c94d88ec3484cf49d707c83c5c6f43c : SUCCESS in 19m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0fb0159d318747bfaf49c00c3e1fec49 : SUCCESS in 19m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4d640a49340e4e0b9c7b7faad5858644 : SUCCESS in 19m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/218bb74e8c0d402892e555b309f3f2b3 : SUCCESS in 21m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/edc988bc846e4549a14b512488ffa6b9 : SUCCESS in 19m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b292e5d084fe4b99b4a9b4583fa38a97 : SUCCESS in 18m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2b831636be4640bb8c2096dd58c2f7fc : SUCCESS in 20m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ceaf1f7456e94d28b04b3f444000c9f4 : SUCCESS in 8m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d05fdc85f0224e959be6e766291985b9 : SUCCESS in 28m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7060239f968f41dda759a2db375995cf : SUCCESS in 10m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/036b248d284a4fcc8a71b0cb43ff944c : SUCCESS in 1h 04m 51s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/80250e486fe4457b91f44831121d2fed : SUCCESS in 51m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/727c1c184e2b48a5bd95fbe25d58134a : SUCCESS in 55m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/231a4172b64c479b814cfdd629c71936 : SUCCESS in 44m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5faaaddc8d974159bade40f8973bfbce : SUCCESS in 24m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9c4382cb5a448c8aa5958b2b75c0da7 : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c07944c13068488b862c8924a11e2879 : SUCCESS in 30m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb784d78b3e943328bc221447f02b93c : SUCCESS in 1h 01m 24s","accounts_in_message":[],"_revision_number":5},{"id":"9637db6907e4b91aa85b2f3cc311d230f9d72da4","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-08-31 06:39:52.000000000","message":"Patch Set 5: Code-Review+2\n\nNice, a handy tool to have in our pocket next time we need to manually insert some shard ranges.","accounts_in_message":[],"_revision_number":5},{"id":"9802c6d6a11557a413eb82ba8c869c0ff8d05173","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2022-09-09 04:37:33.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a8aea8a9f90538c9d0f64ccfb83fcaf3778503ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 04:38:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5eb0cda10282294d6064acf4840e0bd9095038c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 05:45:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9321124a8894c8a8956db57305d2629\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf7ed116cecf4fb19d5a6357ee3c0ed1 : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/36e1999a2a054927a286c98e71a6df73 : SUCCESS in 46m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/748b95b126244412a21db6bd2f4f835b : SUCCESS in 39m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84a5cbbc5b23452e8505f5b0616307eb : SUCCESS in 59m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2a0ff70448734f90ad8676955cd72a32 : SUCCESS in 8m 55s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1ab576e76fb94609b5fb5e6a3fd8f56a : SUCCESS in 6m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/17815d1e45aa4070a5a3b8f3f7e7208a : SUCCESS in 20m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7f844a5f7cdf44448080fe0f4fec303e : SUCCESS in 17m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9a95b604e88a4097ad17c25b12fb8ca7 : SUCCESS in 18m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cde80e3e9f684170a582789c7a415ac7 : SUCCESS in 21m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5f964b54e7304aa791148e1e9516d2b2 : SUCCESS in 21m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b5e6c6e13f5f481d83789363f5108d4f : SUCCESS in 22m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9fc4f35de33b4003a7adfb1c677697e5 : SUCCESS in 19m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0eee696edf8141ea8c248eb99577b5c8 : SUCCESS in 19m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4f9baa749c9745f7bdab9b228d7ce488 : SUCCESS in 19m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6ac009c687c64375a667756a86fe0a6c : SUCCESS in 8m 56s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/67c547f035eb4eba826a464bd7fc299f : SUCCESS in 9m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3298179f6c444e55a8d0ff0f712d0eb6 : SUCCESS in 55m 14s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a6d63f1ea0f744979b76d72019dca5fc : SUCCESS in 1h 07m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/34778bf428a144c391eb5cc617eb99c5 : SUCCESS in 45m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/82394b7b67a5481382f28a0881549f47 : SUCCESS in 44m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d74ec94f8a9e4541a411cfbe14b8d4a7 : SUCCESS in 20m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/578e80152a474f8ab31cd426a716d288 : SUCCESS in 7m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c90838e81d044fecbe245a8a0168cbc1 : SUCCESS in 30m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4d4541ae7544c258cf144188dd96525 : SUCCESS in 47m 37s","accounts_in_message":[],"_revision_number":5},{"id":"a058d7f8db0aa8ed941a8636912a381a7d69b32f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 05:45:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"6a86540ba5116edb377f45ea3434f9291c8c937f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 05:46:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e286d065ef74bf5ac0c20a835129534\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2ccd188f84541fe99b244d77f1fd9b6 : SUCCESS in 40s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/203f64dfdf2a4d0b82805b35186218d2 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/a8ab7db0e93b45b997718631993b01b5 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d2edf6646d670597b247ec3cf57b824a8469fa00","revisions":{"6b6a7695e471a79e5021d855fd952d30a008d2b8":{"kind":"REWORK","_number":1,"created":"2022-08-08 13:56:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/30/852430/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/852430/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/852430/1"}}},"commit":{"parents":[{"commit":"0c8c36b3fa51a3d8d4f06973f9fdd92e812c138e","subject":"Merge \"ring-builder: Better format large weights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c8c36b3fa51a3d8d4f06973f9fdd92e812c138e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-08 13:54:29.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-08 13:54:29.000000000","tz":60},"subject":"swift-manage-shard-ranges: add \u0027merge\u0027 subcommand","message":"swift-manage-shard-ranges: add \u0027merge\u0027 subcommand\n\nAdds a subcommand to swift-manage-shard-ranges that allows arbitrary\nshard ranges to be read from a file and merge into a container DB. The\nfile should be a JSON serialized list of dicts.\n\nChange-Id: Ic9ffcc042399f3834027a7935b64292d1fffe8d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b6a7695e471a79e5021d855fd952d30a008d2b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b6a7695e471a79e5021d855fd952d30a008d2b8"}]},"branch":"refs/heads/master"},"0683d7bb462251791e0893d6266f88a1dd137555":{"kind":"REWORK","_number":2,"created":"2022-08-19 14:43:48.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/30/852430/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/852430/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/852430/2"}}},"commit":{"parents":[{"commit":"0c8c36b3fa51a3d8d4f06973f9fdd92e812c138e","subject":"Merge \"ring-builder: Better format large weights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c8c36b3fa51a3d8d4f06973f9fdd92e812c138e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-08 13:54:29.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-19 14:43:39.000000000","tz":60},"subject":"swift-manage-shard-ranges: add \u0027merge\u0027 subcommand","message":"swift-manage-shard-ranges: add \u0027merge\u0027 subcommand\n\nAdds a subcommand to swift-manage-shard-ranges that allows arbitrary\nshard ranges to be read from a file and merge into a container DB. The\nfile should be a JSON serialized list of dicts.\n\nChange-Id: Ic9ffcc042399f3834027a7935b64292d1fffe8d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0683d7bb462251791e0893d6266f88a1dd137555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0683d7bb462251791e0893d6266f88a1dd137555"}]},"branch":"refs/heads/master"},"cbbc7003b16d7bb0864390e416e10ca11effa978":{"kind":"REWORK","_number":3,"created":"2022-08-19 14:53:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/30/852430/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/852430/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/852430/3"}}},"commit":{"parents":[{"commit":"0c8c36b3fa51a3d8d4f06973f9fdd92e812c138e","subject":"Merge \"ring-builder: Better format large weights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c8c36b3fa51a3d8d4f06973f9fdd92e812c138e"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-08 13:54:29.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-19 14:53:43.000000000","tz":60},"subject":"swift-manage-shard-ranges: add \u0027merge\u0027 subcommand","message":"swift-manage-shard-ranges: add \u0027merge\u0027 subcommand\n\nAdds a subcommand to swift-manage-shard-ranges that allows arbitrary\nshard ranges to be read from a file and merge into a container DB. The\nfile should be a JSON serialized list of dicts.\n\nChange-Id: Ic9ffcc042399f3834027a7935b64292d1fffe8d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbbc7003b16d7bb0864390e416e10ca11effa978"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbbc7003b16d7bb0864390e416e10ca11effa978"}]},"branch":"refs/heads/master"},"854bd5f051d442b4e65fdb99fb6382f7a9da78d1":{"kind":"REWORK","_number":4,"created":"2022-08-24 15:08:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/30/852430/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/852430/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/852430/4"}}},"commit":{"parents":[{"commit":"6504a5f59e84c3d56409836af522866c414142ca","subject":"Merge \"swift-manage-shard-ranges repair: ignore recent overlaps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6504a5f59e84c3d56409836af522866c414142ca"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-08 13:54:29.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-24 15:07:40.000000000","tz":60},"subject":"swift-manage-shard-ranges: add \u0027merge\u0027 subcommand","message":"swift-manage-shard-ranges: add \u0027merge\u0027 subcommand\n\nAdds a subcommand to swift-manage-shard-ranges that allows arbitrary\nshard ranges to be read from a file and merged into a container\nDB. The file should be a JSON serialized list of dicts.\n\nThe merge subcommand is only recommended for emergency shard range\nmanipulation by expert users.\n\nChange-Id: Ic9ffcc042399f3834027a7935b64292d1fffe8d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/854bd5f051d442b4e65fdb99fb6382f7a9da78d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/854bd5f051d442b4e65fdb99fb6382f7a9da78d1"}]},"branch":"refs/heads/master"},"d2edf6646d670597b247ec3cf57b824a8469fa00":{"kind":"REWORK","_number":5,"created":"2022-08-30 12:00:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/30/852430/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/30/852430/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/30/852430/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/30/852430/5"}}},"commit":{"parents":[{"commit":"6504a5f59e84c3d56409836af522866c414142ca","subject":"Merge \"swift-manage-shard-ranges repair: ignore recent overlaps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6504a5f59e84c3d56409836af522866c414142ca"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-08 13:54:29.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2022-08-30 12:00:32.000000000","tz":60},"subject":"swift-manage-shard-ranges: add \u0027merge\u0027 subcommand","message":"swift-manage-shard-ranges: add \u0027merge\u0027 subcommand\n\nAdds a subcommand to swift-manage-shard-ranges that allows arbitrary\nshard ranges to be read from a file and merged into a container\nDB. The file should be a JSON serialized list of dicts.\n\nThe merge subcommand is only recommended for emergency shard range\nmanipulation by expert users.\n\nChange-Id: Ic9ffcc042399f3834027a7935b64292d1fffe8d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2edf6646d670597b247ec3cf57b824a8469fa00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2edf6646d670597b247ec3cf57b824a8469fa00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[]}
