)]}'
{"id":"openstack%2Fswift~852517","triplet_id":"openstack%2Fswift~master~I270a098de8e6bd7ab2996c118f7a0a761626e7a6","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I270a098de8e6bd7ab2996c118f7a0a761626e7a6","subject":"Remove unicode literal from code","status":"ABANDONED","created":"2022-08-09 06:48:14.000000000","updated":"2022-08-09 10:46:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"93d661546057e4eddf58df56cc6ac9f13095be8b","_number":852517,"virtual_id_number":852517,"owner":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-09 07:02:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-08-09 07:02:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"95494a42d00b3a058be87b0e7c860769345257a4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-08-09 06:48:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c94ea779364cd455756095159710aefca1e4db1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 07:02:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41bba7f94b2a45bb876d6352360dbe63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaaa537385994999aa1a5daf04aceea0 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b307bc10390b4870ae5cc94632484514 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"93d661546057e4eddf58df56cc6ac9f13095be8b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-08-09 10:46:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"884f2c965b289599dd9214b3b0539ebca0b9e901","revisions":{"884f2c965b289599dd9214b3b0539ebca0b9e901":{"kind":"REWORK","_number":1,"created":"2022-08-09 06:48:14.000000000","uploader":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"ref":"refs/changes/17/852517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/852517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/852517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/852517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/852517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/852517/1"}}},"commit":{"parents":[{"commit":"0c8c36b3fa51a3d8d4f06973f9fdd92e812c138e","subject":"Merge \"ring-builder: Better format large weights\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c8c36b3fa51a3d8d4f06973f9fdd92e812c138e"}]}],"author":{"name":"chenwei","email":"chen.wei@99cloud.net","date":"2022-08-09 06:47:42.000000000","tz":480},"committer":{"name":"chenwei","email":"chen.wei@99cloud.net","date":"2022-08-09 06:47:55.000000000","tz":480},"subject":"Remove unicode literal from code","message":"Remove unicode literal from code\n\nAll strings are considered as unicode literal string from Python 3.\n\nCloses bug: #1984045\n\nChange-Id: I270a098de8e6bd7ab2996c118f7a0a761626e7a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/884f2c965b289599dd9214b3b0539ebca0b9e901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/884f2c965b289599dd9214b3b0539ebca0b9e901"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
